)]}'
{"id":"openstack%2Foctavia~668901","triplet_id":"openstack%2Foctavia~stable%2Focata~I48e82f39e3980d6d50e3b710326b8e437eddb08b","project":"openstack/octavia","branch":"stable/ocata","topic":"bindep-fallback","hashtags":[],"change_id":"I48e82f39e3980d6d50e3b710326b8e437eddb08b","subject":"Add bindep.txt and ignore sha1 warning","status":"MERGED","created":"2019-07-03 13:21:28.000000000","updated":"2019-07-05 19:27:03.000000000","submitted":"2019-07-05 19:27:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668901-1562354823357-6efaa91b","meta_rev_id":"0bbd7ca7ee97d5fd3a41f54bccec9fd464731d32","_number":668901,"virtual_id_number":668901,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-07-05 19:27:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-07-05 17:04:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-07-05 17:32:37.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-07-05 17:32:37.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"date":"2019-07-03 13:21:28.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 13:21:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-07-05 17:04:47.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-07-05 17:32:37.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2019-07-05 19:27:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8337d4cd3a161b1b0d467e7e39328caea7f64941","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-03 13:21:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fa30a2a5bfc12844d0bab4d79ffa9b5886dbd2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 14:59:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/01/668901/1/check/openstack-tox-pep8/5052e3a/ : SUCCESS in 4m 42s\n- openstack-tox-py27 http://logs.openstack.org/01/668901/1/check/openstack-tox-py27/32f8659/ : SUCCESS in 4m 30s\n- openstack-tox-py35 http://logs.openstack.org/01/668901/1/check/openstack-tox-py35/2e3f627/ : SUCCESS in 5m 01s\n- openstack-tox-docs http://logs.openstack.org/01/668901/1/check/openstack-tox-docs/5b023b5/html/ : SUCCESS in 4m 19s\n- build-openstack-releasenotes http://logs.openstack.org/01/668901/1/check/build-openstack-releasenotes/47bb1c0/html/ : SUCCESS in 3m 55s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/01/668901/1/check/neutron-lbaasv2-dsvm-api/b7ad8d5/ : SUCCESS in 1h 03m 13s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/01/668901/1/check/neutron-lbaasv2-dsvm-scenario/d40f9bd/ : SUCCESS in 1h 21m 55s (non-voting)\n- octavia-tox-functional http://logs.openstack.org/01/668901/1/check/octavia-tox-functional/d6ba2f4/ : SUCCESS in 4m 39s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/01/668901/1/check/octavia-v1-dsvm-scenario/4fccaeb/ : SUCCESS in 54m 53s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/01/668901/1/check/octavia-v1-dsvm-scenario-multinode/74a7842/ : SUCCESS in 1h 03m 25s\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/01/668901/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/a7bb3ff/ : SUCCESS in 48m 59s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/01/668901/1/check/openstack-tox-cover/ce37bff/cover/ : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ccd9919481c94a7243cb5cf1acc46029532bd289","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-05 17:04:47.000000000","message":"Patch Set 1: Code-Review+2\n\nUmm... Ok. I\u0027m not sure how anyone is even using octavia pre-Pike at this point, and this is so old we don\u0027t even support it, but it can\u0027t hurt to let the code pass testing I guess.\n*shrug*","accounts_in_message":[],"_revision_number":1},{"id":"69494d173d580ae592f2690e8825d5f64aba83e4","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2019-07-05 17:32:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d00e05fd860102e7a24d79e9b9e6c0eabd5cb765","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 17:32:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"55156813a38c595459e179236ad0dc9efe23948d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 19:27:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"feb4ece0ba758f43bd6f666f356426238d84327c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 19:27:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/01/668901/1/gate/openstack-tox-pep8/7bdfd8d/ : SUCCESS in 5m 36s\n- openstack-tox-py27 http://logs.openstack.org/01/668901/1/gate/openstack-tox-py27/b83a619/ : SUCCESS in 5m 03s\n- openstack-tox-py35 http://logs.openstack.org/01/668901/1/gate/openstack-tox-py35/156f316/ : SUCCESS in 5m 44s\n- openstack-tox-docs http://logs.openstack.org/01/668901/1/gate/openstack-tox-docs/11874ba/html/ : SUCCESS in 5m 22s\n- build-openstack-releasenotes http://logs.openstack.org/01/668901/1/gate/build-openstack-releasenotes/ec7b876/html/ : SUCCESS in 4m 11s\n- octavia-tox-functional http://logs.openstack.org/01/668901/1/gate/octavia-tox-functional/41b2fcd/ : SUCCESS in 4m 29s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/01/668901/1/gate/octavia-v1-dsvm-scenario/2d3066b/ : SUCCESS in 50m 50s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/01/668901/1/gate/octavia-v1-dsvm-scenario-multinode/f76e917/ : SUCCESS in 1h 05m 44s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/01/668901/1/gate/neutron-lbaasv2-dsvm-api/af29ca6/ : SUCCESS in 1h 06m 10s\n- openstack-tox-cover http://logs.openstack.org/01/668901/1/gate/openstack-tox-cover/8812095/cover/ : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"168cdce13cd951743edf1bd4aa898f447ce14cd5","revisions":{"168cdce13cd951743edf1bd4aa898f447ce14cd5":{"kind":"REWORK","_number":1,"created":"2019-07-03 13:21:28.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/01/668901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/668901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/668901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/668901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/668901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/668901/1"}}},"commit":{"parents":[{"commit":"20e2916d2f54b0df41aa5fbd05a995c0fcae416c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20e2916d2f54b0df41aa5fbd05a995c0fcae416c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-24 19:57:55.000000000","tz":-420},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-07-03 13:19:39.000000000","tz":120},"subject":"Add bindep.txt and ignore sha1 warning","message":"Add bindep.txt and ignore sha1 warning\n\nThis is a combination of 2 commits to pass the check and gate jobs.\n\n1. Add bindep.txt for Octavia\n\nSome base packages are no longer automatically included in zuul\njob runs. This is good as we need to call out our platform package\nrequirements.\nThis patch adds a bindep.txt to Octavia to capture those\npackage requirements.\n\n(cherry picked from commit d9c459a8686817bdaf4967ffb45b2c180a5e4265)\n\n2. \"Resolve\" bandit issue with sha1 hashes\n\nWe only use them as a consistent hash, not for anything security\nrelated. Even if someone could manage a hash collision or something,\nthe worst that could happen is their own LB stops working...\nSo, just tell bandit to ignore them.\nAdded a docstring to one function to attempt to be extra clear.\n\nConflicts:\n      octavia/common/tls_utils/cert_parser.py\nNote(elod.illes): conflict is due to not having patch\nI039ed0b40df8b72a1238f8896548fe77086c530c on stable pike.\n\n(cherry picked from commit 91fae05a7d53362f5484154fc35c921f3482bc2f)\n(cherry picked from commit 12bf4409faabfd4437d4f09a06cd216b24064ee7)\n\nChange-Id: I48e82f39e3980d6d50e3b710326b8e437eddb08b\n(cherry picked from commit af80b8f719bda324a75507b266dcd2f0cf2aeedc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/168cdce13cd951743edf1bd4aa898f447ce14cd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/168cdce13cd951743edf1bd4aa898f447ce14cd5"}]},"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
