)]}'
{"id":"openstack%2Foctavia~669359","triplet_id":"openstack%2Foctavia~master~I477d2d12c64c1e38e3917333e5d47f612cab93b0","project":"openstack/octavia","branch":"master","topic":"fix_wsgi_config_user_permissions","hashtags":[],"change_id":"I477d2d12c64c1e38e3917333e5d47f612cab93b0","subject":"Fix user permission for WSGI configuration Task: 35692 Story: 2006172","status":"MERGED","created":"2019-07-05 13:43:58.000000000","updated":"2020-09-01 15:53:02.000000000","submitted":"2020-09-01 15:49:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669359-1598975354191-1189cc01","meta_rev_id":"71457379c09b82a484a426238464d85f8d90082b","_number":669359,"virtual_id_number":669359,"owner":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 15:49:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":2,"date":"2020-09-01 11:56:18.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-09-01 11:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-01 11:56:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":1,"date":"2020-09-01 11:56:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"recommended":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"all":[{"value":1,"date":"2020-09-01 11:56:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":2,"date":"2020-09-01 11:56:18.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 13:44:10.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-08-28 22:54:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-08-29 16:03:42.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-01 05:17:54.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-09-01 11:56:26.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-01 15:49:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da25c20cdb11748c0832ce4abac94a01b2444af7","author":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"date":"2019-07-05 13:43:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c52a4459835274e725789e862b5875b134d5709","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-07-05 13:44:10.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"9e3da4705b9ecbc3d849aecbe96608524465414b","author":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"date":"2019-07-05 13:51:12.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"85c00d3f87e3a932a68baf6884cea4621b14c43e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 16:04:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/59/669359/2/check/openstack-tox-cover/c2cab7b/cover/ : SUCCESS in 7m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/669359/2/check/openstack-tox-lower-constraints/47e6d26/ : SUCCESS in 7m 51s\n- openstack-tox-pep8 http://logs.openstack.org/59/669359/2/check/openstack-tox-pep8/3757832/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/59/669359/2/check/openstack-tox-py27/060b4ac/ : SUCCESS in 5m 42s\n- openstack-tox-py36 http://logs.openstack.org/59/669359/2/check/openstack-tox-py36/9e5365e/ : SUCCESS in 5m 02s\n- openstack-tox-py37 http://logs.openstack.org/59/669359/2/check/openstack-tox-py37/960f7a0/ : SUCCESS in 4m 53s\n- openstack-tox-docs http://logs.openstack.org/59/669359/2/check/openstack-tox-docs/8ca48e5/html/ : SUCCESS in 5m 54s\n- openstack-tox-functional http://logs.openstack.org/59/669359/2/check/openstack-tox-functional/181da0e/ : SUCCESS in 5m 22s\n- openstack-tox-functional-py36 http://logs.openstack.org/59/669359/2/check/openstack-tox-functional-py36/3c6fe8a/ : SUCCESS in 5m 05s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/59/669359/2/check/octavia-v2-dsvm-noop-api/826ad37/ : SUCCESS in 39m 25s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/59/669359/2/check/octavia-v2-dsvm-noop-py2-api/332aada/ : SUCCESS in 44m 29s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/59/669359/2/check/octavia-v2-dsvm-scenario/8a53c46/ : SUCCESS in 1h 12m 11s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/59/669359/2/check/octavia-v2-dsvm-py2-scenario/7302f22/ : SUCCESS in 1h 21m 04s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/59/669359/2/check/octavia-v2-dsvm-py2-scenario-centos-7/8268b1c/ : FAILURE in 2h 06m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/59/669359/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/c315f68/ : SUCCESS in 1h 28m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/59/669359/2/check/octavia-v2-dsvm-scenario-two-node/d0da314/ : SUCCESS in 1h 44m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/59/669359/2/check/octavia-v2-dsvm-py2-scenario-two-node/c001e3f/ : SUCCESS in 1h 30m 51s (non-voting)\n- octavia-grenade http://logs.openstack.org/59/669359/2/check/octavia-grenade/99849fd/ : SUCCESS in 1h 05m 28s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/59/669359/2/check/octavia-v2-dsvm-tls-barbican/645292a/ : SUCCESS in 52m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/59/669359/2/check/octavia-v2-dsvm-scenario-amphora-v2/ef7527b/ : SUCCESS in 1h 13m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b83682ca308bd3d57565df656951e7c41033f6df","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-28 22:54:39.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"e7407bb0a5184f4d0c4e750276cea93c91677be7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-29 16:03:42.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ae48096ca9d5899f7e622731b4cc75578765be52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 16:03:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0730dd08415bd65edb0ea2e119ea70734e68e945","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 16:10:30.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6f323a0f425463aa7829fa384c02236 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/900111add1cd4e85953c81698bd2dce4 : CANCELED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2b61c4ca6ff443eba3a360504bbce1d : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fab84835bec1427fb803bfc273b27be0 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce1ec3d7e79247c2aea90daae28c416a : CANCELED\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5c47d9d5c7094035a6bfbd1d0c323979 : FAILURE in 5m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/47a1fc3d12a54d3388d6dc9f894aebb7 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3356e810158f4c44a7b8a2175ddea16c : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dc45437f1bd24104a45a6eb5d40363f0 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7b96734b552846c48647c2d0bc0d9d42 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ac39acb74e14231ba9d21c45fd9d64a : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8fe464f4301d4ddf97cdf306228edf35 : CANCELED","accounts_in_message":[],"_revision_number":2},{"id":"994768dff9f99ff93b2930704084cf4a8835d9b9","author":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"date":"2020-08-31 10:13:05.000000000","message":"Patch Set 2: Backport-Candidate+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"42f6f14d584b36bc0c7216f328f5fa4b802e5b9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 11:36:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9de595de0aa43bda22a3051169b3d66 : FAILURE in 6m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34927f24468a4f2a902cfa23e2832cc9 : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab3640c86fcf4040b7c5a774e2672c66 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a61a9ad93bc24421a010691ed1c70d8e : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30f479a92d1b4c789ce9f9f50327fdd2 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e18d941167764e1bb0ba5e6adb589a2d : SUCCESS in 10m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4f489076b78f490e93b675869a0496f7 : SUCCESS in 5m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f96aa0d9af524b0e8a62ceee71ca93f5 : FAILURE in 9m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9402a7fccab9477d971cac2e5bd9c376 : FAILURE in 6m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/86e097c7c04c4d04aa3a7a46fbdf8980 : FAILURE in 15m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b15a82425c1043ca981b0ff3a5a7b39f : FAILURE in 15m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/172764e0dffc45639a20f1ccdcabed3d : FAILURE in 16m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6ac361b8f20144d28e16d461be9084e8 : FAILURE in 23m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/61d9696065304256872761c14930d57a : FAILURE in 15m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c856484489f8436da501b01d808d808c : FAILURE in 24m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e9ec71dee1004b6f87d0bbc4ca827372 : FAILURE in 12m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/30ba5036e603454c8fff4abd2196a368 : FAILURE in 17m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c4a2efd926dc4568986fb41a36eed679 : FAILURE in 18m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/77f9fd5f60254375b03aa71bd21ca0a2 : FAILURE in 24m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/12b0ea1c8a994cd680ecd27007c13776 : FAILURE in 14m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ecc6d1a811d34e15b98721b2a5b39d35 : SUCCESS in 1h 21m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2abb71e691a6f62816379d7df04532b5cc51af87","author":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"date":"2020-08-31 12:26:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b3052bea0d536898e4dc8ab5700499243d5dc4f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:03:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40af24e9e0cb42608cd1fd042711238f : FAILURE in 8m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a857923537224b6f94eefe6ef11444f6 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2ca10236a4c4cc180c64fb2683203e3 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/890713aab5ab42f18f45becb0dde127d : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a837c5ed3f041a481c75a6968a52ec2 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32b934fb4d9c4749818176240138d960 : SUCCESS in 11m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/63e09d822e014965ac45020fff1d66b7 : SUCCESS in 5m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/797e2e48601b4cf1a32cb6ea5d0a5c2f : FAILURE in 8m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/303f8d9eba064d4985efdb5c70bddb55 : FAILURE in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2e70680d447f44df9d5d5fb4897436a5 : FAILURE in 22m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe8418003cc3454e97bbbacfb29465d2 : FAILURE in 15m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1a38eb16266d46dab71c7437f84d2ef4 : FAILURE in 16m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/99be874e387b498ab011a13dd575a9d1 : FAILURE in 21m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d1740cc7ffa45b58ab86af9cd313bd0 : FAILURE in 21m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fc6338f530764fc68b6b206ba53390af : FAILURE in 12m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/94d068eb8a444a589e38a2646eaf4bed : FAILURE in 15m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/69a62447b96c468faa5c341ad88a6217 : FAILURE in 18m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c34fb4f9fa9d4297af11b3de7103a1bc : FAILURE in 14m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/58b0023c0b354765a68d12092870fd42 : FAILURE in 14m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4c8a2d5bf1a54152a5f8748ae04d70a5 : FAILURE in 15m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/744e1aee6579439d9ecfb12fe16e562c : FAILURE in 35m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"727704b76634fd90dec79d900cc252da274489fa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-01 05:12:16.000000000","message":"Patch Set 2:\n\nrecheck\ndevstack should be fixed now","accounts_in_message":[],"_revision_number":2},{"id":"63e4856f048662fc31a1eb07d2353c50c7c904a5","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-09-01 05:17:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3b761e829e5a50854827a599d442dc643c223df7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 06:43:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bb84f7a5f0b4640bb03a8ba36b201f2 : FAILURE in 9m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70f350fadf444567abfe7e5151449dff : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63096d562ce042b8ba39b218f83ac42b : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5ecd5d9bca04b14a97adcb4536eda7f : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23a386ceac28426187d560a47aa2ca76 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f98dadde86b347068e29273dc0500854 : SUCCESS in 10m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7cfef56d4b0c4f3a87295970bd9ff755 : SUCCESS in 5m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0810f9324c614e4cb09b5f02d971b766 : FAILURE in 7m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a4b33daf94914bebbb3000911cb5e31e : FAILURE in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/98bb954bba264df1bc55a6145b1a3287 : SUCCESS in 38m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fdffaea5e83445dba64e756434b1b32c : SUCCESS in 40m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/05a79d45281848d2ac54b0a51b15e21a : SUCCESS in 48m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a0c372f3f979492a9b6b42455db77e15 : SUCCESS in 29m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18637d305ac44b4c92b28de3d596f729 : SUCCESS in 46m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/303b0b81f37f4e83b8b143527d8b0de9 : SUCCESS in 1h 02m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1fd6ecc2fb6644b3aede77ca5721d66b : SUCCESS in 1h 11m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fed242816143400e9a2f0afe292a1b1d : SUCCESS in 1h 19m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0c0616dfa9184f9fb88cbf4301369aaa : SUCCESS in 1h 11m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/706aea7e7dd742fa80884ec2feb7c6f3 : SUCCESS in 42m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/88bdc4724ffe4517a5814331e70c686e : SUCCESS in 1h 14m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/44dbbe7b4fcd43889dee7fa439dec34d : SUCCESS in 1h 28m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"859adba5be56a279e1c83c26044113c7c5419a7d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-01 11:56:18.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b5329829381d7f1928dd29ee6a43ee52c159ed59","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-01 11:56:26.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8e93f16d70ba9617e991a2efe2fdd3603b43a55c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 13:26:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a907b079130e43468902a8bd237bcf55 : SUCCESS in 8m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10e91d796d3c4ea0bd52e8545ec40c18 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b4a967cbade41fcb1f08e4aa745f249 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10d76a3566b94a6b9b2228d6a0f9e404 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2f326e55a484ab59bb16d97e1309bc7 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255e2e2e2872464a984bdd98b0f21e13 : SUCCESS in 11m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d67ef261d7f840c999cbeddacd8d1ef1 : SUCCESS in 6m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b4de7761c9a34639a70b4cb86e808757 : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b41573e67c444dd29a02bc094efdb671 : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9cbe0033f8b741939dcd32f9df9151f6 : SUCCESS in 43m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4fd547ab2e64637991d267a7c87e02f : SUCCESS in 49m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/86573f9268ee4711a33f634230fe6bf9 : SUCCESS in 43m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/39b0f331297946759be8a38b486b7e8e : SUCCESS in 32m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/403f0ae419f9407cbd26997c3d52c0dc : SUCCESS in 1h 00m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f8e0f286740d41fe9659bef72ec6cb31 : SUCCESS in 1h 05m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c3bfccd410b64b6d85c3867abae60de6 : SUCCESS in 1h 27m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/61ab410d1e8a4fb698ca4167431e99a8 : SUCCESS in 54m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/11d7b3129e494ffc828435a6ef4ceab7 : SUCCESS in 1h 24m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8da5c4e37f5e4212a53ddddee620a366 : SUCCESS in 44m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/032a6f23ee394944996c2766b3b1c4d9 : SUCCESS in 49m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/47e5dbea1e0e4568a513ba950c9a5199 : FAILURE in 37m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8c0f2d9a667978ebf5575de4d328dee010c36d00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 13:26:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"42f494a5e8856387dc726ed0f958135965241d23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 15:49:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17d2fa9fea55405bac1c7ab355ffa0f5 : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5032e60d099d478d9df60643770e85d6 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b023e214e02740edaa3394c1d75a800b : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/419b39636789445aba8a22cde018068d : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09f9facfeeb14a8e9338e5297c310cbd : SUCCESS in 11m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9e263c1414ea43d7b3743fd7fefc0a29 : SUCCESS in 8m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fdbcebfa3bc3427cb15d587f55bd9b60 : SUCCESS in 46m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d0806ecf0bf4117b417fec36c16dd44 : SUCCESS in 1h 08m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ab10d712badf46c0aa3e5becfebde231 : SUCCESS in 55m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0b45ecc15bff436ea9e15734f7da9ac7 : SUCCESS in 41m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f485effddd04412a3b752203ac8057c : SUCCESS in 55m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fc51e735dbe94e2d9db6ec51aa733e83 : SUCCESS in 2h 19m 37s","accounts_in_message":[],"_revision_number":3},{"id":"1e1c2abac10085c980ec723ea5ae9794dba1d353","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 15:49:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0253d46bb9a865b946758ce6cb31fc9932d35787","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 15:53:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25e192c970444db1888662cdade0e8f6 : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6df1ce8e28eb76cb782dc06f913598cfa2ecc156","revisions":{"ba3e68ebc63935cc07d20876d8b50d725810db05":{"kind":"REWORK","_number":1,"created":"2019-07-05 13:43:58.000000000","uploader":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"ref":"refs/changes/59/669359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/669359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/669359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/669359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/669359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/669359/1"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","date":"2019-07-05 13:43:49.000000000","tz":120},"committer":{"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","date":"2019-07-05 13:43:49.000000000","tz":120},"subject":"Task: 35692 Story: 2006172 Fix user permission for wsgi config","message":"Task: 35692 Story: 2006172 Fix user permission for wsgi config\n\nChange-Id: I477d2d12c64c1e38e3917333e5d47f612cab93b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba3e68ebc63935cc07d20876d8b50d725810db05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba3e68ebc63935cc07d20876d8b50d725810db05"}]},"branch":"refs/heads/master"},"6f6c1a13ed8b0755f4bb278f08cd104732504c16":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-05 13:51:12.000000000","uploader":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"ref":"refs/changes/59/669359/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/669359/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/669359/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/669359/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/669359/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/669359/2"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","date":"2019-07-05 13:43:49.000000000","tz":120},"committer":{"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","date":"2019-07-05 13:49:20.000000000","tz":120},"subject":"Fix user permission for WSGI configuration Task: 35692 Story: 2006172","message":"Fix user permission for WSGI configuration\nTask: 35692\nStory: 2006172\n\nChange-Id: I477d2d12c64c1e38e3917333e5d47f612cab93b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f6c1a13ed8b0755f4bb278f08cd104732504c16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f6c1a13ed8b0755f4bb278f08cd104732504c16"}]},"branch":"refs/heads/master"},"6df1ce8e28eb76cb782dc06f913598cfa2ecc156":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-01 11:56:18.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/59/669359/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/669359/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/669359/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/669359/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/669359/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/669359/3"}}},"commit":{"parents":[{"commit":"a5f0524fd00b1fdb9d47bafba81155eeb769f3ad","subject":"Add ALPN support for TLS-terminated HTTPS LBs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5f0524fd00b1fdb9d47bafba81155eeb769f3ad"}]}],"author":{"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","date":"2019-07-05 13:43:49.000000000","tz":120},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-09-01 11:56:18.000000000","tz":0},"subject":"Fix user permission for WSGI configuration Task: 35692 Story: 2006172","message":"Fix user permission for WSGI configuration\nTask: 35692\nStory: 2006172\n\nChange-Id: I477d2d12c64c1e38e3917333e5d47f612cab93b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6df1ce8e28eb76cb782dc06f913598cfa2ecc156"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6df1ce8e28eb76cb782dc06f913598cfa2ecc156"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
