)]}'
{"id":"openstack%2Foctavia~669440","triplet_id":"openstack%2Foctavia~master~I76cdc49e482458a2beefb26b9301023559d16029","project":"openstack/octavia","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I76cdc49e482458a2beefb26b9301023559d16029","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 18:03:43.000000000","updated":"2019-07-13 10:20:46.000000000","submitted":"2019-07-13 10:20:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669440-1563013246168-5870c8eb","meta_rev_id":"7dfcfc9c358bd5895ba677d7604206b817057072","_number":669440,"virtual_id_number":669440,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-07-13 10:20:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-08 18:59:12.000000000","_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2019-07-13 01:02:02.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-07-13 08:39:33.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-07-13 09:22:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":2,"date":"2019-07-12 22:05:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":1,"date":"2019-07-13 08:51:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2019-07-13 08:39:33.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 22:05:10.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-07-13 01:02:02.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-07-13 08:39:33.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-07-13 08:51:31.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-07-13 09:22:18.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-07-13 10:20:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0fc9488769900358072539df7b46a7b702e27514","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 18:03:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9001ae04694610b8a926045662a3e396b5040638","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 20:27:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/40/669440/1/check/build-openstack-api-ref/c623527/html/ : SUCCESS in 4m 01s\n- openstack-tox-cover http://logs.openstack.org/40/669440/1/check/openstack-tox-cover/de78550/cover/ : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/669440/1/check/openstack-tox-lower-constraints/584e58c/ : SUCCESS in 6m 10s\n- openstack-tox-pep8 http://logs.openstack.org/40/669440/1/check/openstack-tox-pep8/16f2fc4/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/40/669440/1/check/openstack-tox-py27/9df909b/ : SUCCESS in 3m 36s\n- openstack-tox-py36 http://logs.openstack.org/40/669440/1/check/openstack-tox-py36/5035cd6/ : SUCCESS in 3m 52s\n- openstack-tox-py37 http://logs.openstack.org/40/669440/1/check/openstack-tox-py37/9777377/ : SUCCESS in 6m 51s\n- openstack-tox-docs http://logs.openstack.org/40/669440/1/check/openstack-tox-docs/57f3443/html/ : SUCCESS in 5m 18s\n- build-openstack-releasenotes http://logs.openstack.org/40/669440/1/check/build-openstack-releasenotes/4143861/html/ : SUCCESS in 4m 55s\n- openstack-tox-functional http://logs.openstack.org/40/669440/1/check/openstack-tox-functional/0cb79d3/ : SUCCESS in 5m 09s\n- openstack-tox-functional-py36 http://logs.openstack.org/40/669440/1/check/openstack-tox-functional-py36/18ba3f0/ : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/40/669440/1/check/octavia-v2-dsvm-noop-api/2494f85/ : SUCCESS in 51m 11s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/40/669440/1/check/octavia-v2-dsvm-noop-py2-api/74218f8/ : SUCCESS in 47m 58s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/40/669440/1/check/octavia-v2-dsvm-scenario/2c95396/ : SUCCESS in 1h 38m 30s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/40/669440/1/check/octavia-v2-dsvm-py2-scenario/c76fd6a/ : SUCCESS in 1h 33m 06s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/40/669440/1/check/octavia-v2-dsvm-py2-scenario-centos-7/7f665b3/ : FAILURE in 2h 16m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/40/669440/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/ffd6a51/ : SUCCESS in 1h 31m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/40/669440/1/check/octavia-v2-dsvm-scenario-two-node/50e879a/ : SUCCESS in 1h 39m 04s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/40/669440/1/check/octavia-v2-dsvm-py2-scenario-two-node/6f33661/ : SUCCESS in 1h 08m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/40/669440/1/check/octavia-v2-dsvm-tls-barbican/a5f1b66/ : SUCCESS in 48m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/40/669440/1/check/octavia-v2-dsvm-scenario-amphora-v2/312515a/ : SUCCESS in 1h 24m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1f26d21d7aa75f138cb2f10594218cf076c31b9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-08 18:20:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2bf2f06d4b77ce031d8f098600250c80158165b","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-08 18:59:12.000000000","message":"Patch Set 1:\n\nThe idea is for a default \u0027tox\u0027 run to just run the latest py37 rather than spending cycles running py36, py37, and py27 etc. Then if a developer wants to run py36 they can run \u0027tox -e py36\u0027. Most projects are moving to just py37 for defaults.","accounts_in_message":[],"_revision_number":1},{"id":"718e9e6c94b866bc7123494ab53f53e7083fb07b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-12 22:05:10.000000000","message":"Patch Set 1: Code-Review+2\n\nThere used to be an issue with the order of those py27 and py3x tests, but I hope that is resolved now. Easy to fix if not.","accounts_in_message":[],"_revision_number":1},{"id":"b28f4478ef9a9c5d636a270b66c2f865830284fb","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-07-13 01:02:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b95e542eed2db86b176458f198cb36d5859c1b01","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-13 08:39:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"52b0e298fc729dfabb867f6c944d0f94f9ce6769","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 08:39:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"527ea8038b64cc1c598b50d03fcc9d89a9b965d3","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-07-13 08:51:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4ecee103c36a523697c3a063bddeac0310ca8751","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-07-13 09:22:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"838792534ae9027e49b178ae482c8ea46f029be3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 10:20:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/40/669440/1/gate/build-openstack-api-ref/b4b2def/html/ : SUCCESS in 3m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/669440/1/gate/openstack-tox-lower-constraints/5a8e457/ : SUCCESS in 7m 05s\n- openstack-tox-pep8 http://logs.openstack.org/40/669440/1/gate/openstack-tox-pep8/90828c9/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/40/669440/1/gate/openstack-tox-py27/1859859/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/40/669440/1/gate/openstack-tox-py36/456417c/ : SUCCESS in 4m 00s\n- openstack-tox-py37 http://logs.openstack.org/40/669440/1/gate/openstack-tox-py37/e7f211e/ : SUCCESS in 4m 25s\n- openstack-tox-docs http://logs.openstack.org/40/669440/1/gate/openstack-tox-docs/0f52f77/html/ : SUCCESS in 4m 58s\n- build-openstack-releasenotes http://logs.openstack.org/40/669440/1/gate/build-openstack-releasenotes/b642a18/html/ : SUCCESS in 4m 01s\n- openstack-tox-functional http://logs.openstack.org/40/669440/1/gate/openstack-tox-functional/123f9ac/ : SUCCESS in 5m 12s\n- openstack-tox-functional-py36 http://logs.openstack.org/40/669440/1/gate/openstack-tox-functional-py36/92d2cf7/ : SUCCESS in 5m 07s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/40/669440/1/gate/octavia-v2-dsvm-noop-api/a9e3fba/ : SUCCESS in 46m 01s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/40/669440/1/gate/octavia-v2-dsvm-noop-py2-api/09c2310/ : SUCCESS in 41m 15s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/40/669440/1/gate/octavia-v2-dsvm-scenario/465873a/ : SUCCESS in 1h 31m 45s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/40/669440/1/gate/octavia-v2-dsvm-py2-scenario/be886f3/ : SUCCESS in 1h 35m 15s","accounts_in_message":[],"_revision_number":1},{"id":"2dfa1f7b14c04ae709c55bd44967793fdc885be2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 10:20:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8a2560ed905fc27c4511de9af2ac3f96e61b18b","revisions":{"f8a2560ed905fc27c4511de9af2ac3f96e61b18b":{"kind":"REWORK","_number":1,"created":"2019-07-05 18:03:43.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/40/669440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/669440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/669440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/669440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/669440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/669440/1"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:48:17.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 18:03:42.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I76cdc49e482458a2beefb26b9301023559d16029\nStory: #2005924\nTask: #34227\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f8a2560ed905fc27c4511de9af2ac3f96e61b18b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f8a2560ed905fc27c4511de9af2ac3f96e61b18b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
