)]}'
{"id":"openstack%2Foctavia~669725","triplet_id":"openstack%2Foctavia~master~Ia05d7e9f8ac2692e5769648255ef0e4e10a18c03","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ia05d7e9f8ac2692e5769648255ef0e4e10a18c03","subject":"Switch the default Ubuntu kernel to -virtual","status":"MERGED","created":"2019-07-08 17:30:44.000000000","updated":"2019-07-09 08:25:47.000000000","submitted":"2019-07-09 08:25:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669725-1562660747808-5d2650a7","meta_rev_id":"26a2c3d10fecf8663ad8f347f430af8c0502d242","_number":669725,"virtual_id_number":669725,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-07-09 08:25:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":2,"date":"2019-07-08 18:20:30.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2019-07-09 06:34:12.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":1,"date":"2019-07-09 06:34:12.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 18:20:30.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2019-07-09 06:34:12.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-07-09 08:25:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e0cf8762fb85a813fcd77560d8118932a67bd03","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-08 17:30:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cff1194319560dc107dddf117769e0eaed34179d","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2019-07-08 18:20:14.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"1f1be7002353fd132a2042ce99cd365d25f9a5b7","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2019-07-08 18:20:30.000000000","message":"Patch Set 1: -Backport-Candidate Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"29752310d0281f2fe4014f2909fdf9cb8b08d21c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 19:26:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/25/669725/1/check/openstack-tox-cover/0d0168b/cover/ : SUCCESS in 10m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/669725/1/check/openstack-tox-lower-constraints/8168a92/ : SUCCESS in 8m 07s\n- openstack-tox-pep8 http://logs.openstack.org/25/669725/1/check/openstack-tox-pep8/e13707f/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/25/669725/1/check/openstack-tox-py27/e8823a2/ : SUCCESS in 5m 46s\n- openstack-tox-py36 http://logs.openstack.org/25/669725/1/check/openstack-tox-py36/8ac34db/ : SUCCESS in 6m 01s\n- openstack-tox-py37 http://logs.openstack.org/25/669725/1/check/openstack-tox-py37/450d041/ : SUCCESS in 5m 14s\n- openstack-tox-docs http://logs.openstack.org/25/669725/1/check/openstack-tox-docs/7a77bc1/html/ : SUCCESS in 4m 50s\n- build-openstack-releasenotes http://logs.openstack.org/25/669725/1/check/build-openstack-releasenotes/f90873b/html/ : SUCCESS in 3m 52s\n- openstack-tox-functional http://logs.openstack.org/25/669725/1/check/openstack-tox-functional/5a51644/ : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 http://logs.openstack.org/25/669725/1/check/openstack-tox-functional-py36/5a20864/ : SUCCESS in 4m 48s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/25/669725/1/check/octavia-v2-dsvm-noop-api/0684672/ : SUCCESS in 47m 00s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/25/669725/1/check/octavia-v2-dsvm-noop-py2-api/f9ee494/ : SUCCESS in 45m 12s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/25/669725/1/check/octavia-v2-dsvm-scenario/b59d1d4/ : SUCCESS in 1h 15m 15s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/25/669725/1/check/octavia-v2-dsvm-py2-scenario/6e49990/ : SUCCESS in 1h 44m 29s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/25/669725/1/check/octavia-v2-dsvm-py2-scenario-centos-7/e8e01c3/ : FAILURE in 1h 48m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/25/669725/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/c310871/ : FAILURE in 1h 35m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/25/669725/1/check/octavia-v2-dsvm-scenario-two-node/a9883ee/ : SUCCESS in 1h 47m 47s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/25/669725/1/check/octavia-v2-dsvm-py2-scenario-two-node/08c0a42/ : SUCCESS in 1h 25m 02s (non-voting)\n- octavia-grenade http://logs.openstack.org/25/669725/1/check/octavia-grenade/cd0b81c/ : SUCCESS in 1h 32m 31s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/25/669725/1/check/octavia-v2-dsvm-tls-barbican/9d1486c/ : SUCCESS in 50m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/25/669725/1/check/octavia-v2-dsvm-scenario-amphora-v2/c2a9a66/ : SUCCESS in 1h 34m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4e999b477ac50046004a586c47fb1f37a65196a","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-07-09 06:34:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"37ae636367563fecdc5ebc55ca8466fef85c8d58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 06:34:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aefe4c97daffd1c3b61cf044cc112ec34c9c8a76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 08:25:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"26a2c3d10fecf8663ad8f347f430af8c0502d242","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 08:25:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/25/669725/1/gate/openstack-tox-lower-constraints/bacb0f7/ : SUCCESS in 9m 14s\n- openstack-tox-pep8 http://logs.openstack.org/25/669725/1/gate/openstack-tox-pep8/c13d8c1/ : SUCCESS in 5m 44s\n- openstack-tox-py27 http://logs.openstack.org/25/669725/1/gate/openstack-tox-py27/b16cd05/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/25/669725/1/gate/openstack-tox-py36/5b905a6/ : SUCCESS in 5m 13s\n- openstack-tox-py37 http://logs.openstack.org/25/669725/1/gate/openstack-tox-py37/722cf9e/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/25/669725/1/gate/openstack-tox-docs/63d9d63/html/ : SUCCESS in 5m 32s\n- build-openstack-releasenotes http://logs.openstack.org/25/669725/1/gate/build-openstack-releasenotes/060183c/html/ : SUCCESS in 4m 25s\n- openstack-tox-functional http://logs.openstack.org/25/669725/1/gate/openstack-tox-functional/b6bd84b/ : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 http://logs.openstack.org/25/669725/1/gate/openstack-tox-functional-py36/08850cf/ : SUCCESS in 4m 56s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/25/669725/1/gate/octavia-v2-dsvm-noop-api/2596e93/ : SUCCESS in 40m 20s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/25/669725/1/gate/octavia-v2-dsvm-noop-py2-api/c832c4c/ : SUCCESS in 52m 18s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/25/669725/1/gate/octavia-v2-dsvm-scenario/29ea580/ : SUCCESS in 1h 45m 40s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/25/669725/1/gate/octavia-v2-dsvm-py2-scenario/ef719df/ : SUCCESS in 1h 45m 08s\n- octavia-grenade http://logs.openstack.org/25/669725/1/gate/octavia-grenade/a420f66/ : SUCCESS in 1h 42m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3aacb67ada526501a8c080c219ee2be869b959b","revisions":{"e3aacb67ada526501a8c080c219ee2be869b959b":{"kind":"REWORK","_number":1,"created":"2019-07-08 17:30:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/25/669725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/669725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/669725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/669725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/669725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/669725/1"}}},"commit":{"parents":[{"commit":"31428bf25a247a4d9803871c49ca6e64b7856198","subject":"Merge \"only rollback DB when we have a connection to the DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31428bf25a247a4d9803871c49ca6e64b7856198"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-08 17:28:01.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-08 17:30:43.000000000","tz":-420},"subject":"Switch the default Ubuntu kernel to -virtual","message":"Switch the default Ubuntu kernel to -virtual\n\nA previous patch changed the default Ubuntu kernel for the amphora\nimages to the -kvm kernel[1]. This kernel is missing a module required\nfor cryptsetup.\nThis patch switches the default to the -virtual kernel, which is\nslightly larger than the -kvm kernel, but includes the required\nkernel module for cryptsetup.\n\n[1] https://review.opendev.org/665861\n\nChange-Id: Ia05d7e9f8ac2692e5769648255ef0e4e10a18c03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3aacb67ada526501a8c080c219ee2be869b959b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3aacb67ada526501a8c080c219ee2be869b959b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
