)]}'
{"id":"openstack%2Foctavia~670342","triplet_id":"openstack%2Foctavia~master~I7793d7a9113b23ac88e7c53d5dc292a70b9453b5","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I7793d7a9113b23ac88e7c53d5dc292a70b9453b5","subject":"Add warning log if auth_strategy is not keystone","status":"MERGED","created":"2019-07-11 16:09:32.000000000","updated":"2019-08-26 13:50:48.000000000","submitted":"2019-07-16 05:31:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670342-1563255104372-9977374c","meta_rev_id":"c8004c8c9635d5eb4f83ed5873d0ae159b61d481","_number":670342,"virtual_id_number":670342,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":2,"date":"2019-07-16 05:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"all":[{"value":2,"date":"2019-07-14 10:38:20.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":1,"date":"2019-07-14 13:30:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":2,"date":"2019-07-16 03:52:57.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2019-07-14 08:17:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":1,"date":"2019-07-16 03:52:57.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"all":[{"value":2,"date":"2019-07-14 10:38:20.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-14 08:17:21.000000000","updated_by":{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},"reviewer":{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},"state":"REVIEWER"},{"updated":"2019-07-14 10:38:20.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-07-14 13:30:07.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-07-16 03:52:57.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-07-16 05:31:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cea30ec82103f2b65c3cc3552cf494aebdb2b0e1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-11 16:09:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e090fbe491df5d20c474c9b7f66f34e024c7184","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-11 16:12:54.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4cbfc0818f6aec0432e18616256e02de1834236d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-11 16:15:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d7d691a837edac2da87d4e8065b1bec5d21a735c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 18:13:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/42/670342/2/check/openstack-tox-cover/218e041/cover/ : SUCCESS in 7m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/42/670342/2/check/openstack-tox-lower-constraints/da59071/ : SUCCESS in 7m 37s\n- openstack-tox-pep8 http://logs.openstack.org/42/670342/2/check/openstack-tox-pep8/3a6c8b3/ : SUCCESS in 6m 24s\n- openstack-tox-py27 http://logs.openstack.org/42/670342/2/check/openstack-tox-py27/8f66273/ : SUCCESS in 4m 48s\n- openstack-tox-py36 http://logs.openstack.org/42/670342/2/check/openstack-tox-py36/d7e393b/ : SUCCESS in 3m 51s\n- openstack-tox-py37 http://logs.openstack.org/42/670342/2/check/openstack-tox-py37/ea1e544/ : SUCCESS in 5m 30s\n- openstack-tox-docs http://logs.openstack.org/42/670342/2/check/openstack-tox-docs/d34e6fb/html/ : SUCCESS in 5m 10s\n- openstack-tox-functional http://logs.openstack.org/42/670342/2/check/openstack-tox-functional/733bc46/ : SUCCESS in 5m 22s\n- openstack-tox-functional-py36 http://logs.openstack.org/42/670342/2/check/openstack-tox-functional-py36/c324c34/ : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/42/670342/2/check/octavia-v2-dsvm-noop-api/360a641/ : SUCCESS in 45m 14s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/42/670342/2/check/octavia-v2-dsvm-noop-py2-api/742c617/ : SUCCESS in 51m 24s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/42/670342/2/check/octavia-v2-dsvm-scenario/2ac9060/ : SUCCESS in 1h 36m 23s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/42/670342/2/check/octavia-v2-dsvm-py2-scenario/bc2c615/ : SUCCESS in 1h 12m 04s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/42/670342/2/check/octavia-v2-dsvm-py2-scenario-centos-7/775094d/ : SUCCESS in 1h 39m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/42/670342/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/9e84018/ : SUCCESS in 1h 31m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/42/670342/2/check/octavia-v2-dsvm-scenario-two-node/9bc165b/ : SUCCESS in 1h 42m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/42/670342/2/check/octavia-v2-dsvm-py2-scenario-two-node/05dbf1d/ : SUCCESS in 1h 45m 13s (non-voting)\n- octavia-grenade http://logs.openstack.org/42/670342/2/check/octavia-grenade/f6ed1d4/ : SUCCESS in 1h 13m 56s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/42/670342/2/check/octavia-v2-dsvm-tls-barbican/96c3950/ : SUCCESS in 45m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/42/670342/2/check/octavia-v2-dsvm-scenario-amphora-v2/15c65ef/ : SUCCESS in 1h 20m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1a22426247443d41e8d0b5e86d817d76ffbede01","author":{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},"date":"2019-07-14 08:17:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"be0ef0079e96ea11d5da7b955248b23628e9a3ff","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-14 10:38:20.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2\n\nYup. we default auth_strategy to Keystone since pike, and this warning looks valuable.\nLGTM.","accounts_in_message":[],"_revision_number":2},{"id":"8718a07a7ece9990e609d34cfcc37ce5dbb04a75","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-07-14 13:30:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ea1acd244627cbfb5ab5c8a2dca555b6b1b32b75","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-07-16 03:52:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"49a6a70ecd577ccf1a65e33614dbe14c8ac3aae2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 03:53:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bd659d927ea8c8b8d3617b023fbb9bf2abf8ace5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 05:31:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/42/670342/2/gate/openstack-tox-lower-constraints/7f36d7d/ : SUCCESS in 6m 07s\n- openstack-tox-pep8 http://logs.openstack.org/42/670342/2/gate/openstack-tox-pep8/cc06bd4/ : SUCCESS in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/42/670342/2/gate/openstack-tox-py27/2489f74/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/42/670342/2/gate/openstack-tox-py36/9955976/ : SUCCESS in 4m 47s\n- openstack-tox-py37 http://logs.openstack.org/42/670342/2/gate/openstack-tox-py37/58fb66c/ : SUCCESS in 4m 32s\n- openstack-tox-docs http://logs.openstack.org/42/670342/2/gate/openstack-tox-docs/84829f6/html/ : SUCCESS in 5m 05s\n- openstack-tox-functional http://logs.openstack.org/42/670342/2/gate/openstack-tox-functional/77e8f69/ : SUCCESS in 4m 32s\n- openstack-tox-functional-py36 http://logs.openstack.org/42/670342/2/gate/openstack-tox-functional-py36/737255d/ : SUCCESS in 5m 15s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/42/670342/2/gate/octavia-v2-dsvm-noop-api/c35214b/ : SUCCESS in 42m 11s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/42/670342/2/gate/octavia-v2-dsvm-noop-py2-api/d9a7ad7/ : SUCCESS in 37m 15s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/42/670342/2/gate/octavia-v2-dsvm-scenario/e094139/ : SUCCESS in 1h 31m 35s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/42/670342/2/gate/octavia-v2-dsvm-py2-scenario/68b5d93/ : SUCCESS in 1h 18m 14s\n- octavia-grenade http://logs.openstack.org/42/670342/2/gate/octavia-grenade/c0c7934/ : SUCCESS in 1h 13m 43s","accounts_in_message":[],"_revision_number":2},{"id":"a7086568c61422272cfdba5805f74907e6ca7c20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 05:31:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c8004c8c9635d5eb4f83ed5873d0ae159b61d481","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 13:50:48.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit b88f104116a9df74e8d224cc69f69c0d668c2e76","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"48371c98ee49807b5c2b73a442d649bad40b73ec","revisions":{"bf21a7e41fe7b837a1b468df43e1681dcdd76c46":{"kind":"REWORK","_number":1,"created":"2019-07-11 16:09:32.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/42/670342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/670342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/670342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/670342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/670342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/670342/1"}}},"commit":{"parents":[{"commit":"e3aacb67ada526501a8c080c219ee2be869b959b","subject":"Switch the default Ubuntu kernel to -virtual","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3aacb67ada526501a8c080c219ee2be869b959b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-11 16:06:23.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-11 16:09:31.000000000","tz":-420},"subject":"Add warning log if auth_strategy is not keystone","message":"Add warning log if auth_strategy is not keystone\n\nA user came to the IRC channel with CLI errors:\n\"Client-side error: Validation failure: Missing project ID in\nrequest where one is required.\"\n\nThe root cause was the [api_settings] auth_strategy was set to\n\"noauth\" instead of \"keystone\".\n\nThis patch adds a warning log message to the API process that\nwarns users that typically the auth_strategy should be set to\nkeystone.\n\nChange-Id: I7793d7a9113b23ac88e7c53d5dc292a70b9453b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf21a7e41fe7b837a1b468df43e1681dcdd76c46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf21a7e41fe7b837a1b468df43e1681dcdd76c46"}]},"branch":"refs/heads/master"},"48371c98ee49807b5c2b73a442d649bad40b73ec":{"kind":"REWORK","_number":2,"created":"2019-07-11 16:15:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/42/670342/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/670342/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/670342/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/670342/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/670342/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/670342/2"}}},"commit":{"parents":[{"commit":"e3aacb67ada526501a8c080c219ee2be869b959b","subject":"Switch the default Ubuntu kernel to -virtual","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3aacb67ada526501a8c080c219ee2be869b959b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-11 16:06:23.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-11 16:15:00.000000000","tz":-420},"subject":"Add warning log if auth_strategy is not keystone","message":"Add warning log if auth_strategy is not keystone\n\nA user came to the IRC channel with CLI errors:\n\"Client-side error: Validation failure: Missing project ID in\nrequest where one is required.\"\n\nThe root cause was the [api_settings] auth_strategy was set to\n\"noauth\" instead of \"keystone\".\n\nThis patch adds a warning log message to the API process that\nwarns users that typically the auth_strategy should be set to\nkeystone.\nIt also points the user to have an administrator check the keystone\nsettings in the octavia.conf.\n\nChange-Id: I7793d7a9113b23ac88e7c53d5dc292a70b9453b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48371c98ee49807b5c2b73a442d649bad40b73ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48371c98ee49807b5c2b73a442d649bad40b73ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
