)]}'
{"id":"openstack%2Foctavia~670986","triplet_id":"openstack%2Foctavia~master~I25030e46ced9b1c77fad543aa9285c053a388f14","project":"openstack/octavia","branch":"master","topic":"openstackdocstheme","hashtags":[],"change_id":"I25030e46ced9b1c77fad543aa9285c053a388f14","subject":"Bump the openstackdocstheme extension to 1.20","status":"MERGED","created":"2019-07-16 09:10:36.000000000","updated":"2019-07-30 12:38:48.000000000","submitted":"2019-07-30 12:38:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670986-1564490328519-622e802a","meta_rev_id":"6086bdabe7c423b6b2f8999572efaed4e07256a0","_number":670986,"virtual_id_number":670986,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"date":"2019-07-23 22:04:53.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-07-30 12:38:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-07-26 21:55:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-07-29 10:05:58.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-07-30 12:26:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 22:04:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-07-29 10:05:58.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-07-30 12:26:00.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-07-30 12:38:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6cfd4c45cee5791133f7cc0a4d360dcd064bbfe6","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-16 09:10:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8e341c499373ec3f2c98c0cad2d7805679d41aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 09:19:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/86/670986/1/check/build-openstack-api-ref/9bca9c7/html/ : SUCCESS in 3m 29s\n- requirements-check http://logs.openstack.org/86/670986/1/check/requirements-check/20873ab/ : SUCCESS in 2m 36s\n- openstack-tox-pep8 http://logs.openstack.org/86/670986/1/check/openstack-tox-pep8/08fd9a6/ : SUCCESS in 5m 02s\n- openstack-tox-docs http://logs.openstack.org/86/670986/1/check/openstack-tox-docs/f07a5ac/html/ : SUCCESS in 5m 28s\n- build-openstack-releasenotes http://logs.openstack.org/86/670986/1/check/build-openstack-releasenotes/a2ad822/html/ : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"45be20f75d6c52c770f3a2f1b66554ad5e42a1bb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-19 22:41:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nNot sure if the missing version is important or not, but calling it out as a change with this patch.","accounts_in_message":[],"_revision_number":1},{"id":"57b1fa92eba1ae106354c24b0dcddac75675bb9d","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-22 01:44:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"814ad355f8611439d25cfcf57a2349788d7437bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 01:54:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref finger://ze02.openstack.org/63d3f5c117ea48d5a20148b885aa2c3d : POST_FAILURE in 3m 09s\n- requirements-check finger://ze01.openstack.org/c5c75da2b8de4e9bb1d9e4e9cce37849 : POST_FAILURE in 2m 00s\n- openstack-tox-pep8 finger://ze03.openstack.org/dda16d28eefd450d8c555cec1b7d0b28 : POST_FAILURE in 4m 39s\n- openstack-tox-docs finger://ze02.openstack.org/b771563caf2d42aca0b49d3cf4d58267 : POST_FAILURE in 4m 36s\n- build-openstack-releasenotes finger://ze02.openstack.org/020272bea7ab40c2a798b49c9bb8dff2 : POST_FAILURE in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"3decd1405991bfb697481870409fc2268a9144a7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-23 22:04:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2ed57faed3245cb60715adb27c5ea83789223b7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 22:13:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/86/670986/2/check/build-openstack-api-ref/18290b7/html/ : SUCCESS in 4m 34s\n- requirements-check http://logs.openstack.org/86/670986/2/check/requirements-check/b07897f/ : SUCCESS in 2m 32s\n- openstack-tox-pep8 http://logs.openstack.org/86/670986/2/check/openstack-tox-pep8/e65da6a/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/86/670986/2/check/openstack-tox-docs/570a64d/html/ : SUCCESS in 5m 22s\n- build-openstack-releasenotes http://logs.openstack.org/86/670986/2/check/build-openstack-releasenotes/6b4820a/html/ : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":2},{"id":"6a481f832cf9e05ed9fbde98f7d064fbf96def21","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-26 21:55:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dfeefee85c330f2b30f5e79b792038c178da720b","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-07-29 10:05:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"21b7f3103611d3b440535d3529efa7e0fe29fff9","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-30 12:26:00.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"09521f8b2d99117275a69bc1929796627c897913","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 12:26:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"591b02e7f53ba25c1a628dd82fde776899e90665","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 12:38:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/86/670986/2/gate/build-openstack-api-ref/ee9b8bb/html/ : SUCCESS in 4m 17s\n- requirements-check http://logs.openstack.org/86/670986/2/gate/requirements-check/c4abd80/ : SUCCESS in 3m 05s\n- openstack-tox-pep8 http://logs.openstack.org/86/670986/2/gate/openstack-tox-pep8/8222643/ : SUCCESS in 5m 58s\n- openstack-tox-docs http://logs.openstack.org/86/670986/2/gate/openstack-tox-docs/60a0fef/html/ : SUCCESS in 6m 37s\n- build-openstack-releasenotes http://logs.openstack.org/86/670986/2/gate/build-openstack-releasenotes/fb6815d/html/ : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"1eb3ab8f882cf0710edde20bfd2957156df82443","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 12:38:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6b056dac251350add9e09002577899cb26bb914b","revisions":{"ab7fa2b40d64e02bd8472e16ee6903d3e9b4278c":{"kind":"REWORK","_number":1,"created":"2019-07-16 09:10:36.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/86/670986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/670986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/670986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/670986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/670986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/670986/1"}}},"commit":{"parents":[{"commit":"58a4490bbf44df30bcf1c38adfedf1923fdbaeeb","subject":"Merge \"Add warning log if auth_strategy is not keystone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58a4490bbf44df30bcf1c38adfedf1923fdbaeeb"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-16 09:07:09.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-16 09:10:43.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: I25030e46ced9b1c77fad543aa9285c053a388f14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab7fa2b40d64e02bd8472e16ee6903d3e9b4278c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab7fa2b40d64e02bd8472e16ee6903d3e9b4278c"}]},"branch":"refs/heads/master"},"6b056dac251350add9e09002577899cb26bb914b":{"kind":"REWORK","_number":2,"created":"2019-07-22 01:44:43.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/86/670986/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/670986/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/670986/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/670986/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/670986/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/670986/2"}}},"commit":{"parents":[{"commit":"58a4490bbf44df30bcf1c38adfedf1923fdbaeeb","subject":"Merge \"Add warning log if auth_strategy is not keystone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58a4490bbf44df30bcf1c38adfedf1923fdbaeeb"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-16 09:07:09.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-22 01:44:14.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: I25030e46ced9b1c77fad543aa9285c053a388f14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6b056dac251350add9e09002577899cb26bb914b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6b056dac251350add9e09002577899cb26bb914b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
