)]}'
{"id":"openstack%2Foctavia~672102","triplet_id":"openstack%2Foctavia~master~I9f617159eddc8af532403e47c325048d92cc9ade","project":"openstack/octavia","branch":"master","topic":"retire-api-site","hashtags":[],"change_id":"I9f617159eddc8af532403e47c325048d92cc9ade","subject":"Update api-ref location","status":"MERGED","created":"2019-07-22 16:49:36.000000000","updated":"2019-07-23 08:25:38.000000000","submitted":"2019-07-23 08:25:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672102-1563870338685-358cd872","meta_rev_id":"ec6dfff2c93be45cfcf1a8f346fcbf1242689c7a","_number":672102,"virtual_id_number":672102,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-07-23 08:25:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-07-23 06:33:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-07-22 19:48:05.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-07-23 06:33:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 16:53:50.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-07-22 19:48:05.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-07-23 06:33:55.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-07-23 08:25:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86641f2547ebcb4df7be4eb1cd76951b29c8d47a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-22 16:49:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f3cc2d6090f67ec394dea1c27fd1bb7925d380b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-22 16:53:50.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"9aa181f00ba0001cee81f0942bffac040cecc954","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-22 17:18:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c0c914676cdba0c03903d6f819a1095eee6c7be5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 19:26:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/02/672102/2/check/build-openstack-api-ref/903d8fe/html/ : SUCCESS in 4m 47s\n- openstack-tox-cover http://logs.openstack.org/02/672102/2/check/openstack-tox-cover/5ede215/cover/ : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/672102/2/check/openstack-tox-lower-constraints/f16994d/ : SUCCESS in 7m 30s\n- openstack-tox-pep8 http://logs.openstack.org/02/672102/2/check/openstack-tox-pep8/613363b/ : SUCCESS in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/02/672102/2/check/openstack-tox-py27/a9b242c/ : SUCCESS in 3m 46s\n- openstack-tox-py36 http://logs.openstack.org/02/672102/2/check/openstack-tox-py36/97cf8e3/ : SUCCESS in 3m 28s\n- openstack-tox-py37 http://logs.openstack.org/02/672102/2/check/openstack-tox-py37/9c06e02/ : SUCCESS in 6m 09s\n- openstack-tox-docs http://logs.openstack.org/02/672102/2/check/openstack-tox-docs/65f1dd7/html/ : SUCCESS in 7m 01s\n- build-openstack-releasenotes http://logs.openstack.org/02/672102/2/check/build-openstack-releasenotes/8c86492/html/ : SUCCESS in 5m 54s\n- openstack-tox-functional http://logs.openstack.org/02/672102/2/check/openstack-tox-functional/8ead7e5/ : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 http://logs.openstack.org/02/672102/2/check/openstack-tox-functional-py36/de16975/ : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/02/672102/2/check/octavia-v2-dsvm-noop-api/2f4a0c5/ : SUCCESS in 50m 44s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/02/672102/2/check/octavia-v2-dsvm-noop-py2-api/c72eeea/ : SUCCESS in 44m 36s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/02/672102/2/check/octavia-v2-dsvm-scenario/f4149b7/ : SUCCESS in 1h 30m 34s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/02/672102/2/check/octavia-v2-dsvm-py2-scenario/e5133e5/ : SUCCESS in 1h 37m 06s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/02/672102/2/check/octavia-v2-dsvm-py2-scenario-centos-7/a212db5/ : FAILURE in 1h 59m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/02/672102/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/422aba4/ : SUCCESS in 1h 42m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/02/672102/2/check/octavia-v2-dsvm-scenario-two-node/e8ab4eb/ : SUCCESS in 1h 57m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/02/672102/2/check/octavia-v2-dsvm-py2-scenario-two-node/a254199/ : SUCCESS in 1h 51m 04s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/02/672102/2/check/octavia-v2-dsvm-tls-barbican/43dbb17/ : SUCCESS in 57m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/02/672102/2/check/octavia-v2-dsvm-scenario-amphora-v2/b1f02c6/ : SUCCESS in 1h 34m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0058c83f5ee841a9800f9c711da44feb0dbe8388","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-22 19:48:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ff88cc4d444eefd4262c9ce8c497e038bdd830df","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-23 06:33:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"25160d1d994d34c8febf68c83a32a65fac1e8020","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 06:34:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b241368a2378e770c56b468f1c85fb5f61e6863e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 08:25:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/02/672102/2/gate/build-openstack-api-ref/b675c14/html/ : SUCCESS in 4m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/672102/2/gate/openstack-tox-lower-constraints/85ef15f/ : SUCCESS in 8m 10s\n- openstack-tox-pep8 http://logs.openstack.org/02/672102/2/gate/openstack-tox-pep8/26af8ac/ : SUCCESS in 5m 58s\n- openstack-tox-py27 http://logs.openstack.org/02/672102/2/gate/openstack-tox-py27/26f209a/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/02/672102/2/gate/openstack-tox-py36/2663169/ : SUCCESS in 5m 04s\n- openstack-tox-py37 http://logs.openstack.org/02/672102/2/gate/openstack-tox-py37/5cf66b5/ : SUCCESS in 4m 19s\n- openstack-tox-docs http://logs.openstack.org/02/672102/2/gate/openstack-tox-docs/f1b2d47/html/ : SUCCESS in 5m 13s\n- build-openstack-releasenotes http://logs.openstack.org/02/672102/2/gate/build-openstack-releasenotes/7b73831/html/ : SUCCESS in 4m 25s\n- openstack-tox-functional http://logs.openstack.org/02/672102/2/gate/openstack-tox-functional/54dffb9/ : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 http://logs.openstack.org/02/672102/2/gate/openstack-tox-functional-py36/2538f2c/ : SUCCESS in 6m 10s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/02/672102/2/gate/octavia-v2-dsvm-noop-api/e2088e3/ : SUCCESS in 44m 33s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/02/672102/2/gate/octavia-v2-dsvm-noop-py2-api/58a5cbc/ : SUCCESS in 46m 56s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/02/672102/2/gate/octavia-v2-dsvm-scenario/92ae742/ : SUCCESS in 1h 45m 44s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/02/672102/2/gate/octavia-v2-dsvm-py2-scenario/a899497/ : SUCCESS in 1h 42m 08s","accounts_in_message":[],"_revision_number":2},{"id":"05521b140e7d402aadc9f70d5f29ac8fadd70f29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 08:25:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bc19e07923f25c34243491aedc177df6c4b962fe","revisions":{"8331e076fdbc034cbb1ecdcb7acfc348e462b7df":{"kind":"REWORK","_number":1,"created":"2019-07-22 16:49:36.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/02/672102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/672102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/672102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/672102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/672102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/672102/1"}}},"commit":{"parents":[{"commit":"2bb2f918ee4fb38b7a17fb6f8781042d204b05e8","subject":"Fix cryptsetup --pbkdf-memory failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2bb2f918ee4fb38b7a17fb6f8781042d204b05e8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-22 16:49:27.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-22 16:49:27.000000000","tz":120},"subject":"Update api-ref location","message":"Update api-ref location\n\nThe api documentation is now published on docs.openstack.org instead\nof developer.openstack.org. Update all links that are changed to the\nnew location.\n\nNote that redirects will be set up as well but let\u0027s point now to the\nnew location.\n\nFor details, see:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-July/007828.html\n\nChange-Id: I9f617159eddc8af532403e47c325048d92cc9ade\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8331e076fdbc034cbb1ecdcb7acfc348e462b7df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8331e076fdbc034cbb1ecdcb7acfc348e462b7df"}]},"branch":"refs/heads/master"},"bc19e07923f25c34243491aedc177df6c4b962fe":{"kind":"REWORK","_number":2,"created":"2019-07-22 17:18:49.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/02/672102/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/672102/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/672102/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/672102/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/672102/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/672102/2"}}},"commit":{"parents":[{"commit":"2bb2f918ee4fb38b7a17fb6f8781042d204b05e8","subject":"Fix cryptsetup --pbkdf-memory failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2bb2f918ee4fb38b7a17fb6f8781042d204b05e8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-22 16:49:27.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-22 17:18:40.000000000","tz":120},"subject":"Update api-ref location","message":"Update api-ref location\n\nThe api documentation is now published on docs.openstack.org instead\nof developer.openstack.org. Update all links that are changed to the\nnew location.\n\nNote that redirects will be set up as well but let\u0027s point now to the\nnew location.\n\nFor details, see:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-July/007828.html\n\nChange-Id: I9f617159eddc8af532403e47c325048d92cc9ade\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc19e07923f25c34243491aedc177df6c4b962fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc19e07923f25c34243491aedc177df6c4b962fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
