)]}'
{"id":"openstack%2Foctavia~672477","triplet_id":"openstack%2Foctavia~master~Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc","project":"openstack/octavia","branch":"master","topic":"source_ip_port_octavia_ovn","hashtags":[],"change_id":"Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc","subject":"Validate supported LB algorithm in Amphora provider drivers","status":"MERGED","created":"2019-07-24 09:17:17.000000000","updated":"2019-09-26 14:23:07.000000000","submitted":"2019-09-26 14:21:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672477-1569507673633-9cafdce8","meta_rev_id":"43cf2a3ec34f37b9dab9dfdd4fe4cadbf83c7432","_number":672477,"virtual_id_number":672477,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 14:21:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-26 07:03:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-26 07:03:18.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-09-26 12:28:25.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-13 07:37:43.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-09-19 17:44:06.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-23 20:55:03.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-26 12:28:25.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-26 14:21:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"003ddba87c7543e457856a2e2b1c5ca1ad632439","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-24 09:17:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e82a086871bcdc4de22ba195022cbc1d72755d37","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-24 09:17:39.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c75d0a8b6c967759ec4555a9cc8eef950a9fe343","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 10:11:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/77/672477/2/check/openstack-tox-cover/8f5c9bb/ : FAILURE in 4m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/672477/2/check/openstack-tox-lower-constraints/c12ba73/ : FAILURE in 5m 05s\n- openstack-tox-pep8 http://logs.openstack.org/77/672477/2/check/openstack-tox-pep8/ee1b7ec/ : SUCCESS in 5m 15s\n- openstack-tox-py27 http://logs.openstack.org/77/672477/2/check/openstack-tox-py27/fa19055/ : FAILURE in 5m 56s\n- openstack-tox-py36 http://logs.openstack.org/77/672477/2/check/openstack-tox-py36/3a1bca4/ : FAILURE in 3m 32s\n- openstack-tox-py37 http://logs.openstack.org/77/672477/2/check/openstack-tox-py37/4ca8e28/ : FAILURE in 5m 14s\n- openstack-tox-docs http://logs.openstack.org/77/672477/2/check/openstack-tox-docs/6cce438/ : POST_FAILURE in 5m 17s\n- openstack-tox-functional http://logs.openstack.org/77/672477/2/check/openstack-tox-functional/76fa2c0/ : FAILURE in 3m 24s\n- openstack-tox-functional-py36 http://logs.openstack.org/77/672477/2/check/openstack-tox-functional-py36/90c6762/ : FAILURE in 4m 28s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/77/672477/2/check/octavia-v2-dsvm-noop-api/aee74c1/ : SUCCESS in 49m 06s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/77/672477/2/check/octavia-v2-dsvm-noop-py2-api/2716694/ : SUCCESS in 49m 02s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/77/672477/2/check/octavia-v2-dsvm-scenario/4b20e2b/ : FAILURE in 30m 50s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/77/672477/2/check/octavia-v2-dsvm-py2-scenario/671930f/ : FAILURE in 31m 14s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/77/672477/2/check/octavia-v2-dsvm-py2-scenario-centos-7/bdf18fb/ : FAILURE in 16m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/77/672477/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/2658080/ : FAILURE in 38m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/77/672477/2/check/octavia-v2-dsvm-scenario-two-node/b439021/ : FAILURE in 39m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/77/672477/2/check/octavia-v2-dsvm-py2-scenario-two-node/12cbc8c/ : FAILURE in 38m 13s (non-voting)\n- octavia-grenade http://logs.openstack.org/77/672477/2/check/octavia-grenade/fc7caf9/ : FAILURE in 44m 26s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/77/672477/2/check/octavia-v2-dsvm-tls-barbican/0dfd3e5/ : FAILURE in 35m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/77/672477/2/check/octavia-v2-dsvm-scenario-amphora-v2/ef8b28c/ : FAILURE in 31m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f7315a1c119bfd4258da14393ec94207d8a0939b","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-07-26 11:46:36.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"7c74fdae960e60f772361bed4546eab897b72a96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 14:11:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/77/672477/3/check/openstack-tox-cover/ed7c995/cover/ : SUCCESS in 8m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/672477/3/check/openstack-tox-lower-constraints/caf10ea/ : SUCCESS in 7m 17s\n- openstack-tox-pep8 http://logs.openstack.org/77/672477/3/check/openstack-tox-pep8/5e075fe/ : SUCCESS in 6m 52s\n- openstack-tox-py27 http://logs.openstack.org/77/672477/3/check/openstack-tox-py27/fe7d164/ : SUCCESS in 5m 55s\n- openstack-tox-py36 http://logs.openstack.org/77/672477/3/check/openstack-tox-py36/00ca680/ : SUCCESS in 4m 58s\n- openstack-tox-py37 http://logs.openstack.org/77/672477/3/check/openstack-tox-py37/448421b/ : SUCCESS in 5m 15s\n- openstack-tox-docs http://logs.openstack.org/77/672477/3/check/openstack-tox-docs/85c2b8b/html/ : SUCCESS in 6m 21s\n- openstack-tox-functional http://logs.openstack.org/77/672477/3/check/openstack-tox-functional/fb851af/ : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 http://logs.openstack.org/77/672477/3/check/openstack-tox-functional-py36/6ee8fbc/ : SUCCESS in 6m 33s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/77/672477/3/check/octavia-v2-dsvm-noop-api/f10c77a/ : SUCCESS in 45m 45s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/77/672477/3/check/octavia-v2-dsvm-noop-py2-api/9e27267/ : SUCCESS in 46m 26s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/77/672477/3/check/octavia-v2-dsvm-scenario/b355aa8/ : SUCCESS in 1h 47m 50s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/77/672477/3/check/octavia-v2-dsvm-py2-scenario/4db3685/ : SUCCESS in 1h 49m 50s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/77/672477/3/check/octavia-v2-dsvm-py2-scenario-centos-7/53422bd/ : FAILURE in 2h 18m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/77/672477/3/check/octavia-v2-dsvm-scenario-ubuntu-bionic/a553d74/ : SUCCESS in 1h 33m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/77/672477/3/check/octavia-v2-dsvm-scenario-two-node/f5eb8f5/ : SUCCESS in 2h 03m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/77/672477/3/check/octavia-v2-dsvm-py2-scenario-two-node/aa686f0/ : SUCCESS in 2h 06m 15s (non-voting)\n- octavia-grenade http://logs.openstack.org/77/672477/3/check/octavia-grenade/9ff0820/ : SUCCESS in 1h 16m 05s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/77/672477/3/check/octavia-v2-dsvm-tls-barbican/c5d338c/ : SUCCESS in 1h 01m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/77/672477/3/check/octavia-v2-dsvm-scenario-amphora-v2/f8f75fc/ : SUCCESS in 1h 32m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fc6e78ef3b78779313cf41d2bb33cd402d73f579","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 13:40:01.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0d9b6f40bcfec31cf9e48cdbd6c5810d6f0a2ad0","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-05 08:19:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"db3f554d24832673a38502952e7c9acd89374cab","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-05 08:19:15.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ce162da25af8f87b197902d4bb19ae59f8e71a8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 10:40:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/77/672477/4/check/openstack-tox-cover/a13b94d/ : FAILURE in 7m 12s\n- openstack-tox-lower-constraints https://logs.opendev.org/77/672477/4/check/openstack-tox-lower-constraints/027e0a4/ : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://logs.opendev.org/77/672477/4/check/openstack-tox-pep8/964782a/ : SUCCESS in 7m 21s\n- openstack-tox-py27 https://logs.opendev.org/77/672477/4/check/openstack-tox-py27/5af555f/ : FAILURE in 8m 05s\n- openstack-tox-py36 https://logs.opendev.org/77/672477/4/check/openstack-tox-py36/fec6810/ : FAILURE in 6m 53s\n- openstack-tox-py37 https://logs.opendev.org/77/672477/4/check/openstack-tox-py37/85b5c4a/ : FAILURE in 12m 25s\n- openstack-tox-docs https://logs.opendev.org/77/672477/4/check/openstack-tox-docs/f557f75/html/ : SUCCESS in 6m 57s\n- openstack-tox-functional https://logs.opendev.org/77/672477/4/check/openstack-tox-functional/81dfb6c/ : SUCCESS in 5m 53s\n- openstack-tox-functional-py36 https://logs.opendev.org/77/672477/4/check/openstack-tox-functional-py36/dc45ab0/ : SUCCESS in 6m 33s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/77/672477/4/check/octavia-v2-dsvm-noop-api/4aaf0a5/ : SUCCESS in 40m 26s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/77/672477/4/check/octavia-v2-dsvm-noop-py2-api/658ff96/ : SUCCESS in 45m 35s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/77/672477/4/check/octavia-v2-dsvm-scenario/ca5c6ee/ : SUCCESS in 1h 42m 46s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/77/672477/4/check/octavia-v2-dsvm-py2-scenario/aec2a4f/ : SUCCESS in 1h 51m 39s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/77/672477/4/check/octavia-v2-dsvm-py2-scenario-centos-7/41e1c21/ : FAILURE in 2h 15m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/77/672477/4/check/octavia-v2-dsvm-scenario-two-node/b53cc03/ : SUCCESS in 1h 54m 01s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/77/672477/4/check/octavia-v2-dsvm-py2-scenario-two-node/8ef90d0/ : SUCCESS in 1h 55m 31s (non-voting)\n- octavia-grenade https://logs.opendev.org/77/672477/4/check/octavia-grenade/3c8a774/ : SUCCESS in 1h 21m 26s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/77/672477/4/check/octavia-v2-dsvm-tls-barbican/0c0f391/ : SUCCESS in 1h 25m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/77/672477/4/check/octavia-v2-dsvm-scenario-amphora-v2/06e9116/ : SUCCESS in 1h 37m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/77/672477/4/check/octavia-v2-act-stdby-iptables-dsvm-scenario/d2c13cb/ : FAILURE in 1h 24m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/77/672477/4/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/f166fa8/ : FAILURE in 1h 20m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/77/672477/4/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/2280407/ : FAILURE in 1h 16m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"df12b1fe200a0fc98d79b279c197159d4127ab59","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-07 07:51:23.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"913eaeef25d51f4ab4886d1dfe8cf3b50447944b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 10:34:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/77/672477/5/check/openstack-tox-cover/0b6f24b/ : FAILURE in 7m 34s\n- openstack-tox-lower-constraints https://logs.opendev.org/77/672477/5/check/openstack-tox-lower-constraints/5260207/ : FAILURE in 8m 25s\n- openstack-tox-pep8 https://logs.opendev.org/77/672477/5/check/openstack-tox-pep8/d12289a/ : SUCCESS in 6m 12s\n- openstack-tox-py27 https://logs.opendev.org/77/672477/5/check/openstack-tox-py27/0510fc9/ : SUCCESS in 4m 34s\n- openstack-tox-py36 https://logs.opendev.org/77/672477/5/check/openstack-tox-py36/1aa8811/ : SUCCESS in 6m 29s\n- openstack-tox-py37 https://logs.opendev.org/77/672477/5/check/openstack-tox-py37/0090d6a/ : SUCCESS in 4m 25s\n- openstack-tox-docs https://logs.opendev.org/77/672477/5/check/openstack-tox-docs/75f5bbc/html/ : SUCCESS in 5m 42s\n- openstack-tox-functional https://logs.opendev.org/77/672477/5/check/openstack-tox-functional/2f7cf4c/ : FAILURE in 5m 10s\n- openstack-tox-functional-py36 https://logs.opendev.org/77/672477/5/check/openstack-tox-functional-py36/0a76944/ : FAILURE in 6m 03s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/77/672477/5/check/octavia-v2-dsvm-noop-api/8c35031/ : SUCCESS in 1h 23m 35s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/77/672477/5/check/octavia-v2-dsvm-noop-py2-api/9ff95fb/ : SUCCESS in 49m 30s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/77/672477/5/check/octavia-v2-dsvm-scenario/993b379/ : SUCCESS in 1h 41m 09s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/77/672477/5/check/octavia-v2-dsvm-py2-scenario/ebcea5a/ : SUCCESS in 1h 49m 40s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/77/672477/5/check/octavia-v2-dsvm-py2-scenario-centos-7/5242f30/ : TIMED_OUT in 2h 34m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/77/672477/5/check/octavia-v2-dsvm-scenario-two-node/862f4dd/ : SUCCESS in 1h 21m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/77/672477/5/check/octavia-v2-dsvm-py2-scenario-two-node/f2663c4/ : SUCCESS in 1h 55m 05s (non-voting)\n- octavia-grenade https://logs.opendev.org/77/672477/5/check/octavia-grenade/4df6e80/ : SUCCESS in 1h 29m 03s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/77/672477/5/check/octavia-v2-dsvm-tls-barbican/f74b1cf/ : SUCCESS in 57m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/77/672477/5/check/octavia-v2-dsvm-scenario-amphora-v2/2d70bc2/ : SUCCESS in 1h 34m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/77/672477/5/check/octavia-v2-act-stdby-iptables-dsvm-scenario/21386d5/ : SUCCESS in 1h 11m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/77/672477/5/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/1fa62ab/ : FAILURE in 1h 19m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/77/672477/5/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/cd5fc7d/ : FAILURE in 1h 23m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2fead0e4b7c136d8b7cb9ed1fe2e15bf886429ee","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-07 12:01:11.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"86d1e73beb8528ee85ca71c64461dd0be6925e82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 14:43:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/77/672477/6/check/openstack-tox-cover/324c722/cover/ : SUCCESS in 9m 57s\n- openstack-tox-lower-constraints https://logs.opendev.org/77/672477/6/check/openstack-tox-lower-constraints/7fad026/ : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://logs.opendev.org/77/672477/6/check/openstack-tox-pep8/224f82f/ : SUCCESS in 6m 32s\n- openstack-tox-py27 https://logs.opendev.org/77/672477/6/check/openstack-tox-py27/5227eac/ : SUCCESS in 6m 09s\n- openstack-tox-py36 https://logs.opendev.org/77/672477/6/check/openstack-tox-py36/5a016a2/ : SUCCESS in 5m 12s\n- openstack-tox-py37 https://logs.opendev.org/77/672477/6/check/openstack-tox-py37/6fd40f4/ : SUCCESS in 5m 03s\n- openstack-tox-docs https://logs.opendev.org/77/672477/6/check/openstack-tox-docs/866fe2b/html/ : SUCCESS in 7m 04s\n- openstack-tox-functional https://logs.opendev.org/77/672477/6/check/openstack-tox-functional/7e38b7b/ : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://logs.opendev.org/77/672477/6/check/openstack-tox-functional-py36/7bf080e/ : SUCCESS in 8m 03s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/77/672477/6/check/octavia-v2-dsvm-noop-api/8fed5bf/ : SUCCESS in 44m 18s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/77/672477/6/check/octavia-v2-dsvm-noop-py2-api/ef0c7e3/ : SUCCESS in 50m 50s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/77/672477/6/check/octavia-v2-dsvm-scenario/094653a/ : SUCCESS in 1h 49m 03s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/77/672477/6/check/octavia-v2-dsvm-py2-scenario/2192b2b/ : SUCCESS in 1h 55m 15s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/77/672477/6/check/octavia-v2-dsvm-py2-scenario-centos-7/1be76e8/ : FAILURE in 2h 38m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/77/672477/6/check/octavia-v2-dsvm-scenario-two-node/b773808/ : SUCCESS in 1h 59m 05s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/77/672477/6/check/octavia-v2-dsvm-py2-scenario-two-node/829f0be/ : TIMED_OUT in 2h 35m 00s (non-voting)\n- octavia-grenade https://logs.opendev.org/77/672477/6/check/octavia-grenade/78395d1/ : SUCCESS in 1h 04m 44s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/77/672477/6/check/octavia-v2-dsvm-tls-barbican/f9ae875/ : SUCCESS in 57m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/77/672477/6/check/octavia-v2-dsvm-scenario-amphora-v2/21042ba/ : SUCCESS in 1h 42m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/77/672477/6/check/octavia-v2-act-stdby-iptables-dsvm-scenario/5c36fcb/ : SUCCESS in 1h 08m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/77/672477/6/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/4beff0d/ : SUCCESS in 1h 04m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/77/672477/6/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/c2ad762/ : FAILURE in 1h 17m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"737931e3653642ff2041b5beae1937a9f3750fba","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-28 08:08:02.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"9e4b1b57304bf0f3f08cf92e377f206817e599a2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-04 20:32:59.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"d382fd4c1b995f9346d258aaa0b6e68a94b29489","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 01:16:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d936e47622d44588a309522bc257b1bd : SUCCESS in 11m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61b0737c56994d3482f29c42b0d96c08 : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1baa6cbbabde450a8c0fcd54444a1218 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b750586e1d2346e797903592c0416b07 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffbbc58171704b46bfa712ce68d7bd89 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8cc58176c7e47b5a349a2edf759fb46 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97a4ad0ff9264c4eb22656f5c39f46f6 : SUCCESS in 7m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bc2c0f2d3f2849b2bda9eec8bd19e874 : SUCCESS in 5m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/52c0ce21bf94423f9f706dfdd89365ae : SUCCESS in 7m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d7252957cb6a4c8dbbdcaaf17f2fb356 : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a4c5d36989142778218c9234b400517 : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/81219140baf94bf6aad0422cc30208e0 : SUCCESS in 46m 16s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e486164c47fe4b8cb38698e4e382da24 : SUCCESS in 47m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c9dbc57214e4319b027d9d34945ccf2 : RETRY_LIMIT in 1h 04m 26s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8782885319da47f7ad27c1be826434b4 : SUCCESS in 1h 46m 17s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/258548d0dd2a4d299cad9237467ca93c : SUCCESS in 2h 26m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/84a06353aefa459d837e869f95031876 : SUCCESS in 2h 14m 41s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/14d364c249e749618f871c346deec3e8 : SUCCESS in 2h 07m 27s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7b841641b5b34f24baec323e0bf7fdc6 : SUCCESS in 1h 24m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a78bec5a4f494ea9a2b09c11041bc62b : SUCCESS in 56m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5f2878db9b294e76b393003b08cda536 : SUCCESS in 1h 41m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/302789c7e42e4c53805d491d4c4e1f94 : SUCCESS in 1h 18m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/713ac83d336845d2b17550b496adc6e9 : FAILURE in 1h 07m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/dfdf5d6dcf2c4623828f5442a155700c : FAILURE in 1h 08m 07s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"56bfda727c0374400783f81b617fe4dded9e01c6","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-05 07:40:21.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"dc7568afc532ba401ec3e0232b9b3d0427f7c89c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 12:35:21.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd7078a31c9045cca9217bc72f0c2282 : SUCCESS in 7m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/754b7df29322482a9c88156d22c5a08f : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf78ff4b4285468b9b8f232abf22474f : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9d4839fcaebb42d9b7c5bfeb84db1241 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56cebe1ec1f247ca9af87e5295a6f17e : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9740a1549ed04655b491bdb20856e035 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5464bf7f2d24403cbbbe0d8d3d19624b : SUCCESS in 6m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/867575a7dcb947cb856dc3d219076e82 : SUCCESS in 5m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7be5fa983275480080be23e9c25cef24 : SUCCESS in 6m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/776ddb5e524c4f32baa3a95e87d41441 : SUCCESS in 6m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e968a3e750db4cdf9072ef4f45db30b3 : SUCCESS in 6m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5d859dfb4faf45f5837071d7bdfc2afc : SUCCESS in 47m 18s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/26eaaa026a0b41d798f5701b9e5eb15f : SUCCESS in 51m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6fa064022446401ea0ec687aaa7b0682 : RETRY_LIMIT in 59m 48s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2619985b41fb47e8b3bdd3d5aa264743 : SUCCESS in 1h 31m 31s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0029acb220e6453c81463c78c02a283c : SUCCESS in 2h 20m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cca524c9496849fe9c8e37d82f4c0346 : SUCCESS in 1h 57m 02s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/962e9cc1a22f4421bf22aaaad0dd4077 : SUCCESS in 2h 15m 48s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b3defcddd5ec4dd78e2ba1f243fa81f7 : FAILURE in 40m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/09eb158527f24cfb9c9f4652045715aa : SUCCESS in 56m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1e59ae24c21745eaac9bf5b9c60b8834 : FAILURE in 36m 40s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a7fa16934fdf43e89203964143e6017e : SUCCESS in 1h 14m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2490e5fa70244c4ba756856ef88e8c61 : SUCCESS in 1h 05m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f0f106dc74854e3887c954dede3209ca : FAILURE in 1h 01m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"35e6dc512363b8eed292c1afdc96d3941ab4582f","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-05 12:38:18.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"da5c94da448f62ab93563dd4463ab7481fcda0ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 19:12:59.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/160a21f4f3e44df8a31f079c6071557c : SUCCESS in 9m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dba7fab36b574bf7932d29401ac1d0d8 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d685c7573bef4a4bb18cad4e0584d45f : SUCCESS in 9m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/36a29f357fc543298f221c842ce4c0f4 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b09bce0e634049248b4edda59170d83a : SUCCESS in 8m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7a018d0847c414caf78e5f681d7f2d2 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6d2120e238a48a1902745650c5ed790 : SUCCESS in 9m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/edaa8177dd254cfa97d8000dad74dd6b : SUCCESS in 5m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/30d8b313784649ccabd4a11b43d9b026 : SUCCESS in 6m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/32e1b781033c47c098b128f03d00280a : SUCCESS in 5m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8046b311f00f478faef2089f6859f10d : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f6b9cea9ce064cd883c2f50569876239 : SUCCESS in 51m 08s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/73dccd400d134509a0e0e1d10295c464 : SUCCESS in 53m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c63e57e7f984bba85b829bea497c10b : RETRY_LIMIT in 1h 39m 07s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/904d2cf7e1ae4e0196f8b4dbc4c51f1f : RETRY_LIMIT in 1h 17m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/86b557a93bd44a28a030f2288812f372 : TIMED_OUT in 2h 42m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1dd94af9bb7b4072bcf539f4599ac511 : SUCCESS in 2h 01m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/50af0244d25f402085f60481a7ee90b7 : SUCCESS in 2h 03m 26s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1fd8413f362e4a589e30f4ddeb651b0f : FAILURE in 43m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9eedd636d90f4c459279cec12143d8a4 : SUCCESS in 59m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/52b6ddcd518e48e9b37b7d89e78ef64b : RETRY_LIMIT in 56m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/772d46b8712b497bb627e019254f5f2d : SUCCESS in 1h 05m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ccf5025fe118485ea596b2e9f0f1df46 : SUCCESS in 1h 18m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a22b7d48067b492d8e3973b517adfa44 : FAILURE in 1h 32m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2faf3bfcd804798bb0a93139dc532ceb3b48d5f6","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-10 11:23:15.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"ae9299a5973750d95eacbf6264a89a42f024c251","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 15:02:04.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c066663bf00c49a29d79898dc0d3ac6f : FAILURE in 4m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a12d66cfa52045b0af0f41306be4b34c : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa6416f7823a410e9cbfac74d76a84f4 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75d2baa53e5a47df82046a05510e6a71 : FAILURE in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e49645d28ab74674b03677d8816a320a : FAILURE in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29bc23a2eae5429ca6f3c99da9b3d714 : FAILURE in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adb14885f65c4b5c9c15f72b29bf059c : SUCCESS in 6m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/202fb1000b70466abbbf938d76f2c91f : FAILURE in 4m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bc90bdd0c1c94c7fb539a521771ca482 : SUCCESS in 5m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a1509f687e7448428473e606438359f3 : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/98080c98002d416cb650e48b7b2b53d8 : SUCCESS in 7m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/44f20eea0bc843c58f195052f03768c9 : SUCCESS in 45m 51s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/bb4c0430e88443398a6b5ccefce8b0ee : SUCCESS in 46m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fce2fb2f9dbe4f11a936b0a31e0f3758 : RETRY_LIMIT in 1h 00m 36s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ef464665776c4a65866df38763bdb6c9 : FAILURE in 36m 05s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0832f0150bbe46d798602efd1a6417ac : SUCCESS in 2h 04m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/941330538fd64986b9ae6088def9baa6 : SUCCESS in 2h 05m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/cddfe465d5a8413db223225fb4063196 : SUCCESS in 1h 26m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6f47f34a73734d6c87de7f83ee35495f : SUCCESS in 1h 24m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0c88983d41ff46ce8c1aaf4ccb525e2d : SUCCESS in 51m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/be3bad1c41cd4d56bdd8f8abc9f39687 : SUCCESS in 1h 43m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/89134c7b574d417ba0e962cce0ccd944 : FAILURE in 1h 12m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b4df2a72fb3948ab9d9f77cc1e4c04cb : SUCCESS in 1h 13m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f16b7100343d49549e1027df2ef91584 : FAILURE in 1h 06m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b3d97af273aaacc4f05f13227f1a5133815b759c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-11 10:07:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"acf5bb90b33fef34e485aadfe8ffc6501b298bd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 12:54:07.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e130bf2ffaf4879af3fb10749df4c0c : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffb8b04604c344859309672c66d05d6a : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c99143925fa94a1ea62b0c82e7d6361f : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ff2c860cf404c08ac9bab22131f8248 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edada718c9864e6b86fdb81ff3c0cfbb : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0817ca0ce3de41fcabd93630c0ac9e7b : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/793e27d227f245adb83d305cadb62916 : SUCCESS in 6m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/55e76e9361ad47cc8d67a8f8c51a8123 : SUCCESS in 6m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7ee722c4e7254d0e9fe37e2c4d2a2b81 : SUCCESS in 6m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/18440605118a4336afb0740af851eecb : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7b76bd3eb7484b6f90a1c6d623adfd50 : SUCCESS in 6m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/26d8f2293ebe4b68a01cf7a98fb3e8ad : SUCCESS in 48m 08s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/efc691da795d489d853cfcf0dfd2d671 : SUCCESS in 48m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8a4e0c95a7e4a0189655d197aa19191 : SUCCESS in 1h 35m 11s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/72bcfebfcab147a4850cd876fd57be5f : SUCCESS in 1h 47m 47s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/795eefbb0fce4badb927cc4c6b4835a3 : SUCCESS in 2h 30m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/db53230124c445d2a16e85c339e37496 : SUCCESS in 1h 53m 23s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/59299556357e461eb3d54dde67afcada : SUCCESS in 1h 59m 56s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c5edf92c7315462d95cfe920f3b24bd5 : SUCCESS in 1h 21m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b5c8faa90cae49e8ba8868fdac114ada : SUCCESS in 1h 01m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e98be9ffa7aa482b823dfb7d3162c4ff : SUCCESS in 1h 45m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0bc3e1ff49ae474f8d38c386454e1a01 : SUCCESS in 1h 10m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/74c163dbdd064e41bd804df0c69d5612 : SUCCESS in 1h 12m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9801e8c6fefd4ccdb938fd88f3c7fdf2 : SUCCESS in 1h 26m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e5211df331961ccae8aed59162e832446bc1c9fd","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-12 14:00:48.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"4574867943048ebd1b49bc7723d12e9b73a01ec5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 17:38:46.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80813b24a77f42cc91d2d77a59712b8c : FAILURE in 7m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff0efc120ec74ee1aa77e5d2f1dd619f : FAILURE in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86300a89fc5e4b7e950ce7f7fe95880f : FAILURE in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/38e1e6a02c854c85ac826ec024b0b115 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17108e1bc0e8414b9c5a1e8e39ad6256 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09bed2703fab4514ab63dac0a65ab175 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/439dec3bba7d446ead36ed83f2a2d7f8 : SUCCESS in 7m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f6b5e6cf3f2d43bdaa19bf78ddc4a60e : SUCCESS in 4m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/88e1ff5a43474c219795840f81306760 : FAILURE in 5m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/293708aea7624a0797d4172f220744b6 : FAILURE in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c261e45f98cf4f4dab6b7bf3784179c0 : FAILURE in 6m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/444851c459fc44578259f45178417f17 : SUCCESS in 53m 09s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b1ca4ac49a1c4e3d88dc1df24b44dfc3 : SUCCESS in 49m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5fa2a6b58e647ceb9031730eb1f7f6d : SUCCESS in 1h 45m 28s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5684ca19bcda4869acfa6316db77d647 : SUCCESS in 1h 33m 18s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8dad0bbb36794c8d86995fa7a2bceb6f : SUCCESS in 2h 16m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/51912c30b69b48a686f2ef098561c121 : FAILURE in 1h 56m 49s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0ab584bdc8734340adb731087db6bfbd : SUCCESS in 1h 12m 36s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9509541c908147608ec33f66120e96d5 : SUCCESS in 1h 13m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4b4e78f9ce12495aa53dee3605678133 : SUCCESS in 1h 03m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bdd670e05a524a319162e89c108cd9b2 : SUCCESS in 1h 47m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7e548ff9e26348fd9de993efdf234318 : SUCCESS in 1h 08m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/67e839b8cff3437cad982a4371a29ff5 : SUCCESS in 1h 14m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/851606e974ff4ae6ad1754d81a5850ab : FAILURE in 49m 39s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6f5c346727b3a5bf983aea49e19186d66b995091","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-13 07:37:43.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"d28a96762c45464243067d4f616bc64262f567c2","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-13 10:08:49.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"a17c9ae233255eaec84dd209cf54fcc247a1ef44","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-13 12:53:02.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"75d7548eb221d09e760bf0c908965c2b24f83adb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 14:14:59.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a094c9ee92cf426991685cb2446d3ca2 : SUCCESS in 8m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9629cb3bbc04b8cac990ff66cb34649 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c43bb8098a0485b9c156bc26aeb5305 : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1486b72ef258497daf9a2d2bf3fc6207 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/187e574d95bc4e3fad677e5829ff6258 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86a1df30fa8e48e1b8f73b831e399d43 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658ad5553cb344ad858e85d22d66a2e6 : SUCCESS in 7m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0c4867d355cc4b459c99ed66c65ae022 : SUCCESS in 5m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/219ccd89189c4995b9b470391437eff8 : SUCCESS in 6m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/651331557b744b9fb175fb70382d398a : SUCCESS in 5m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/de3247adaf574337bdea47f0ccc63ba3 : SUCCESS in 6m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/47d852e0c96e4319bf271fffc1107e21 : FAILURE in 26m 58s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f216d22cfa864088b2663c4ce9fc395e : FAILURE in 24m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b6a7e6d90f64baebc75126d9516fc6b : FAILURE in 25m 40s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/cf825663ede64bdfbb5bb5a13702b8b0 : FAILURE in 25m 22s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f488b1f7a25640149b4c6e1ba0391ec1 : FAILURE in 30m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/29f145ee9d55480b8e9e3142b840aeb0 : FAILURE in 31m 19s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/632a91903d5140378d5cb3aa12e70eec : FAILURE in 31m 09s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7de10dfb53394832aa87960ca06f3288 : FAILURE in 1h 08m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7d8b48b8675b4d189d81b0213b8e165b : FAILURE in 36m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b270be7e9db04320abb26965729792d1 : FAILURE in 31m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a1da57ff3815493da45cf705d18c4a79 : FAILURE in 25m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8caa1bd8c26b46c688f67bcf72784fb6 : FAILURE in 32m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/cb43d243a30742e3910a255a2729804d : FAILURE in 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"dbc4f353c44743585a96c14d565962a4ac915f8d","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-13 14:54:36.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"1a91bc461c3da96cacef99b5f2d92a3745026aa8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-13 15:25:09.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"dfb5d1d80c3768e24425b16a38758a33965da4df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 17:49:11.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a236d0fe5e34bcf833e6a1660aa59ad : RETRY_LIMIT in 2m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bc1d059b1bf43e081e697a78a6c6fd4 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ac220115c7b4645862dd44a9a5abbbe : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/789794f09ef04f0a8a7a9a3fcf325c96 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a33d3538e9e4cdba8e04c6647b98b28 : SUCCESS in 7m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88bb20bb623845f0b8a0940daa8c6573 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2ecb164df314419825f8893e7e7369d : SUCCESS in 6m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b28530adc3934781a688533b835059f6 : SUCCESS in 6m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/24ff6dacfe3644458d7e77ed25335825 : SUCCESS in 7m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ee7865f634614971bdd3b061ebacd2af : SUCCESS in 6m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/434b93272ac8443e959bb4d343ca0056 : SUCCESS in 5m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/073dc93108ff405e8b4c7a0ea6ba7148 : SUCCESS in 59m 50s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/92d2f3e0b267469fae247e82fef8aed8 : SUCCESS in 55m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/39e6d126e279468e9d2c558b8e19d9dd : SUCCESS in 1h 58m 27s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/752490a0bf6445fdacb2c8575e3c59b9 : SUCCESS in 1h 48m 03s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1def69416946445d839698613fc22f4e : SUCCESS in 2h 32m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4828d84445ca426abf2ea53b74c2e645 : SUCCESS in 2h 06m 47s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/29dba7bd877e4d8a8f87e468b004d6b9 : SUCCESS in 2h 19m 46s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1e6484c3f2aa416982b6266ac49827c9 : SUCCESS in 1h 09m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4711381923334398aa587e44e8ffde53 : SUCCESS in 56m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9b482cfde05b4b65af0e1626224d81ab : SUCCESS in 1h 46m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec6e183166e64c08877296e782a6a455 : SUCCESS in 1h 12m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b8ecc1a258624b2084d87788255212b8 : SUCCESS in 1h 23m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/354fbb5deff1433981bf934191786357 : FAILURE in 1h 31m 23s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a3c009f97f4c0109f30e8d679a822caaac096ceb","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-16 09:21:24.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"b1d416536f31154c4410c94f2260c2fe6ecc84e4","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-16 09:22:01.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"3f86dbfb1bfddc15628007d540f8ad37d2202f74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 11:05:34.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/215664e31c7240e9bb1551d66b144b7b : SUCCESS in 9m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a40f4790b3de466785557eeae5cd547c : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc78353547184810b139d3f01f746c74 : SUCCESS in 6m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ffbd7e582f0042e08fb61457fecfc2b8 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea37facf784d49c5b25b652fbf091480 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/82ff83e1376b4c63b7b9fa843a2f0659 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d88f36bc7764004b4e7fa33e2f635fc : SUCCESS in 6m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3cc949b5b61d4bfb8eb858ca36a8b497 : SUCCESS in 5m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/553a0166f10f4331a5cc8123cbfd635d : SUCCESS in 7m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ceff300b8d75481f862fbe2b954bda15 : SUCCESS in 5m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/66f683f7fdad464b8e0bb70fa7129ef5 : SUCCESS in 5m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/38fe656acae646338fbcafd57502e661 : FAILURE in 30m 17s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/bcfb04bad0284212a2919d4460c7c6f7 : FAILURE in 33m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0f9a11b78944920b9eb150c5e19cde6 : FAILURE in 38m 24s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9d395b0197304b59ada37b60d1269cd2 : FAILURE in 35m 15s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fae92f7851f84e8b83dba793602baa5c : FAILURE in 55m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a845c0e06e6246f784cd7528f464f97a : FAILURE in 36m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/7b15030f36ae432a90df790df03aa99b : FAILURE in 41m 51s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e8a9e059aadd40bab25e02171afadb3a : SUCCESS in 1h 27m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/90016f6215f1427799a568a8101c4388 : FAILURE in 40m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0c7f871b34af4c9699b5c262971b77ec : FAILURE in 39m 40s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1dcf6e9cb7874de986892f80df94bb52 : FAILURE in 32m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4fec7f17079349dda809411b70db9deb : FAILURE in 38m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7920e81871fd46af8e0be435b0f2b67c : FAILURE in 51m 37s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9536a8ec1a81920c716a1b5ab2eb6d01e10c6e5f","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-17 07:05:45.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"6565956f75d4695383c0f5f6f53e4c36f2715e6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 10:55:02.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0aa9e94b26f144599071126c47995a33 : SUCCESS in 8m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ece113c4e194788b87e5065675cdbb7 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ae2eae884554eb6b8cd45c8bf916fc2 : SUCCESS in 6m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/01633bc901d9406099f68abf6e2f9e28 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5823d6bad6d6457fb7f59426b2ec4c82 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/148b072d8f8140e3ad0423fd5a17938d : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/276219063cff406a9431eee392bb20fa : SUCCESS in 8m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1ab15b05a0a545af8e60e549731cb9b7 : SUCCESS in 7m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d38f0a2ad7ae4a7085f0847968f93337 : SUCCESS in 7m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3c9dcc54909f4782b7d419d6b6addc28 : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8519d86d9371496798d46e41f5e9c8cf : SUCCESS in 5m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df22afef575c4f7b829377d53b940d9d : SUCCESS in 47m 18s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2b34274999d0475b95e7df818105866d : SUCCESS in 46m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e42e47707faa40b5a74e016c3e6844f3 : SUCCESS in 1h 28m 25s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/427bba6575b8433c9e8514f4b8390399 : FAILURE in 29m 32s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/16d5edcff9ba48e5bdf43a99cfa59596 : SUCCESS in 2h 14m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/36e2ca8456894b4198cf8042dfd2018c : FAILURE in 1h 55m 32s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/fc8fba55f44b48b696ae85df3b3dab04 : SUCCESS in 2h 13m 19s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8e3d181ebe0f40f6b89ed0965d6d4a99 : SUCCESS in 1h 29m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e1237280c5594c6d8e8227afba78972b : SUCCESS in 1h 01m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9c4d018d4c96407fa8d09b3a6accb516 : SUCCESS in 1h 36m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c277d5efecc408fae0835cc348d8e4e : SUCCESS in 1h 12m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2a147a0b8a2a420ab9c52551d432a0d3 : RETRY_LIMIT in 3m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/bbafce3602fa451da79f76d2c17a8903 : SUCCESS in 1h 28m 47s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e554346b616afbb3f998dfac8bcd4474859a7012","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-18 06:03:59.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"9540d6d8bb7f39f7c012a7c16e110e0cfdf4f8da","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-18 06:05:16.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"56820c4ef50b7014140fbb5cdf96bb5357a4d5f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 10:01:23.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec7320d558e84683aa00e3e14fc18130 : SUCCESS in 8m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3d61e7197194afe8f5d1ac3b85d2981 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15b88b8e2a5e4f30909540a5c4c3010c : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9cccddd55bba48a2b5c2ebaf213904a9 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a4b00e77302401c936b9b3acf367883 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69a8c8dbc34f46b5a2c23b7f4863f35d : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98189f7fa7874fa0850cc347705c39d6 : SUCCESS in 7m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1b548ceb6b41438e8ac58c256940fe78 : SUCCESS in 5m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/66883e4400044d66b8ad324d65c25bd3 : SUCCESS in 8m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5c6cd99b9f7f4b6c89aa69e39e671d96 : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/28cdb90e12d9418ba1b523bee32dfb20 : SUCCESS in 5m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/217f33f75b3146bba72c67e579aee33d : SUCCESS in 52m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/28b524e2b5314ec2813a037581a08559 : SUCCESS in 48m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dcefbcde9430408892e7e1374063860d : SUCCESS in 2h 07m 21s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7bc63af813214defb25315b238c26557 : SUCCESS in 2h 04m 45s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1446941154b14f63afd03af85d82bfa0 : TIMED_OUT in 2h 36m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/57b5cc3659ad408fa30229a4730c26d7 : SUCCESS in 2h 16m 23s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/69903f62279543a4b729b23583bdcb2c : SUCCESS in 2h 03m 33s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d526f8c548694e2fb34347b44ffcae4b : SUCCESS in 1h 30m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9ab4dc8ef6d74e91b02a8ef4f83ce54f : SUCCESS in 1h 01m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a97d3b40d2cb46d28b52440f82d45be6 : SUCCESS in 1h 38m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c358e6822d094ab5a39af27641b48e29 : SUCCESS in 1h 12m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/09edcb5135fa49f590bea82e9e61228e : SUCCESS in 1h 13m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/44ced86bed7542aa8f097fd36545e415 : SUCCESS in 1h 38m 19s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"d7f90c6a3f1266939948008cd4bbd392df8f2781","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-18 17:22:31.000000000","message":"Patch Set 14: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"c092119cf376fcca36385db4cd3849b6a1bbd029","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-19 10:06:10.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"da83c3ac0dc3644fe332dff11aadf9b175fd52a9","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-19 10:06:11.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"d866b7e613466d673d85fdc9c53fa46804e9f7c2","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-19 10:06:41.000000000","message":"Patch Set 15:\n\n\u003e (2 comments)\n\nThank You for the review!","accounts_in_message":[],"_revision_number":15},{"id":"07c0f320ca1c9096cc60cc502ba3d9aed1b94b52","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-19 10:10:57.000000000","message":"Patch Set 15: Code-Review+2\n\n(2 comments)\n\nJust a nit to consider if you have to update the patch.","accounts_in_message":[],"_revision_number":15},{"id":"bd727a5b8b7ac5f2db6f624c2c761f1613843065","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-19 17:06:18.000000000","message":"Patch Set 15: -Code-Review","accounts_in_message":[],"_revision_number":15},{"id":"ee154e1a922c65e3eaf2f97035c955a912db0647","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-19 17:44:06.000000000","message":"Patch Set 15: Code-Review-1\n\n(2 comments)\n\nI think we can quickly make that more DRY, so -1","accounts_in_message":[],"_revision_number":15},{"id":"8e57fa6d8c3636abab7af12efbcaa4366cd72f14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:45:16.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7932e3e7bbc94bf6bb57cce7b083ab74 : SUCCESS in 7m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78197fac6e394749bee04731304786c8 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/840d45b351e143b9a9316d82bac6a7b6 : SUCCESS in 7m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6893f73ef7a4d0ba20cdfa7cb209795 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a5e491b37ab48119d28e0525594dd65 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e498e7560e594e9ba2d8a1abdfc8d46a : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1fbbb5bc1544e9dafa0910b008c9b86 : SUCCESS in 9m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c5e3f1495d684096b73193cf141a488a : FAILURE in 25m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5372b8be3d654640a35c94cdf6ff9cf7 : SUCCESS in 7m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5c6b230c7f03433eaf7c89db36ebfb8b : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/82adba2f859642fbb4e49d86d7017560 : SUCCESS in 7m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1bd074e548f144ecb6b7499ee275763e : FAILURE in 1h 22m 51s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/11021f81e3e248c4af248645204d79ea : FAILURE in 1h 23m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/57dc63fb22c940599eb87999277d0d2c : FAILURE in 39m 23s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ecce7b32f2d647de8aae0c11b68e4a16 : SUCCESS in 1h 41m 58s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8948a8e594a44b07b230aebcc55f2064 : FAILURE in 20m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d1d3a6e47c104b58aa8edeed30a12c76 : SUCCESS in 2h 06m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a5522cc2320a422a8dad88c275054beb : FAILURE in 1h 39m 26s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4e001144e3fc478692ccd1a2f36a06e9 : FAILURE in 1h 34m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6a84142e1229434da048f9780c470f3c : SUCCESS in 1h 00m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0746dd4bff244fc6885739a025461d5e : SUCCESS in 1h 32m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7679c30a64a542abbee862bdafc855d6 : FAILURE in 38m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a004dba8fc3d4fe1bc3592d5dda2328e : SUCCESS in 1h 17m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/825ed8514e964ded825470d71b861757 : FAILURE in 24m 46s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"62cf8abe010be3ed2bf037ef341327c2c3443b7a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-19 17:50:33.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"b34526beaf6669f0f40ed66c01df1a17d7b48082","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-20 11:20:33.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"c08cdef73a28720160001147e81a5fedbf1f7271","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-20 11:20:43.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"e273411eb3741c6fb476d3c3c6200183d0bc5c28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 14:10:15.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a4fb70c7a8c40a79ec802cedd3a50ec : SUCCESS in 7m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7c0688d64a14f4288ecce04bfd1f0fb : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4661e3fcc9f41448ed52bc40f6df9a5 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/366ec18ed3f4418888255de519dac8d9 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b41bb1d8ea0d42efbca7dbd67fe7a4e8 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5026ba76409147cc8879504384742e6b : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f51592ef43c84e5cbada3249d24be02d : SUCCESS in 13m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0cf04d356ab3471bb675f48c15f5fd10 : SUCCESS in 5m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/963eaf18aab44460bbfbfb4138da0c86 : SUCCESS in 6m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fb1d420951f540d0baaf60d88decc5bf : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e3855dd0e23b4dc3ba67525e24b3a27a : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/70a601ce0f0645c89010cde7da368a6c : SUCCESS in 49m 14s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/11a48b6e0b78466cb0bb78b53e527627 : SUCCESS in 41m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49ee69241a944f6f8eb5c9dd6e5051d8 : SUCCESS in 1h 33m 42s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e35f2d28984a4c5db47b55abc5b52107 : SUCCESS in 1h 59m 35s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b81b0b2389fd4a628dd584405cb8aaba : SUCCESS in 2h 11m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/79feb7082e2c469497bcf5989650933e : SUCCESS in 1h 20m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/bf58909bc6a94207847df96c4e5ba95a : SUCCESS in 2h 07m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2fdc9dafe30d4494883e63b8a865e51c : SUCCESS in 1h 15m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d3868f040f8042b8a5a6ef8e30bf1455 : SUCCESS in 53m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d33cab95df2944109608bf9b89febf8a : SUCCESS in 1h 36m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e98f1774d54f476caf9ccbf29e3ebfc3 : SUCCESS in 1h 06m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e98a04cb6ab342e485835c9e556d7a28 : SUCCESS in 1h 15m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e27d1859eb57440e91a7652349b901a6 : FAILURE in 28m 30s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"a7d8e28a465f0162d8d2beedd5320d09a8bfc02e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-20 21:13:26.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"52e3aab078a78c117f6df95b937c17bb94639021","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-23 20:55:03.000000000","message":"Patch Set 16: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"5ea8148cc8cc1eac7179046bc6125a39b958f4f0","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-23 20:58:01.000000000","message":"Patch Set 16: Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"70eb47a2bba238494cc33f2343fd22cb6836f734","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 20:58:16.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"2485c0261fd0fb96dd170df2d9eb49e60f91ef07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 23:01:52.000000000","message":"Patch Set 16: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9a55c1eb21c44c9b840a8eae97e8de0 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fae41d38684341b58ac16522ab9ca30c : SUCCESS in 6m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b76a63d15a141fe9e447647dc956e76 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17630e73e544444fbf126015d56bd54c : SUCCESS in 6m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da40381e7d3b4cdb916a43fb244aee0f : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b3c67d155c44fd0bd8e3804259e4847 : SUCCESS in 11m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/216bf7fea3a74e3e8536f6995d693f0d : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c2c59cea78324c3cbd4a627046ea80b7 : SUCCESS in 6m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b64f851d56194f91b164807e9eeccc16 : SUCCESS in 55m 39s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3672936883644e80b487976fd9d7b7d9 : SUCCESS in 47m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5fa2d3d1addf41e88edabc50e508f386 : CANCELED in 1h 24m 23s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/098c01e6393c44b49272812e5ee321de : CANCELED in 1h 24m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bd753d02a9274fbbb7b616509bae43fa : FAILURE in 1h 22m 28s","accounts_in_message":[],"_revision_number":16},{"id":"93fcbe4e54d7a3db0e2b473841afe5cdcab06e7f","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-24 07:27:22.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"828f1336cc3d052d125b3918b54c29da96e7dc51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 10:45:31.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fd036d3c2424074b7b8e1a839135278 : SUCCESS in 9m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e5120f0cb224c65ab71bebfe2aaec15 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dff8f5607db045ad90b9992048b750cd : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb2400be87044f2189b92e6a20524999 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24dae1f6cd3546e9b0e8bd7ead9661f5 : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a8e9684e04f4107816fd80f4ef3d75e : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed5e241a46504d939dc601f495d24a0f : SUCCESS in 11m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7c11a4c3c7204dcea91db21003818858 : SUCCESS in 4m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ae1f5df72c5c4df0a1c6cefee3f04c5b : SUCCESS in 6m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/789555560d054ab39fbe21ab6491b8f6 : SUCCESS in 7m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bf6cb5614f7a4e628e7c683fa73350e4 : SUCCESS in 5m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/95cec051a44c4f67885f391dd67e7da5 : SUCCESS in 44m 35s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0b7780bdc1834c5996260ac58c7738a0 : SUCCESS in 45m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/87573a3085a8493db83fa0d9bff75042 : SUCCESS in 1h 53m 11s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/271b196ae0344ca1bb77421c683d829f : SUCCESS in 1h 54m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ec5fcbdfcc98460aa4e26f605552db97 : SUCCESS in 2h 18m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/74b040cabe6b4062a3cb5fd30a7c6343 : SUCCESS in 1h 21m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3c259829de584d47bd469683fbc9b49c : FAILURE in 1h 32m 06s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fed0a998bc8c4dbab6526bf1a2f234f8 : SUCCESS in 1h 22m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0a21d80e7dd94a9ab3cc64eb518fe50f : SUCCESS in 56m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/291004ac1fd14352a1eca7b490ef19a5 : SUCCESS in 1h 28m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad78ce76873b4376a3c15b85c3dc9c82 : SUCCESS in 1h 03m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c5f6a1d57d1c496b8463c9e874e2f654 : SUCCESS in 1h 06m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2ff1f8638b07459385d7ed36ff28d9f0 : FAILURE in 1h 14m 55s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"f41ce4e0b6d29db32bc37a2e43ba70c24b9bcbf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 10:45:45.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"335a610f75479b158f296913ac495291a5ba13fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 10:58:45.000000000","message":"Patch Set 16: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59f04153ce78454e87a46f2c9257d611 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69cbc52e689245f684065ead89243e6c : FAILURE in 6m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/973a2e0f805a457ca64225fcaacf953f : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91c94dff47a5488bae56cbc9c8e4a9e7 : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f7fdb2db3b1475385d29a611375c77c : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03c084fa5d2f4d43a43e73465dec8ef1 : CANCELED\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d8d70e81de8b4fb2bb701a968edb05df : SUCCESS in 5m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/957b51bb0a004e25b9495a17e00ab263 : SUCCESS in 6m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/52d28160469649e0a9a71ea837591fbf : CANCELED\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/89ecabb1d21b4b179d57a52483f468cf : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad323657e0694a46b214ef20c5b65726 : CANCELED\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e175bfa65e7d4bcabd7f93bdb57a5879 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c70e9ab3903246c88a16dde87e69e234 : CANCELED","accounts_in_message":[],"_revision_number":16},{"id":"cc7c2c290fdb0662be170370f012af123391a353","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-24 12:50:00.000000000","message":"Patch Set 16:\n\npretty strange\n\ncheck stage passed pep8\ngate stage failed with:\nYour code has been rated at 9.99/10\n\nI\u0027m rechecking this...","accounts_in_message":[],"_revision_number":16},{"id":"ddde1b60e920002b8a7651f44b7c3c81164bba94","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-24 12:50:15.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"9ecf1da2fadbc00aba331e74388d61e4e8610eb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 15:21:15.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d977e4969869411b8990e948f1ec7976 : SUCCESS in 10m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d7dc03f55584cdaa59b32397bf293a7 : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83040b3c83a445699a71fffa643f462a : FAILURE in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/90489ca6d9624673ac3c098604c22bc5 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a72cbbce91c4582ab2e052a47818850 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46e58b3bf6244edebdb6aa987bfbb337 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ce4282346f9493ea175328101fdc3fb : SUCCESS in 9m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a52cef5910864353a3e2b5ee56102843 : SUCCESS in 4m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/37f88941896a4436902c6224665b279c : SUCCESS in 6m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0c7144b578994fb4b079907eb1e48267 : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6423e4bf81be4af0aeb703bdbcb4d11d : SUCCESS in 5m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ffecbda73f49420db13e018075eec983 : SUCCESS in 44m 22s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e02939aad7b74f19a3cc7ebc5733e938 : SUCCESS in 45m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/007a1dee174249caa7367ba069e30b31 : FAILURE in 2h 06m 41s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0bf6146652b340789a3041b3b9cd8dc4 : SUCCESS in 1h 50m 06s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/67f3080113b3412db8dc51f3d4b0ff0d : SUCCESS in 2h 19m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5c0e532315fd421f9b37a9261d5367d0 : SUCCESS in 2h 09m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c279d664a85045d88c4f34a3d9d89933 : SUCCESS in 1h 50m 43s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/10fa76adbdf34574ae87c65f7cf089a7 : SUCCESS in 1h 19m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/658108c7d6244bcbbe9ecc80ebc8e6fa : SUCCESS in 53m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/990827f189844e4fb4ff7c514560b487 : SUCCESS in 1h 27m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c1e8e0b43cd437f886c9c7f6c3a3e81 : SUCCESS in 1h 07m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0a7a16b5e9aa4f87bfca0f7e16a40072 : FAILURE in 1h 07m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2ca059bd029d4b4bb1398d5ba5eac716 : SUCCESS in 1h 29m 10s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"c8ca78427bf1e87867a84a5d5eb4b8fda1566dc6","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-09-26 07:03:18.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"de064d6c40de835c37bffa091b4657bfb4604add","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:40:46.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0767e9a4c5649f485e2fc4ddc4a418e : SUCCESS in 8m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be7b4722e7174eddb426d510e5ad2f15 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b666a55e813446ba6681148730cbaee : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7cd7c090682474e9135970de1abf3eb : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87b4b85ff7b44ab79c1fdb5d2cc3b020 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fd6edbd24b54dfa969dab17fcbfe5df : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/859059e133924877a80bf68a5599950b : SUCCESS in 10m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/227ce068166c4ed7b4a6b20ae804d2db : SUCCESS in 4m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3a361a359bea481aa9ea4911c34cc648 : SUCCESS in 6m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0bd45d2cbc804e19a7671c5acc03a593 : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4824eed62b44486fb64ed07657211555 : SUCCESS in 6m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0c66be5b1a2747b088ee35507724dc05 : SUCCESS in 48m 51s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c110cc19c0a04a3ea88ca4b0358d97a4 : SUCCESS in 57m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/47aca5841bcf472bb519896fe2327b68 : SUCCESS in 1h 45m 20s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6c2fa6e54b0c46f8ad1906dabc3cd610 : SUCCESS in 1h 56m 06s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ef9e8a8cc7b64055be07cf0ff65d2fa4 : SUCCESS in 2h 01m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b85482f4adaa4905be88524581d70567 : TIMED_OUT in 2h 33m 33s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/78375964b2cd41ac91a8508498ec4614 : SUCCESS in 1h 28m 22s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1150d3b5bbb3454b910c4d80c3843cd9 : SUCCESS in 1h 22m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f9b3742c360e457bbf2824e9af5fb2e7 : SUCCESS in 53m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/14c7046c1e664f9d8bacf9ca5f3cac9c : SUCCESS in 1h 45m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68be3f8e32804269abfc44572384ff5a : SUCCESS in 1h 11m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/194b1c2c382c43e2ab31c3ffc8cc2650 : SUCCESS in 1h 15m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/bee63995914f4521926568227d162498 : SUCCESS in 1h 27m 05s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"d2f57b1ba79de8b9be3c4467b6988d3c9c2d1f22","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-26 12:28:25.000000000","message":"Patch Set 17: Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"11e1c9b6bfb152d102c660b5f5217062418bd8cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 12:30:08.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"c7f0221d23f32bede5433f4be2a973090aa0a460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 14:21:13.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/699856d5618744f7bcba9f5b473b5adb : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63b8fb847457454c9e4117b7b9c9d707 : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b349664ceaee444da0f573acb7761c2e : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5484baf7c47c4151b8a2f30fb4db8bbe : SUCCESS in 7m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4f303506d704637baeaa54e03a52564 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85109544178c4bca83395eeb46cca88a : SUCCESS in 15m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3b124197b22e470e8711b60fe34cd988 : SUCCESS in 6m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/93542a1c86414da69560a76a87c6018a : SUCCESS in 8m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bfa0b24fc6b04875994f449e61a2daef : SUCCESS in 46m 13s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e48fd9ca87d542938cf19dbeac7deb1a : SUCCESS in 53m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/052bb7cc462a4760bff755434735abef : SUCCESS in 1h 43m 51s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5d751073366541c68a5b55e6e4d0cacf : SUCCESS in 1h 44m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b0d0194a1597422bb2c556ecfa6f8d4e : SUCCESS in 1h 20m 28s","accounts_in_message":[],"_revision_number":17},{"id":"697c51786b7609638878abd49e8ea40639d4da11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 14:21:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17},{"id":"621cbf4336b87c8432667f93f974ed4a5c01f5dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 14:23:07.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bec6648ea68d47eea5b58870ae0a7ab7 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"d7f0c819e376297989773c35f417fb3e0f2cd1e8","revisions":{"5a388dc849dcf0dc0befcb9f8ad7fd4c75a56e0f":{"kind":"REWORK","_number":1,"created":"2019-07-24 09:17:17.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/77/672477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/1"}}},"commit":{"parents":[{"commit":"3912ed20be9707d53b827731874b99849e3cef5c","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3912ed20be9707d53b827731874b99849e3cef5c"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:15:05.000000000","tz":0},"subject":"Validate supported LB algorithm in Amphora provider divers","message":"Validate supported LB algorithm in Amphora provider divers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a388dc849dcf0dc0befcb9f8ad7fd4c75a56e0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a388dc849dcf0dc0befcb9f8ad7fd4c75a56e0f"}]},"branch":"refs/heads/master"},"507cc3bda601ec05629d99817888e6ff17a192e4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-24 09:17:39.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/77/672477/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/2"}}},"commit":{"parents":[{"commit":"3912ed20be9707d53b827731874b99849e3cef5c","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3912ed20be9707d53b827731874b99849e3cef5c"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:17:25.000000000","tz":0},"subject":"Validate supported LB algorithm in Amphora provider drivers","message":"Validate supported LB algorithm in Amphora provider drivers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/507cc3bda601ec05629d99817888e6ff17a192e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/507cc3bda601ec05629d99817888e6ff17a192e4"}]},"branch":"refs/heads/master"},"7bb0ab1f5f6dc680fc6c9727cdd25ce80263f77b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-26 11:46:36.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/77/672477/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/3"}}},"commit":{"parents":[{"commit":"10d9f529de31b1b964fb19a9033e3576bf3f610e","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/10d9f529de31b1b964fb19a9033e3576bf3f610e"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-26 11:46:18.000000000","tz":0},"subject":"Validate supported LB algorithm in Amphora provider drivers","message":"Validate supported LB algorithm in Amphora provider drivers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7bb0ab1f5f6dc680fc6c9727cdd25ce80263f77b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7bb0ab1f5f6dc680fc6c9727cdd25ce80263f77b"}]},"branch":"refs/heads/master"},"2492c39e7bfc792f9f8806be156d01209747cafb":{"kind":"REWORK","_number":4,"created":"2019-08-05 08:19:13.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/77/672477/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/4"}}},"commit":{"parents":[{"commit":"10d9f529de31b1b964fb19a9033e3576bf3f610e","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/10d9f529de31b1b964fb19a9033e3576bf3f610e"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-05 08:12:51.000000000","tz":0},"subject":"Validate supported LB algorithm in Amphora provider drivers","message":"Validate supported LB algorithm in Amphora provider drivers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2492c39e7bfc792f9f8806be156d01209747cafb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2492c39e7bfc792f9f8806be156d01209747cafb"}]},"branch":"refs/heads/master"},"77a1dea89d96f1ff9a1cf33012be40da8bc1d95e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-07 07:51:23.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/77/672477/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/5"}}},"commit":{"parents":[{"commit":"3671985c47093c61bea0022c35b2f2010a138592","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3671985c47093c61bea0022c35b2f2010a138592"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-07 07:51:23.000000000","tz":0},"subject":"Validate supported LB algorithm in Amphora provider drivers","message":"Validate supported LB algorithm in Amphora provider drivers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77a1dea89d96f1ff9a1cf33012be40da8bc1d95e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77a1dea89d96f1ff9a1cf33012be40da8bc1d95e"}]},"branch":"refs/heads/master"},"5fdd0bdfd513706281b216af5c628812315079e5":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-08-07 12:01:11.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/77/672477/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/6"}}},"commit":{"parents":[{"commit":"eee38136e8adf285f6565300fc89792523e9732c","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eee38136e8adf285f6565300fc89792523e9732c"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-08-07 11:58:58.000000000","tz":0},"subject":"Validate supported LB algorithm in Amphora provider drivers","message":"Validate supported LB algorithm in Amphora provider drivers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5fdd0bdfd513706281b216af5c628812315079e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5fdd0bdfd513706281b216af5c628812315079e5"}]},"branch":"refs/heads/master"},"1e199ffbb0b19cd946001e3d3e54e457b4e6a105":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-09-04 20:32:59.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/77/672477/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/7"}}},"commit":{"parents":[{"commit":"acfd4783407e4186233d1a30faa7dce159136d7e","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/acfd4783407e4186233d1a30faa7dce159136d7e"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-04 20:32:59.000000000","tz":0},"subject":"Validate supported LB algorithm in Amphora provider drivers","message":"Validate supported LB algorithm in Amphora provider drivers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e199ffbb0b19cd946001e3d3e54e457b4e6a105"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e199ffbb0b19cd946001e3d3e54e457b4e6a105"}]},"branch":"refs/heads/master"},"d6af814b4c53b3e60d2d0291f0112763d59e9424":{"kind":"REWORK","_number":8,"created":"2019-09-11 10:07:10.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/77/672477/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/8"}}},"commit":{"parents":[{"commit":"0654a4d4d78e4cc270b3204a720cfd4cb8a748d5","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0654a4d4d78e4cc270b3204a720cfd4cb8a748d5"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-11 10:04:10.000000000","tz":0},"subject":"Validate supported LB algorithm in Amphora provider drivers","message":"Validate supported LB algorithm in Amphora provider drivers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6af814b4c53b3e60d2d0291f0112763d59e9424"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6af814b4c53b3e60d2d0291f0112763d59e9424"}]},"branch":"refs/heads/master"},"3435982443706ebc63b6e075831a7db7ed5a8e7b":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-09-12 14:00:48.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/77/672477/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/9"}}},"commit":{"parents":[{"commit":"fff0dc2636b661449ace9fe3f358c5ff3787ddec","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fff0dc2636b661449ace9fe3f358c5ff3787ddec"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-12 14:00:34.000000000","tz":0},"subject":"Validate supported LB algorithm in Amphora provider drivers","message":"Validate supported LB algorithm in Amphora provider drivers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3435982443706ebc63b6e075831a7db7ed5a8e7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3435982443706ebc63b6e075831a7db7ed5a8e7b"}]},"branch":"refs/heads/master"},"b64c452d48183f7860b1adeabdcd6ef197f6543b":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-09-13 10:08:49.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/77/672477/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/10"}}},"commit":{"parents":[{"commit":"dccd9a3dd462eb05452377cc6db0005e03d2095d","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dccd9a3dd462eb05452377cc6db0005e03d2095d"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-13 10:08:51.000000000","tz":120},"subject":"Validate supported LB algorithm in Amphora provider drivers","message":"Validate supported LB algorithm in Amphora provider drivers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b64c452d48183f7860b1adeabdcd6ef197f6543b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b64c452d48183f7860b1adeabdcd6ef197f6543b"}]},"branch":"refs/heads/master"},"06091252a8b4e200ff8774dc4ac1b0401566b7e4":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-09-13 12:53:02.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/77/672477/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/11"}}},"commit":{"parents":[{"commit":"0de571a5998ceb035562443347fcd1339f10a255","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0de571a5998ceb035562443347fcd1339f10a255"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-13 12:53:14.000000000","tz":120},"subject":"Validate supported LB algorithm in Amphora provider drivers","message":"Validate supported LB algorithm in Amphora provider drivers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/06091252a8b4e200ff8774dc4ac1b0401566b7e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/06091252a8b4e200ff8774dc4ac1b0401566b7e4"}]},"branch":"refs/heads/master"},"dd4a9b5aa927fa43f6c33a9ed9673f5e9485bc8a":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-09-13 14:54:36.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/77/672477/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/12"}}},"commit":{"parents":[{"commit":"dbdc8a48f62b07b4ad83ae2aa2215ffc2d2ec885","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dbdc8a48f62b07b4ad83ae2aa2215ffc2d2ec885"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-13 14:54:16.000000000","tz":0},"subject":"Validate supported LB algorithm in Amphora provider drivers","message":"Validate supported LB algorithm in Amphora provider drivers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd4a9b5aa927fa43f6c33a9ed9673f5e9485bc8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd4a9b5aa927fa43f6c33a9ed9673f5e9485bc8a"}]},"branch":"refs/heads/master"},"b84c40aaa528d65c23b347132781ab86e2b00b2b":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-09-16 09:21:24.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/77/672477/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/13"}}},"commit":{"parents":[{"commit":"2eac7a7862330746ad77d9b28760c99fa706e179","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2eac7a7862330746ad77d9b28760c99fa706e179"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-16 09:21:24.000000000","tz":0},"subject":"Validate supported LB algorithm in Amphora provider drivers","message":"Validate supported LB algorithm in Amphora provider drivers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b84c40aaa528d65c23b347132781ab86e2b00b2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b84c40aaa528d65c23b347132781ab86e2b00b2b"}]},"branch":"refs/heads/master"},"e39d14c1af2aa7c2d0148298a5a27304f301f5a5":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-09-17 07:05:45.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/77/672477/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/14"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-17 07:05:45.000000000","tz":0},"subject":"Validate supported LB algorithm in Amphora provider drivers","message":"Validate supported LB algorithm in Amphora provider drivers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e39d14c1af2aa7c2d0148298a5a27304f301f5a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e39d14c1af2aa7c2d0148298a5a27304f301f5a5"}]},"branch":"refs/heads/master"},"f006e481648e0dbf6c9bbca9c6d8d79c24f3c6f0":{"kind":"REWORK","_number":15,"created":"2019-09-19 10:06:11.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/77/672477/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/15"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-19 10:04:08.000000000","tz":0},"subject":"Validate supported LB algorithm in Amphora provider drivers","message":"Validate supported LB algorithm in Amphora provider drivers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f006e481648e0dbf6c9bbca9c6d8d79c24f3c6f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f006e481648e0dbf6c9bbca9c6d8d79c24f3c6f0"}]},"branch":"refs/heads/master"},"34ad52e49223940e8d383fbb3aa3865214373791":{"kind":"REWORK","_number":16,"created":"2019-09-20 11:20:33.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/77/672477/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/16"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-20 11:18:22.000000000","tz":0},"subject":"Validate supported LB algorithm in Amphora provider drivers","message":"Validate supported LB algorithm in Amphora provider drivers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34ad52e49223940e8d383fbb3aa3865214373791"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34ad52e49223940e8d383fbb3aa3865214373791"}]},"branch":"refs/heads/master"},"d7f0c819e376297989773c35f417fb3e0f2cd1e8":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2019-09-26 07:03:18.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/77/672477/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/672477/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/672477/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/672477/17"}}},"commit":{"parents":[{"commit":"d5ffd2ca4048fe5ec40b63a4afd5109997834594","subject":"Fix new pep8/pylint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5ffd2ca4048fe5ec40b63a4afd5109997834594"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-07-24 09:02:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-09-26 07:03:18.000000000","tz":0},"subject":"Validate supported LB algorithm in Amphora provider drivers","message":"Validate supported LB algorithm in Amphora provider drivers\n\nThis patch adds supported LB algorithm validation in\nAmphora provider drivers.\n\nChange-Id: Ifc995ebe6165dbb57b8eb40bcc3c7e2a0eba94dc\nStory: 2006264\nTask: 35971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7f0c819e376297989773c35f417fb3e0f2cd1e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7f0c819e376297989773c35f417fb3e0f2cd1e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
