)]}'
{"id":"openstack%2Foctavia~673173","triplet_id":"openstack%2Foctavia~master~Ib7bb36264892b4dc5ac14f972e5bb2b6b1fa5c54","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ib7bb36264892b4dc5ac14f972e5bb2b6b1fa5c54","subject":"WIP: Switch Fedora-based amphora to fedora-minimal","status":"ABANDONED","created":"2019-07-28 14:32:35.000000000","updated":"2019-11-18 10:09:26.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bf2e88bf537f6ce8e3412bebe68300a024889749","_number":673173,"virtual_id_number":673173,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":-1,"date":"2019-09-18 10:40:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-03 00:29:39.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-18 10:40:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a28b94ca50b10c8d773ad12bb5c0613189fa1f03","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-28 14:32:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e12f363fa6ef95ab770c9164df7d21063a0d87c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-28 17:11:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/73/673173/1/check/openstack-tox-cover/3496061/cover/ : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/673173/1/check/openstack-tox-lower-constraints/5f3c195/ : SUCCESS in 5m 42s\n- openstack-tox-pep8 http://logs.openstack.org/73/673173/1/check/openstack-tox-pep8/cfb393d/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/73/673173/1/check/openstack-tox-py27/06e6a57/ : SUCCESS in 4m 29s\n- openstack-tox-py36 http://logs.openstack.org/73/673173/1/check/openstack-tox-py36/e478c1b/ : SUCCESS in 4m 54s\n- openstack-tox-py37 http://logs.openstack.org/73/673173/1/check/openstack-tox-py37/fc8782d/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/73/673173/1/check/openstack-tox-docs/4b2c956/html/ : SUCCESS in 5m 18s\n- openstack-tox-functional http://logs.openstack.org/73/673173/1/check/openstack-tox-functional/18bece3/ : SUCCESS in 4m 35s\n- openstack-tox-functional-py36 http://logs.openstack.org/73/673173/1/check/openstack-tox-functional-py36/22fb381/ : SUCCESS in 5m 10s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/73/673173/1/check/octavia-v2-dsvm-noop-api/de69412/ : SUCCESS in 47m 44s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/73/673173/1/check/octavia-v2-dsvm-noop-py2-api/7b90219/ : SUCCESS in 47m 35s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/73/673173/1/check/octavia-v2-dsvm-scenario/7f33267/ : SUCCESS in 1h 29m 08s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/73/673173/1/check/octavia-v2-dsvm-py2-scenario/1cde5a2/ : SUCCESS in 1h 26m 26s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/73/673173/1/check/octavia-v2-dsvm-py2-scenario-centos-7/64c8d2e/ : FAILURE in 2h 33m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/73/673173/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/4936aa8/ : SUCCESS in 1h 34m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/73/673173/1/check/octavia-v2-dsvm-scenario-two-node/2b0f23f/ : FAILURE in 1h 58m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/73/673173/1/check/octavia-v2-dsvm-py2-scenario-two-node/3c6dc97/ : SUCCESS in 1h 18m 35s (non-voting)\n- octavia-grenade http://logs.openstack.org/73/673173/1/check/octavia-grenade/3906bdf/ : SUCCESS in 1h 15m 20s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/73/673173/1/check/octavia-v2-dsvm-tls-barbican/13ab83e/ : SUCCESS in 55m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/73/673173/1/check/octavia-v2-dsvm-scenario-amphora-v2/e378bca/ : SUCCESS in 1h 30m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/73/673173/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/d84d2c0/ : SUCCESS in 1h 01m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/73/673173/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/0a62fbb/ : SUCCESS in 1h 05m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/73/673173/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/c85066f/ : FAILURE in 1h 13m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43e4941e4a496416d22d945c5979db30d28de1b2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 00:29:39.000000000","message":"Patch Set 1: Code-Review+2\n\nWe don\u0027t have a gate that tests fedora amps? Which ... Really, we should. I don\u0027t know if I can trust a change like this until we have a Fedora gate...\nThat said, it isn\u0027t an officially supported amp image, and I don\u0027t really HAVE to trust it, I just trust you, so seems good to me. :D","accounts_in_message":[],"_revision_number":1},{"id":"00e39f6d2d4f41571159001e59e461ee00f2c2e4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 00:30:14.000000000","message":"Patch Set 1:\n\nI guess we\u0027ll see when the publish job runs, if it works or not. :D","accounts_in_message":[],"_revision_number":1},{"id":"320f188356eba157d47ea993ab45b4ba0b404301","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-03 09:26:41.000000000","message":"Patch Set 1: Workflow-1\n\nFair point. We can have it tested in CI before merging.","accounts_in_message":[],"_revision_number":1},{"id":"c4bfba01ef0a969fe54b33190023d36b2c623478","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-05 16:09:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8b9a8913778b6f438ee71d4f233a6e4402516abb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-05 16:10:23.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"45ef6478123bed883a7e1efe6c391149160f351d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 17:07:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/73/673173/2/check/openstack-tox-cover/1e8d77d/ : FAILURE in 7m 52s\n- openstack-tox-lower-constraints https://logs.opendev.org/73/673173/2/check/openstack-tox-lower-constraints/9d71bc6/ : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://logs.opendev.org/73/673173/2/check/openstack-tox-pep8/326cd5b/ : SUCCESS in 6m 28s\n- openstack-tox-py27 https://logs.opendev.org/73/673173/2/check/openstack-tox-py27/91d9c88/ : FAILURE in 8m 17s\n- openstack-tox-py36 https://logs.opendev.org/73/673173/2/check/openstack-tox-py36/c9e7ab5/ : FAILURE in 6m 30s\n- openstack-tox-py37 https://logs.opendev.org/73/673173/2/check/openstack-tox-py37/8fd4014/ : FAILURE in 6m 41s\n- openstack-tox-docs https://logs.opendev.org/73/673173/2/check/openstack-tox-docs/49b8cc4/html/ : SUCCESS in 6m 00s\n- publish-openstack-octavia-amphora-image-fedora29 https://logs.opendev.org/73/673173/2/check/publish-openstack-octavia-amphora-image-fedora29/ae8de47/ : FAILURE in 2m 22s\n- octavia-v2-dsvm-scenario-fedora29 https://logs.opendev.org/73/673173/2/check/octavia-v2-dsvm-scenario-fedora29/67e38fc/ : FAILURE in 45m 29s","accounts_in_message":[],"_revision_number":2},{"id":"91ab372186f3d4c446b70fc38358e65ee79477d1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-05 17:21:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"455d0c73696779fbb08bcc4d02fddc50a8fb73ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 20:06:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/73/673173/3/check/openstack-tox-cover/8f14bbb/ : FAILURE in 6m 49s\n- openstack-tox-lower-constraints https://logs.opendev.org/73/673173/3/check/openstack-tox-lower-constraints/4b19c01/ : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://logs.opendev.org/73/673173/3/check/openstack-tox-pep8/35e0aff/ : SUCCESS in 5m 09s\n- openstack-tox-py27 https://logs.opendev.org/73/673173/3/check/openstack-tox-py27/638692d/ : FAILURE in 6m 35s\n- openstack-tox-py36 https://logs.opendev.org/73/673173/3/check/openstack-tox-py36/6d1ca7b/ : FAILURE in 6m 31s\n- openstack-tox-py37 https://logs.opendev.org/73/673173/3/check/openstack-tox-py37/da2a2b8/ : FAILURE in 6m 35s\n- openstack-tox-docs https://logs.opendev.org/73/673173/3/check/openstack-tox-docs/05e07c2/html/ : SUCCESS in 6m 34s\n- publish-openstack-octavia-amphora-image-fedora29 https://logs.opendev.org/73/673173/3/check/publish-openstack-octavia-amphora-image-fedora29/9bbe813/ : SUCCESS in 15m 17s\n- octavia-v2-dsvm-scenario-fedora29 https://logs.opendev.org/73/673173/3/check/octavia-v2-dsvm-scenario-fedora29/6e3313b/ : TIMED_OUT in 2h 33m 20s","accounts_in_message":[],"_revision_number":3},{"id":"721dea9891bef02a393074318e1237410a94c58a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-06 09:43:27.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"7e157bdfc4bea95b2ae8d7f8fb2eab896d8876b1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-06 09:44:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"def021cdfc65d017e762b3d7a7f312d635996211","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-14 10:40:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f55ee216625e31d6738e430620b7df0e026058b7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-14 10:42:39.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"a73e3a42e82e50e138c5eb1b5d54902e94340ba7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-14 10:44:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f025962172101d1769adb0947957789c872bf2ac","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-14 10:44:12.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b027c9df83d3367ada3581c2f542bb12eee3a545","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 11:25:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2aeb21a3a2814653b2ef7620d6589f89 : SUCCESS in 6m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2bceba0684d48eeaf762fb1ce095c17 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec7d01edcf8d4d029231edf6c2e7a737 : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b23710ff8e34172bd6bc3db753c9b7a : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/460b8a552fab45ea8f7763f917f55851 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41ddb692f76843d1a260f7cc86115f83 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aae8b365addf4e9f97e80ba547626c43 : SUCCESS in 7m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d9e6b5cce66546f6a89b2db12b205e13 : SUCCESS in 4m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c9d265c3f58243c5b5ee0ca8db898740 : SUCCESS in 6m 56s\n- publish-openstack-octavia-amphora-image-fedora30 https://zuul.opendev.org/t/openstack/build/cc37379c2be7445c8afd60c0dba12c12 : FAILURE in 10m 38s\n- octavia-v2-dsvm-scenario-fedora30 https://zuul.opendev.org/t/openstack/build/1271e7d8727842a3b8fb50c089d04cb0 : FAILURE in 36m 21s","accounts_in_message":[],"_revision_number":5},{"id":"2a4110b6351260a56464b9330a5fafab7e9fae73","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-16 16:22:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2174b065236d5dc329fcf24ba9576dd2279e9d7d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-16 17:17:18.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"ea639d4d0538c95bdcf72fd2073bcbcf022091c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 17:21:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ac312dadde446d78f937916c888fb29 : SUCCESS in 9m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f40c573edd44347826d6a8536da3bc1 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08fe6535246f4de3bf52c8233d9e020e : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e480df6a5fbd4e50892313b1a93acc26 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b9203d7b81244a39127461fbd3e0b22 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/586e5c75bfb4411d91c9299766845cca : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b57dd3cd4504de3b89e76aa5a371bfb : SUCCESS in 7m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1e322befd53c4be5a2e182e60f2c1ede : SUCCESS in 4m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4c5ebbde2baf46ceaeb074fda78851b1 : SUCCESS in 6m 43s\n- publish-openstack-octavia-amphora-image-fedora30 https://zuul.opendev.org/t/openstack/build/6cc8a340f89b454eac7fde9cd559df32 : SUCCESS in 13m 56s\n- octavia-v2-dsvm-scenario-fedora30 https://zuul.opendev.org/t/openstack/build/3ec701d463044b63bfa4dfcc5f07c9d6 : FAILURE in 49m 01s","accounts_in_message":[],"_revision_number":6},{"id":"c291acdcc540d8a56153fb08ef38d4165b32fb66","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-16 18:27:45.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"c1d91d326c36372a4afcf89da18b895eee84b5bf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-16 18:27:50.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"25a4ab0a2bfb30ab62f1702c39cb9c6212909cb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 21:06:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5cbb5fa72a24bfa804ad7b97a1697cd : SUCCESS in 7m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e5243acdc69498d9b4866768a3d4f82 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bf83854537a4845b72fb66d3df1b7d4 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/34261ab3fa9e448c8e721b626312a53d : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c73fc6584c4410086ce6c8d2610ab28 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/889c8600e83241c1b4eebac996c2e2b5 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6e78c7c16454532bfece1869ccfa4af : SUCCESS in 8m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ee84de97f61546838a4007e6c95706d6 : SUCCESS in 5m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/95b032ae7e1148be99e3f3192cb1b016 : SUCCESS in 6m 56s\n- publish-openstack-octavia-amphora-image-fedora30 https://zuul.opendev.org/t/openstack/build/4b3ee57966f84186be4623c02be1e971 : SUCCESS in 16m 15s\n- octavia-v2-dsvm-scenario-fedora30 https://zuul.opendev.org/t/openstack/build/8e4de078a6ca4705b5323e89add6c3da : TIMED_OUT in 2h 32m 48s","accounts_in_message":[],"_revision_number":7},{"id":"623d50e50a913bf41d57f6665e98c2b80ffd42c1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-18 06:36:47.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"e8cffe1ea118e93a5ed6d9fa28bb5805d5515a55","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-18 06:36:57.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"ebbfa3b7506bc5a52d59b06a78f0f14cb1067d2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 10:40:26.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2c5a37083a24bb4af1e002aa2f2264c : SUCCESS in 9m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6451b090bc0e4ba0b4090e5ad72abb58 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a16f2055155c41afa72c969cdcde772e : SUCCESS in 6m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fdcb28b60414910b888e6e832205584 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d640ef7623364de0b8481a6aecdeda04 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9cd0fd48aa9644849eaf0ac7b59c0509 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf889d313acd4b6f85e70d29b99a80ad : SUCCESS in 8m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1d0d1b7f03534352a862c15885e0fa97 : SUCCESS in 5m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/042fca512421417a89c785d144d457fb : SUCCESS in 6m 53s\n- publish-openstack-octavia-amphora-image-fedora30 https://zuul.opendev.org/t/openstack/build/9251ba497d8441ec94bd10b6fe9d0cba : SUCCESS in 15m 27s\n- octavia-v2-dsvm-scenario-fedora30 https://zuul.opendev.org/t/openstack/build/01e204340994448b924967efc402c498 : TIMED_OUT in 2h 33m 58s","accounts_in_message":[],"_revision_number":9},{"id":"bf2e88bf537f6ce8e3412bebe68300a024889749","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-18 10:09:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"eeddec11ea5c5fd7d5bfe5d0d313b78dd363a998","revisions":{"deb7abfdfd698c99dbce6ff75f012b2fe8934943":{"kind":"REWORK","_number":1,"created":"2019-07-28 14:32:35.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/73/673173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/673173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/673173/1"}}},"commit":{"parents":[{"commit":"262ffaa9d71feb1b79b59f7c6a9ea8b7f977415e","subject":"Build amphora image using centos-minimal element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/262ffaa9d71feb1b79b59f7c6a9ea8b7f977415e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 14:29:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 14:32:13.000000000","tz":120},"subject":"Switch Fedora-based amphora to fedora-minimal","message":"Switch Fedora-based amphora to fedora-minimal\n\nThis patch also upgrades to Fedora 29 (28 reached end of life).\n\nChange-Id: Ib7bb36264892b4dc5ac14f972e5bb2b6b1fa5c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/deb7abfdfd698c99dbce6ff75f012b2fe8934943"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/deb7abfdfd698c99dbce6ff75f012b2fe8934943"}]},"branch":"refs/heads/master"},"d8e3892bf6f63818c14be95a12336bfd57a2e318":{"kind":"REWORK","_number":2,"created":"2019-08-05 16:09:57.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/73/673173/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/673173/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/673173/2"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 14:29:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-05 16:09:18.000000000","tz":120},"subject":"WIP: Switch Fedora-based amphora to fedora-minimal","message":"WIP: Switch Fedora-based amphora to fedora-minimal\n\nThis patch also upgrades to Fedora 29 (28 reached end of life).\n\nChange-Id: Ib7bb36264892b4dc5ac14f972e5bb2b6b1fa5c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d8e3892bf6f63818c14be95a12336bfd57a2e318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d8e3892bf6f63818c14be95a12336bfd57a2e318"}]},"branch":"refs/heads/master"},"34ceac4922cc30af41605d2eb935cc29e941f4d6":{"kind":"REWORK","_number":3,"created":"2019-08-05 17:21:48.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/73/673173/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/673173/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/673173/3"}}},"commit":{"parents":[{"commit":"0c30ffdb814148e7fb82817a187a25b2a0676c97","subject":"Install missing packages in nodepool instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c30ffdb814148e7fb82817a187a25b2a0676c97"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 14:29:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-05 17:21:27.000000000","tz":120},"subject":"WIP: Switch Fedora-based amphora to fedora-minimal","message":"WIP: Switch Fedora-based amphora to fedora-minimal\n\nThis patch also upgrades to Fedora 29 (28 reached end of life).\n\nChange-Id: Ib7bb36264892b4dc5ac14f972e5bb2b6b1fa5c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34ceac4922cc30af41605d2eb935cc29e941f4d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34ceac4922cc30af41605d2eb935cc29e941f4d6"}]},"branch":"refs/heads/master"},"13d7070f9df1be5d1af6384cb9440a289aba59fb":{"kind":"REWORK","_number":4,"created":"2019-09-14 10:40:56.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/73/673173/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/673173/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/673173/4"}}},"commit":{"parents":[{"commit":"2eac7a7862330746ad77d9b28760c99fa706e179","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2eac7a7862330746ad77d9b28760c99fa706e179"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 14:29:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-14 10:39:30.000000000","tz":120},"subject":"WIP: Switch Fedora-based amphora to fedora-minimal","message":"WIP: Switch Fedora-based amphora to fedora-minimal\n\nThis patch also upgrades to Fedora 30 (28 reached end of life).\n\nChange-Id: Ib7bb36264892b4dc5ac14f972e5bb2b6b1fa5c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13d7070f9df1be5d1af6384cb9440a289aba59fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13d7070f9df1be5d1af6384cb9440a289aba59fb"}]},"branch":"refs/heads/master"},"df0a38a3ef55a418b5bac227006b3666bd1d9c00":{"kind":"REWORK","_number":5,"created":"2019-09-14 10:44:08.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/73/673173/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/673173/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/673173/5"}}},"commit":{"parents":[{"commit":"2eac7a7862330746ad77d9b28760c99fa706e179","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2eac7a7862330746ad77d9b28760c99fa706e179"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 14:29:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-14 10:43:59.000000000","tz":120},"subject":"WIP: Switch Fedora-based amphora to fedora-minimal","message":"WIP: Switch Fedora-based amphora to fedora-minimal\n\nThis patch also upgrades to Fedora 30 (28 reached end of life).\n\nChange-Id: Ib7bb36264892b4dc5ac14f972e5bb2b6b1fa5c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df0a38a3ef55a418b5bac227006b3666bd1d9c00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df0a38a3ef55a418b5bac227006b3666bd1d9c00"}]},"branch":"refs/heads/master"},"87557fc9933f10fdd434bd238d80feea4ed5c63f":{"kind":"REWORK","_number":6,"created":"2019-09-16 16:22:58.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/73/673173/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/673173/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/673173/6"}}},"commit":{"parents":[{"commit":"2eac7a7862330746ad77d9b28760c99fa706e179","subject":"Add new algorithm SOURCE_IP_PORT","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2eac7a7862330746ad77d9b28760c99fa706e179"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 14:29:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-16 16:22:49.000000000","tz":120},"subject":"WIP: Switch Fedora-based amphora to fedora-minimal","message":"WIP: Switch Fedora-based amphora to fedora-minimal\n\nThis patch also upgrades to Fedora 30 (28 reached end of life).\n\nChange-Id: Ib7bb36264892b4dc5ac14f972e5bb2b6b1fa5c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/87557fc9933f10fdd434bd238d80feea4ed5c63f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/87557fc9933f10fdd434bd238d80feea4ed5c63f"}]},"branch":"refs/heads/master"},"27fd0e810d278af9834bb361df76c5830422096d":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-09-16 18:27:45.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/73/673173/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/673173/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/673173/7"}}},"commit":{"parents":[{"commit":"774f202068c0f8c57f3c9906d07c650f37937a3b","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/774f202068c0f8c57f3c9906d07c650f37937a3b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 14:29:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-16 18:27:45.000000000","tz":0},"subject":"WIP: Switch Fedora-based amphora to fedora-minimal","message":"WIP: Switch Fedora-based amphora to fedora-minimal\n\nThis patch also upgrades to Fedora 30 (28 reached end of life).\n\nChange-Id: Ib7bb36264892b4dc5ac14f972e5bb2b6b1fa5c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27fd0e810d278af9834bb361df76c5830422096d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27fd0e810d278af9834bb361df76c5830422096d"}]},"branch":"refs/heads/master"},"0a7b3c86250b28b24afb53a2d138463cc404b89d":{"kind":"REWORK","_number":8,"created":"2019-09-18 06:36:47.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/73/673173/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/673173/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/673173/8"}}},"commit":{"parents":[{"commit":"774f202068c0f8c57f3c9906d07c650f37937a3b","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/774f202068c0f8c57f3c9906d07c650f37937a3b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 14:29:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-18 06:36:43.000000000","tz":0},"subject":"WIP: Switch Fedora-based amphora to fedora-minimal","message":"WIP: Switch Fedora-based amphora to fedora-minimal\n\nThis patch also upgrades to Fedora 30 (28 reached end of life).\n\nChange-Id: Ib7bb36264892b4dc5ac14f972e5bb2b6b1fa5c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a7b3c86250b28b24afb53a2d138463cc404b89d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a7b3c86250b28b24afb53a2d138463cc404b89d"}]},"branch":"refs/heads/master"},"eeddec11ea5c5fd7d5bfe5d0d313b78dd363a998":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-09-18 06:36:57.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/73/673173/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/673173/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/673173/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/673173/9"}}},"commit":{"parents":[{"commit":"926179c97dd2f86ad3476a7f63b10ed121d9bed1","subject":"Fix openstack port show content handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926179c97dd2f86ad3476a7f63b10ed121d9bed1"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 14:29:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-18 06:36:57.000000000","tz":0},"subject":"WIP: Switch Fedora-based amphora to fedora-minimal","message":"WIP: Switch Fedora-based amphora to fedora-minimal\n\nThis patch also upgrades to Fedora 30 (28 reached end of life).\n\nChange-Id: Ib7bb36264892b4dc5ac14f972e5bb2b6b1fa5c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eeddec11ea5c5fd7d5bfe5d0d313b78dd363a998"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eeddec11ea5c5fd7d5bfe5d0d313b78dd363a998"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
