)]}'
{"id":"openstack%2Foctavia~673181","triplet_id":"openstack%2Foctavia~master~I349f2c0ea7498474209d9d38218184ca115dde75","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I349f2c0ea7498474209d9d38218184ca115dde75","subject":"Add DIB_PYTHON_VERSION to override Python version","status":"ABANDONED","created":"2019-07-28 16:49:01.000000000","updated":"2019-08-26 14:44:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"af030e06c72df6d64ebc1d8f86f30aefaae9c2d9","_number":673181,"virtual_id_number":673181,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-07-28 18:45:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-28 18:45:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7145ae8c87b89ee752343fcaa884dd0b4fde44b6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-28 16:49:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e69a61644c4f6b36dcf65f262299d51a5ead4f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-28 18:45:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/81/673181/1/check/openstack-tox-cover/9c58ff7/cover/ : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/673181/1/check/openstack-tox-lower-constraints/d8c46d1/ : SUCCESS in 6m 58s\n- openstack-tox-pep8 http://logs.openstack.org/81/673181/1/check/openstack-tox-pep8/b5b1a59/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/81/673181/1/check/openstack-tox-py27/77e1cd9/ : SUCCESS in 4m 13s\n- openstack-tox-py36 http://logs.openstack.org/81/673181/1/check/openstack-tox-py36/040a9c2/ : SUCCESS in 4m 46s\n- openstack-tox-py37 http://logs.openstack.org/81/673181/1/check/openstack-tox-py37/8d014a4/ : SUCCESS in 5m 20s\n- openstack-tox-docs http://logs.openstack.org/81/673181/1/check/openstack-tox-docs/de5c3d8/html/ : SUCCESS in 5m 11s\n- openstack-tox-functional http://logs.openstack.org/81/673181/1/check/openstack-tox-functional/2a81d6d/ : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 http://logs.openstack.org/81/673181/1/check/openstack-tox-functional-py36/f1558a1/ : SUCCESS in 5m 03s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/81/673181/1/check/octavia-v2-dsvm-noop-api/fb0edf7/ : SUCCESS in 41m 52s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/81/673181/1/check/octavia-v2-dsvm-noop-py2-api/3e42996/ : SUCCESS in 44m 56s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/81/673181/1/check/octavia-v2-dsvm-scenario/f61bee6/ : SUCCESS in 1h 36m 28s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/81/673181/1/check/octavia-v2-dsvm-py2-scenario/1c634ad/ : SUCCESS in 1h 29m 19s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/81/673181/1/check/octavia-v2-dsvm-py2-scenario-centos-7/6e09818/ : SUCCESS in 1h 55m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/81/673181/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/c555753/ : FAILURE in 1h 43m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/81/673181/1/check/octavia-v2-dsvm-scenario-two-node/5153c05/ : FAILURE in 1h 41m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/81/673181/1/check/octavia-v2-dsvm-py2-scenario-two-node/f3eb4cc/ : SUCCESS in 1h 12m 00s (non-voting)\n- octavia-grenade http://logs.openstack.org/81/673181/1/check/octavia-grenade/3d9d3d8/ : SUCCESS in 1h 21m 28s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/81/673181/1/check/octavia-v2-dsvm-tls-barbican/5409bdf/ : SUCCESS in 45m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/81/673181/1/check/octavia-v2-dsvm-scenario-amphora-v2/cbc8aa6/ : SUCCESS in 1h 26m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/81/673181/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/cf17776/ : SUCCESS in 1h 00m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/81/673181/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/1ac53ca/ : SUCCESS in 1h 16m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/81/673181/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/7b2e6fb/ : SUCCESS in 1h 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af030e06c72df6d64ebc1d8f86f30aefaae9c2d9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 14:44:29.000000000","message":"Abandoned\n\nSee https://review.opendev.org/#/c/673805/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a97833c43ae51c78fdb7bcc1ae0a59ae6fbb121d","revisions":{"a97833c43ae51c78fdb7bcc1ae0a59ae6fbb121d":{"kind":"REWORK","_number":1,"created":"2019-07-28 16:49:01.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/81/673181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/673181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/673181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/673181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/673181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/673181/1"}}},"commit":{"parents":[{"commit":"f4094381fdefffa2c23dfb6555765c48e5b77ba1","subject":"Merge \"Make Ubuntu bionic the default for amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4094381fdefffa2c23dfb6555765c48e5b77ba1"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 16:39:22.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-28 16:48:04.000000000","tz":120},"subject":"Add DIB_PYTHON_VERSION to override Python version","message":"Add DIB_PYTHON_VERSION to override Python version\n\nThe dib-python DIB element automatically picks the default Python\nversion for each distro. This patch allows Devstack users to override\nthe Python version the amphora-agent will run on.\n\nChange-Id: I349f2c0ea7498474209d9d38218184ca115dde75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a97833c43ae51c78fdb7bcc1ae0a59ae6fbb121d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a97833c43ae51c78fdb7bcc1ae0a59ae6fbb121d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
