)]}'
{"id":"openstack%2Foctavia~673337","triplet_id":"openstack%2Foctavia~master~I5bc21d5abdeb67a3a8ae88456735643463f15694","project":"openstack/octavia","branch":"master","topic":"lib-get","hashtags":[],"change_id":"I5bc21d5abdeb67a3a8ae88456735643463f15694","subject":"Lookup interfaces by MAC directly","status":"MERGED","created":"2019-07-29 16:51:52.000000000","updated":"2019-08-21 19:01:25.000000000","submitted":"2019-08-21 18:59:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673337-1566413953852-462ff7de","meta_rev_id":"e44a698ef763833de12eb3591891ea52255b448e","_number":673337,"virtual_id_number":673337,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-08-21 18:59:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-08-21 06:54:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-08-12 17:27:28.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-08-21 06:54:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-08-16 15:43:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-12 17:27:28.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-08-21 06:54:31.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-21 18:59:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6801d3ce2be145eab5109bdaea0c0b2176b9c81b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-29 16:51:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71e81c370a13b73eacc13f2d1ec3dd22dd60c51a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-29 16:54:18.000000000","message":"Patch Set 1: Workflow-1\n\nThis is \"experimental\"","accounts_in_message":[],"_revision_number":1},{"id":"0f18b147d3d0657ce1fc046f86287115cfc43e55","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-29 17:06:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7f52a311878ff9d9f1259f7a8353e20edd749cfd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-29 17:15:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4d68832c3ee2e39da6a14bd1419baa93c8f2d8d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 19:17:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/37/673337/2/check/openstack-tox-cover/d0b0626/ : FAILURE in 6m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/37/673337/2/check/openstack-tox-lower-constraints/70a94bc/ : FAILURE in 8m 07s\n- openstack-tox-pep8 http://logs.openstack.org/37/673337/2/check/openstack-tox-pep8/60397d3/ : SUCCESS in 6m 06s\n- openstack-tox-py27 http://logs.openstack.org/37/673337/2/check/openstack-tox-py27/d3c78c1/ : SUCCESS in 6m 33s\n- openstack-tox-py36 http://logs.openstack.org/37/673337/2/check/openstack-tox-py36/4e303f5/ : SUCCESS in 7m 07s\n- openstack-tox-py37 http://logs.openstack.org/37/673337/2/check/openstack-tox-py37/636c47f/ : SUCCESS in 5m 31s\n- openstack-tox-docs http://logs.openstack.org/37/673337/2/check/openstack-tox-docs/519de7e/html/ : SUCCESS in 4m 53s\n- openstack-tox-functional http://logs.openstack.org/37/673337/2/check/openstack-tox-functional/bf8e239/ : FAILURE in 5m 38s\n- openstack-tox-functional-py36 http://logs.openstack.org/37/673337/2/check/openstack-tox-functional-py36/b29a674/ : FAILURE in 6m 22s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/37/673337/2/check/octavia-v2-dsvm-noop-api/35e099e/ : SUCCESS in 46m 20s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/37/673337/2/check/octavia-v2-dsvm-noop-py2-api/5c666ee/ : SUCCESS in 47m 10s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/37/673337/2/check/octavia-v2-dsvm-scenario/9422905/ : SUCCESS in 1h 52m 23s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/37/673337/2/check/octavia-v2-dsvm-py2-scenario/805bc0a/ : SUCCESS in 1h 41m 35s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/37/673337/2/check/octavia-v2-dsvm-py2-scenario-centos-7/f9e8c93/ : FAILURE in 2h 05m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/37/673337/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/c731897/ : SUCCESS in 1h 37m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/37/673337/2/check/octavia-v2-dsvm-scenario-two-node/37b4e24/ : SUCCESS in 2h 02m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/37/673337/2/check/octavia-v2-dsvm-py2-scenario-two-node/ed04d07/ : FAILURE in 1h 24m 56s (non-voting)\n- octavia-grenade http://logs.openstack.org/37/673337/2/check/octavia-grenade/034ee4d/ : SUCCESS in 1h 25m 08s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/37/673337/2/check/octavia-v2-dsvm-tls-barbican/53622eb/ : SUCCESS in 1h 00m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/37/673337/2/check/octavia-v2-dsvm-scenario-amphora-v2/ca9f601/ : SUCCESS in 1h 53m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/37/673337/2/check/octavia-v2-act-stdby-iptables-dsvm-scenario/53962a0/ : SUCCESS in 1h 15m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/37/673337/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/72138bb/ : SUCCESS in 1h 15m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/37/673337/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/5f53375/ : SUCCESS in 1h 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4565b96c186960fca19479a65ba7a8fbb4f30d25","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-29 20:05:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9ed575cc5f834c4b0cb7143f759c817f030f0541","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 22:13:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/37/673337/3/check/openstack-tox-cover/253f56b/cover/ : SUCCESS in 8m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/37/673337/3/check/openstack-tox-lower-constraints/333ee85/ : SUCCESS in 7m 29s\n- openstack-tox-pep8 http://logs.openstack.org/37/673337/3/check/openstack-tox-pep8/e153a95/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/37/673337/3/check/openstack-tox-py27/5ad4096/ : SUCCESS in 4m 40s\n- openstack-tox-py36 http://logs.openstack.org/37/673337/3/check/openstack-tox-py36/83514db/ : SUCCESS in 4m 04s\n- openstack-tox-py37 http://logs.openstack.org/37/673337/3/check/openstack-tox-py37/bfb6d7f/ : SUCCESS in 6m 05s\n- openstack-tox-docs http://logs.openstack.org/37/673337/3/check/openstack-tox-docs/6ebfa73/html/ : SUCCESS in 6m 54s\n- openstack-tox-functional http://logs.openstack.org/37/673337/3/check/openstack-tox-functional/384b8b2/ : SUCCESS in 6m 04s\n- openstack-tox-functional-py36 http://logs.openstack.org/37/673337/3/check/openstack-tox-functional-py36/ef10a0f/ : SUCCESS in 6m 58s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/37/673337/3/check/octavia-v2-dsvm-noop-api/f83567c/ : SUCCESS in 48m 34s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/37/673337/3/check/octavia-v2-dsvm-noop-py2-api/6ad66fd/ : SUCCESS in 43m 52s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/37/673337/3/check/octavia-v2-dsvm-scenario/914448f/ : SUCCESS in 1h 35m 40s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/37/673337/3/check/octavia-v2-dsvm-py2-scenario/9892e94/ : FAILURE in 1h 30m 56s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/37/673337/3/check/octavia-v2-dsvm-py2-scenario-centos-7/b787db4/ : FAILURE in 2h 04m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/37/673337/3/check/octavia-v2-dsvm-scenario-two-node/8554045/ : FAILURE in 1h 32m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/37/673337/3/check/octavia-v2-dsvm-py2-scenario-two-node/476aa92/ : SUCCESS in 1h 46m 48s (non-voting)\n- octavia-grenade http://logs.openstack.org/37/673337/3/check/octavia-grenade/37ccf7a/ : SUCCESS in 1h 18m 48s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/37/673337/3/check/octavia-v2-dsvm-tls-barbican/904a8ed/ : SUCCESS in 55m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/37/673337/3/check/octavia-v2-dsvm-scenario-amphora-v2/0ade359/ : SUCCESS in 1h 39m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/37/673337/3/check/octavia-v2-act-stdby-iptables-dsvm-scenario/6c9cf14/ : SUCCESS in 1h 07m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/37/673337/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/0ca8161/ : SUCCESS in 1h 17m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/37/673337/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/110f892/ : SUCCESS in 1h 27m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f4d7abff38eb865d44a715f6c51745f5f1c0ba91","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-30 07:09:47.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6182c7e68a11102c65a004fa10aca8856da0ce00","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-30 23:40:00.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"4642da54d6dc34f2b6caf2d32e8f7fef37ce964f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 01:38:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/37/673337/4/check/openstack-tox-cover/ea364bb/ : FAILURE in 5m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/37/673337/4/check/openstack-tox-lower-constraints/0f7c75c/ : FAILURE in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/37/673337/4/check/openstack-tox-pep8/4c9d06f/ : SUCCESS in 5m 43s\n- openstack-tox-py27 http://logs.openstack.org/37/673337/4/check/openstack-tox-py27/d389047/ : FAILURE in 3m 56s\n- openstack-tox-py36 http://logs.openstack.org/37/673337/4/check/openstack-tox-py36/733be40/ : FAILURE in 5m 34s\n- openstack-tox-py37 http://logs.openstack.org/37/673337/4/check/openstack-tox-py37/70f04c9/ : FAILURE in 6m 35s\n- openstack-tox-docs http://logs.openstack.org/37/673337/4/check/openstack-tox-docs/fd83eba/html/ : SUCCESS in 6m 53s\n- openstack-tox-functional http://logs.openstack.org/37/673337/4/check/openstack-tox-functional/693502e/ : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 http://logs.openstack.org/37/673337/4/check/openstack-tox-functional-py36/ea76d8e/ : SUCCESS in 5m 57s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/37/673337/4/check/octavia-v2-dsvm-noop-api/d82ab9b/ : SUCCESS in 49m 26s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/37/673337/4/check/octavia-v2-dsvm-noop-py2-api/d3d2be0/ : SUCCESS in 45m 01s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/37/673337/4/check/octavia-v2-dsvm-scenario/47b2b34/ : SUCCESS in 1h 42m 35s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/37/673337/4/check/octavia-v2-dsvm-py2-scenario/afe3c65/ : SUCCESS in 1h 43m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/37/673337/4/check/octavia-v2-dsvm-py2-scenario-centos-7/52f5cc0/ : FAILURE in 1h 49m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/37/673337/4/check/octavia-v2-dsvm-scenario-two-node/6eb17b2/ : FAILURE in 1h 32m 28s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/37/673337/4/check/octavia-v2-dsvm-py2-scenario-two-node/eb71a1a/ : FAILURE in 1h 48m 21s (non-voting)\n- octavia-grenade http://logs.openstack.org/37/673337/4/check/octavia-grenade/9675e98/ : SUCCESS in 1h 14m 50s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/37/673337/4/check/octavia-v2-dsvm-tls-barbican/bd9fbf3/ : SUCCESS in 54m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/37/673337/4/check/octavia-v2-dsvm-scenario-amphora-v2/358e71d/ : SUCCESS in 1h 31m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/37/673337/4/check/octavia-v2-act-stdby-iptables-dsvm-scenario/f5d228d/ : FAILURE in 1h 15m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/37/673337/4/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/6d5e1f7/ : FAILURE in 1h 06m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/37/673337/4/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/f5a840c/ : FAILURE in 1h 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d755a818b9cf1995d81163a545b78b8a13c18a10","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-02 00:27:17.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"7a5bc3d92d74c9331a11685a2f93d4e23e90d6e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 02:26:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/37/673337/5/check/openstack-tox-cover/8969883/ : FAILURE in 5m 22s\n- openstack-tox-lower-constraints https://logs.opendev.org/37/673337/5/check/openstack-tox-lower-constraints/990ae12/ : FAILURE in 5m 51s\n- openstack-tox-pep8 https://logs.opendev.org/37/673337/5/check/openstack-tox-pep8/3d65700/ : SUCCESS in 6m 21s\n- openstack-tox-py27 https://logs.opendev.org/37/673337/5/check/openstack-tox-py27/1fb05d6/ : FAILURE in 4m 52s\n- openstack-tox-py36 https://logs.opendev.org/37/673337/5/check/openstack-tox-py36/6728e00/ : FAILURE in 4m 13s\n- openstack-tox-py37 https://logs.opendev.org/37/673337/5/check/openstack-tox-py37/fc48478/ : FAILURE in 5m 56s\n- openstack-tox-docs https://logs.opendev.org/37/673337/5/check/openstack-tox-docs/d063d00/html/ : SUCCESS in 6m 54s\n- openstack-tox-functional https://logs.opendev.org/37/673337/5/check/openstack-tox-functional/6d37a31/ : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://logs.opendev.org/37/673337/5/check/openstack-tox-functional-py36/7285bb6/ : SUCCESS in 6m 35s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/37/673337/5/check/octavia-v2-dsvm-noop-api/b898f3e/ : SUCCESS in 45m 30s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/37/673337/5/check/octavia-v2-dsvm-noop-py2-api/5b0c039/ : SUCCESS in 42m 45s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/37/673337/5/check/octavia-v2-dsvm-scenario/a727597/ : SUCCESS in 1h 37m 16s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/37/673337/5/check/octavia-v2-dsvm-py2-scenario/2a718ee/ : SUCCESS in 1h 47m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/37/673337/5/check/octavia-v2-dsvm-py2-scenario-centos-7/4e3a8b2/ : FAILURE in 35m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/37/673337/5/check/octavia-v2-dsvm-scenario-two-node/1b359d1/ : SUCCESS in 1h 51m 35s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/37/673337/5/check/octavia-v2-dsvm-py2-scenario-two-node/77aa53b/ : FAILURE in 1h 24m 17s (non-voting)\n- octavia-grenade https://logs.opendev.org/37/673337/5/check/octavia-grenade/015eb82/ : SUCCESS in 1h 23m 00s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/37/673337/5/check/octavia-v2-dsvm-tls-barbican/5491eba/ : SUCCESS in 54m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/37/673337/5/check/octavia-v2-dsvm-scenario-amphora-v2/db6b54c/ : SUCCESS in 1h 26m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/37/673337/5/check/octavia-v2-act-stdby-iptables-dsvm-scenario/cc73e27/ : FAILURE in 1h 08m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/37/673337/5/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/42034ff/ : FAILURE in 1h 19m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/37/673337/5/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/c4b57d1/ : FAILURE in 1h 14m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f1b53cad1aa1a448c04ace29a22df7019251cf79","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 13:41:50.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"223371f10ffde964a216bd8571fa8b79bda24880","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 14:15:03.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"0b7abeb0fd1b63b88184a496f6284beeb6e955fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 16:47:01.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/37/673337/7/check/openstack-tox-cover/8da39eb/cover/ : SUCCESS in 7m 01s\n- openstack-tox-lower-constraints https://logs.opendev.org/37/673337/7/check/openstack-tox-lower-constraints/f531bd3/ : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://logs.opendev.org/37/673337/7/check/openstack-tox-pep8/13c3959/ : SUCCESS in 5m 16s\n- openstack-tox-py27 https://logs.opendev.org/37/673337/7/check/openstack-tox-py27/a846fe0/ : SUCCESS in 4m 26s\n- openstack-tox-py36 https://logs.opendev.org/37/673337/7/check/openstack-tox-py36/fe6dd11/ : SUCCESS in 3m 16s\n- openstack-tox-py37 https://logs.opendev.org/37/673337/7/check/openstack-tox-py37/7883bca/ : SUCCESS in 4m 08s\n- openstack-tox-docs https://logs.opendev.org/37/673337/7/check/openstack-tox-docs/8acf910/ : FAILURE in 4m 08s\n- openstack-tox-functional https://logs.opendev.org/37/673337/7/check/openstack-tox-functional/2682c5f/ : SUCCESS in 4m 31s\n- openstack-tox-functional-py36 https://logs.opendev.org/37/673337/7/check/openstack-tox-functional-py36/3afd46f/ : SUCCESS in 4m 42s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/37/673337/7/check/octavia-v2-dsvm-noop-api/86870ba/ : SUCCESS in 41m 47s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/37/673337/7/check/octavia-v2-dsvm-noop-py2-api/fa1363f/ : SUCCESS in 44m 01s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/37/673337/7/check/octavia-v2-dsvm-scenario/ea40733/ : SUCCESS in 1h 36m 01s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/37/673337/7/check/octavia-v2-dsvm-py2-scenario/c200f0a/ : SUCCESS in 1h 37m 46s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/37/673337/7/check/octavia-v2-dsvm-py2-scenario-centos-7/511b1c9/ : FAILURE in 2h 26m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/37/673337/7/check/octavia-v2-dsvm-scenario-two-node/f1dde90/ : SUCCESS in 1h 58m 02s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/37/673337/7/check/octavia-v2-dsvm-py2-scenario-two-node/195d7f9/ : SUCCESS in 1h 47m 20s (non-voting)\n- octavia-grenade https://logs.opendev.org/37/673337/7/check/octavia-grenade/bd17fdd/ : SUCCESS in 1h 17m 52s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/37/673337/7/check/octavia-v2-dsvm-tls-barbican/5b656fb/ : SUCCESS in 50m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/37/673337/7/check/octavia-v2-dsvm-scenario-amphora-v2/1e112e3/ : SUCCESS in 1h 32m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/37/673337/7/check/octavia-v2-act-stdby-iptables-dsvm-scenario/70edbe2/ : FAILURE in 1h 23m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/37/673337/7/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/5ccd78a/ : FAILURE in 1h 25m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/37/673337/7/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/6d4dcd6/ : FAILURE in 1h 03m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9dac62dd6b79c8746d39ee4c7ddbc04062ea103b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-05 18:17:22.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"5a60412e9ef3132d6539405921a1a71669d38503","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-05 20:50:08.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"008f5f2802e57bf03816193005a59735847eef06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 20:55:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/37/673337/8/check/openstack-tox-cover/c91698a/cover/ : SUCCESS in 8m 44s\n- openstack-tox-lower-constraints https://logs.opendev.org/37/673337/8/check/openstack-tox-lower-constraints/46026bc/ : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://logs.opendev.org/37/673337/8/check/openstack-tox-pep8/7bf34ca/ : SUCCESS in 6m 47s\n- openstack-tox-py27 https://logs.opendev.org/37/673337/8/check/openstack-tox-py27/25eac36/ : SUCCESS in 6m 50s\n- openstack-tox-py36 https://logs.opendev.org/37/673337/8/check/openstack-tox-py36/0944960/ : SUCCESS in 6m 10s\n- openstack-tox-py37 https://logs.opendev.org/37/673337/8/check/openstack-tox-py37/e5e5049/ : SUCCESS in 5m 04s\n- openstack-tox-docs https://logs.opendev.org/37/673337/8/check/openstack-tox-docs/d95a295/html/ : SUCCESS in 5m 47s\n- openstack-tox-functional https://logs.opendev.org/37/673337/8/check/openstack-tox-functional/f4f4d2c/ : SUCCESS in 5m 45s\n- openstack-tox-functional-py36 https://logs.opendev.org/37/673337/8/check/openstack-tox-functional-py36/e327a21/ : SUCCESS in 6m 41s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/37/673337/8/check/octavia-v2-dsvm-noop-api/13d0c71/ : SUCCESS in 40m 13s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/37/673337/8/check/octavia-v2-dsvm-noop-py2-api/0048ab9/ : SUCCESS in 42m 24s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/37/673337/8/check/octavia-v2-dsvm-scenario/442693c/ : FAILURE in 38m 28s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/37/673337/8/check/octavia-v2-dsvm-py2-scenario/128ffab/ : SUCCESS in 1h 39m 40s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/37/673337/8/check/octavia-v2-dsvm-py2-scenario-centos-7/dabd8f4/ : TIMED_OUT in 2h 34m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/37/673337/8/check/octavia-v2-dsvm-scenario-two-node/67e2aa3/ : SUCCESS in 1h 59m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/37/673337/8/check/octavia-v2-dsvm-py2-scenario-two-node/69f46d8/ : FAILURE in 28m 36s (non-voting)\n- octavia-grenade https://logs.opendev.org/37/673337/8/check/octavia-grenade/13ccc3f/ : SUCCESS in 1h 26m 34s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/37/673337/8/check/octavia-v2-dsvm-tls-barbican/d74aebe/ : SUCCESS in 1h 01m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/37/673337/8/check/octavia-v2-dsvm-scenario-amphora-v2/c1fc09c/ : SUCCESS in 1h 34m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/37/673337/8/check/octavia-v2-act-stdby-iptables-dsvm-scenario/a094293/ : FAILURE in 1h 25m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/37/673337/8/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/9f88f98/ : FAILURE in 1h 22m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/37/673337/8/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/1444979/ : FAILURE in 29m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"60bed7082f546345deeb488c20862852e64399db","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-09 17:18:22.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8b8cf84b2859c37b2e0aecee9d11d52c417e361b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-09 17:18:50.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"3d0d29bd6e951da744149686eb53187b95469285","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-09 22:53:52.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"2b76482ab20bfb8ef72c521d6a32d3252ac9f4c0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-09 23:08:33.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"be0e6b5c4de7cc9931821f7756bdf75e9def8454","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 01:29:25.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/37/673337/12/check/openstack-tox-cover/8cc1932/cover/ : SUCCESS in 6m 46s\n- openstack-tox-lower-constraints https://logs.opendev.org/37/673337/12/check/openstack-tox-lower-constraints/c36d451/ : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://logs.opendev.org/37/673337/12/check/openstack-tox-pep8/1a7b04b/ : SUCCESS in 6m 26s\n- openstack-tox-py27 https://logs.opendev.org/37/673337/12/check/openstack-tox-py27/e25bb98/ : SUCCESS in 4m 44s\n- openstack-tox-py36 https://logs.opendev.org/37/673337/12/check/openstack-tox-py36/74a963c/ : SUCCESS in 4m 15s\n- openstack-tox-py37 https://logs.opendev.org/37/673337/12/check/openstack-tox-py37/5defbe2/ : SUCCESS in 5m 16s\n- openstack-tox-docs https://logs.opendev.org/37/673337/12/check/openstack-tox-docs/f2de73d/html/ : SUCCESS in 5m 21s\n- openstack-tox-functional https://logs.opendev.org/37/673337/12/check/openstack-tox-functional/b6b890f/ : SUCCESS in 5m 13s\n- openstack-tox-functional-py36 https://logs.opendev.org/37/673337/12/check/openstack-tox-functional-py36/3616e05/ : SUCCESS in 4m 39s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-noop-api/932456b/ : SUCCESS in 49m 26s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-noop-py2-api/95e67e0/ : SUCCESS in 50m 31s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-scenario/742b1d8/ : SUCCESS in 1h 43m 16s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-py2-scenario/71ef427/ : SUCCESS in 1h 28m 18s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-py2-scenario-centos-7/10b292f/ : FAILURE in 2h 07m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-scenario-two-node/3312940/ : FAILURE in 2h 09m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-py2-scenario-two-node/ed1c6a5/ : FAILURE in 2h 13m 32s (non-voting)\n- octavia-grenade https://logs.opendev.org/37/673337/12/check/octavia-grenade/8a2c3ab/ : SUCCESS in 1h 19m 43s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-tls-barbican/1551901/ : SUCCESS in 52m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-scenario-amphora-v2/4bed352/ : SUCCESS in 1h 30m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/37/673337/12/check/octavia-v2-act-stdby-iptables-dsvm-scenario/97a31ba/ : SUCCESS in 1h 05m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/37/673337/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/440c965/ : FAILURE in 43m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/37/673337/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/21ff4ff/ : SUCCESS in 1h 28m 10s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"056186e00f1da6bc905186de57b611737c922d34","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-10 01:43:38.000000000","message":"Patch Set 12:\n\nrecheck\n\nmaybe?","accounts_in_message":[],"_revision_number":12},{"id":"02925c003b3f419eee3c93500f820db0d0dcab79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 04:00:27.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/37/673337/12/check/openstack-tox-cover/79760ae/cover/ : SUCCESS in 6m 07s\n- openstack-tox-lower-constraints https://logs.opendev.org/37/673337/12/check/openstack-tox-lower-constraints/21e5823/ : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://logs.opendev.org/37/673337/12/check/openstack-tox-pep8/e391a1e/ : SUCCESS in 4m 52s\n- openstack-tox-py27 https://logs.opendev.org/37/673337/12/check/openstack-tox-py27/ae558db/ : SUCCESS in 3m 41s\n- openstack-tox-py36 https://logs.opendev.org/37/673337/12/check/openstack-tox-py36/0dd82c4/ : SUCCESS in 3m 21s\n- openstack-tox-py37 https://logs.opendev.org/37/673337/12/check/openstack-tox-py37/0bd84e8/ : SUCCESS in 4m 27s\n- openstack-tox-docs https://logs.opendev.org/37/673337/12/check/openstack-tox-docs/2803008/html/ : SUCCESS in 5m 01s\n- openstack-tox-functional https://logs.opendev.org/37/673337/12/check/openstack-tox-functional/32f92f2/ : SUCCESS in 5m 17s\n- openstack-tox-functional-py36 https://logs.opendev.org/37/673337/12/check/openstack-tox-functional-py36/e285002/ : SUCCESS in 4m 51s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-noop-api/4950dd3/ : SUCCESS in 43m 21s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-noop-py2-api/f665268/ : SUCCESS in 42m 36s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-scenario/dc62f4c/ : SUCCESS in 1h 43m 43s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-py2-scenario/91e90ca/ : SUCCESS in 1h 37m 13s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-py2-scenario-centos-7/5b088bc/ : FAILURE in 2h 15m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-scenario-two-node/c6dad50/ : SUCCESS in 1h 19m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-py2-scenario-two-node/bc05d69/ : SUCCESS in 1h 15m 17s (non-voting)\n- octavia-grenade https://logs.opendev.org/37/673337/12/check/octavia-grenade/3cc2953/ : SUCCESS in 1h 12m 10s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-tls-barbican/b512ea7/ : FAILURE in 1h 04m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/37/673337/12/check/octavia-v2-dsvm-scenario-amphora-v2/777c459/ : SUCCESS in 1h 31m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/37/673337/12/check/octavia-v2-act-stdby-iptables-dsvm-scenario/e8ca875/ : SUCCESS in 1h 03m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/37/673337/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/f952353/ : SUCCESS in 1h 16m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/37/673337/12/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/b18a095/ : FAILURE in 1h 05m 24s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"89b2adcdda7eca972c0fb9d157620df7c6d5291b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-10 08:38:18.000000000","message":"Patch Set 12:\n\nDid this break cent7 or do both of those happen to be failing on the same thing? \u003e_\u003e","accounts_in_message":[],"_revision_number":12},{"id":"b8ff1cb873138ed9d3fcf3eeb3885ed5688e6a6c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-12 17:27:28.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"0e2574cf8b27934f3fb8db52c45f8cb6fd5ded72","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-13 21:16:46.000000000","message":"Patch Set 12:\n\nIs is also used on line 187 where it does use the return for a log message.","accounts_in_message":[],"_revision_number":12},{"id":"d62fdb066bfba87e2feef0194209abfedd7d381c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-16 15:43:07.000000000","message":"Patch Set 12: Backport-Candidate+1","accounts_in_message":[],"_revision_number":12},{"id":"13f6c9024434e2a673bd222436b9790c38f594a5","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-20 22:49:38.000000000","message":"Patch Set 12:\n\nrecheck\n\nNo one answered my question, guess a recheck is the best way to easily check.","accounts_in_message":[],"_revision_number":12},{"id":"bd58b25073c97ab2f7b24530bb71a77254a70eda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 01:02:53.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d175c3f73027499b822a6fa8c92559b9 : SUCCESS in 7m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f2f8b3e426141cea8a626d2f31d3bd2 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d0d29bff7b1412488ad59195cf6a645 : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/781e0f4beb8c480a92a756065c75f978 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3d78f97a0ca42fb81ffa51718b2612c : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c0c4783ed2f4b579a21a31777b6e3e9 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d06acee67c04477b634742d5f9be420 : SUCCESS in 9m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d1c9c2d8107b40f78588e58092544e0d : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b2820aeca6c1453da3955d84ff97dd88 : SUCCESS in 6m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9c381d63976c420d8721e595ad9652ba : SUCCESS in 50m 29s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6c037b57b5e649da9d5d1efe3f4b5aae : SUCCESS in 49m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9668938068a54590837b254a2d1738b5 : SUCCESS in 1h 37m 30s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/955e2193878f4eaca333e902b94fb909 : SUCCESS in 1h 36m 47s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1889ef1336684c6f93abcacd56c3e94c : SUCCESS in 2h 04m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/af35113f2a994e4db908c273c59a1fbb : SUCCESS in 1h 13m 34s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4295c89840514004b67c8ce68ed2eafd : SUCCESS in 1h 20m 20s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aa92c6e8c20d40c48d25c3788f101542 : SUCCESS in 1h 21m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6088128bab8a4ca4ada8fcc8f279fc40 : SUCCESS in 58m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7fbddbc0d2fa424f84c2c919e1a3e657 : SUCCESS in 1h 40m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef03a00220214689ba472b8e2aa1a4ff : SUCCESS in 1h 14m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4a6591f7cbdf4faa9f0e97597f093ef6 : SUCCESS in 1h 21m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7f2e6c1502e84b99badced84a94e1360 : FAILURE in 1h 22m 25s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d41af79d2b41a51dbcc67a6fa3ebe5ee08851aa0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-21 06:53:13.000000000","message":"Patch Set 12:\n\ncentos7 job looks good.","accounts_in_message":[],"_revision_number":12},{"id":"da6b1e22b88b51f68fffdae0a028574ef4909d3c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-21 06:54:31.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"3cf8d3d01ca7b45e6130857afe5828d0282fc6be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 06:54:59.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"7d2c0f00e9b1fa020e788a5f13af2bb887cd6a2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 07:19:55.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37fe128c7ce048c0b08a464c6d818f98 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7feb016bc0774fc4be37a80299fdb6eb : SUCCESS in 8m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f85f937c71a439090c4714e7a8f1046 : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c94f6b60d554e5faf4e13662824df31 : SUCCESS in 7m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b92a9da24d44f91978490af96f006df : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae333f0a4f7243519d1073a5b46eb3d5 : SUCCESS in 8m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f18ab2b7f77e4970b0cd5fb31dbe29e6 : SUCCESS in 8m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6feb6372a6ac46eba9c2cfa2a3963885 : SUCCESS in 9m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc87863fbea44bd4b978eca3316a4e08 : CANCELED in 20m 43s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/45c794995ae3437d81cfe5b8e19e64b7 : CANCELED in 18m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a2686027e2fe43c989e032c56341cc85 : CANCELED in 23m 32s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9beb69e077ac416b989037a24d9b1bfa : FAILURE in 23m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e28f4e9be30144b2ac322c8407557d2f : CANCELED","accounts_in_message":[],"_revision_number":12},{"id":"4c026b08dedcdac6116871bee1c4aefa1c2e45c1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-21 07:27:32.000000000","message":"Patch Set 12:\n\nrecheck\n\nUgh, canceled jobs?","accounts_in_message":[],"_revision_number":12},{"id":"3d84a8d239ea95d599f69c5edde48de03e957983","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 09:14:21.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c43ba3eeba804eec9eda41b0861df804 : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77ef4cfd51a04b7aae3dedca34857ba3 : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b3b965bdba47269f715113328e4ac5 : SUCCESS in 7m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a15ee76459194883a37b2d5022fb66f6 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/385557ff660b4c8b8911b0d739b6d509 : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9699be30053842649b54cb6a40381f43 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65980ee3c6c84e8dbb9229ebf61f2217 : SUCCESS in 7m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/131f30ea9e684caf9de13496579da354 : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9fe796729f7d49ebaee60b603fb5c4ca : SUCCESS in 5m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ceeac4cc40434fe4a464631698ad52d6 : SUCCESS in 53m 34s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/eea0f12ee42748b4986870629cb893e2 : SUCCESS in 42m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bd1a54c5f5434ef99b15f387143a3dad : FAILURE in 28m 39s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9b58aadfeaae4585bea918ba8dcd6e42 : FAILURE in 28m 38s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e269c0d1227d4e65a3c21b61d4a535e7 : FAILURE in 37m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/24deb261682e470b99a817d2fb9acf34 : FAILURE in 1h 31m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/38909904bba44ebcaff3f096c7613659 : FAILURE in 1h 33m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5f98e6b0a964465e93601d0cd872f87f : SUCCESS in 1h 21m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/42c220566b7f43bba9429b561571a8b4 : FAILURE in 37m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c22a464fb1ab47d7af0d1da2ac3503fd : FAILURE in 33m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da701d58b5664442baa85f79146a8a80 : FAILURE in 32m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f980a910e4d2486eb68f4487fa818386 : FAILURE in 30m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ebdabc3e0a1542e9a861b0dc28c0d198 : FAILURE in 37m 35s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4eddc2e1766eeb406183ffbd2f69610c836c5974","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-21 09:26:40.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"0b109106930100be5d882b0152eba6b33b583e3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 11:45:52.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30648cb72dca4c3da7eba416ba3c2af9 : SUCCESS in 8m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc2342cddf7740ef83e8bf59c311e77a : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eefb88d3e3c4f83af19420ff0b72115 : SUCCESS in 6m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e20b17fdedc46ab9f64e6d6b116aed5 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ad61c9dd06343a881a565c2790b4110 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/93f2a31cdf3143fa9ff0161b08f42465 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c192ff60bc264d3dbf53fe45741817a5 : SUCCESS in 8m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3963922a510845c3ae17e6811dc3ca8e : SUCCESS in 8m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/06ebd2b7ad394b2b8957d961cdec560a : SUCCESS in 9m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bb68a07c02554a74ab0a20411c9b42ec : SUCCESS in 49m 05s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/aeea02ce320f4b759d105645ad50eb6e : SUCCESS in 46m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5234aca8bfb74fed81ea37492b128d55 : SUCCESS in 2h 11m 42s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/10e659ad1bb2470e9419885be5c3c9ce : SUCCESS in 2h 12m 04s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fc56882f4c9747a196317a6143722598 : SUCCESS in 2h 07m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/89b98e38524f438fa1f94e7441e31f5f : SUCCESS in 1h 59m 59s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/39a015ce27d14cf99ef639f81e473e42 : SUCCESS in 1h 53m 42s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/deec4d5d7dc5428782d609fb2af794c2 : SUCCESS in 1h 19m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6383b093629c4a60b60af0411d46311a : SUCCESS in 1h 03m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1cc019143ce941d0a3ac886ed493b10c : SUCCESS in 1h 47m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3e89cb3d050248b08f2edec5349f2f27 : SUCCESS in 1h 11m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/73b58a832f9f4c398500b8e8c5830d00 : SUCCESS in 1h 19m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1b0c2db2096a4cce97e496a252585cdd : SUCCESS in 1h 32m 44s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3d87798d760d12b8317eb4b7b6ef45fe8f8fcbbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 11:46:04.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"71455688f7942c90feb728322280a434fe832d2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 12:22:00.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8d383e11ee241a5986c2aff3408194d : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbb64365f1de45b29a0e725f2161f79a : SUCCESS in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2018bc6a7774165b11cf226273e2a5c : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6793e4b5015a4c6fb0ab3207f21e31d6 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60668346036f497dae8cfcb2f13f6c92 : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dded5efcb104912ab1a2de4344172f8 : SUCCESS in 8m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2fae2196f72640d29c57c1d78cf2b8ae : SUCCESS in 7m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5400ca9f2e834f9da1a993c62bc4ced6 : SUCCESS in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/99b7becb47ac41cdb53089d62715fb16 : CANCELED\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7519e3b3179849ccb89122d2480a8ddc : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/402ce829d2214934958753d71ba0e37b : FAILURE in 33m 41s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3f1cb59d1a2b4eaaa98ed9849a7aa586 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e584d643e7ae460bb86ea994d2086ee2 : CANCELED","accounts_in_message":[],"_revision_number":12},{"id":"c9a566650739ea3c73cf2544e3dd4cce2c9e358a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-21 14:42:17.000000000","message":"Patch Set 12:\n\nrecheck\nTemporary failure resolving \u0027mirror.regionone.fortnebula.opendev.org\u0027","accounts_in_message":[],"_revision_number":12},{"id":"200e8fc3c900ac3beca23aec34c263cd2bb648cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 17:06:00.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aedd303117584c02a3fbfb5824afad20 : SUCCESS in 7m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3c831bc8da14fe5835d87a93b6f0994 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0b1cec087044378813338dad43e4324 : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/471a8211b59c42b294a16677421e9284 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1792a13b19634ddaa6be4dc3b3acc39f : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff698685f8a64b9aabc91fef27992e24 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c1218a0b42142a6964d8f2e6bd259d6 : SUCCESS in 8m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/55c3ea589ec143d785ad0a06baef6e48 : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/04d214b67cd84666812a16af785e560d : SUCCESS in 7m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bee3e5be52884830a4ecacd0e114862a : SUCCESS in 50m 10s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c57e95afb9ce4b0598a6599f2ff374d2 : SUCCESS in 55m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0fa0aa085320472caed901c1a430e4f1 : SUCCESS in 1h 43m 38s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7d9309969794403ea51f945edbea98ca : SUCCESS in 1h 53m 43s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/41952f2197394dd1b26c5b313a2f51dd : SUCCESS in 2h 13m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c3e6d994a98d4476bd39a913385f4fae : SUCCESS in 2h 06m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/42238c611c404056b2594fad9673068e : SUCCESS in 2h 11m 10s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e9c5d80cebbd42e4ada707455263208f : SUCCESS in 1h 35m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/58de45f0c9b144b29c8e6f1f3df367b2 : SUCCESS in 47m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bb3ef785085342c2a4ae97ad07a788af : SUCCESS in 2h 13m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a545aeb5aa9e4d099a33a1d6457d29e1 : SUCCESS in 1h 29m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/29e61cf23e2149358cdfcbd0d8e9cccf : FAILURE in 23m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ad435332d22b46098b6cb61612ae1049 : SUCCESS in 1h 28m 37s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"9d051a3e0aeb77de007b8825ca713fe5e13e44e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 17:06:12.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"e15e7273983b4b2fde0048ff0109a7bc8f8bffe3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 18:59:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"35c7c83e014c7cd44e6229b89a1611100772deab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 18:59:13.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04e9f390750a45fba9addcf5928481a8 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e62545059234bcba9f21adfd41738c5 : SUCCESS in 6m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21ad09849abf49b7a9451f38d938d32b : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dcc6b54cf914f719a957410dbd40707 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e04f377c2bfb4da48dca2514c1a7862b : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6690fbf9c044458cb668ec5b1c6ce40b : SUCCESS in 8m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/336a4063f88b4bc1900d01d31ad909fa : SUCCESS in 7m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e8038a730ecd48b291b412698154cade : SUCCESS in 6m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3791f585a10b442988dd5e7410aa6b8c : SUCCESS in 46m 51s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ffc73ead89734b2786051f7b11716c79 : SUCCESS in 55m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20d42ea3bd284768b992d46f00a839dd : SUCCESS in 1h 41m 40s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/afbe8ec8e3434922a4c8319c732efd9e : SUCCESS in 1h 44m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1af5a460d8e446b2b1911aed042db073 : SUCCESS in 1h 41m 58s","accounts_in_message":[],"_revision_number":12},{"id":"3b16bcf7a780af84b69c88e2a9a169708423f7a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 19:01:25.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c8d6b00a22c4578a063710ba887f9c8 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"2529fa33abfe17c645c4fdf23b0098b5908f11aa","revisions":{"219f90938fe907aa586dba59580bcf492fccefc8":{"kind":"REWORK","_number":1,"created":"2019-07-29 16:51:52.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/673337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/673337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/673337/1"}}},"commit":{"parents":[{"commit":"f4094381fdefffa2c23dfb6555765c48e5b77ba1","subject":"Merge \"Make Ubuntu bionic the default for amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4094381fdefffa2c23dfb6555765c48e5b77ba1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 16:49:24.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 16:51:51.000000000","tz":-420},"subject":"Lookup interfaces by MAC directly","message":"Lookup interfaces by MAC directly\n\nCurrently the amphora agent will lookup interfaces using the\ninterface name determined earlier in the plug method. This can\nlead to a race condition with the udev interface renaming rule.\nThis patch changes the interface lookup to use the MAC address\ndirectly and not rely on the interface name.\n\nChange-Id: I5bc21d5abdeb67a3a8ae88456735643463f15694\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/219f90938fe907aa586dba59580bcf492fccefc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/219f90938fe907aa586dba59580bcf492fccefc8"}]},"branch":"refs/heads/master"},"d63bdfd475dc6292b059d0686538e8b8989494c7":{"kind":"REWORK","_number":2,"created":"2019-07-29 17:06:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/673337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/673337/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/673337/2"}}},"commit":{"parents":[{"commit":"f4094381fdefffa2c23dfb6555765c48e5b77ba1","subject":"Merge \"Make Ubuntu bionic the default for amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4094381fdefffa2c23dfb6555765c48e5b77ba1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 16:49:24.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 17:05:14.000000000","tz":-420},"subject":"Lookup interfaces by MAC directly","message":"Lookup interfaces by MAC directly\n\nCurrently the amphora agent will lookup interfaces using the\ninterface name determined earlier in the plug method. This can\nlead to a race condition with the udev interface renaming rule.\nThis patch changes the interface lookup to use the MAC address\ndirectly and not rely on the interface name.\n\nStory: 2006300\nTask: 36013\n\nChange-Id: I5bc21d5abdeb67a3a8ae88456735643463f15694\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d63bdfd475dc6292b059d0686538e8b8989494c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d63bdfd475dc6292b059d0686538e8b8989494c7"}]},"branch":"refs/heads/master"},"c021c8ff5b829c57f9bf31f1581def970ed2d00e":{"kind":"REWORK","_number":3,"created":"2019-07-29 20:05:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/673337/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/673337/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/673337/3"}}},"commit":{"parents":[{"commit":"f4094381fdefffa2c23dfb6555765c48e5b77ba1","subject":"Merge \"Make Ubuntu bionic the default for amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4094381fdefffa2c23dfb6555765c48e5b77ba1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 16:49:24.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 20:05:44.000000000","tz":-420},"subject":"Lookup interfaces by MAC directly","message":"Lookup interfaces by MAC directly\n\nCurrently the amphora agent will lookup interfaces using the\ninterface name determined earlier in the plug method. This can\nlead to a race condition with the udev interface renaming rule.\nThis patch changes the interface lookup to use the MAC address\ndirectly and not rely on the interface name.\n\nStory: 2006300\nTask: 36013\n\nChange-Id: I5bc21d5abdeb67a3a8ae88456735643463f15694\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c021c8ff5b829c57f9bf31f1581def970ed2d00e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c021c8ff5b829c57f9bf31f1581def970ed2d00e"}]},"branch":"refs/heads/master"},"c8eef8d1e84aafdbbfdc17e8349b9175a4b02ff9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-30 23:40:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/673337/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/673337/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/673337/4"}}},"commit":{"parents":[{"commit":"21f4f6b850154fd1a90c2f91f7daabd98612745d","subject":"Fix a unit test for new octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21f4f6b850154fd1a90c2f91f7daabd98612745d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 16:49:24.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 23:40:00.000000000","tz":0},"subject":"Lookup interfaces by MAC directly","message":"Lookup interfaces by MAC directly\n\nCurrently the amphora agent will lookup interfaces using the\ninterface name determined earlier in the plug method. This can\nlead to a race condition with the udev interface renaming rule.\nThis patch changes the interface lookup to use the MAC address\ndirectly and not rely on the interface name.\n\nStory: 2006300\nTask: 36013\n\nChange-Id: I5bc21d5abdeb67a3a8ae88456735643463f15694\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8eef8d1e84aafdbbfdc17e8349b9175a4b02ff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8eef8d1e84aafdbbfdc17e8349b9175a4b02ff9"}]},"branch":"refs/heads/master"},"0fb7901e502c289ce76b6c6822346940cd273b2f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-02 00:27:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/673337/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/673337/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/673337/5"}}},"commit":{"parents":[{"commit":"bd4a8b9261de635cb4f6da6cc377330a5be2ca39","subject":"Fix a unit test for new octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd4a8b9261de635cb4f6da6cc377330a5be2ca39"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 16:49:24.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-02 00:20:37.000000000","tz":-420},"subject":"Lookup interfaces by MAC directly","message":"Lookup interfaces by MAC directly\n\nCurrently the amphora agent will lookup interfaces using the\ninterface name determined earlier in the plug method. This can\nlead to a race condition with the udev interface renaming rule.\nThis patch changes the interface lookup to use the MAC address\ndirectly and not rely on the interface name.\n\nStory: 2006300\nTask: 36013\n\nChange-Id: I5bc21d5abdeb67a3a8ae88456735643463f15694\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0fb7901e502c289ce76b6c6822346940cd273b2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0fb7901e502c289ce76b6c6822346940cd273b2f"}]},"branch":"refs/heads/master"},"b272624f8a8e8d77b82b99f90ec62d9a851a28f9":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-08-03 13:41:50.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/37/673337/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/673337/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/673337/6"}}},"commit":{"parents":[{"commit":"179f9558c9cc40d02acc0f7ea64c3625523b3edc","subject":"Fix a unit test for new octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/179f9558c9cc40d02acc0f7ea64c3625523b3edc"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 16:49:24.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-03 13:41:50.000000000","tz":0},"subject":"Lookup interfaces by MAC directly","message":"Lookup interfaces by MAC directly\n\nCurrently the amphora agent will lookup interfaces using the\ninterface name determined earlier in the plug method. This can\nlead to a race condition with the udev interface renaming rule.\nThis patch changes the interface lookup to use the MAC address\ndirectly and not rely on the interface name.\n\nStory: 2006300\nTask: 36013\n\nChange-Id: I5bc21d5abdeb67a3a8ae88456735643463f15694\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b272624f8a8e8d77b82b99f90ec62d9a851a28f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b272624f8a8e8d77b82b99f90ec62d9a851a28f9"}]},"branch":"refs/heads/master"},"7c3c76422ccc8119f07a71223908216759c3a328":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-03 14:15:03.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/37/673337/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/673337/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/673337/7"}}},"commit":{"parents":[{"commit":"c58ef30d291f679ea2aac0b4f24cf1afe48f82eb","subject":"Fix a unit test for new octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c58ef30d291f679ea2aac0b4f24cf1afe48f82eb"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 16:49:24.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-03 14:15:03.000000000","tz":0},"subject":"Lookup interfaces by MAC directly","message":"Lookup interfaces by MAC directly\n\nCurrently the amphora agent will lookup interfaces using the\ninterface name determined earlier in the plug method. This can\nlead to a race condition with the udev interface renaming rule.\nThis patch changes the interface lookup to use the MAC address\ndirectly and not rely on the interface name.\n\nStory: 2006300\nTask: 36013\n\nChange-Id: I5bc21d5abdeb67a3a8ae88456735643463f15694\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c3c76422ccc8119f07a71223908216759c3a328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c3c76422ccc8119f07a71223908216759c3a328"}]},"branch":"refs/heads/master"},"c73dcc0a6cdc335b14a029b3f27a7ed64ad81027":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-08-05 18:17:22.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/37/673337/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/673337/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/673337/8"}}},"commit":{"parents":[{"commit":"b32064cdf2d5c70140a84e2497931941c4ee6887","subject":"Fix a unit test for new octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b32064cdf2d5c70140a84e2497931941c4ee6887"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 16:49:24.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-05 18:17:22.000000000","tz":0},"subject":"Lookup interfaces by MAC directly","message":"Lookup interfaces by MAC directly\n\nCurrently the amphora agent will lookup interfaces using the\ninterface name determined earlier in the plug method. This can\nlead to a race condition with the udev interface renaming rule.\nThis patch changes the interface lookup to use the MAC address\ndirectly and not rely on the interface name.\n\nStory: 2006300\nTask: 36013\n\nChange-Id: I5bc21d5abdeb67a3a8ae88456735643463f15694\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c73dcc0a6cdc335b14a029b3f27a7ed64ad81027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c73dcc0a6cdc335b14a029b3f27a7ed64ad81027"}]},"branch":"refs/heads/master"},"4a1e94fee9391d2e01085e07873d86036f8aca38":{"kind":"REWORK","_number":9,"created":"2019-08-09 17:18:22.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/673337/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/673337/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/673337/9"}}},"commit":{"parents":[{"commit":"b32064cdf2d5c70140a84e2497931941c4ee6887","subject":"Fix a unit test for new octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b32064cdf2d5c70140a84e2497931941c4ee6887"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 16:49:24.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-09 17:18:21.000000000","tz":-420},"subject":"Lookup interfaces by MAC directly","message":"Lookup interfaces by MAC directly\n\nCurrently the amphora agent will lookup interfaces using the\ninterface name determined earlier in the plug method. This can\nlead to a race condition with the udev interface renaming rule.\nThis patch changes the interface lookup to use the MAC address\ndirectly and not rely on the interface name.\n\nStory: 2006300\nTask: 36013\n\nChange-Id: I5bc21d5abdeb67a3a8ae88456735643463f15694\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a1e94fee9391d2e01085e07873d86036f8aca38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a1e94fee9391d2e01085e07873d86036f8aca38"}]},"branch":"refs/heads/master"},"77c1812808832aebc4310138256fb21248b418aa":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-08-09 17:18:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/673337/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/673337/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/673337/10"}}},"commit":{"parents":[{"commit":"6e4da8506469de634a4ef57538c7f3ffed3cc937","subject":"Merge \"Re-enable the py36 tox environment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e4da8506469de634a4ef57538c7f3ffed3cc937"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 16:49:24.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-09 17:18:50.000000000","tz":0},"subject":"Lookup interfaces by MAC directly","message":"Lookup interfaces by MAC directly\n\nCurrently the amphora agent will lookup interfaces using the\ninterface name determined earlier in the plug method. This can\nlead to a race condition with the udev interface renaming rule.\nThis patch changes the interface lookup to use the MAC address\ndirectly and not rely on the interface name.\n\nStory: 2006300\nTask: 36013\n\nChange-Id: I5bc21d5abdeb67a3a8ae88456735643463f15694\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77c1812808832aebc4310138256fb21248b418aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77c1812808832aebc4310138256fb21248b418aa"}]},"branch":"refs/heads/master"},"2d326446fdc750fe3eae2fc5502426372ec58fd7":{"kind":"REWORK","_number":11,"created":"2019-08-09 22:53:52.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/673337/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/673337/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/673337/11"}}},"commit":{"parents":[{"commit":"b32064cdf2d5c70140a84e2497931941c4ee6887","subject":"Fix a unit test for new octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b32064cdf2d5c70140a84e2497931941c4ee6887"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 16:49:24.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-09 22:53:46.000000000","tz":-420},"subject":"Lookup interfaces by MAC directly","message":"Lookup interfaces by MAC directly\n\nCurrently the amphora agent will lookup interfaces using the\ninterface name determined earlier in the plug method. This can\nlead to a race condition with the udev interface renaming rule.\nThis patch changes the interface lookup to use the MAC address\ndirectly and not rely on the interface name.\n\nStory: 2006300\nTask: 36013\n\nChange-Id: I5bc21d5abdeb67a3a8ae88456735643463f15694\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d326446fdc750fe3eae2fc5502426372ec58fd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d326446fdc750fe3eae2fc5502426372ec58fd7"}]},"branch":"refs/heads/master"},"2529fa33abfe17c645c4fdf23b0098b5908f11aa":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-08-09 23:08:33.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/673337/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/673337/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/673337/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/673337/12"}}},"commit":{"parents":[{"commit":"6e4da8506469de634a4ef57538c7f3ffed3cc937","subject":"Merge \"Re-enable the py36 tox environment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e4da8506469de634a4ef57538c7f3ffed3cc937"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 16:49:24.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-09 23:08:33.000000000","tz":0},"subject":"Lookup interfaces by MAC directly","message":"Lookup interfaces by MAC directly\n\nCurrently the amphora agent will lookup interfaces using the\ninterface name determined earlier in the plug method. This can\nlead to a race condition with the udev interface renaming rule.\nThis patch changes the interface lookup to use the MAC address\ndirectly and not rely on the interface name.\n\nStory: 2006300\nTask: 36013\n\nChange-Id: I5bc21d5abdeb67a3a8ae88456735643463f15694\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2529fa33abfe17c645c4fdf23b0098b5908f11aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2529fa33abfe17c645c4fdf23b0098b5908f11aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
