)]}'
{"id":"openstack%2Foctavia~673344","triplet_id":"openstack%2Foctavia~master~I6be0a593b1deb43f8aba982043ebf427be57d937","project":"openstack/octavia","branch":"master","topic":"fix-story-2006304","hashtags":[],"change_id":"I6be0a593b1deb43f8aba982043ebf427be57d937","subject":"Correcting typo in healthmonitors-list-response.json - http_vesion to http_version.","status":"MERGED","created":"2019-07-29 17:37:40.000000000","updated":"2019-07-29 19:17:42.000000000","submitted":"2019-07-29 19:17:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673344-1564427862725-bedbf1ba","meta_rev_id":"5a9c9964fd25d584316d7de0a8ff0bcec8b04dd4","_number":673344,"virtual_id_number":673344,"owner":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-07-29 19:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-07-29 19:00:01.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-29 18:13:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2019-07-29 19:00:01.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2019-07-29 18:13:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-29 18:13:45.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-07-29 19:00:01.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-07-29 19:17:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4d56d4d43d918f4246abfd713f2405d5a94f77c3","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2019-07-29 17:37:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de8d23cd86f2bc46363b33189e4c72a97fc76eab","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-29 17:47:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThanks for the patch. For future reference, there\u0027s no need to open a story for fixing typos.","accounts_in_message":[],"_revision_number":1},{"id":"348231e0ce835bc2ac3ec46e9b99682885f95146","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2019-07-29 17:48:51.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c48a942c3dac627ae2e0c9b4f1e2372213604504","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 18:01:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/44/673344/2/check/build-openstack-api-ref/0738cd3/html/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/44/673344/2/check/openstack-tox-pep8/7c28ccc/ : SUCCESS in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/44/673344/2/check/openstack-tox-docs/4d8e8a6/html/ : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":2},{"id":"f7d6f01917fb19000532d873b7ec5bc699913a00","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2019-07-29 18:05:06.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bdc1c9dcdaaa31d7c6034c0913045188ead579d3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-29 18:13:45.000000000","message":"Patch Set 3: Backport-Candidate-2 Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"925ca6baa969e27ffc1893c84ae9437885507043","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 18:35:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/44/673344/3/check/build-openstack-api-ref/94f821e/html/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/44/673344/3/check/openstack-tox-pep8/15a003e/ : SUCCESS in 5m 25s\n- openstack-tox-docs http://logs.openstack.org/44/673344/3/check/openstack-tox-docs/a1db20c/html/ : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":3},{"id":"0b255a3b7461ee7ca03911853cf6f8745ddbe3c2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-29 19:00:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"96c3da62d2aa662cb896c9ea25316af523658712","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 19:00:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2a73a636789fe637b79ab1af7418adcadaec4fcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 19:17:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/44/673344/3/gate/build-openstack-api-ref/cd8741a/html/ : SUCCESS in 5m 57s\n- openstack-tox-pep8 http://logs.openstack.org/44/673344/3/gate/openstack-tox-pep8/a4e8a91/ : SUCCESS in 5m 26s\n- openstack-tox-docs http://logs.openstack.org/44/673344/3/gate/openstack-tox-docs/7f143d1/html/ : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":3},{"id":"6811236923b56705b3700613006ed2cf5458d3b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 19:17:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"52485738b3772156273132e7a8b9d67eb0b5e91e","revisions":{"e4c38176f6b11e95659128df9b689650fa610ab5":{"kind":"REWORK","_number":1,"created":"2019-07-29 17:37:40.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/44/673344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/673344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/673344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/673344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/673344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/673344/1"}}},"commit":{"parents":[{"commit":"f4094381fdefffa2c23dfb6555765c48e5b77ba1","subject":"Merge \"Make Ubuntu bionic the default for amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4094381fdefffa2c23dfb6555765c48e5b77ba1"}]}],"author":{"name":"root","email":"sandyada@redhat.com","date":"2019-07-29 12:06:28.000000000","tz":330},"committer":{"name":"root","email":"sandyada@redhat.com","date":"2019-07-29 12:08:31.000000000","tz":330},"subject":"Task: 36020 Story: 2006304","message":"Task: 36020\nStory: 2006304\n\nCorrecting typographical error in healthmonitors-list-response.json from http_vesion to http_version.\n\nChange-Id: I6be0a593b1deb43f8aba982043ebf427be57d937\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4c38176f6b11e95659128df9b689650fa610ab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4c38176f6b11e95659128df9b689650fa610ab5"}]},"branch":"refs/heads/master"},"66bd81aa864899c8e11407a63f982e2af621d549":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-29 17:48:51.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/44/673344/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/673344/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/673344/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/673344/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/673344/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/673344/2"}}},"commit":{"parents":[{"commit":"f4094381fdefffa2c23dfb6555765c48e5b77ba1","subject":"Merge \"Make Ubuntu bionic the default for amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4094381fdefffa2c23dfb6555765c48e5b77ba1"}]}],"author":{"name":"root","email":"sandyada@redhat.com","date":"2019-07-29 12:06:28.000000000","tz":330},"committer":{"name":"root","email":"sandyada@redhat.com","date":"2019-07-29 12:17:27.000000000","tz":330},"subject":"Correcting typo in healthmonitors-list-response.json - http_vesion to http_version.","message":"Correcting typo in healthmonitors-list-response.json - http_vesion to http_version.\n\nCorrecting typographical error in api-ref/source/v2/examples/healthmonitors-list-response.json from http_vesion to http_version.\n\nTask: 36020\nStory: 2006304\nChange-Id: I6be0a593b1deb43f8aba982043ebf427be57d937\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66bd81aa864899c8e11407a63f982e2af621d549"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66bd81aa864899c8e11407a63f982e2af621d549"}]},"branch":"refs/heads/master"},"52485738b3772156273132e7a8b9d67eb0b5e91e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-07-29 18:05:06.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/44/673344/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/673344/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/673344/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/673344/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/673344/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/673344/3"}}},"commit":{"parents":[{"commit":"f4094381fdefffa2c23dfb6555765c48e5b77ba1","subject":"Merge \"Make Ubuntu bionic the default for amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4094381fdefffa2c23dfb6555765c48e5b77ba1"}]}],"author":{"name":"root","email":"sandyada@redhat.com","date":"2019-07-29 12:06:28.000000000","tz":330},"committer":{"name":"root","email":"sandyada@redhat.com","date":"2019-07-29 12:34:53.000000000","tz":330},"subject":"Correcting typo in healthmonitors-list-response.json - http_vesion to http_version.","message":"Correcting typo in healthmonitors-list-response.json - http_vesion to http_version.\n\nThere is a typographical error in healthmonitors-list-response.json.\nCorrecting spelling from http_vesion to http_version.\n\nTask: 36020\nStory: 2006304\nChange-Id: I6be0a593b1deb43f8aba982043ebf427be57d937\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/52485738b3772156273132e7a8b9d67eb0b5e91e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/52485738b3772156273132e7a8b9d67eb0b5e91e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
