)]}'
{"id":"openstack%2Foctavia~673359","triplet_id":"openstack%2Foctavia~master~I6dc4a7626cd5244b445893b96a7d7351ca528dc1","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I6dc4a7626cd5244b445893b96a7d7351ca528dc1","subject":"Add missing lib dependencies to requirements.txt","status":"MERGED","created":"2019-07-29 19:54:43.000000000","updated":"2019-08-07 07:30:42.000000000","submitted":"2019-08-07 07:30:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673359-1565163042183-25fc65c6","meta_rev_id":"9c8ff5d1d56363bb6c32a450765a8b396ddc6f29","_number":673359,"virtual_id_number":673359,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-08-07 07:30:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-08-06 15:42:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-08-07 01:04:03.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-08-07 01:04:03.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 15:42:24.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-07 01:04:03.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-08-07 07:30:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec21ca066c778239fac3640b8d2f868b0ca16927","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-29 19:54:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa0de80873e570ed968cb94f6fe6c0ee53439724","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 22:07:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/59/673359/1/check/requirements-check/040658d/ : SUCCESS in 2m 07s\n- openstack-tox-cover http://logs.openstack.org/59/673359/1/check/openstack-tox-cover/aadfa70/cover/ : SUCCESS in 7m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/59/673359/1/check/openstack-tox-lower-constraints/d5a2310/ : SUCCESS in 7m 01s\n- openstack-tox-pep8 http://logs.openstack.org/59/673359/1/check/openstack-tox-pep8/020ddcd/ : SUCCESS in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/59/673359/1/check/openstack-tox-py27/3ae4a06/ : SUCCESS in 3m 51s\n- openstack-tox-py36 http://logs.openstack.org/59/673359/1/check/openstack-tox-py36/605b2d2/ : SUCCESS in 4m 12s\n- openstack-tox-py37 http://logs.openstack.org/59/673359/1/check/openstack-tox-py37/a040684/ : SUCCESS in 4m 29s\n- openstack-tox-docs http://logs.openstack.org/59/673359/1/check/openstack-tox-docs/3455a47/html/ : SUCCESS in 6m 54s\n- openstack-tox-functional http://logs.openstack.org/59/673359/1/check/openstack-tox-functional/ee5174d/ : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 http://logs.openstack.org/59/673359/1/check/openstack-tox-functional-py36/65b0a3c/ : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/59/673359/1/check/octavia-v2-dsvm-noop-api/ab31d00/ : SUCCESS in 45m 27s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/59/673359/1/check/octavia-v2-dsvm-noop-py2-api/7617af9/ : SUCCESS in 52m 35s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/59/673359/1/check/octavia-v2-dsvm-scenario/cee2bd9/ : SUCCESS in 1h 52m 55s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/59/673359/1/check/octavia-v2-dsvm-py2-scenario/0aa54a8/ : SUCCESS in 1h 48m 43s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/59/673359/1/check/octavia-v2-dsvm-py2-scenario-centos-7/928fcb7/ : SUCCESS in 2h 09m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/59/673359/1/check/octavia-v2-dsvm-scenario-two-node/01aa2d8/ : SUCCESS in 2h 08m 04s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/59/673359/1/check/octavia-v2-dsvm-py2-scenario-two-node/67f2906/ : SUCCESS in 2h 05m 24s (non-voting)\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/59/673359/1/check/octavia-v2-dsvm-tls-barbican/31b7024/ : SUCCESS in 1h 01m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/59/673359/1/check/octavia-v2-dsvm-scenario-amphora-v2/be831b8/ : SUCCESS in 1h 29m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/59/673359/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/9c337c2/ : SUCCESS in 1h 04m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/59/673359/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/9f0dd99/ : FAILURE in 56m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/59/673359/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/fe9f9c3/ : FAILURE in 52m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd0b27062f9b6b3a4c1a7385628be094c1b21589","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 00:38:24.000000000","message":"Patch Set 1: Code-Review+2\n\nAh yes, I guess for all three of those we assumed they would be required as a dependency of another package, but that\u0027s not the correct way to do dependencies. :D","accounts_in_message":[],"_revision_number":1},{"id":"6f20fd4cf0f335fec6285ddf11df289d0d71be28","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-08-03 23:15:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d20b2afb1e61dfddc135623c005cefbedef17afc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 23:15:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee7509551efb2368a8c7bbd8986bd7f817eda8d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 23:20:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/59/673359/1/gate/requirements-check/be722f8/ : SUCCESS in 1m 52s\n- openstack-tox-lower-constraints finger://ze01.openstack.org/a7aac522bdee413d972e0852ea6b2dd5 : CANCELED in 4m 25s\n- openstack-tox-pep8 https://logs.opendev.org/59/673359/1/gate/openstack-tox-pep8/14d9a75/ : SUCCESS in 4m 08s\n- openstack-tox-py27 finger://ze03.openstack.org/ffa5c470624d46e58c7a68e45195e187 : CANCELED in 4m 28s\n- openstack-tox-py36 finger://ze01.openstack.org/8811cc539ca349a0aad861cf7a377596 : CANCELED in 4m 25s\n- openstack-tox-py37 finger://ze02.openstack.org/875ada41842d4bac93c6b086b9a61156 : CANCELED in 4m 26s\n- openstack-tox-docs https://logs.opendev.org/59/673359/1/gate/openstack-tox-docs/7134b8c/ : FAILURE in 4m 26s\n- openstack-tox-functional https://logs.opendev.org/59/673359/1/gate/openstack-tox-functional/0875e8c/ : SUCCESS in 4m 25s\n- openstack-tox-functional-py36 https://logs.opendev.org/59/673359/1/gate/openstack-tox-functional-py36/07f1178/ : SUCCESS in 4m 16s\n- octavia-v2-dsvm-noop-api finger://ze02.openstack.org/41c230ad003745028ffe4213ad252501 : CANCELED in 4m 12s\n- octavia-v2-dsvm-noop-py2-api finger://ze03.openstack.org/9e300d1575644897b1bc3813665c8775 : CANCELED in 4m 04s\n- octavia-v2-dsvm-scenario finger://ze05.openstack.org/37747bb11a3749ec987efeb7b9de8996 : CANCELED in 4m 11s\n- octavia-v2-dsvm-py2-scenario finger://ze01.openstack.org/75a016c060094aa7af541909fd1316f3 : CANCELED in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b5d6ec50bfea542c8caa998ee14c31e77939fe77","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-08-05 01:21:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"76deb5bce5e0762b4487b91ca114aa2f947d7330","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 03:51:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/59/673359/1/check/requirements-check/eb75acd/ : SUCCESS in 3m 05s\n- openstack-tox-cover https://logs.opendev.org/59/673359/1/check/openstack-tox-cover/4e5c39e/ : FAILURE in 6m 20s\n- openstack-tox-lower-constraints https://logs.opendev.org/59/673359/1/check/openstack-tox-lower-constraints/3efa5a4/ : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://logs.opendev.org/59/673359/1/check/openstack-tox-pep8/dc1e1f0/ : SUCCESS in 4m 53s\n- openstack-tox-py27 https://logs.opendev.org/59/673359/1/check/openstack-tox-py27/85c75c0/ : FAILURE in 7m 18s\n- openstack-tox-py36 https://logs.opendev.org/59/673359/1/check/openstack-tox-py36/5932c03/ : FAILURE in 6m 03s\n- openstack-tox-py37 https://logs.opendev.org/59/673359/1/check/openstack-tox-py37/8d8437e/ : FAILURE in 7m 31s\n- openstack-tox-docs https://logs.opendev.org/59/673359/1/check/openstack-tox-docs/320036b/ : FAILURE in 4m 21s\n- openstack-tox-functional https://logs.opendev.org/59/673359/1/check/openstack-tox-functional/f144ef2/ : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://logs.opendev.org/59/673359/1/check/openstack-tox-functional-py36/817409b/ : SUCCESS in 5m 11s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/59/673359/1/check/octavia-v2-dsvm-noop-api/5d192f1/ : SUCCESS in 48m 38s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/59/673359/1/check/octavia-v2-dsvm-noop-py2-api/614e4be/ : SUCCESS in 41m 51s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/59/673359/1/check/octavia-v2-dsvm-scenario/a99f5d4/ : SUCCESS in 1h 32m 45s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/59/673359/1/check/octavia-v2-dsvm-py2-scenario/420eb57/ : SUCCESS in 1h 35m 12s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/673359/1/check/octavia-v2-dsvm-py2-scenario-centos-7/0e4b9aa/ : FAILURE in 2h 27m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/59/673359/1/check/octavia-v2-dsvm-scenario-two-node/b7ca1fc/ : SUCCESS in 1h 09m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/59/673359/1/check/octavia-v2-dsvm-py2-scenario-two-node/6379948/ : SUCCESS in 1h 04m 45s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/59/673359/1/check/octavia-v2-dsvm-tls-barbican/7065b4f/ : SUCCESS in 45m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/59/673359/1/check/octavia-v2-dsvm-scenario-amphora-v2/cc63cd2/ : SUCCESS in 1h 23m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/59/673359/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/f5b2dfd/ : FAILURE in 1h 10m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/59/673359/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/7d1007e/ : FAILURE in 1h 11m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/673359/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/308ec14/ : FAILURE in 1h 03m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8ab567a2f02935ca8338c62a63f75d28850df6e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-06 15:42:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3fd834f7b9be16da8853936235121172e9769a76","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-06 15:42:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7fbe7c4c39b0b91fe50c6e5ce3d0984639415115","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 18:36:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://logs.opendev.org/59/673359/2/check/requirements-check/b6e93f4/ : SUCCESS in 2m 32s\n- openstack-tox-cover https://logs.opendev.org/59/673359/2/check/openstack-tox-cover/3f924f8/cover/ : SUCCESS in 8m 05s\n- openstack-tox-lower-constraints https://logs.opendev.org/59/673359/2/check/openstack-tox-lower-constraints/b45dfe8/ : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://logs.opendev.org/59/673359/2/check/openstack-tox-pep8/402cb61/ : SUCCESS in 7m 04s\n- openstack-tox-py27 https://logs.opendev.org/59/673359/2/check/openstack-tox-py27/ff5254e/ : SUCCESS in 4m 22s\n- openstack-tox-py36 https://logs.opendev.org/59/673359/2/check/openstack-tox-py36/7cb85ac/ : SUCCESS in 6m 01s\n- openstack-tox-py37 https://logs.opendev.org/59/673359/2/check/openstack-tox-py37/9c4fc5f/ : SUCCESS in 4m 38s\n- openstack-tox-docs https://logs.opendev.org/59/673359/2/check/openstack-tox-docs/34d7e3a/html/ : SUCCESS in 6m 38s\n- openstack-tox-functional https://logs.opendev.org/59/673359/2/check/openstack-tox-functional/67b393e/ : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://logs.opendev.org/59/673359/2/check/openstack-tox-functional-py36/b7614ff/ : SUCCESS in 5m 02s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-noop-api/f6dea1a/ : SUCCESS in 46m 04s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-noop-py2-api/ac07697/ : SUCCESS in 46m 01s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-scenario/6efe09f/ : SUCCESS in 1h 32m 25s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-py2-scenario/2fc8763/ : SUCCESS in 1h 43m 00s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-py2-scenario-centos-7/22a8dd2/ : TIMED_OUT in 2h 34m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-scenario-two-node/dea64db/ : FAILURE in 1h 38m 23s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-py2-scenario-two-node/ad5b3c0/ : SUCCESS in 1h 21m 55s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-tls-barbican/48fac1e/ : SUCCESS in 50m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-scenario-amphora-v2/0e224fe/ : SUCCESS in 1h 33m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/59/673359/2/check/octavia-v2-act-stdby-iptables-dsvm-scenario/bb6901e/ : FAILURE in 1h 09m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/59/673359/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/93bb223/ : FAILURE in 1h 21m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/673359/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/39c578d/ : FAILURE in 1h 16m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"42c5aeb3cb3657a56a354953b0735baceed54322","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-08-07 01:04:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"011d55f2dc63e781e9564748bc21a7e718b222c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 01:04:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e40a02979454a4381d185aeae32bd1c788653363","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 02:34:59.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/59/673359/2/gate/requirements-check/22aa09a/ : SUCCESS in 2m 10s\n- openstack-tox-lower-constraints https://logs.opendev.org/59/673359/2/gate/openstack-tox-lower-constraints/2109863/ : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://logs.opendev.org/59/673359/2/gate/openstack-tox-pep8/b701830/ : SUCCESS in 5m 04s\n- openstack-tox-py27 https://logs.opendev.org/59/673359/2/gate/openstack-tox-py27/fb60222/ : SUCCESS in 4m 08s\n- openstack-tox-py36 https://logs.opendev.org/59/673359/2/gate/openstack-tox-py36/e1a8d84/ : SUCCESS in 4m 45s\n- openstack-tox-py37 https://logs.opendev.org/59/673359/2/gate/openstack-tox-py37/fe9cd3f/ : SUCCESS in 5m 54s\n- openstack-tox-docs https://logs.opendev.org/59/673359/2/gate/openstack-tox-docs/4c3cdbd/html/ : SUCCESS in 5m 19s\n- openstack-tox-functional https://logs.opendev.org/59/673359/2/gate/openstack-tox-functional/d199af4/ : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://logs.opendev.org/59/673359/2/gate/openstack-tox-functional-py36/83737c5/ : SUCCESS in 5m 40s\n- octavia-v2-dsvm-noop-api finger://ze03.openstack.org/3ffcc9abb03c4ddc9c0aee484c15c6e1 : CANCELED in 22m 46s\n- octavia-v2-dsvm-noop-py2-api finger://ze02.openstack.org/4aca4b64131744ef9e90650a2221846e : CANCELED in 21m 11s\n- octavia-v2-dsvm-scenario finger://ze05.openstack.org/79ae8e2ca57243a580a457e94ca72657 : CANCELED in 22m 39s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/59/673359/2/gate/octavia-v2-dsvm-py2-scenario/d6c5c53/ : FAILURE in 23m 08s","accounts_in_message":[],"_revision_number":2},{"id":"55411a754e7235b9bfd0ad31c9e0dd20947d52da","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-08-07 02:50:05.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8faf2a20ccb167473ffab03ee675fc3025d61014","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 05:41:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://logs.opendev.org/59/673359/2/check/requirements-check/5dffba9/ : SUCCESS in 2m 11s\n- openstack-tox-cover https://logs.opendev.org/59/673359/2/check/openstack-tox-cover/c31331c/cover/ : SUCCESS in 6m 07s\n- openstack-tox-lower-constraints https://logs.opendev.org/59/673359/2/check/openstack-tox-lower-constraints/7a1f607/ : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://logs.opendev.org/59/673359/2/check/openstack-tox-pep8/9ea46c9/ : SUCCESS in 4m 42s\n- openstack-tox-py27 https://logs.opendev.org/59/673359/2/check/openstack-tox-py27/250d149/ : SUCCESS in 4m 06s\n- openstack-tox-py36 https://logs.opendev.org/59/673359/2/check/openstack-tox-py36/dd69d0b/ : SUCCESS in 3m 43s\n- openstack-tox-py37 https://logs.opendev.org/59/673359/2/check/openstack-tox-py37/5efd2b0/ : SUCCESS in 5m 06s\n- openstack-tox-docs https://logs.opendev.org/59/673359/2/check/openstack-tox-docs/25b6d6f/html/ : SUCCESS in 5m 03s\n- openstack-tox-functional https://logs.opendev.org/59/673359/2/check/openstack-tox-functional/8710972/ : SUCCESS in 5m 12s\n- openstack-tox-functional-py36 https://logs.opendev.org/59/673359/2/check/openstack-tox-functional-py36/fbd9b0e/ : SUCCESS in 4m 42s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-noop-api/f03e96f/ : SUCCESS in 48m 53s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-noop-py2-api/c598205/ : SUCCESS in 43m 12s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-scenario/c559b88/ : SUCCESS in 1h 37m 45s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-py2-scenario/650d185/ : SUCCESS in 1h 38m 12s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-py2-scenario-centos-7/d035d66/ : TIMED_OUT in 2h 36m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-scenario-two-node/3259a6f/ : SUCCESS in 1h 18m 00s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-py2-scenario-two-node/18926c4/ : SUCCESS in 1h 21m 37s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-tls-barbican/4218781/ : SUCCESS in 54m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/59/673359/2/check/octavia-v2-dsvm-scenario-amphora-v2/02da5a5/ : SUCCESS in 1h 35m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/59/673359/2/check/octavia-v2-act-stdby-iptables-dsvm-scenario/7291a19/ : SUCCESS in 1h 12m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/59/673359/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/1e8d734/ : FAILURE in 1h 16m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/673359/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/824f8d6/ : FAILURE in 1h 16m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"83ca78e82623f0ffc3d7b5fe437744f57f7eb9a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 05:41:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5d5a1f922b0735975ba94816a699e7caa7910d44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 07:30:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://logs.opendev.org/59/673359/2/gate/requirements-check/e4a2a33/ : SUCCESS in 2m 26s\n- openstack-tox-lower-constraints https://logs.opendev.org/59/673359/2/gate/openstack-tox-lower-constraints/40b2450/ : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://logs.opendev.org/59/673359/2/gate/openstack-tox-pep8/5f04973/ : SUCCESS in 5m 28s\n- openstack-tox-py27 https://logs.opendev.org/59/673359/2/gate/openstack-tox-py27/fe0306b/ : SUCCESS in 4m 40s\n- openstack-tox-py36 https://logs.opendev.org/59/673359/2/gate/openstack-tox-py36/13ad7b6/ : SUCCESS in 4m 41s\n- openstack-tox-py37 https://logs.opendev.org/59/673359/2/gate/openstack-tox-py37/8944ec4/ : SUCCESS in 5m 30s\n- openstack-tox-docs https://logs.opendev.org/59/673359/2/gate/openstack-tox-docs/32e3ae4/html/ : SUCCESS in 6m 04s\n- openstack-tox-functional https://logs.opendev.org/59/673359/2/gate/openstack-tox-functional/17fd667/ : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://logs.opendev.org/59/673359/2/gate/openstack-tox-functional-py36/acd6c33/ : SUCCESS in 5m 07s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/59/673359/2/gate/octavia-v2-dsvm-noop-api/8d2e15d/ : SUCCESS in 50m 37s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/59/673359/2/gate/octavia-v2-dsvm-noop-py2-api/b5afd2e/ : SUCCESS in 45m 06s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/59/673359/2/gate/octavia-v2-dsvm-scenario/df41a08/ : SUCCESS in 1h 39m 12s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/59/673359/2/gate/octavia-v2-dsvm-py2-scenario/ebbe130/ : SUCCESS in 1h 46m 33s","accounts_in_message":[],"_revision_number":2},{"id":"9c8ff5d1d56363bb6c32a450765a8b396ddc6f29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 07:30:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"262a89c7ed5d3da113e4f1c62130efe12b8defb8","revisions":{"d7523d82e5eaef5dda5b032b3481e0bf91ab64ce":{"kind":"REWORK","_number":1,"created":"2019-07-29 19:54:43.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/59/673359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/673359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/673359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/673359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/673359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/673359/1"}}},"commit":{"parents":[{"commit":"f4094381fdefffa2c23dfb6555765c48e5b77ba1","subject":"Merge \"Make Ubuntu bionic the default for amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4094381fdefffa2c23dfb6555765c48e5b77ba1"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-29 19:51:50.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-29 19:54:29.000000000","tz":120},"subject":"Add missing lib dependencies to requirements.txt","message":"Add missing lib dependencies to requirements.txt\n\nSome Python libraries were missing in requirements.txt. This patch adds\nthem to the list of dependencies.\n\nChange-Id: I6dc4a7626cd5244b445893b96a7d7351ca528dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7523d82e5eaef5dda5b032b3481e0bf91ab64ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7523d82e5eaef5dda5b032b3481e0bf91ab64ce"}]},"branch":"refs/heads/master"},"262a89c7ed5d3da113e4f1c62130efe12b8defb8":{"kind":"REWORK","_number":2,"created":"2019-08-06 15:42:01.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/59/673359/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/673359/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/673359/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/673359/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/673359/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/673359/2"}}},"commit":{"parents":[{"commit":"b32064cdf2d5c70140a84e2497931941c4ee6887","subject":"Fix a unit test for new octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b32064cdf2d5c70140a84e2497931941c4ee6887"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-29 19:51:50.000000000","tz":120},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-06 15:41:34.000000000","tz":-420},"subject":"Add missing lib dependencies to requirements.txt","message":"Add missing lib dependencies to requirements.txt\n\nSome Python libraries were missing in requirements.txt. This patch adds\nthem to the list of dependencies.\n\nChange-Id: I6dc4a7626cd5244b445893b96a7d7351ca528dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/262a89c7ed5d3da113e4f1c62130efe12b8defb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/262a89c7ed5d3da113e4f1c62130efe12b8defb8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
