)]}'
{"id":"openstack%2Foctavia~673518","triplet_id":"openstack%2Foctavia~stable%2Fstein~Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b","project":"openstack/octavia","branch":"stable/stein","topic":"single-process","hashtags":[],"change_id":"Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b","subject":"Fix multi-listener load balancers","status":"MERGED","created":"2019-07-30 13:28:11.000000000","updated":"2019-08-30 20:34:13.000000000","submitted":"2019-08-30 20:32:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673518-1567197134755-1d8aaa17","meta_rev_id":"afbff17b6e27a9e9a36ad272d64a1fbea0ed6745","_number":673518,"virtual_id_number":673518,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-08-30 20:32:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-08-20 06:20:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-08-15 17:15:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-31 09:52:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-08-30 18:27:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2019-08-01 07:09:06.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-31 09:52:28.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-08-01 07:09:06.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-08-20 06:20:57.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-30 18:27:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-30 20:32:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f45b95a8d08cf2911673c504299b4672656334b9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-30 13:28:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6015ad501b43148ca5cfc91ca3265a52999c4bdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 16:04:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/18/673518/1/check/openstack-tox-cover/d32ba38/cover/ : SUCCESS in 9m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/673518/1/check/openstack-tox-lower-constraints/b934bc7/ : SUCCESS in 7m 15s\n- openstack-tox-pep8 http://logs.openstack.org/18/673518/1/check/openstack-tox-pep8/eeb3c69/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/18/673518/1/check/openstack-tox-py27/5fb27ff/ : SUCCESS in 4m 09s\n- openstack-tox-py35 http://logs.openstack.org/18/673518/1/check/openstack-tox-py35/bedc738/ : SUCCESS in 5m 31s\n- openstack-tox-py36 http://logs.openstack.org/18/673518/1/check/openstack-tox-py36/a699870/ : SUCCESS in 4m 06s\n- openstack-tox-docs http://logs.openstack.org/18/673518/1/check/openstack-tox-docs/1220511/html/ : SUCCESS in 5m 14s\n- build-openstack-releasenotes http://logs.openstack.org/18/673518/1/check/build-openstack-releasenotes/3067a6d/html/ : SUCCESS in 4m 46s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/18/673518/1/check/neutron-lbaasv2-dsvm-api/406a021/ : SUCCESS in 1h 25m 45s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/18/673518/1/check/neutron-lbaasv2-dsvm-py3x-api/74250e8/ : SUCCESS in 1h 27m 03s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/18/673518/1/check/neutron-lbaasv2-dsvm-scenario/c142a44/ : SUCCESS in 1h 31m 00s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/18/673518/1/check/neutron-lbaasv2-dsvm-py3x-scenario/356cd1b/ : FAILURE in 1h 13m 16s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/18/673518/1/check/openstack-tox-functional/df0f04a/ : SUCCESS in 6m 46s\n- openstack-tox-functional-py35 http://logs.openstack.org/18/673518/1/check/openstack-tox-functional-py35/01da32c/ : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 http://logs.openstack.org/18/673518/1/check/openstack-tox-functional-py36/e1a3fe4/ : SUCCESS in 6m 50s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/18/673518/1/check/octavia-v1-dsvm-scenario/70d9b3b/ : SUCCESS in 1h 14m 38s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/18/673518/1/check/octavia-v1-dsvm-py3x-scenario/9e9b4de/ : SUCCESS in 1h 14m 14s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/18/673518/1/check/octavia-v2-dsvm-noop-api/46fcc8f/ : SUCCESS in 55m 48s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/18/673518/1/check/octavia-v2-dsvm-noop-py2-api/e21cea2/ : SUCCESS in 48m 44s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/18/673518/1/check/octavia-v2-dsvm-scenario/f5472dc/ : SUCCESS in 1h 29m 45s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/18/673518/1/check/octavia-v2-dsvm-py2-scenario/1283c9d/ : SUCCESS in 1h 46m 14s\n- octavia-v2-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/18/673518/1/check/octavia-v2-dsvm-scenario-ubuntu-xenial/97c7227/ : SUCCESS in 1h 54m 57s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/18/673518/1/check/octavia-v2-dsvm-py2-scenario-centos-7/9ca644b/ : FAILURE in 2h 25m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/18/673518/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/25f604e/ : FAILURE in 41m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/18/673518/1/check/octavia-v2-dsvm-scenario-two-node/cf710f9/ : SUCCESS in 1h 56m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/18/673518/1/check/octavia-v2-dsvm-py2-scenario-two-node/5893abc/ : FAILURE in 1h 37m 45s (non-voting)\n- octavia-grenade http://logs.openstack.org/18/673518/1/check/octavia-grenade/18e719a/ : SUCCESS in 1h 13m 35s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/18/673518/1/check/octavia-v2-dsvm-tls-barbican/c54463d/ : SUCCESS in 50m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f9d0fd8f8de0e628f23a9fdc0ee59df8b604ffd6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-07-31 09:21:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"533e3fa9ec9736a21bbd702e50bc91ba3208ef5b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-31 09:41:42.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3134288b571ffe206a3f0309817068b391c15d95","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-31 09:44:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c98da6582ec95021eb52b744ba12107f09c42f47","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-07-31 09:52:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"875bed12a68532a2c362a33c1c411968bfc6788a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 12:09:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/18/673518/2/check/openstack-tox-cover/8c70608/cover/ : SUCCESS in 7m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/673518/2/check/openstack-tox-lower-constraints/466f1c3/ : SUCCESS in 7m 27s\n- openstack-tox-pep8 http://logs.openstack.org/18/673518/2/check/openstack-tox-pep8/d392ca4/ : SUCCESS in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/18/673518/2/check/openstack-tox-py27/ba6d20d/ : SUCCESS in 4m 46s\n- openstack-tox-py35 http://logs.openstack.org/18/673518/2/check/openstack-tox-py35/e7544b7/ : SUCCESS in 4m 37s\n- openstack-tox-py36 http://logs.openstack.org/18/673518/2/check/openstack-tox-py36/45f607c/ : SUCCESS in 4m 12s\n- openstack-tox-docs http://logs.openstack.org/18/673518/2/check/openstack-tox-docs/0292b52/html/ : SUCCESS in 4m 43s\n- build-openstack-releasenotes http://logs.openstack.org/18/673518/2/check/build-openstack-releasenotes/c3b96c6/html/ : SUCCESS in 4m 44s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/18/673518/2/check/neutron-lbaasv2-dsvm-api/b4b71b0/ : SUCCESS in 1h 18m 31s\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/18/673518/2/check/neutron-lbaasv2-dsvm-py3x-api/4a3718d/ : SUCCESS in 1h 19m 24s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/18/673518/2/check/neutron-lbaasv2-dsvm-scenario/68bc30f/ : SUCCESS in 1h 25m 42s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/18/673518/2/check/neutron-lbaasv2-dsvm-py3x-scenario/ff8ab23/ : SUCCESS in 1h 20m 42s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/18/673518/2/check/openstack-tox-functional/bb8d00c/ : SUCCESS in 7m 00s\n- openstack-tox-functional-py35 http://logs.openstack.org/18/673518/2/check/openstack-tox-functional-py35/fafac7f/ : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 http://logs.openstack.org/18/673518/2/check/openstack-tox-functional-py36/2469462/ : SUCCESS in 6m 57s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/18/673518/2/check/octavia-v1-dsvm-scenario/8a19eb7/ : SUCCESS in 1h 02m 52s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/18/673518/2/check/octavia-v1-dsvm-py3x-scenario/500a3ea/ : SUCCESS in 1h 03m 12s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/18/673518/2/check/octavia-v2-dsvm-noop-api/e4ba83a/ : SUCCESS in 45m 23s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/18/673518/2/check/octavia-v2-dsvm-noop-py2-api/aba71fe/ : SUCCESS in 44m 08s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/18/673518/2/check/octavia-v2-dsvm-scenario/0024cc6/ : SUCCESS in 1h 34m 58s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/18/673518/2/check/octavia-v2-dsvm-py2-scenario/4a56a3b/ : SUCCESS in 1h 49m 02s\n- octavia-v2-dsvm-scenario-ubuntu-xenial http://logs.openstack.org/18/673518/2/check/octavia-v2-dsvm-scenario-ubuntu-xenial/7d02630/ : SUCCESS in 1h 40m 39s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/18/673518/2/check/octavia-v2-dsvm-py2-scenario-centos-7/e40137f/ : SUCCESS in 2h 18m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic http://logs.openstack.org/18/673518/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/ada4465/ : FAILURE in 46m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/18/673518/2/check/octavia-v2-dsvm-scenario-two-node/f3f7e11/ : SUCCESS in 1h 43m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/18/673518/2/check/octavia-v2-dsvm-py2-scenario-two-node/39d5698/ : FAILURE in 1h 41m 20s (non-voting)\n- octavia-grenade http://logs.openstack.org/18/673518/2/check/octavia-grenade/05c4b70/ : SUCCESS in 1h 19m 15s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/18/673518/2/check/octavia-v2-dsvm-tls-barbican/b4fc922/ : SUCCESS in 55m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6abdbc875591d5be1ee1ec63a0bbf4d6c781576d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-09 18:53:24.000000000","message":"Patch Set 2: Code-Review-1\n\nKinda want to sit on this for a little bit until we\u0027re sure we worked all the kinks out of this patch on master... We know it did break some functionality (thought you\u0027ve got that followup already proposed) so I\u0027d like to give it a little longer to simmer first before we backport to a stable branch.\n-1 to indicate that I\u0027d like to wait a bit, nothing you need to do in the meantime.","accounts_in_message":[],"_revision_number":2},{"id":"dbabdc8cb9da3600a0b56fec7355658f99e80d6e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-15 17:15:22.000000000","message":"Patch Set 2: Code-Review+2\n\nI am ok moving forward with this backport as long as we pull the related fix back at the same time.\nI think the issue it resolves is severe and the longer we wait, the harder it will be to pull back other bug fixes.","accounts_in_message":[],"_revision_number":2},{"id":"5951fec199e5d9dff73d9ff395dbadfc3979fd05","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-20 06:20:57.000000000","message":"Patch Set 2: Code-Review+2\n\nDo we have a lot more patches waiting for backports? I guess that\u0027s fair? I do wish we could spend some time letting it get more fully tested, but I\u0027m ok with this if you want. I\u0027ll let you +A.","accounts_in_message":[],"_revision_number":2},{"id":"93d74d5389574ac29faf1210cb65750bf2461c8e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-30 18:27:57.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"660633369e393600c1f29763066b5aca602d3b79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 18:28:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"de8f96c74a3fa2bca9f8b51b85843f90c9f49a94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 20:32:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"11f54410a3c5238ad5cab97e6ccc0d03dbf07842","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 20:32:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af9b0dc1297a4680b142de36897d2a4d : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0067ac6839c8419698fcf8a61fbd77b7 : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e9f3e036f6642deb9c4fd5471313417 : SUCCESS in 4m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/aa81b791697b480b98547f9120a63814 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff64f72cff4d477eb41a5b7734f8537d : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c95a6305d054b609441fa2f36974b7d : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c44371e987024f6dbb99d88ff65d882e : SUCCESS in 5m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7882985772154635a4db2be830c7e84b : SUCCESS in 8m 11s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/83d07e5cb1984bb79b582adcaded23ef : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e9ddabff11af487aaa90447ffc2b91f1 : SUCCESS in 6m 17s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aeb3ce7f9eaa4df2b898f7084159d046 : SUCCESS in 1h 05m 24s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/850c843897d8474389bfef24883852a5 : SUCCESS in 1h 00m 07s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d9948c4785ac4ae0b838cf327ead1587 : SUCCESS in 1h 25m 16s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/778562a879bd48f9982b645e2a19486c : SUCCESS in 1h 29m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e51d5126f0d64db9b62e7bbd6e832b2d : SUCCESS in 42m 23s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/35403889ad2f40af87a0c2c839af7a75 : SUCCESS in 44m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ce019366bc94db1864520843eb70961 : SUCCESS in 1h 38m 58s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6695151612234750945fb4c780a79d1f : SUCCESS in 1h 25m 33s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/3fc723f325f7438da410ae9fbfdb8d69 : SUCCESS in 1h 52m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aada51cc7d2c43f7a39f27166297c9cb : SUCCESS in 1h 12m 42s","accounts_in_message":[],"_revision_number":2},{"id":"a2524189b4e369acbeeb19b36e1c67d75d231070","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 20:34:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88a27e9f813e465e802aa3ec30924898 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"646071d85298a3f54630afb986917b5d0c05cdd8","revisions":{"7afc98e5ab677dd6c66bf4c0bfb5bf2ce73ed78f":{"kind":"REWORK","_number":1,"created":"2019-07-30 13:28:11.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/18/673518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/18/673518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/673518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/673518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/673518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/18/673518/1"}}},"commit":{"parents":[{"commit":"15358a71e4aeffee8a3283ed08137e3b6daab52e","subject":"Merge \"Update tox.ini for new upper constraints strategy\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15358a71e4aeffee8a3283ed08137e3b6daab52e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-30 13:18:06.000000000","tz":120},"subject":"Fix multi-listener load balancers","message":"Fix multi-listener load balancers\n\nLoad balancers with multiple listeners, running on an amphora image\nwith HAProxy 1.8 or newer can experience excessive memory usage that\nmay lead to an ERROR provisioning_status.\nThis patch resolves this issue by consolidating the listeners into\na single haproxy process inside the amphora.\n\nConflicts:\n    octavia/amphorae/backends/agent/api_server/loadbalancer.py\n    octavia/amphorae/backends/agent/api_server/udp_listener_base.py\n    octavia/amphorae/backends/agent/api_server/util.py\n    octavia/amphorae/drivers/driver_base.py\n    octavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n    octavia/common/jinja/haproxy/combined_listeners/templates/base.j2\n    octavia/common/jinja/haproxy/combined_listeners/templates/macros.j2\n    octavia/controller/worker/tasks/amphora_driver_tasks.py\n    octavia/controller/worker/v1/flows/amphora_flows.py\n    octavia/controller/worker/v1/flows/l7policy_flows.py\n    octavia/controller/worker/v1/flows/listener_flows.py\n    octavia/controller/worker/v1/flows/load_balancer_flows.py\n    octavia/controller/worker/v1/flows/member_flows.py\n    octavia/controller/worker/v1/flows/pool_flows.py\n    octavia/controller/worker/v2/controller_worker.py\n    octavia/controller/worker/v2/flows/health_monitor_flows.py\n    octavia/controller/worker/v2/flows/l7rule_flows.py\n    octavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n    octavia/tests/unit/common/jinja/haproxy/combined_listeners/test_jinja_cfg.py\n    octavia/tests/unit/common/jinja/haproxy/split_listeners/test_jinja_cfg.py\n    octavia/tests/unit/common/sample_configs/sample_configs_combined.py\n    octavia/tests/unit/common/sample_configs/sample_configs_split.py\n    octavia/tests/unit/controller/worker/v1/tasks/test_amphora_driver_tasks.py\n    octavia/tests/unit/controller/worker/v2/flows/test_load_balancer_flows.py\n    octavia/tests/unit/controller/worker/v2/test_controller_worker.py\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n(cherry picked from commit 06ce4777c3bad2868e906101bcd2cd7c753b4859)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7afc98e5ab677dd6c66bf4c0bfb5bf2ce73ed78f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7afc98e5ab677dd6c66bf4c0bfb5bf2ce73ed78f"}]},"branch":"refs/heads/stable/stein"},"646071d85298a3f54630afb986917b5d0c05cdd8":{"kind":"REWORK","_number":2,"created":"2019-07-31 09:44:10.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/18/673518/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/18/673518/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/673518/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/673518/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/18/673518/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/18/673518/2"}}},"commit":{"parents":[{"commit":"15358a71e4aeffee8a3283ed08137e3b6daab52e","subject":"Merge \"Update tox.ini for new upper constraints strategy\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15358a71e4aeffee8a3283ed08137e3b6daab52e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-27 23:00:22.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-31 09:43:59.000000000","tz":120},"subject":"Fix multi-listener load balancers","message":"Fix multi-listener load balancers\n\nLoad balancers with multiple listeners, running on an amphora image\nwith HAProxy 1.8 or newer can experience excessive memory usage that\nmay lead to an ERROR provisioning_status.\nThis patch resolves this issue by consolidating the listeners into\na single haproxy process inside the amphora.\n\nConflicts:\n    octavia/amphorae/backends/agent/api_server/loadbalancer.py\n    octavia/amphorae/backends/agent/api_server/udp_listener_base.py\n    octavia/amphorae/backends/agent/api_server/util.py\n    octavia/amphorae/drivers/driver_base.py\n    octavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n    octavia/common/jinja/haproxy/combined_listeners/templates/base.j2\n    octavia/common/jinja/haproxy/combined_listeners/templates/macros.j2\n    octavia/controller/worker/tasks/amphora_driver_tasks.py\n    octavia/controller/worker/v1/flows/amphora_flows.py\n    octavia/controller/worker/v1/flows/l7policy_flows.py\n    octavia/controller/worker/v1/flows/listener_flows.py\n    octavia/controller/worker/v1/flows/load_balancer_flows.py\n    octavia/controller/worker/v1/flows/member_flows.py\n    octavia/controller/worker/v1/flows/pool_flows.py\n    octavia/controller/worker/v2/controller_worker.py\n    octavia/controller/worker/v2/flows/health_monitor_flows.py\n    octavia/controller/worker/v2/flows/l7rule_flows.py\n    octavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n    octavia/tests/unit/common/jinja/haproxy/combined_listeners/test_jinja_cfg.py\n    octavia/tests/unit/common/jinja/haproxy/split_listeners/test_jinja_cfg.py\n    octavia/tests/unit/common/sample_configs/sample_configs_combined.py\n    octavia/tests/unit/common/sample_configs/sample_configs_split.py\n    octavia/tests/unit/controller/worker/v1/tasks/test_amphora_driver_tasks.py\n    octavia/tests/unit/controller/worker/v2/flows/test_load_balancer_flows.py\n    octavia/tests/unit/controller/worker/v2/test_controller_worker.py\n\nStory: 2005412\nTask: 34744\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Idaccbcfa0126f1e26fbb3ad770c65c9266cfad5b\n(cherry picked from commit 06ce4777c3bad2868e906101bcd2cd7c753b4859)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/646071d85298a3f54630afb986917b5d0c05cdd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/646071d85298a3f54630afb986917b5d0c05cdd8"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
