)]}'
{"id":"openstack%2Foctavia~673687","triplet_id":"openstack%2Foctavia~master~Ife93834654385430d6a333d3d6ee67b719b6c37e","project":"openstack/octavia","branch":"master","topic":"lib-get","hashtags":[],"change_id":"Ife93834654385430d6a333d3d6ee67b719b6c37e","subject":"Fix a unit test for new octavia-lib","status":"MERGED","created":"2019-07-30 20:52:38.000000000","updated":"2019-08-06 15:18:57.000000000","submitted":"2019-08-06 15:18:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673687-1565104737233-27e79566","meta_rev_id":"5ee52d8a877f9ceec7638d185e1ab106f55618a2","_number":673687,"virtual_id_number":673687,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2019-08-06 08:48:03.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-08-06 15:18:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-08-05 15:50:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-05 16:32:02.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-08-05 16:32:02.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":-2,"date":"2019-08-05 16:32:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 15:50:20.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-05 16:32:02.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-08-06 08:48:03.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-08-06 15:18:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa1f4b0fb4c8da91aa403fc68d2f1cca1b4e1778","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-30 20:52:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6af1b104d3c69c2e7e226adec9ae5d70715abc6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 23:21:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/87/673687/1/check/openstack-tox-cover/72f70ba/ : FAILURE in 4m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/673687/1/check/openstack-tox-lower-constraints/df43f17/ : FAILURE in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/87/673687/1/check/openstack-tox-pep8/03f3dd1/ : SUCCESS in 4m 36s\n- openstack-tox-py27 http://logs.openstack.org/87/673687/1/check/openstack-tox-py27/a2d6074/ : FAILURE in 4m 24s\n- openstack-tox-py36 http://logs.openstack.org/87/673687/1/check/openstack-tox-py36/53d5b55/ : FAILURE in 3m 37s\n- openstack-tox-py37 http://logs.openstack.org/87/673687/1/check/openstack-tox-py37/e07b629/ : FAILURE in 5m 14s\n- openstack-tox-docs http://logs.openstack.org/87/673687/1/check/openstack-tox-docs/d3c592e/html/ : SUCCESS in 5m 10s\n- openstack-tox-functional http://logs.openstack.org/87/673687/1/check/openstack-tox-functional/ca21ac6/ : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 http://logs.openstack.org/87/673687/1/check/openstack-tox-functional-py36/e3c84e1/ : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/87/673687/1/check/octavia-v2-dsvm-noop-api/f455ef6/ : SUCCESS in 51m 55s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/87/673687/1/check/octavia-v2-dsvm-noop-py2-api/ff1361d/ : SUCCESS in 49m 19s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/87/673687/1/check/octavia-v2-dsvm-scenario/e3aada2/ : SUCCESS in 1h 33m 32s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/87/673687/1/check/octavia-v2-dsvm-py2-scenario/70b40d0/ : SUCCESS in 2h 25m 13s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/87/673687/1/check/octavia-v2-dsvm-py2-scenario-centos-7/73bf2bb/ : SUCCESS in 2h 06m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/87/673687/1/check/octavia-v2-dsvm-scenario-two-node/09fed06/ : SUCCESS in 2h 00m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/87/673687/1/check/octavia-v2-dsvm-py2-scenario-two-node/04325c1/ : SUCCESS in 2h 11m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/87/673687/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/bd354ee/ : FAILURE in 1h 18m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/87/673687/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/f556375/ : FAILURE in 1h 50m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/87/673687/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/61ed46e/ : FAILURE in 1h 37m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e567cda34fa1f9ddb2c7f2a21fbd8520797c3ee2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-02 00:27:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"09b1207821e6e1364616bb8ebd34d642729299fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 03:14:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/87/673687/2/check/openstack-tox-cover/2800777/ : FAILURE in 5m 07s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/673687/2/check/openstack-tox-lower-constraints/f3233d7/ : FAILURE in 5m 35s\n- openstack-tox-pep8 https://logs.opendev.org/87/673687/2/check/openstack-tox-pep8/b8ab544/ : SUCCESS in 6m 07s\n- openstack-tox-py27 https://logs.opendev.org/87/673687/2/check/openstack-tox-py27/3179ab8/ : FAILURE in 5m 27s\n- openstack-tox-py36 https://logs.opendev.org/87/673687/2/check/openstack-tox-py36/d36750a/ : FAILURE in 4m 01s\n- openstack-tox-py37 https://logs.opendev.org/87/673687/2/check/openstack-tox-py37/1aa223c/ : FAILURE in 4m 18s\n- openstack-tox-docs https://logs.opendev.org/87/673687/2/check/openstack-tox-docs/b61f2b6/html/ : SUCCESS in 5m 59s\n- openstack-tox-functional https://logs.opendev.org/87/673687/2/check/openstack-tox-functional/e0394d2/ : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/673687/2/check/openstack-tox-functional-py36/6620144/ : SUCCESS in 5m 35s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/673687/2/check/octavia-v2-dsvm-noop-api/1503e68/ : SUCCESS in 45m 17s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/673687/2/check/octavia-v2-dsvm-noop-py2-api/effbff0/ : SUCCESS in 44m 56s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/673687/2/check/octavia-v2-dsvm-scenario/1c8f0a9/ : SUCCESS in 1h 35m 28s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/673687/2/check/octavia-v2-dsvm-py2-scenario/0051d2f/ : SUCCESS in 1h 44m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/2/check/octavia-v2-dsvm-py2-scenario-centos-7/1be2da4/ : FAILURE in 2h 27m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/87/673687/2/check/octavia-v2-dsvm-scenario-two-node/0f366b3/ : FAILURE in 1h 32m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/87/673687/2/check/octavia-v2-dsvm-py2-scenario-two-node/c77d5a3/ : FAILURE in 1h 47m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/87/673687/2/check/octavia-v2-act-stdby-iptables-dsvm-scenario/4fdd6e2/ : FAILURE in 1h 22m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/87/673687/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/71241ec/ : FAILURE in 1h 37m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/48673c9/ : FAILURE in 1h 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8697f56c1b657cef2ae6b7e264eb7f78ff7ffd92","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 13:41:28.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"8f8f17d7b11f9d82b31a36793b381e9d13ab6ba7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 14:14:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2189ea76b4400d68c48fd5124988164f531620fc","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 14:26:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"db2a76143e424962822c76db74ec5a7977a8bcee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 17:56:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/87/673687/5/check/requirements-check/62a75e8/ : SUCCESS in 2m 11s\n- openstack-tox-cover https://logs.opendev.org/87/673687/5/check/openstack-tox-cover/657a72a/cover/ : SUCCESS in 7m 27s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/673687/5/check/openstack-tox-lower-constraints/2b04ec4/ : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://logs.opendev.org/87/673687/5/check/openstack-tox-pep8/2741af6/ : SUCCESS in 5m 48s\n- openstack-tox-py27 https://logs.opendev.org/87/673687/5/check/openstack-tox-py27/512d63a/ : SUCCESS in 4m 20s\n- openstack-tox-py36 https://logs.opendev.org/87/673687/5/check/openstack-tox-py36/313511f/ : SUCCESS in 4m 54s\n- openstack-tox-py37 https://logs.opendev.org/87/673687/5/check/openstack-tox-py37/465b434/ : SUCCESS in 4m 21s\n- openstack-tox-docs https://logs.opendev.org/87/673687/5/check/openstack-tox-docs/7310a39/ : FAILURE in 3m 40s\n- openstack-tox-functional https://logs.opendev.org/87/673687/5/check/openstack-tox-functional/cbf05c1/ : SUCCESS in 4m 25s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/673687/5/check/openstack-tox-functional-py36/a28ff2c/ : SUCCESS in 6m 14s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-noop-api/c040ba3/ : SUCCESS in 43m 54s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-noop-py2-api/556e77e/ : SUCCESS in 50m 34s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-scenario/805fcae/ : SUCCESS in 1h 30m 46s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-py2-scenario/98bf6f0/ : FAILURE in 2h 06m 54s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-py2-scenario-centos-7/a95fd3c/ : FAILURE in 2h 26m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-scenario-two-node/309dc23/ : SUCCESS in 1h 10m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-py2-scenario-two-node/8998f5c/ : FAILURE in 1h 37m 16s (non-voting)\n- octavia-grenade https://logs.opendev.org/87/673687/5/check/octavia-grenade/634c16a/ : SUCCESS in 1h 37m 17s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-tls-barbican/6809adb/ : SUCCESS in 47m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-scenario-amphora-v2/f897512/ : SUCCESS in 1h 26m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/87/673687/5/check/octavia-v2-act-stdby-iptables-dsvm-scenario/e5bab62/ : FAILURE in 1h 12m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/87/673687/5/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/0d6ebc3/ : FAILURE in 1h 27m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/5/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/4a5b62e/ : FAILURE in 1h 06m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b4ad7fa4cdab316db18c7092767967f42db525d1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-05 07:20:55.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"79ca1f3c037f919105f504d262af1a17812cb228","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 09:59:21.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/87/673687/5/check/requirements-check/346935b/ : SUCCESS in 2m 29s\n- openstack-tox-cover https://logs.opendev.org/87/673687/5/check/openstack-tox-cover/15499d5/cover/ : SUCCESS in 8m 18s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/673687/5/check/openstack-tox-lower-constraints/b836f15/ : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://logs.opendev.org/87/673687/5/check/openstack-tox-pep8/dbbb3c8/ : SUCCESS in 6m 32s\n- openstack-tox-py27 https://logs.opendev.org/87/673687/5/check/openstack-tox-py27/15a368b/ : SUCCESS in 4m 07s\n- openstack-tox-py36 https://logs.opendev.org/87/673687/5/check/openstack-tox-py36/7af3ceb/ : SUCCESS in 4m 46s\n- openstack-tox-py37 https://logs.opendev.org/87/673687/5/check/openstack-tox-py37/0b2cb58/ : SUCCESS in 5m 49s\n- openstack-tox-docs https://logs.opendev.org/87/673687/5/check/openstack-tox-docs/2cefbd4/html/ : SUCCESS in 6m 46s\n- openstack-tox-functional https://logs.opendev.org/87/673687/5/check/openstack-tox-functional/9e2dde5/ : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/673687/5/check/openstack-tox-functional-py36/f459d15/ : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-noop-api/08c9ff3/ : SUCCESS in 45m 20s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-noop-py2-api/d39f229/ : SUCCESS in 47m 21s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-scenario/39436df/ : FAILURE in 34m 12s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-py2-scenario/9247735/ : FAILURE in 1h 29m 51s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-py2-scenario-centos-7/4c46f7d/ : FAILURE in 2h 36m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-scenario-two-node/14345e1/ : SUCCESS in 2h 05m 49s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-py2-scenario-two-node/90fe875/ : SUCCESS in 1h 50m 37s (non-voting)\n- octavia-grenade https://logs.opendev.org/87/673687/5/check/octavia-grenade/cc7e3e2/ : SUCCESS in 1h 21m 38s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-tls-barbican/6831045/ : SUCCESS in 59m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-scenario-amphora-v2/051381d/ : SUCCESS in 2h 00m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/87/673687/5/check/octavia-v2-act-stdby-iptables-dsvm-scenario/6153a89/ : FAILURE in 1h 17m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/87/673687/5/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/106750a/ : FAILURE in 1h 16m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/5/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/c242397/ : FAILURE in 1h 13m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"affb899a87161810480265b4dc07c67e96f5620a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-05 10:02:48.000000000","message":"Patch Set 5:\n\nrecheck\n\n\u0027Connection reset by peer\" on pip install","accounts_in_message":[],"_revision_number":5},{"id":"dfd25da1af6de1b29470e33db24152dc0fae2bc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 12:41:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://logs.opendev.org/87/673687/5/check/requirements-check/1f2836d/ : SUCCESS in 2m 27s\n- openstack-tox-cover https://logs.opendev.org/87/673687/5/check/openstack-tox-cover/f98593f/cover/ : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/673687/5/check/openstack-tox-lower-constraints/60891c6/ : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://logs.opendev.org/87/673687/5/check/openstack-tox-pep8/0c3ea7c/ : SUCCESS in 4m 32s\n- openstack-tox-py27 https://logs.opendev.org/87/673687/5/check/openstack-tox-py27/1a6dfd2/ : SUCCESS in 3m 52s\n- openstack-tox-py36 https://logs.opendev.org/87/673687/5/check/openstack-tox-py36/37129a8/ : SUCCESS in 3m 39s\n- openstack-tox-py37 https://logs.opendev.org/87/673687/5/check/openstack-tox-py37/ba270a9/ : SUCCESS in 4m 52s\n- openstack-tox-docs https://logs.opendev.org/87/673687/5/check/openstack-tox-docs/c417669/html/ : SUCCESS in 5m 12s\n- openstack-tox-functional https://logs.opendev.org/87/673687/5/check/openstack-tox-functional/57619cf/ : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/673687/5/check/openstack-tox-functional-py36/a8a751d/ : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-noop-api/f02222d/ : SUCCESS in 45m 42s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-noop-py2-api/483f4a0/ : SUCCESS in 44m 12s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-scenario/2cb6fb4/ : SUCCESS in 1h 41m 35s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-py2-scenario/4315c0b/ : SUCCESS in 1h 48m 27s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-py2-scenario-centos-7/25cd4c3/ : FAILURE in 2h 26m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-scenario-two-node/7094198/ : FAILURE in 1h 39m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-py2-scenario-two-node/3ebdc04/ : SUCCESS in 1h 53m 15s (non-voting)\n- octavia-grenade https://logs.opendev.org/87/673687/5/check/octavia-grenade/f64bedb/ : SUCCESS in 1h 04m 44s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-tls-barbican/18fee53/ : SUCCESS in 51m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/87/673687/5/check/octavia-v2-dsvm-scenario-amphora-v2/0022fed/ : SUCCESS in 1h 40m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/87/673687/5/check/octavia-v2-act-stdby-iptables-dsvm-scenario/b5b41f5/ : FAILURE in 1h 21m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/87/673687/5/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/e5d774b/ : FAILURE in 1h 30m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/5/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/c1592b8/ : FAILURE in 1h 16m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f49338f98d5bb180d130074d2afc492f6cdc2ee9","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-05 15:27:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4bf201c4aedea9e3834bf52a95aad5ca6f16e7ad","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-05 15:50:20.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"1d8703f1b7716553830ce40d8c0783ea239007f5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-05 16:32:02.000000000","message":"Patch Set 6: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0adbfb9946e7fe7c0933e45ca20ac0363d0b5d7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 18:09:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://logs.opendev.org/87/673687/6/check/requirements-check/72ddcaf/ : SUCCESS in 3m 42s\n- openstack-tox-cover https://logs.opendev.org/87/673687/6/check/openstack-tox-cover/c253a68/cover/ : SUCCESS in 8m 52s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/673687/6/check/openstack-tox-lower-constraints/ee58c01/ : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://logs.opendev.org/87/673687/6/check/openstack-tox-pep8/b632a55/ : SUCCESS in 5m 12s\n- openstack-tox-py27 https://logs.opendev.org/87/673687/6/check/openstack-tox-py27/419a7fa/ : SUCCESS in 5m 58s\n- openstack-tox-py36 https://logs.opendev.org/87/673687/6/check/openstack-tox-py36/2be3680/ : SUCCESS in 5m 09s\n- openstack-tox-py37 https://logs.opendev.org/87/673687/6/check/openstack-tox-py37/264dde5/ : SUCCESS in 6m 10s\n- openstack-tox-docs https://logs.opendev.org/87/673687/6/check/openstack-tox-docs/83a4c59/html/ : SUCCESS in 5m 48s\n- openstack-tox-functional https://logs.opendev.org/87/673687/6/check/openstack-tox-functional/6be45c9/ : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/673687/6/check/openstack-tox-functional-py36/d76e468/ : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-noop-api/55223d4/ : SUCCESS in 49m 26s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-noop-py2-api/b1b621a/ : SUCCESS in 44m 43s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-scenario/0b983ac/ : SUCCESS in 1h 48m 05s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-py2-scenario/9383576/ : SUCCESS in 1h 52m 38s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-py2-scenario-centos-7/be84ecf/ : FAILURE in 2h 34m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-scenario-two-node/85bc388/ : SUCCESS in 1h 26m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-py2-scenario-two-node/02d16a4/ : SUCCESS in 2h 03m 38s (non-voting)\n- octavia-grenade https://logs.opendev.org/87/673687/6/check/octavia-grenade/18caaec/ : SUCCESS in 1h 07m 47s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-tls-barbican/1186187/ : SUCCESS in 1h 09m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-scenario-amphora-v2/febc72a/ : SUCCESS in 1h 38m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-act-stdby-iptables-dsvm-scenario/673c4ea/ : FAILURE in 1h 18m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/82c03c5/ : FAILURE in 1h 19m 40s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/6/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/9a5cfe3/ : FAILURE in 7m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8f8498114c3eaa32823a163bc60d612cef5f5577","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 18:09:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"835beae8bfa3150c7b274a338992796292269d46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 19:30:50.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/87/673687/6/gate/requirements-check/ada3bcc/ : SUCCESS in 3m 09s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/673687/6/gate/openstack-tox-lower-constraints/cc47e22/ : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://logs.opendev.org/87/673687/6/gate/openstack-tox-pep8/8e7dfc3/ : SUCCESS in 5m 42s\n- openstack-tox-py27 https://logs.opendev.org/87/673687/6/gate/openstack-tox-py27/42cb322/ : SUCCESS in 4m 10s\n- openstack-tox-py36 https://logs.opendev.org/87/673687/6/gate/openstack-tox-py36/d157bc2/ : SUCCESS in 4m 41s\n- openstack-tox-py37 https://logs.opendev.org/87/673687/6/gate/openstack-tox-py37/77faab1/ : SUCCESS in 4m 49s\n- openstack-tox-docs https://logs.opendev.org/87/673687/6/gate/openstack-tox-docs/c64c3cd/html/ : SUCCESS in 5m 55s\n- openstack-tox-functional https://logs.opendev.org/87/673687/6/gate/openstack-tox-functional/f5d1387/ : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/673687/6/gate/openstack-tox-functional-py36/a530b73/ : SUCCESS in 6m 05s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/673687/6/gate/octavia-v2-dsvm-noop-api/6e3115a/ : SUCCESS in 44m 20s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/673687/6/gate/octavia-v2-dsvm-noop-py2-api/6b20ce5/ : SUCCESS in 47m 16s\n- octavia-v2-dsvm-scenario finger://ze01.openstack.org/56de6b3653f247fbb62e4add591ac65d : CANCELED in 55m 47s\n- octavia-v2-dsvm-py2-scenario finger://ze05.openstack.org/fc2f6ecb269a40fd88da9f79f2df7482 : CANCELED in 56m 19s\n- octavia-grenade https://logs.opendev.org/87/673687/6/gate/octavia-grenade/5b69221/ : FAILURE in 1h 00m 01s","accounts_in_message":[],"_revision_number":6},{"id":"2db08d76788790a5308a1d3f1e92a2eb6b4af5a1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-05 19:50:04.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"c823708e30aee5d5b82d1dbeb6331273e2cc2c3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 22:28:05.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/87/673687/6/check/requirements-check/23d7602/ : SUCCESS in 3m 39s\n- openstack-tox-cover https://logs.opendev.org/87/673687/6/check/openstack-tox-cover/ac07180/cover/ : SUCCESS in 6m 56s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/673687/6/check/openstack-tox-lower-constraints/237f956/ : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://logs.opendev.org/87/673687/6/check/openstack-tox-pep8/d270f04/ : SUCCESS in 4m 55s\n- openstack-tox-py27 https://logs.opendev.org/87/673687/6/check/openstack-tox-py27/d110409/ : SUCCESS in 4m 14s\n- openstack-tox-py36 https://logs.opendev.org/87/673687/6/check/openstack-tox-py36/20e1ee1/ : SUCCESS in 4m 28s\n- openstack-tox-py37 https://logs.opendev.org/87/673687/6/check/openstack-tox-py37/d4da1f3/ : SUCCESS in 6m 24s\n- openstack-tox-docs https://logs.opendev.org/87/673687/6/check/openstack-tox-docs/410d240/html/ : SUCCESS in 6m 53s\n- openstack-tox-functional https://logs.opendev.org/87/673687/6/check/openstack-tox-functional/d402ce7/ : SUCCESS in 7m 00s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/673687/6/check/openstack-tox-functional-py36/7e7f60d/ : SUCCESS in 6m 11s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-noop-api/3360e75/ : SUCCESS in 38m 13s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-noop-py2-api/870bfa9/ : SUCCESS in 49m 03s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-scenario/b5a8ce2/ : SUCCESS in 1h 51m 15s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-py2-scenario/53c1c73/ : FAILURE in 2h 25m 50s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-py2-scenario-centos-7/cffafdf/ : TIMED_OUT in 2h 34m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-scenario-two-node/dcd9d6b/ : SUCCESS in 2h 01m 56s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-py2-scenario-two-node/84a2f2c/ : SUCCESS in 1h 53m 01s (non-voting)\n- octavia-grenade https://logs.opendev.org/87/673687/6/check/octavia-grenade/06c5c6f/ : SUCCESS in 1h 20m 18s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-tls-barbican/35280a9/ : SUCCESS in 55m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-scenario-amphora-v2/cad5210/ : SUCCESS in 1h 37m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-act-stdby-iptables-dsvm-scenario/ebcc038/ : FAILURE in 1h 11m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/beb3a1d/ : FAILURE in 1h 19m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/6/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/856e103/ : FAILURE in 53m 43s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ca9a139929521fb909c3650026b312af97bf7b18","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-05 22:32:46.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"ffc523fc8dc3db5693f91aab818d65cf4ec72262","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 01:12:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://logs.opendev.org/87/673687/6/check/requirements-check/6fb15f0/ : SUCCESS in 2m 03s\n- openstack-tox-cover https://logs.opendev.org/87/673687/6/check/openstack-tox-cover/a51d3a4/cover/ : SUCCESS in 7m 51s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/673687/6/check/openstack-tox-lower-constraints/6b5ccf1/ : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://logs.opendev.org/87/673687/6/check/openstack-tox-pep8/f0cc34e/ : SUCCESS in 5m 14s\n- openstack-tox-py27 https://logs.opendev.org/87/673687/6/check/openstack-tox-py27/584bce8/ : SUCCESS in 3m 55s\n- openstack-tox-py36 https://logs.opendev.org/87/673687/6/check/openstack-tox-py36/17342ad/ : SUCCESS in 4m 03s\n- openstack-tox-py37 https://logs.opendev.org/87/673687/6/check/openstack-tox-py37/1e05c4a/ : SUCCESS in 3m 55s\n- openstack-tox-docs https://logs.opendev.org/87/673687/6/check/openstack-tox-docs/3bb3de3/html/ : SUCCESS in 5m 35s\n- openstack-tox-functional https://logs.opendev.org/87/673687/6/check/openstack-tox-functional/c56235f/ : SUCCESS in 4m 15s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/673687/6/check/openstack-tox-functional-py36/1e2e002/ : SUCCESS in 6m 21s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-noop-api/d32524b/ : SUCCESS in 42m 00s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-noop-py2-api/fa4d7ca/ : SUCCESS in 1h 02m 54s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-scenario/0255b62/ : SUCCESS in 1h 38m 23s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-py2-scenario/971faaa/ : SUCCESS in 1h 48m 38s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-py2-scenario-centos-7/78046ca/ : TIMED_OUT in 2h 36m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-scenario-two-node/3173747/ : SUCCESS in 2h 28m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-py2-scenario-two-node/6a24ea4/ : SUCCESS in 1h 13m 52s (non-voting)\n- octavia-grenade https://logs.opendev.org/87/673687/6/check/octavia-grenade/ea42db7/ : SUCCESS in 1h 21m 53s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-tls-barbican/e32f28b/ : FAILURE in 38m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-scenario-amphora-v2/c20e5a5/ : SUCCESS in 1h 27m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-act-stdby-iptables-dsvm-scenario/0d6f3cb/ : FAILURE in 1h 04m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/5a834ae/ : FAILURE in 23m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/6/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/5db2c29/ : FAILURE in 1h 12m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"513e17209b0497957820c594b58236aabe0c92d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 01:12:32.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"442a6f13c0db084ca613bc68e5cd9c950bb71d9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 01:50:37.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/87/673687/6/gate/requirements-check/f58af5e/ : SUCCESS in 1m 53s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/673687/6/gate/openstack-tox-lower-constraints/d6ff4aa/ : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://logs.opendev.org/87/673687/6/gate/openstack-tox-pep8/165aec9/ : SUCCESS in 4m 05s\n- openstack-tox-py27 https://logs.opendev.org/87/673687/6/gate/openstack-tox-py27/6956351/ : SUCCESS in 3m 20s\n- openstack-tox-py36 https://logs.opendev.org/87/673687/6/gate/openstack-tox-py36/219affc/ : SUCCESS in 3m 20s\n- openstack-tox-py37 https://logs.opendev.org/87/673687/6/gate/openstack-tox-py37/4a77ff8/ : SUCCESS in 4m 11s\n- openstack-tox-docs https://logs.opendev.org/87/673687/6/gate/openstack-tox-docs/eb93cc9/html/ : SUCCESS in 4m 43s\n- openstack-tox-functional https://logs.opendev.org/87/673687/6/gate/openstack-tox-functional/c07ca04/ : SUCCESS in 4m 49s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/673687/6/gate/openstack-tox-functional-py36/d3d714b/ : SUCCESS in 5m 25s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/673687/6/gate/octavia-v2-dsvm-noop-api/cec5d2c/ : SUCCESS in 36m 21s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/673687/6/gate/octavia-v2-dsvm-noop-py2-api/e53e651/ : SUCCESS in 35m 10s\n- octavia-v2-dsvm-scenario finger://ze02.openstack.org/7ca2f82d396248bf8145c1819e45df3b : CANCELED in 36m 48s\n- octavia-v2-dsvm-py2-scenario finger://ze02.openstack.org/86f1d9e2834b46969343b1f35d114518 : CANCELED in 37m 05s\n- octavia-grenade https://logs.opendev.org/87/673687/6/gate/octavia-grenade/1480f3b/ : FAILURE in 36m 35s","accounts_in_message":[],"_revision_number":6},{"id":"7999a001fced564a21a3e8c73de1accaaf2fa4d7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-06 05:53:29.000000000","message":"Patch Set 6:\n\nrecheck - OpenStack CI log server was done","accounts_in_message":[],"_revision_number":6},{"id":"aa4826aa68891e88f1f48985679ba8a4af85b81c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 08:44:35.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/87/673687/6/check/requirements-check/c4fed10/ : SUCCESS in 4m 01s\n- openstack-tox-cover https://logs.opendev.org/87/673687/6/check/openstack-tox-cover/8ec160a/cover/ : SUCCESS in 7m 41s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/673687/6/check/openstack-tox-lower-constraints/adbd0ea/ : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://logs.opendev.org/87/673687/6/check/openstack-tox-pep8/d72511b/ : SUCCESS in 4m 52s\n- openstack-tox-py27 finger://ze02.openstack.org/b0374c7c33154ac3be3e4b6078c97a86 : POST_FAILURE in 7m 40s\n- openstack-tox-py36 https://logs.opendev.org/87/673687/6/check/openstack-tox-py36/4f63dcd/ : SUCCESS in 11m 51s\n- openstack-tox-py37 https://logs.opendev.org/87/673687/6/check/openstack-tox-py37/5eb4513/ : SUCCESS in 8m 12s\n- openstack-tox-docs https://logs.opendev.org/87/673687/6/check/openstack-tox-docs/4460f3c/html/ : SUCCESS in 10m 51s\n- openstack-tox-functional https://logs.opendev.org/87/673687/6/check/openstack-tox-functional/efd0e51/ : SUCCESS in 7m 06s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/673687/6/check/openstack-tox-functional-py36/6632012/ : SUCCESS in 12m 12s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-noop-api/dca2b41/ : SUCCESS in 45m 26s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-noop-py2-api/6b98282/ : SUCCESS in 48m 41s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-scenario/7b801ad/ : SUCCESS in 1h 26m 55s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-py2-scenario/6c7fad3/ : SUCCESS in 1h 36m 14s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-py2-scenario-centos-7/2c2896c/ : TIMED_OUT in 2h 34m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-scenario-two-node/189d2c7/ : SUCCESS in 1h 53m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-py2-scenario-two-node/ad9763a/ : SUCCESS in 1h 52m 06s (non-voting)\n- octavia-grenade https://logs.opendev.org/87/673687/6/check/octavia-grenade/38ecf96/ : SUCCESS in 1h 24m 57s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-tls-barbican/2fee000/ : SUCCESS in 1h 00m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-scenario-amphora-v2/b6fa81a/ : SUCCESS in 1h 37m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-act-stdby-iptables-dsvm-scenario/226322d/ : FAILURE in 1h 06m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/3d59f5e/ : FAILURE in 56m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/6/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/83ced1f/ : FAILURE in 1h 24m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"05de0861ca5b7baac49b726fe486172ecd43147b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-06 08:48:03.000000000","message":"Patch Set 6:\n\nrecheck - another CI failure","accounts_in_message":[],"_revision_number":6},{"id":"9b25a39278b413d2b99be0e984a451a5df4882ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 13:15:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://logs.opendev.org/87/673687/6/check/requirements-check/b318d5c/ : SUCCESS in 3m 29s\n- openstack-tox-cover https://logs.opendev.org/87/673687/6/check/openstack-tox-cover/0aec2c7/cover/ : SUCCESS in 6m 47s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/673687/6/check/openstack-tox-lower-constraints/9d7964c/ : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://logs.opendev.org/87/673687/6/check/openstack-tox-pep8/6bcf415/ : SUCCESS in 5m 08s\n- openstack-tox-py27 https://logs.opendev.org/87/673687/6/check/openstack-tox-py27/bca9129/ : SUCCESS in 5m 16s\n- openstack-tox-py36 https://logs.opendev.org/87/673687/6/check/openstack-tox-py36/fde129f/ : SUCCESS in 5m 13s\n- openstack-tox-py37 https://logs.opendev.org/87/673687/6/check/openstack-tox-py37/711d0c1/ : SUCCESS in 5m 13s\n- openstack-tox-docs https://logs.opendev.org/87/673687/6/check/openstack-tox-docs/99dc176/html/ : SUCCESS in 6m 42s\n- openstack-tox-functional https://logs.opendev.org/87/673687/6/check/openstack-tox-functional/1c770f5/ : SUCCESS in 8m 28s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/673687/6/check/openstack-tox-functional-py36/26ab931/ : SUCCESS in 5m 28s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-noop-api/c239b78/ : SUCCESS in 48m 30s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-noop-py2-api/950f9c0/ : SUCCESS in 49m 48s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-scenario/e735305/ : SUCCESS in 1h 51m 33s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-py2-scenario/7bd1ea3/ : SUCCESS in 1h 39m 54s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-py2-scenario-centos-7/2b6fff4/ : TIMED_OUT in 2h 37m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-scenario-two-node/67e6cfa/ : FAILURE in 1h 45m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-py2-scenario-two-node/29c7c43/ : SUCCESS in 2h 07m 48s (non-voting)\n- octavia-grenade https://logs.opendev.org/87/673687/6/check/octavia-grenade/e690c29/ : SUCCESS in 1h 17m 28s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-tls-barbican/7359424/ : SUCCESS in 55m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/87/673687/6/check/octavia-v2-dsvm-scenario-amphora-v2/0df998b/ : SUCCESS in 1h 40m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-act-stdby-iptables-dsvm-scenario/09d8387/ : FAILURE in 1h 17m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/87/673687/6/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/abcaf70/ : FAILURE in 2h 13m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/673687/6/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/532f31f/ : FAILURE in 1h 17m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6778bd48cf6f40d1ed7b7cb460c8646d5f6979eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 13:16:09.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0ebed884ce74e53e1f1968a9d77c103f48636843","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 15:18:53.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://logs.opendev.org/87/673687/6/gate/requirements-check/2eccf6b/ : SUCCESS in 3m 26s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/673687/6/gate/openstack-tox-lower-constraints/78411ee/ : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://logs.opendev.org/87/673687/6/gate/openstack-tox-pep8/3bd0960/ : SUCCESS in 6m 10s\n- openstack-tox-py27 https://logs.opendev.org/87/673687/6/gate/openstack-tox-py27/38e0d40/ : SUCCESS in 4m 58s\n- openstack-tox-py36 https://logs.opendev.org/87/673687/6/gate/openstack-tox-py36/d5e43c8/ : SUCCESS in 4m 08s\n- openstack-tox-py37 https://logs.opendev.org/87/673687/6/gate/openstack-tox-py37/0fc6d3f/ : SUCCESS in 6m 15s\n- openstack-tox-docs https://logs.opendev.org/87/673687/6/gate/openstack-tox-docs/0d164cf/html/ : SUCCESS in 6m 47s\n- openstack-tox-functional https://logs.opendev.org/87/673687/6/gate/openstack-tox-functional/8b99b20/ : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/673687/6/gate/openstack-tox-functional-py36/ca10529/ : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/673687/6/gate/octavia-v2-dsvm-noop-api/cffc005/ : SUCCESS in 53m 40s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/673687/6/gate/octavia-v2-dsvm-noop-py2-api/8da1bfb/ : SUCCESS in 50m 29s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/673687/6/gate/octavia-v2-dsvm-scenario/16e2d62/ : SUCCESS in 1h 55m 13s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/673687/6/gate/octavia-v2-dsvm-py2-scenario/5e646ce/ : SUCCESS in 1h 45m 00s\n- octavia-grenade https://logs.opendev.org/87/673687/6/gate/octavia-grenade/6ed4ee8/ : SUCCESS in 1h 19m 41s","accounts_in_message":[],"_revision_number":6},{"id":"f6b943640099caf8fc93fd6651bfb4dc444fb413","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 15:18:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b32064cdf2d5c70140a84e2497931941c4ee6887","revisions":{"21f4f6b850154fd1a90c2f91f7daabd98612745d":{"kind":"REWORK","_number":1,"created":"2019-07-30 20:52:38.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/87/673687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/673687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/673687/1"}}},"commit":{"parents":[{"commit":"68d9ed58b927dc08cc8cde5dcf1213eb91532d9a","subject":"Merge \"Fix l7rule API handling of None updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/68d9ed58b927dc08cc8cde5dcf1213eb91532d9a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 20:49:30.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 20:52:37.000000000","tz":-420},"subject":"Fix a unit test for new octavia-lib","message":"Fix a unit test for new octavia-lib\n\nA patch[1] introduced a startup check for the socket paths that\nimpacted a unit test checking debtcollector is finding the new\noctavia-lib location of the DriverLibrary.\nThis patch correctly mocks out this check for this unit test.\n\n[1] https://review.opendev.org/#/c/665027/\n\nChange-Id: Ife93834654385430d6a333d3d6ee67b719b6c37e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21f4f6b850154fd1a90c2f91f7daabd98612745d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21f4f6b850154fd1a90c2f91f7daabd98612745d"}]},"branch":"refs/heads/master"},"bd4a8b9261de635cb4f6da6cc377330a5be2ca39":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-02 00:27:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/87/673687/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/673687/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/673687/2"}}},"commit":{"parents":[{"commit":"768fe6ce06f5821a2ef7835f52d8f23243195464","subject":"Merge \"Add failover logging to show the amphora details.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/768fe6ce06f5821a2ef7835f52d8f23243195464"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 20:49:30.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-02 00:20:37.000000000","tz":-420},"subject":"Fix a unit test for new octavia-lib","message":"Fix a unit test for new octavia-lib\n\nA patch[1] introduced a startup check for the socket paths that\nimpacted a unit test checking debtcollector is finding the new\noctavia-lib location of the DriverLibrary.\nThis patch correctly mocks out this check for this unit test.\n\n[1] https://review.opendev.org/#/c/665027/\n\nChange-Id: Ife93834654385430d6a333d3d6ee67b719b6c37e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd4a8b9261de635cb4f6da6cc377330a5be2ca39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd4a8b9261de635cb4f6da6cc377330a5be2ca39"}]},"branch":"refs/heads/master"},"179f9558c9cc40d02acc0f7ea64c3625523b3edc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-03 13:41:28.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/87/673687/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/673687/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/673687/3"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 20:49:30.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-03 13:41:28.000000000","tz":0},"subject":"Fix a unit test for new octavia-lib","message":"Fix a unit test for new octavia-lib\n\nA patch[1] introduced a startup check for the socket paths that\nimpacted a unit test checking debtcollector is finding the new\noctavia-lib location of the DriverLibrary.\nThis patch correctly mocks out this check for this unit test.\n\n[1] https://review.opendev.org/#/c/665027/\n\nChange-Id: Ife93834654385430d6a333d3d6ee67b719b6c37e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/179f9558c9cc40d02acc0f7ea64c3625523b3edc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/179f9558c9cc40d02acc0f7ea64c3625523b3edc"}]},"branch":"refs/heads/master"},"c58ef30d291f679ea2aac0b4f24cf1afe48f82eb":{"kind":"REWORK","_number":4,"created":"2019-08-03 14:14:44.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/87/673687/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/673687/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/673687/4"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 20:49:30.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-03 14:14:18.000000000","tz":-420},"subject":"Fix a unit test for new octavia-lib","message":"Fix a unit test for new octavia-lib\n\nA patch[1] introduced a startup check for the socket paths that\nimpacted a unit test checking debtcollector is finding the new\noctavia-lib location of the DriverLibrary.\nThis patch correctly mocks out this check for this unit test.\n\n[1] https://review.opendev.org/#/c/665027/\n\nChange-Id: Ife93834654385430d6a333d3d6ee67b719b6c37e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c58ef30d291f679ea2aac0b4f24cf1afe48f82eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c58ef30d291f679ea2aac0b4f24cf1afe48f82eb"}]},"branch":"refs/heads/master"},"fb5d7d1120e3d92b5a52f53fcac9e0c30f3b0ddc":{"kind":"REWORK","_number":5,"created":"2019-08-03 14:26:51.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/87/673687/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/673687/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/673687/5"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 20:49:30.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-03 14:21:43.000000000","tz":-420},"subject":"Fix a unit test for new octavia-lib","message":"Fix a unit test for new octavia-lib\n\nA patch[1] introduced a startup check for the socket paths that\nimpacted a unit test checking debtcollector is finding the new\noctavia-lib location of the DriverLibrary.\nThis patch correctly mocks out this check for this unit test.\n\n[1] https://review.opendev.org/#/c/665027/\n\nChange-Id: Ife93834654385430d6a333d3d6ee67b719b6c37e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb5d7d1120e3d92b5a52f53fcac9e0c30f3b0ddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb5d7d1120e3d92b5a52f53fcac9e0c30f3b0ddc"}]},"branch":"refs/heads/master"},"b32064cdf2d5c70140a84e2497931941c4ee6887":{"kind":"REWORK","_number":6,"created":"2019-08-05 15:27:57.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/87/673687/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/673687/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/673687/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/673687/6"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 20:49:30.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-05 15:27:43.000000000","tz":-420},"subject":"Fix a unit test for new octavia-lib","message":"Fix a unit test for new octavia-lib\n\nA patch[1] introduced a startup check for the socket paths that\nimpacted a unit test checking debtcollector is finding the new\noctavia-lib location of the DriverLibrary.\nThis patch correctly mocks out this check for this unit test.\n\n[1] https://review.opendev.org/#/c/665027/\n\nChange-Id: Ife93834654385430d6a333d3d6ee67b719b6c37e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b32064cdf2d5c70140a84e2497931941c4ee6887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b32064cdf2d5c70140a84e2497931941c4ee6887"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
