)]}'
{"id":"openstack%2Foctavia~673712","triplet_id":"openstack%2Foctavia~master~I90c2fa386fab75742d07e89edd3fba8b56bb6aff","project":"openstack/octavia","branch":"master","topic":"lib-get","hashtags":[],"change_id":"I90c2fa386fab75742d07e89edd3fba8b56bb6aff","subject":"Move to using octavia-lib constants","status":"MERGED","created":"2019-07-30 23:52:54.000000000","updated":"2019-09-11 09:42:17.000000000","submitted":"2019-09-11 09:40:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673712-1568194846398-9806e538","meta_rev_id":"79147099cd9e8bae02025ce91416486d1f4209b5","_number":673712,"virtual_id_number":673712,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-09-11 09:40:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-11 02:47:08.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-09-11 07:37:59.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2019-09-11 07:37:59.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":-2,"date":"2019-09-11 07:37:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-03 13:45:24.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-11 02:47:08.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-09-11 07:37:59.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-11 09:40:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"20059f857f7a9ce736a7ad3bf9aa7ffe2a411b9a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-30 23:52:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f8f1a00addf324d68ca503d0c61395d68a5f407","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 02:15:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/12/673712/1/check/openstack-tox-cover/5d5402d/ : FAILURE in 3m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/673712/1/check/openstack-tox-lower-constraints/eec61a8/ : FAILURE in 4m 27s\n- openstack-tox-pep8 http://logs.openstack.org/12/673712/1/check/openstack-tox-pep8/bdd4e0e/ : SUCCESS in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/12/673712/1/check/openstack-tox-py27/bab5ce0/ : FAILURE in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/12/673712/1/check/openstack-tox-py36/e41513f/ : FAILURE in 3m 33s\n- openstack-tox-py37 http://logs.openstack.org/12/673712/1/check/openstack-tox-py37/9b1eed9/ : FAILURE in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/12/673712/1/check/openstack-tox-docs/264ba8d/html/ : SUCCESS in 6m 13s\n- openstack-tox-functional http://logs.openstack.org/12/673712/1/check/openstack-tox-functional/8d47687/ : FAILURE in 6m 18s\n- openstack-tox-functional-py36 http://logs.openstack.org/12/673712/1/check/openstack-tox-functional-py36/4c3fe15/ : FAILURE in 6m 32s\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/12/673712/1/check/octavia-v2-dsvm-noop-api/7311f18/ : SUCCESS in 45m 12s\n- octavia-v2-dsvm-noop-py2-api http://logs.openstack.org/12/673712/1/check/octavia-v2-dsvm-noop-py2-api/fb1fbf3/ : SUCCESS in 44m 56s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/12/673712/1/check/octavia-v2-dsvm-scenario/4c86fea/ : SUCCESS in 1h 40m 07s\n- octavia-v2-dsvm-py2-scenario http://logs.openstack.org/12/673712/1/check/octavia-v2-dsvm-py2-scenario/cdd6dfb/ : SUCCESS in 1h 37m 10s\n- octavia-v2-dsvm-py2-scenario-centos-7 http://logs.openstack.org/12/673712/1/check/octavia-v2-dsvm-py2-scenario-centos-7/2bcd823/ : FAILURE in 2h 14m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node http://logs.openstack.org/12/673712/1/check/octavia-v2-dsvm-scenario-two-node/4281d80/ : SUCCESS in 2h 04m 27s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node http://logs.openstack.org/12/673712/1/check/octavia-v2-dsvm-py2-scenario-two-node/3ca87d6/ : SUCCESS in 1h 59m 26s (non-voting)\n- octavia-grenade http://logs.openstack.org/12/673712/1/check/octavia-grenade/3f4b7b9/ : SUCCESS in 1h 10m 50s\n- octavia-v2-dsvm-tls-barbican http://logs.openstack.org/12/673712/1/check/octavia-v2-dsvm-tls-barbican/18adebb/ : SUCCESS in 59m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 http://logs.openstack.org/12/673712/1/check/octavia-v2-dsvm-scenario-amphora-v2/20453bc/ : SUCCESS in 1h 26m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario http://logs.openstack.org/12/673712/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/14014b8/ : FAILURE in 1h 13m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario http://logs.openstack.org/12/673712/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/9165cd1/ : FAILURE in 1h 20m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 http://logs.openstack.org/12/673712/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/cc8c140/ : FAILURE in 1h 38m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34ae5f372daf8d8f7e38429134455b822cf61ca2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-01 22:48:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fe1b01e83b4dfc7e60d072d80610f35302adb810","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 22:49:31.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"636c24a0947531bb5f41c31cd197cfcd7ff680c2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-01 22:58:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"55ce6eec431fa715a2ce285256ea3b59d077104c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 22:58:40.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"711b5244c01b4a4fa35605262f0d0a86889e1a5d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-01 23:14:36.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"2c24186189e93b2a5b765f25d28ca15dbb2057dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 23:14:52.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"59b75ebb28bd8396116c629d965f9145ff2bd5be","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-02 00:27:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"37ece5f78e76d1411263c7bb5fe7637860085b4f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-02 00:31:39.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"b10a4b0a4ec79f198034f9d8072a23a7c77322a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 03:13:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/12/673712/6/check/openstack-tox-cover/b8e97b9/ : FAILURE in 5m 02s\n- openstack-tox-lower-constraints https://logs.opendev.org/12/673712/6/check/openstack-tox-lower-constraints/50fdcf0/ : FAILURE in 6m 06s\n- openstack-tox-pep8 https://logs.opendev.org/12/673712/6/check/openstack-tox-pep8/c5894ff/ : SUCCESS in 5m 45s\n- openstack-tox-py27 https://logs.opendev.org/12/673712/6/check/openstack-tox-py27/58a10eb/ : FAILURE in 4m 59s\n- openstack-tox-py36 https://logs.opendev.org/12/673712/6/check/openstack-tox-py36/5d008f0/ : FAILURE in 5m 32s\n- openstack-tox-py37 https://logs.opendev.org/12/673712/6/check/openstack-tox-py37/73f0c9a/ : FAILURE in 4m 50s\n- openstack-tox-docs https://logs.opendev.org/12/673712/6/check/openstack-tox-docs/bb3bd5c/html/ : SUCCESS in 6m 08s\n- openstack-tox-functional https://logs.opendev.org/12/673712/6/check/openstack-tox-functional/019813d/ : FAILURE in 5m 34s\n- openstack-tox-functional-py36 https://logs.opendev.org/12/673712/6/check/openstack-tox-functional-py36/83155a5/ : FAILURE in 6m 39s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/12/673712/6/check/octavia-v2-dsvm-noop-api/2caf40b/ : SUCCESS in 50m 14s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/12/673712/6/check/octavia-v2-dsvm-noop-py2-api/0b6f2b8/ : SUCCESS in 45m 23s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/12/673712/6/check/octavia-v2-dsvm-scenario/d45e5f8/ : SUCCESS in 1h 44m 48s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/12/673712/6/check/octavia-v2-dsvm-py2-scenario/9312023/ : SUCCESS in 1h 49m 57s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/12/673712/6/check/octavia-v2-dsvm-py2-scenario-centos-7/168f977/ : FAILURE in 2h 30m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/12/673712/6/check/octavia-v2-dsvm-scenario-two-node/a7e5ba0/ : FAILURE in 1h 38m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/12/673712/6/check/octavia-v2-dsvm-py2-scenario-two-node/91f4491/ : SUCCESS in 1h 50m 16s (non-voting)\n- octavia-grenade https://logs.opendev.org/12/673712/6/check/octavia-grenade/6783fda/ : SUCCESS in 1h 19m 06s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/12/673712/6/check/octavia-v2-dsvm-tls-barbican/33a97af/ : SUCCESS in 1h 04m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/12/673712/6/check/octavia-v2-dsvm-scenario-amphora-v2/07e8c3e/ : SUCCESS in 1h 39m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/12/673712/6/check/octavia-v2-act-stdby-iptables-dsvm-scenario/239c9ad/ : FAILURE in 1h 16m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/12/673712/6/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/420246b/ : FAILURE in 1h 17m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/12/673712/6/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/0fa2657/ : FAILURE in 1h 09m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d07b3ce2e9669f3479f279724beb5e2886ee1005","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 13:45:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a443a7949853f304cb439f208226b8dc72420020","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 14:15:25.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"7ef1d5cde9c87196f083f87f61155ec72dbda184","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 18:04:34.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/12/673712/8/check/openstack-tox-cover/84a5d57/ : FAILURE in 6m 59s\n- openstack-tox-lower-constraints https://logs.opendev.org/12/673712/8/check/openstack-tox-lower-constraints/567df98/ : FAILURE in 6m 53s\n- openstack-tox-pep8 https://logs.opendev.org/12/673712/8/check/openstack-tox-pep8/a9c183b/ : SUCCESS in 6m 07s\n- openstack-tox-py27 https://logs.opendev.org/12/673712/8/check/openstack-tox-py27/4c8fac0/ : FAILURE in 4m 15s\n- openstack-tox-py36 https://logs.opendev.org/12/673712/8/check/openstack-tox-py36/969f891/ : SUCCESS in 4m 11s\n- openstack-tox-py37 https://logs.opendev.org/12/673712/8/check/openstack-tox-py37/ba47595/ : SUCCESS in 4m 23s\n- openstack-tox-docs https://logs.opendev.org/12/673712/8/check/openstack-tox-docs/9a3183a/ : FAILURE in 5m 38s\n- openstack-tox-functional https://logs.opendev.org/12/673712/8/check/openstack-tox-functional/086a582/ : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://logs.opendev.org/12/673712/8/check/openstack-tox-functional-py36/12ad520/ : FAILURE in 7m 16s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/12/673712/8/check/octavia-v2-dsvm-noop-api/b3c1f23/ : SUCCESS in 42m 41s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/12/673712/8/check/octavia-v2-dsvm-noop-py2-api/7af479d/ : SUCCESS in 44m 13s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/12/673712/8/check/octavia-v2-dsvm-scenario/19fc7f3/ : SUCCESS in 1h 55m 57s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/12/673712/8/check/octavia-v2-dsvm-py2-scenario/39568b6/ : SUCCESS in 1h 36m 51s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/12/673712/8/check/octavia-v2-dsvm-py2-scenario-centos-7/4be7f24/ : FAILURE in 2h 34m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/12/673712/8/check/octavia-v2-dsvm-scenario-two-node/af0a3c1/ : FAILURE in 1h 50m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/12/673712/8/check/octavia-v2-dsvm-py2-scenario-two-node/afacd33/ : FAILURE in 1h 22m 10s (non-voting)\n- octavia-grenade https://logs.opendev.org/12/673712/8/check/octavia-grenade/246d887/ : SUCCESS in 1h 14m 47s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/12/673712/8/check/octavia-v2-dsvm-tls-barbican/2a703eb/ : FAILURE in 52m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/12/673712/8/check/octavia-v2-dsvm-scenario-amphora-v2/9596b63/ : SUCCESS in 1h 30m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/12/673712/8/check/octavia-v2-act-stdby-iptables-dsvm-scenario/a95e1ab/ : FAILURE in 1h 16m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/12/673712/8/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/90a7f65/ : FAILURE in 1h 23m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/12/673712/8/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/df4d94b/ : FAILURE in 1h 00m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fb12ee2669006c63d03b9379ba3ca035f01add97","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-05 20:32:52.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"9a410604115ce6f248ed8ac925ad8415a985bc49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 23:14:05.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/12/673712/9/check/openstack-tox-cover/1de9e00/cover/ : SUCCESS in 8m 05s\n- openstack-tox-lower-constraints https://logs.opendev.org/12/673712/9/check/openstack-tox-lower-constraints/e52489f/ : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://logs.opendev.org/12/673712/9/check/openstack-tox-pep8/5eb8f71/ : SUCCESS in 5m 15s\n- openstack-tox-py27 https://logs.opendev.org/12/673712/9/check/openstack-tox-py27/706109b/ : SUCCESS in 4m 06s\n- openstack-tox-py36 https://logs.opendev.org/12/673712/9/check/openstack-tox-py36/dcdff6f/ : SUCCESS in 4m 38s\n- openstack-tox-py37 https://logs.opendev.org/12/673712/9/check/openstack-tox-py37/b076b96/ : SUCCESS in 4m 19s\n- openstack-tox-docs https://logs.opendev.org/12/673712/9/check/openstack-tox-docs/9c4fc6b/html/ : SUCCESS in 7m 06s\n- openstack-tox-functional https://logs.opendev.org/12/673712/9/check/openstack-tox-functional/a59f91b/ : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://logs.opendev.org/12/673712/9/check/openstack-tox-functional-py36/99dbec1/ : SUCCESS in 5m 20s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/12/673712/9/check/octavia-v2-dsvm-noop-api/944690b/ : SUCCESS in 57m 49s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/12/673712/9/check/octavia-v2-dsvm-noop-py2-api/3e8c808/ : SUCCESS in 40m 10s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/12/673712/9/check/octavia-v2-dsvm-scenario/d847a5e/ : FAILURE in 30m 05s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/12/673712/9/check/octavia-v2-dsvm-py2-scenario/6888dec/ : SUCCESS in 1h 49m 29s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/12/673712/9/check/octavia-v2-dsvm-py2-scenario-centos-7/20bea86/ : TIMED_OUT in 2h 34m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/12/673712/9/check/octavia-v2-dsvm-scenario-two-node/1540fe1/ : SUCCESS in 1h 27m 40s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/12/673712/9/check/octavia-v2-dsvm-py2-scenario-two-node/c4c7735/ : SUCCESS in 1h 46m 17s (non-voting)\n- octavia-grenade https://logs.opendev.org/12/673712/9/check/octavia-grenade/4ffa1ab/ : SUCCESS in 1h 10m 40s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/12/673712/9/check/octavia-v2-dsvm-tls-barbican/24df628/ : FAILURE in 34m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/12/673712/9/check/octavia-v2-dsvm-scenario-amphora-v2/40c3432/ : SUCCESS in 1h 25m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/12/673712/9/check/octavia-v2-act-stdby-iptables-dsvm-scenario/3581724/ : FAILURE in 1h 41m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/12/673712/9/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/b7031c1/ : FAILURE in 1h 23m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/12/673712/9/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/f1f56f5/ : FAILURE in 1h 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"55b8297ccc5a820645068f66e293411598cc0570","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-09 17:19:30.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"1dfd3c7ca5e8fde1875d89c3a92276e7fa536943","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-09 23:09:05.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"d71f6e2e8c1391e7033cfcb8b3a118b5965b374a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 01:29:35.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/12/673712/11/check/openstack-tox-cover/1cb0049/cover/ : SUCCESS in 6m 39s\n- openstack-tox-lower-constraints https://logs.opendev.org/12/673712/11/check/openstack-tox-lower-constraints/1d95906/ : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://logs.opendev.org/12/673712/11/check/openstack-tox-pep8/7eb704c/ : SUCCESS in 4m 41s\n- openstack-tox-py27 https://logs.opendev.org/12/673712/11/check/openstack-tox-py27/6c7ccc9/ : SUCCESS in 3m 55s\n- openstack-tox-py36 https://logs.opendev.org/12/673712/11/check/openstack-tox-py36/21cbfc6/ : SUCCESS in 4m 18s\n- openstack-tox-py37 https://logs.opendev.org/12/673712/11/check/openstack-tox-py37/61ec437/ : SUCCESS in 4m 36s\n- openstack-tox-docs https://logs.opendev.org/12/673712/11/check/openstack-tox-docs/81cbbf2/html/ : SUCCESS in 5m 29s\n- openstack-tox-functional https://logs.opendev.org/12/673712/11/check/openstack-tox-functional/99af3b7/ : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 https://logs.opendev.org/12/673712/11/check/openstack-tox-functional-py36/d1357d4/ : SUCCESS in 5m 55s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/12/673712/11/check/octavia-v2-dsvm-noop-api/6c08346/ : SUCCESS in 41m 21s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/12/673712/11/check/octavia-v2-dsvm-noop-py2-api/38014d6/ : SUCCESS in 44m 36s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/12/673712/11/check/octavia-v2-dsvm-scenario/0e6454a/ : SUCCESS in 1h 38m 38s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/12/673712/11/check/octavia-v2-dsvm-py2-scenario/5566b3d/ : SUCCESS in 1h 34m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/12/673712/11/check/octavia-v2-dsvm-py2-scenario-centos-7/25f464e/ : SUCCESS in 2h 07m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/12/673712/11/check/octavia-v2-dsvm-scenario-two-node/8249dfb/ : SUCCESS in 1h 55m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/12/673712/11/check/octavia-v2-dsvm-py2-scenario-two-node/2ff88cd/ : SUCCESS in 2h 05m 38s (non-voting)\n- octavia-grenade https://logs.opendev.org/12/673712/11/check/octavia-grenade/4c4b8dc/ : SUCCESS in 1h 19m 49s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/12/673712/11/check/octavia-v2-dsvm-tls-barbican/f72bd50/ : FAILURE in 32m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/12/673712/11/check/octavia-v2-dsvm-scenario-amphora-v2/e690709/ : SUCCESS in 1h 34m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/12/673712/11/check/octavia-v2-act-stdby-iptables-dsvm-scenario/36c4b3e/ : SUCCESS in 56m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/12/673712/11/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/274dcb7/ : SUCCESS in 1h 07m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/12/673712/11/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/d8af3e0/ : FAILURE in 1h 22m 02s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"821bd2941dbb0fd73608698d74b5a51195e748d1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-10 21:02:36.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"8940cb3db3445a59e75d2f9378b609b917bcb3bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 23:46:16.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3811f9b9f8942df867fefed4c4e2985 : SUCCESS in 7m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ac0c6e669c844a3ad2a3af80f76f747 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c293a253d774f5aa5fa5ae0e89ef534 : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a0dbdccb664240a8af0750fa23ece492 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bfc97ccd11542699c763b85e60811a9 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34d96faf82344a58970e5d93b4a2940d : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d7ca1903dd246aea4902d10d80e991a : SUCCESS in 9m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/07dc8a1d8b98459b93e65ed6b5d5c4a4 : SUCCESS in 5m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f6060f64633d4261b17c2253819b75af : SUCCESS in 7m 10s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5e90a34657a44eaaa3e7fee783409b98 : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/23bc5d11e57d40f1aade1990f3a48b70 : SUCCESS in 5m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5a9e38c7167f4f49a28dc73fa3caed6f : SUCCESS in 44m 19s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2412f80025dd4569a403b4ae8094bd83 : SUCCESS in 44m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62ca65878f4349fcb127b9f07dce1586 : SUCCESS in 1h 42m 05s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1130d04b30b442a095d413ec33014a1d : SUCCESS in 1h 32m 56s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e6a5d82a245146aa8ac5d6aac466c458 : FAILURE in 2h 31m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4f5b52b754af46909d713f19f559a9f7 : SUCCESS in 1h 31m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e4e0f54746de41edb7b4e7ae5df2f7c9 : SUCCESS in 1h 59m 24s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4db354e948994783baa44890e063ff97 : SUCCESS in 1h 28m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cd4df83f0bd440d2a91c8a171d16f604 : SUCCESS in 1h 09m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bf1b5b6dec6e47e389b04774ac0f342d : SUCCESS in 2h 14m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4017d4e7082845a1842b33625b9b2ffa : SUCCESS in 1h 08m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3a7519b1375a4011b0a060a023a3932a : SUCCESS in 1h 13m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7b36a68dbbb942e5bf9ab8fd1c6b85b3 : FAILURE in 1h 26m 26s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"14de7d33ab2377c21ec6b970d8898b0e9d2c5bf9","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-09-11 02:47:08.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"326bed1a46fc8b86db10de740be470b22736bfb1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-11 07:37:59.000000000","message":"Patch Set 12: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"8e88e1914cc37990c39075669241fafe380588a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 07:38:12.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"67ed92384ef352f83c5701b78e2a0bc34b00708c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 09:40:46.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dadfd8d4d1bb4e49aef900f1d0d389d4 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76ba2aced8154b45b24e6818863be1f5 : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fed6d4d2b0054a34aae3f05e1615d5c8 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/052d05c70efb47ef8950de5fd91c9470 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d8262844df4431c81068a9414cd21c1 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/233a3c774a4e48a594e30bafa9ed3c6f : SUCCESS in 6m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/33684caea71c40a18ace577a43334f3a : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf30c495d8c84320bf7446496b39e2fa : SUCCESS in 6m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cdd74e54edd544028aab87cc28a381c8 : SUCCESS in 43m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/dfb607f1b038431ca7c33c637bdb13b7 : SUCCESS in 47m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7e33db9f78054388a3c331b932871ff3 : SUCCESS in 1h 45m 40s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ba0310410a104104afede32aee1372f5 : SUCCESS in 1h 53m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c167d0d65b13497facc3076c6df08aa2 : SUCCESS in 1h 27m 12s","accounts_in_message":[],"_revision_number":12},{"id":"3bb910002b02d3072707d8d9e204d2e2cd8dd55b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 09:40:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"795696eb908f5b812568fa0a16d9541c94600a31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 09:42:17.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9731bca810404396b97b3c6f2cfe105a : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"8577b3af6f49370e4baadf76e8429406ed04598e","revisions":{"2aec7abfd99995943bc3d5256892a262c5a501b2":{"kind":"REWORK","_number":1,"created":"2019-07-30 23:52:54.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/673712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/673712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/673712/1"}}},"commit":{"parents":[{"commit":"86263eacb36118591e431a06a952d82be57ec604","subject":"Add get method support to the driver-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/86263eacb36118591e431a06a952d82be57ec604"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 23:52:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 23:52:22.000000000","tz":-420},"subject":"Move to using octavia-lib constants","message":"Move to using octavia-lib constants\n\nPreviously the samlple data models and driver agent tests used\nstrings for some of the tests. This patch corrects those to use\nconstants from octavia-lib.\n\nDepends-On: https://review.opendev.org/#/c/673681/\nChange-Id: I90c2fa386fab75742d07e89edd3fba8b56bb6aff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aec7abfd99995943bc3d5256892a262c5a501b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aec7abfd99995943bc3d5256892a262c5a501b2"}]},"branch":"refs/heads/master"},"a30ff3c833135c74823412802a040667ba1fd89e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-01 22:48:27.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/673712/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/673712/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/673712/2"}}},"commit":{"parents":[{"commit":"56b3ed7d401e8f3a565b3e367a6f00915e07c4d4","subject":"Add get method support to the driver-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56b3ed7d401e8f3a565b3e367a6f00915e07c4d4"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 23:52:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-01 22:48:27.000000000","tz":0},"subject":"Move to using octavia-lib constants","message":"Move to using octavia-lib constants\n\nPreviously the samlple data models and driver agent tests used\nstrings for some of the tests. This patch corrects those to use\nconstants from octavia-lib.\n\nDepends-On: https://review.opendev.org/#/c/673681/\nChange-Id: I90c2fa386fab75742d07e89edd3fba8b56bb6aff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a30ff3c833135c74823412802a040667ba1fd89e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a30ff3c833135c74823412802a040667ba1fd89e"}]},"branch":"refs/heads/master"},"61d45e81b9a08f1115310a9bc0a9e40e2d94eb2d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-01 22:58:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/673712/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/673712/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/673712/3"}}},"commit":{"parents":[{"commit":"5794c1a891471bed14f3526d50dae1270df3a3f4","subject":"Add get method support to the driver-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5794c1a891471bed14f3526d50dae1270df3a3f4"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 23:52:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-01 22:58:25.000000000","tz":0},"subject":"Move to using octavia-lib constants","message":"Move to using octavia-lib constants\n\nPreviously the samlple data models and driver agent tests used\nstrings for some of the tests. This patch corrects those to use\nconstants from octavia-lib.\n\nDepends-On: https://review.opendev.org/#/c/673681/\nChange-Id: I90c2fa386fab75742d07e89edd3fba8b56bb6aff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61d45e81b9a08f1115310a9bc0a9e40e2d94eb2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61d45e81b9a08f1115310a9bc0a9e40e2d94eb2d"}]},"branch":"refs/heads/master"},"fd945d6294f5489a1553d3d37419ea32ca3e116c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-01 23:14:36.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/673712/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/673712/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/673712/4"}}},"commit":{"parents":[{"commit":"9edbba34ce2f4ea49937a130fa5a070dcd6b2b3b","subject":"Add get method support to the driver-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9edbba34ce2f4ea49937a130fa5a070dcd6b2b3b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 23:52:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-01 23:14:36.000000000","tz":0},"subject":"Move to using octavia-lib constants","message":"Move to using octavia-lib constants\n\nPreviously the samlple data models and driver agent tests used\nstrings for some of the tests. This patch corrects those to use\nconstants from octavia-lib.\n\nDepends-On: https://review.opendev.org/#/c/673681/\nChange-Id: I90c2fa386fab75742d07e89edd3fba8b56bb6aff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd945d6294f5489a1553d3d37419ea32ca3e116c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd945d6294f5489a1553d3d37419ea32ca3e116c"}]},"branch":"refs/heads/master"},"39d405eaf85f9e08b2c1cf78dd81acb33ea81f91":{"kind":"REWORK","_number":5,"created":"2019-08-02 00:27:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/673712/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/673712/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/673712/5"}}},"commit":{"parents":[{"commit":"6c490c703c6d14d3233ce788e033d7bbf5706a6a","subject":"Add get method support to the driver-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c490c703c6d14d3233ce788e033d7bbf5706a6a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 23:52:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-02 00:26:57.000000000","tz":-420},"subject":"Move to using octavia-lib constants","message":"Move to using octavia-lib constants\n\nPreviously the samlple data models and driver agent tests used\nstrings for some of the tests. This patch corrects those to use\nconstants from octavia-lib.\n\nDepends-On: https://review.opendev.org/#/c/673681/\nChange-Id: I90c2fa386fab75742d07e89edd3fba8b56bb6aff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39d405eaf85f9e08b2c1cf78dd81acb33ea81f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39d405eaf85f9e08b2c1cf78dd81acb33ea81f91"}]},"branch":"refs/heads/master"},"54724c31b5648f7de3c52e1252c12b1410f2821e":{"kind":"REWORK","_number":6,"created":"2019-08-02 00:31:39.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/673712/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/673712/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/673712/6"}}},"commit":{"parents":[{"commit":"6c490c703c6d14d3233ce788e033d7bbf5706a6a","subject":"Add get method support to the driver-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c490c703c6d14d3233ce788e033d7bbf5706a6a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 23:52:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-02 00:31:37.000000000","tz":0},"subject":"Move to using octavia-lib constants","message":"Move to using octavia-lib constants\n\nPreviously the samlple data models and driver agent tests used\nstrings for some of the tests. This patch corrects those to use\nconstants from octavia-lib.\n\nDepends-On: https://review.opendev.org/#/c/673681/\nChange-Id: I90c2fa386fab75742d07e89edd3fba8b56bb6aff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54724c31b5648f7de3c52e1252c12b1410f2821e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54724c31b5648f7de3c52e1252c12b1410f2821e"}]},"branch":"refs/heads/master"},"340c21cc45a4c38f655cf9f187338c5650b759bb":{"kind":"REWORK","_number":7,"created":"2019-08-03 13:45:24.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/673712/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/673712/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/673712/7"}}},"commit":{"parents":[{"commit":"86996a1a99aa4c1075eb3ed0cbedc7bdad75d1ed","subject":"Add get method support to the driver-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/86996a1a99aa4c1075eb3ed0cbedc7bdad75d1ed"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 23:52:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-03 13:44:47.000000000","tz":-420},"subject":"Move to using octavia-lib constants","message":"Move to using octavia-lib constants\n\nPreviously the samlple data models and driver agent tests used\nstrings for some of the tests. This patch corrects those to use\nconstants from octavia-lib.\n\nDepends-On: https://review.opendev.org/#/c/673681/\nChange-Id: I90c2fa386fab75742d07e89edd3fba8b56bb6aff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/340c21cc45a4c38f655cf9f187338c5650b759bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/340c21cc45a4c38f655cf9f187338c5650b759bb"}]},"branch":"refs/heads/master"},"44cc0bb393f2b02b8e449a0f9d0d52a5cf985a07":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-08-03 14:15:25.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/673712/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/673712/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/673712/8"}}},"commit":{"parents":[{"commit":"a07fe0b379f647d2294fdd6e7fc061915d7dd8dd","subject":"Add get method support to the driver-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a07fe0b379f647d2294fdd6e7fc061915d7dd8dd"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 23:52:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-03 14:15:25.000000000","tz":0},"subject":"Move to using octavia-lib constants","message":"Move to using octavia-lib constants\n\nPreviously the samlple data models and driver agent tests used\nstrings for some of the tests. This patch corrects those to use\nconstants from octavia-lib.\n\nDepends-On: https://review.opendev.org/#/c/673681/\nChange-Id: I90c2fa386fab75742d07e89edd3fba8b56bb6aff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/44cc0bb393f2b02b8e449a0f9d0d52a5cf985a07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/44cc0bb393f2b02b8e449a0f9d0d52a5cf985a07"}]},"branch":"refs/heads/master"},"e799a47c8306ab18157156da6ad8a1d84ee633c8":{"kind":"REWORK","_number":9,"created":"2019-08-05 20:32:52.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/12/673712/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/673712/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/673712/9"}}},"commit":{"parents":[{"commit":"5971fa53cf294be184e94333c8111a16dd181f11","subject":"Add get method support to the driver-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5971fa53cf294be184e94333c8111a16dd181f11"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 23:52:22.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-05 20:32:30.000000000","tz":-420},"subject":"Move to using octavia-lib constants","message":"Move to using octavia-lib constants\n\nPreviously the samlple data models and driver agent tests used\nstrings for some of the tests. This patch corrects those to use\nconstants from octavia-lib.\n\nDepends-On: https://review.opendev.org/#/c/673681/\nChange-Id: I90c2fa386fab75742d07e89edd3fba8b56bb6aff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e799a47c8306ab18157156da6ad8a1d84ee633c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e799a47c8306ab18157156da6ad8a1d84ee633c8"}]},"branch":"refs/heads/master"},"c43145de356ede818b85c8f18edd3e2de7116f43":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-08-09 17:19:30.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/673712/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/673712/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/673712/10"}}},"commit":{"parents":[{"commit":"935c9bbcab5bb359cad89f3bea4d0652b27b3e0c","subject":"Add get method support to the driver-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/935c9bbcab5bb359cad89f3bea4d0652b27b3e0c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 23:52:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-09 17:19:30.000000000","tz":0},"subject":"Move to using octavia-lib constants","message":"Move to using octavia-lib constants\n\nPreviously the samlple data models and driver agent tests used\nstrings for some of the tests. This patch corrects those to use\nconstants from octavia-lib.\n\nDepends-On: https://review.opendev.org/#/c/673681/\nChange-Id: I90c2fa386fab75742d07e89edd3fba8b56bb6aff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c43145de356ede818b85c8f18edd3e2de7116f43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c43145de356ede818b85c8f18edd3e2de7116f43"}]},"branch":"refs/heads/master"},"71d905c9b8278cc0f5fd6440d8ba924aac56a43a":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-08-09 23:09:05.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/673712/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/673712/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/673712/11"}}},"commit":{"parents":[{"commit":"f6e9e94af07833497cd8bbccbdffe5c0123da4db","subject":"Add get method support to the driver-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f6e9e94af07833497cd8bbccbdffe5c0123da4db"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 23:52:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-09 23:09:05.000000000","tz":0},"subject":"Move to using octavia-lib constants","message":"Move to using octavia-lib constants\n\nPreviously the samlple data models and driver agent tests used\nstrings for some of the tests. This patch corrects those to use\nconstants from octavia-lib.\n\nDepends-On: https://review.opendev.org/#/c/673681/\nChange-Id: I90c2fa386fab75742d07e89edd3fba8b56bb6aff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71d905c9b8278cc0f5fd6440d8ba924aac56a43a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71d905c9b8278cc0f5fd6440d8ba924aac56a43a"}]},"branch":"refs/heads/master"},"8577b3af6f49370e4baadf76e8429406ed04598e":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-09-10 21:02:36.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/673712/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/673712/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/673712/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/673712/12"}}},"commit":{"parents":[{"commit":"09efc2a423c6ab033f9a648549aff87590be5990","subject":"Add get method support to the driver-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09efc2a423c6ab033f9a648549aff87590be5990"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 23:52:22.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-10 21:02:36.000000000","tz":0},"subject":"Move to using octavia-lib constants","message":"Move to using octavia-lib constants\n\nPreviously the samlple data models and driver agent tests used\nstrings for some of the tests. This patch corrects those to use\nconstants from octavia-lib.\n\nDepends-On: https://review.opendev.org/#/c/673681/\nChange-Id: I90c2fa386fab75742d07e89edd3fba8b56bb6aff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8577b3af6f49370e4baadf76e8429406ed04598e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8577b3af6f49370e4baadf76e8429406ed04598e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
