)]}'
{"id":"openstack%2Foctavia~673783","triplet_id":"openstack%2Foctavia~stable%2Frocky~If0eb6e934e8bfb6759c61db3f1f1b89fe02264eb","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"If0eb6e934e8bfb6759c61db3f1f1b89fe02264eb","subject":"Add octavia-v2-dsvm jobs to the gate queue","status":"MERGED","created":"2019-07-31 10:55:46.000000000","updated":"2019-08-08 02:32:29.000000000","submitted":"2019-08-08 02:32:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673783-1565231549084-fcaf60fb","meta_rev_id":"5c7ca29b4dcd137991f220eb6aef4231b44393ac","_number":673783,"virtual_id_number":673783,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-08-08 02:32:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-08-03 00:35:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-08-07 22:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-08-07 22:42:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-03 00:35:41.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-07 22:42:45.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-08 02:32:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5bbb02c20383d8ff3c9e401806b48610e5cd32a0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-31 10:55:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9cf991ea3d05ca19cddf32a437dacfc6c9744e3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 13:19:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/83/673783/1/check/openstack-tox-lower-constraints/50e6a38/ : SUCCESS in 8m 29s\n- openstack-tox-pep8 http://logs.openstack.org/83/673783/1/check/openstack-tox-pep8/186fc96/ : SUCCESS in 6m 42s\n- openstack-tox-py27 http://logs.openstack.org/83/673783/1/check/openstack-tox-py27/162b323/ : SUCCESS in 5m 01s\n- openstack-tox-py35 http://logs.openstack.org/83/673783/1/check/openstack-tox-py35/6fc665e/ : SUCCESS in 5m 14s\n- openstack-tox-docs http://logs.openstack.org/83/673783/1/check/openstack-tox-docs/4a4a208/html/ : SUCCESS in 4m 49s\n- neutron-lbaasv2-dsvm-api http://logs.openstack.org/83/673783/1/check/neutron-lbaasv2-dsvm-api/cccc516/ : SUCCESS in 1h 16m 03s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api http://logs.openstack.org/83/673783/1/check/neutron-lbaasv2-dsvm-py3x-api/7f65e0f/ : SUCCESS in 1h 18m 15s\n- neutron-lbaasv2-dsvm-scenario http://logs.openstack.org/83/673783/1/check/neutron-lbaasv2-dsvm-scenario/226bf8b/ : SUCCESS in 1h 21m 50s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario http://logs.openstack.org/83/673783/1/check/neutron-lbaasv2-dsvm-py3x-scenario/9cf4dab/ : FAILURE in 1h 20m 24s (non-voting)\n- openstack-tox-functional http://logs.openstack.org/83/673783/1/check/openstack-tox-functional/58c9df6/ : SUCCESS in 5m 06s\n- openstack-tox-functional-py35 http://logs.openstack.org/83/673783/1/check/openstack-tox-functional-py35/8268d4f/ : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 http://logs.openstack.org/83/673783/1/check/openstack-tox-functional-py36/9a4b219/ : SUCCESS in 5m 16s\n- octavia-v1-dsvm-scenario http://logs.openstack.org/83/673783/1/check/octavia-v1-dsvm-scenario/e45bb52/ : SUCCESS in 59m 29s\n- octavia-v1-dsvm-py3x-scenario http://logs.openstack.org/83/673783/1/check/octavia-v1-dsvm-py3x-scenario/f5e4205/ : SUCCESS in 59m 07s\n- octavia-v1-dsvm-scenario-multinode http://logs.openstack.org/83/673783/1/check/octavia-v1-dsvm-scenario-multinode/0e4d0f8/ : FAILURE in 1h 29m 31s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode http://logs.openstack.org/83/673783/1/check/octavia-v1-dsvm-py3x-scenario-multinode/203b55e/ : FAILURE in 1h 29m 02s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 http://logs.openstack.org/83/673783/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/35eb554/ : SUCCESS in 1h 04m 55s (non-voting)\n- octavia-v2-dsvm-noop-api http://logs.openstack.org/83/673783/1/check/octavia-v2-dsvm-noop-api/bf0f8bc/ : SUCCESS in 47m 13s\n- octavia-v2-dsvm-noop-py35-api http://logs.openstack.org/83/673783/1/check/octavia-v2-dsvm-noop-py35-api/5d69802/ : SUCCESS in 44m 05s\n- octavia-v2-dsvm-scenario http://logs.openstack.org/83/673783/1/check/octavia-v2-dsvm-scenario/33cba74/ : SUCCESS in 1h 30m 27s\n- octavia-v2-dsvm-py35-scenario http://logs.openstack.org/83/673783/1/check/octavia-v2-dsvm-py35-scenario/fcae001/ : FAILURE in 2h 20m 18s\n- octavia-grenade http://logs.openstack.org/83/673783/1/check/octavia-grenade/e035e6f/ : SUCCESS in 1h 06m 06s\n- openstack-tox-cover http://logs.openstack.org/83/673783/1/check/openstack-tox-cover/d9c2a9d/cover/ : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":1},{"id":"cfba26dcb2f024b74d28b2005651fa5717ac789c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 00:35:41.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck\n\nSeems like this is true, yes.","accounts_in_message":[],"_revision_number":1},{"id":"e00c6824176d61a11047576b5213f4a60861a98e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 02:51:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/83/673783/1/check/openstack-tox-lower-constraints/3203f4b/ : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://logs.opendev.org/83/673783/1/check/openstack-tox-pep8/328a725/ : SUCCESS in 6m 43s\n- openstack-tox-py27 https://logs.opendev.org/83/673783/1/check/openstack-tox-py27/467bbb6/ : SUCCESS in 4m 26s\n- openstack-tox-py35 https://logs.opendev.org/83/673783/1/check/openstack-tox-py35/cf9f2ec/ : SUCCESS in 4m 43s\n- openstack-tox-docs https://logs.opendev.org/83/673783/1/check/openstack-tox-docs/9b383f8/html/ : SUCCESS in 4m 47s\n- neutron-lbaasv2-dsvm-api https://logs.opendev.org/83/673783/1/check/neutron-lbaasv2-dsvm-api/2a1aafc/ : SUCCESS in 1h 13m 42s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://logs.opendev.org/83/673783/1/check/neutron-lbaasv2-dsvm-py3x-api/b07198f/ : SUCCESS in 1h 13m 27s\n- neutron-lbaasv2-dsvm-scenario https://logs.opendev.org/83/673783/1/check/neutron-lbaasv2-dsvm-scenario/569b546/ : FAILURE in 2h 04m 08s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://logs.opendev.org/83/673783/1/check/neutron-lbaasv2-dsvm-py3x-scenario/b9730cd/ : SUCCESS in 1h 24m 05s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/83/673783/1/check/openstack-tox-functional/db1fe1b/ : SUCCESS in 5m 15s\n- openstack-tox-functional-py35 https://logs.opendev.org/83/673783/1/check/openstack-tox-functional-py35/a009a3d/ : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://logs.opendev.org/83/673783/1/check/openstack-tox-functional-py36/9506e6d/ : SUCCESS in 4m 23s\n- octavia-v1-dsvm-scenario https://logs.opendev.org/83/673783/1/check/octavia-v1-dsvm-scenario/57173da/ : SUCCESS in 1h 05m 15s\n- octavia-v1-dsvm-py3x-scenario https://logs.opendev.org/83/673783/1/check/octavia-v1-dsvm-py3x-scenario/2ef20d9/ : SUCCESS in 56m 19s\n- octavia-v1-dsvm-scenario-multinode https://logs.opendev.org/83/673783/1/check/octavia-v1-dsvm-scenario-multinode/bb11920/ : FAILURE in 1h 02m 49s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://logs.opendev.org/83/673783/1/check/octavia-v1-dsvm-py3x-scenario-multinode/ed1f2fd/ : FAILURE in 1h 30m 54s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://logs.opendev.org/83/673783/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/e592cca/ : FAILURE in 32m 42s (non-voting)\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/83/673783/1/check/octavia-v2-dsvm-noop-api/d6ee734/ : SUCCESS in 42m 37s\n- octavia-v2-dsvm-noop-py35-api https://logs.opendev.org/83/673783/1/check/octavia-v2-dsvm-noop-py35-api/4f131c0/ : SUCCESS in 40m 33s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/83/673783/1/check/octavia-v2-dsvm-scenario/b33f86f/ : SUCCESS in 1h 11m 07s\n- octavia-v2-dsvm-py35-scenario https://logs.opendev.org/83/673783/1/check/octavia-v2-dsvm-py35-scenario/89124c4/ : SUCCESS in 1h 40m 28s\n- octavia-grenade https://logs.opendev.org/83/673783/1/check/octavia-grenade/09c9796/ : SUCCESS in 56m 05s\n- openstack-tox-cover https://logs.opendev.org/83/673783/1/check/openstack-tox-cover/c22a76f/cover/ : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"161f622e1c87eacd803d6a08c46d563a1bc3979e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-07 22:42:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"cf59ff21af8531ec1ad93efa8279bc40e73fab5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 22:43:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a78ee7ded4ce94dd1ca4dd04be29bbec0de5967e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 02:32:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/83/673783/1/gate/openstack-tox-lower-constraints/c058d9a/ : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://logs.opendev.org/83/673783/1/gate/openstack-tox-pep8/1e84119/ : SUCCESS in 8m 16s\n- openstack-tox-py27 https://logs.opendev.org/83/673783/1/gate/openstack-tox-py27/6ff5495/ : SUCCESS in 5m 04s\n- openstack-tox-py35 https://logs.opendev.org/83/673783/1/gate/openstack-tox-py35/0313349/ : SUCCESS in 5m 55s\n- openstack-tox-docs https://logs.opendev.org/83/673783/1/gate/openstack-tox-docs/b1d70f8/html/ : SUCCESS in 5m 46s\n- openstack-tox-functional https://logs.opendev.org/83/673783/1/gate/openstack-tox-functional/c4e32fc/ : SUCCESS in 6m 00s\n- openstack-tox-functional-py35 https://logs.opendev.org/83/673783/1/gate/openstack-tox-functional-py35/7f8432b/ : SUCCESS in 7m 53s\n- openstack-tox-functional-py36 https://logs.opendev.org/83/673783/1/gate/openstack-tox-functional-py36/d15dda5/ : SUCCESS in 6m 30s\n- octavia-v1-dsvm-scenario https://logs.opendev.org/83/673783/1/gate/octavia-v1-dsvm-scenario/c10aa7a/ : SUCCESS in 1h 06m 52s\n- octavia-v1-dsvm-py3x-scenario https://logs.opendev.org/83/673783/1/gate/octavia-v1-dsvm-py3x-scenario/b3c524a/ : SUCCESS in 1h 04m 52s\n- octavia-v1-dsvm-scenario-multinode https://logs.opendev.org/83/673783/1/gate/octavia-v1-dsvm-scenario-multinode/d52bd07/ : FAILURE in 1h 22m 06s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://logs.opendev.org/83/673783/1/gate/octavia-v1-dsvm-py3x-scenario-multinode/67d9eb5/ : FAILURE in 1h 23m 27s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://logs.opendev.org/83/673783/1/gate/neutron-lbaasv2-dsvm-py3x-api/b5f2468/ : SUCCESS in 1h 21m 48s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/83/673783/1/gate/octavia-v2-dsvm-noop-api/ca415a5/ : SUCCESS in 58m 50s\n- octavia-v2-dsvm-noop-py35-api https://logs.opendev.org/83/673783/1/gate/octavia-v2-dsvm-noop-py35-api/f604733/ : SUCCESS in 46m 11s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/83/673783/1/gate/octavia-v2-dsvm-scenario/cd1efaa/ : SUCCESS in 1h 41m 24s\n- octavia-v2-dsvm-py35-scenario https://logs.opendev.org/83/673783/1/gate/octavia-v2-dsvm-py35-scenario/c24889d/ : SUCCESS in 1h 39m 23s\n- octavia-grenade https://logs.opendev.org/83/673783/1/gate/octavia-grenade/19ac8a6/ : SUCCESS in 1h 08m 58s\n- openstack-tox-cover https://logs.opendev.org/83/673783/1/gate/openstack-tox-cover/9780390/cover/ : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":1},{"id":"5c7ca29b4dcd137991f220eb6aef4231b44393ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 02:32:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb9ebe4cf7e4fa638d8e3e7b34d38ae9d8debf78","revisions":{"eb9ebe4cf7e4fa638d8e3e7b34d38ae9d8debf78":{"kind":"REWORK","_number":1,"created":"2019-07-31 10:55:46.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/83/673783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/673783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/673783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/673783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/673783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/673783/1"}}},"commit":{"parents":[{"commit":"b8b323ec71642e0833340803e2a2eb5558e54aa3","subject":"Merge \"Fix for utils LB DM transformation function\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8b323ec71642e0833340803e2a2eb5558e54aa3"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-31 10:53:14.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-07-31 10:53:14.000000000","tz":120},"subject":"Add octavia-v2-dsvm jobs to the gate queue","message":"Add octavia-v2-dsvm jobs to the gate queue\n\nOctavia is a standalone project and supports the LBaaSv2.\nWe should be gating on octavia-v2-dsvm jobs in stable/rocky.\n\nChange-Id: If0eb6e934e8bfb6759c61db3f1f1b89fe02264eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb9ebe4cf7e4fa638d8e3e7b34d38ae9d8debf78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb9ebe4cf7e4fa638d8e3e7b34d38ae9d8debf78"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
