)]}'
{"id":"openstack%2Foctavia~674087","triplet_id":"openstack%2Foctavia~master~I3be0fa18578c7c1552f24842a09e18c01e34358a","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I3be0fa18578c7c1552f24842a09e18c01e34358a","subject":"Fix wrong package names or versions for centos-minimal images","status":"MERGED","created":"2019-08-01 16:36:28.000000000","updated":"2019-08-07 11:48:04.000000000","submitted":"2019-08-07 11:48:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674087-1565178484582-0d575412","meta_rev_id":"7c07c7ed38390aaf426ca9863a1248982ce73d8b","_number":674087,"virtual_id_number":674087,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2019-08-07 00:40:03.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2019-08-07 07:28:06.000000000","_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2019-08-07 11:48:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-08-06 15:39:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-08-06 15:39:59.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-08-06 15:39:59.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-08-06 15:40:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 21:58:22.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-08-02 02:34:27.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-08-06 15:40:12.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-07 00:40:03.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-07 11:48:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9105e391fb634901e3176d773572b389bc4fad15","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-01 16:36:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6aa4861f1f6c651ed4a2d0aa2d5b964315784c70","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-01 16:43:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nMea culpa... Thanks a lot, Greg!\n\nCould you please rebase this on top of https://review.opendev.org/#/c/673172/ ? Hopefully this way it will be validated by the centos CI job.","accounts_in_message":[],"_revision_number":1},{"id":"6497c369f98198038c60469e40f921b65c65a618","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-01 16:53:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"973ef065169d2eb59aca9cb6c83d9dcf4e6aa5d4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-01 16:54:52.000000000","message":"Patch Set 1:\n\n(1 comment)\n\ndone and added depends-on 673172","accounts_in_message":[],"_revision_number":1},{"id":"1994dcee8a97e00e81e5a8526d2b8122389833b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 19:29:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/87/674087/2/check/openstack-tox-cover/015225b/cover/ : SUCCESS in 6m 51s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/674087/2/check/openstack-tox-lower-constraints/033c4a7/ : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://logs.opendev.org/87/674087/2/check/openstack-tox-pep8/6f4345c/ : SUCCESS in 6m 10s\n- openstack-tox-py27 https://logs.opendev.org/87/674087/2/check/openstack-tox-py27/eaf7fb7/ : SUCCESS in 5m 44s\n- openstack-tox-py36 https://logs.opendev.org/87/674087/2/check/openstack-tox-py36/952a9ed/ : SUCCESS in 5m 02s\n- openstack-tox-py37 https://logs.opendev.org/87/674087/2/check/openstack-tox-py37/dcb8d89/ : SUCCESS in 4m 44s\n- openstack-tox-docs https://logs.opendev.org/87/674087/2/check/openstack-tox-docs/1c99965/html/ : SUCCESS in 5m 18s\n- openstack-tox-functional https://logs.opendev.org/87/674087/2/check/openstack-tox-functional/1b5e1c4/ : SUCCESS in 5m 31s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/674087/2/check/openstack-tox-functional-py36/8ba8b21/ : SUCCESS in 9m 12s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-noop-api/57032b8/ : SUCCESS in 46m 32s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-noop-py2-api/784d154/ : SUCCESS in 44m 41s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-scenario/00ff0a9/ : SUCCESS in 1h 41m 07s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-py2-scenario/5aa71c5/ : SUCCESS in 2h 07m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-py2-scenario-centos-7/39ec501/ : FAILURE in 2h 29m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-scenario-two-node/ffd5ce5/ : SUCCESS in 1h 53m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-py2-scenario-two-node/b9b0564/ : SUCCESS in 1h 49m 23s (non-voting)\n- octavia-grenade https://logs.opendev.org/87/674087/2/check/octavia-grenade/d0e9c24/ : SUCCESS in 1h 28m 10s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-tls-barbican/f0bf5d1/ : SUCCESS in 57m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-scenario-amphora-v2/df27018/ : FAILURE in 1h 22m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/87/674087/2/check/octavia-v2-act-stdby-iptables-dsvm-scenario/473e427/ : FAILURE in 1h 23m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/87/674087/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/bcd3738/ : FAILURE in 1h 17m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/674087/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/cb040d4/ : FAILURE in 1h 39m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e52ee4f9a2b2040940a7c5c634f7b9cf0074536","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-01 19:36:52.000000000","message":"Patch Set 2:\n\nrecheck\n\ndhclient and haproxy 1.8 are installed. Most tempest tests passed, still rechecking to double-check.\n\nhttps://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-py2-scenario-centos-7/39ec501/controller/logs/dib-build/amphora-x64-haproxy.qcow2_log.txt.gz#_2019-08-01_17_36_46_072","accounts_in_message":[],"_revision_number":2},{"id":"519145de6a6bc59d712b42934e8d1ad886d69887","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 21:49:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/87/674087/2/check/openstack-tox-cover/7d818f5/cover/ : SUCCESS in 7m 20s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/674087/2/check/openstack-tox-lower-constraints/86d2e15/ : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://logs.opendev.org/87/674087/2/check/openstack-tox-pep8/c939c9c/ : SUCCESS in 5m 34s\n- openstack-tox-py27 https://logs.opendev.org/87/674087/2/check/openstack-tox-py27/91a0604/ : SUCCESS in 6m 21s\n- openstack-tox-py36 https://logs.opendev.org/87/674087/2/check/openstack-tox-py36/9835025/ : SUCCESS in 4m 53s\n- openstack-tox-py37 https://logs.opendev.org/87/674087/2/check/openstack-tox-py37/d498bc5/ : SUCCESS in 6m 10s\n- openstack-tox-docs https://logs.opendev.org/87/674087/2/check/openstack-tox-docs/095bc1e/html/ : SUCCESS in 6m 10s\n- openstack-tox-functional https://logs.opendev.org/87/674087/2/check/openstack-tox-functional/5faa95e/ : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/674087/2/check/openstack-tox-functional-py36/07174b8/ : SUCCESS in 8m 08s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-noop-api/f37774a/ : SUCCESS in 44m 58s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-noop-py2-api/cafdc75/ : SUCCESS in 48m 42s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-scenario/daf586b/ : SUCCESS in 1h 24m 42s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-py2-scenario/b54cf13/ : SUCCESS in 1h 37m 00s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-py2-scenario-centos-7/eada151/ : SUCCESS in 1h 59m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-scenario-two-node/b8d00f2/ : SUCCESS in 1h 58m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-py2-scenario-two-node/89155ef/ : FAILURE in 1h 40m 34s (non-voting)\n- octavia-grenade https://logs.opendev.org/87/674087/2/check/octavia-grenade/736bd73/ : SUCCESS in 1h 20m 51s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-tls-barbican/795a91c/ : SUCCESS in 59m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/87/674087/2/check/octavia-v2-dsvm-scenario-amphora-v2/83190ff/ : SUCCESS in 1h 39m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/87/674087/2/check/octavia-v2-act-stdby-iptables-dsvm-scenario/74248dd/ : FAILURE in 1h 19m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/87/674087/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/93dc5a2/ : FAILURE in 52m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/674087/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/a320f0f/ : FAILURE in 56m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"529cbc715ce20fcc8e0bea7c09650b5592616ad2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-01 21:58:22.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"c6c2940f65d8e98e4cc3fbf14282129af8c3dec0","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-08-02 02:34:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"def93b72e1bf6fd669d97445e44fd30dfd30ebe8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 00:18:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"28018417ffaf43b4d72f481b6de071defc1210e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 16:01:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"29a27d6bc7bd52dad49680c8eddab4eea3a346fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 17:56:08.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints openstack-tox-lower-constraints : CANCELED\n- openstack-tox-pep8 openstack-tox-pep8 : CANCELED\n- openstack-tox-py27 openstack-tox-py27 : CANCELED\n- openstack-tox-py36 finger://ze11.openstack.org/8b74c68144384f3e8d07733bede75113 : CANCELED in 1m 11s\n- openstack-tox-py37 https://logs.opendev.org/87/674087/2/gate/openstack-tox-py37/c8489f1/ : FAILURE in 7m 56s\n- openstack-tox-docs https://logs.opendev.org/87/674087/2/gate/openstack-tox-docs/8185f0f/html/ : SUCCESS in 6m 24s\n- openstack-tox-functional https://logs.opendev.org/87/674087/2/gate/openstack-tox-functional/4293d58/ : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 finger://ze05.openstack.org/829ab6f1b7b04217beb62c471d75dc23 : CANCELED in 4m 06s\n- octavia-v2-dsvm-noop-api finger://ze05.openstack.org/984212aabede4d61a765d13e1fac450d : CANCELED in 3m 35s\n- octavia-v2-dsvm-noop-py2-api finger://ze11.openstack.org/da41ebd285064700bde21dadb942b732 : CANCELED in 1m 55s\n- octavia-v2-dsvm-scenario finger://ze02.openstack.org/7520e712f48c49e7929afc24fb0ea374 : CANCELED in 7m 11s\n- octavia-v2-dsvm-py2-scenario finger://ze03.openstack.org/c657a1ab157b4070b6d54867b81363f2 : CANCELED in 6m 33s\n- octavia-grenade finger://ze03.openstack.org/dec65fc0491d4f8b9e5698bf952a530a : CANCELED in 5m 33s","accounts_in_message":[],"_revision_number":2},{"id":"2a14b75baa9a2fca7223fb43368f554dec646523","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-06 15:39:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6990390f193dfe31f4c21b44af0ee14f12ea3b95","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-06 15:40:12.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f6d6d65746d1d3c1d14f9925e1f4202e11c95a8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 18:22:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/87/674087/3/check/openstack-tox-cover/7015bf9/cover/ : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/674087/3/check/openstack-tox-lower-constraints/04818b4/ : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://logs.opendev.org/87/674087/3/check/openstack-tox-pep8/cae0430/ : SUCCESS in 5m 10s\n- openstack-tox-py27 https://logs.opendev.org/87/674087/3/check/openstack-tox-py27/8d0dee7/ : SUCCESS in 6m 26s\n- openstack-tox-py36 https://logs.opendev.org/87/674087/3/check/openstack-tox-py36/cbf3458/ : SUCCESS in 5m 54s\n- openstack-tox-py37 https://logs.opendev.org/87/674087/3/check/openstack-tox-py37/12f176b/ : SUCCESS in 4m 54s\n- openstack-tox-docs https://logs.opendev.org/87/674087/3/check/openstack-tox-docs/b58f2fa/html/ : SUCCESS in 5m 35s\n- openstack-tox-functional https://logs.opendev.org/87/674087/3/check/openstack-tox-functional/d5f7102/ : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/674087/3/check/openstack-tox-functional-py36/b5179c7/ : SUCCESS in 6m 08s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-noop-api/20eb92b/ : SUCCESS in 57m 45s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-noop-py2-api/678a9a5/ : SUCCESS in 45m 25s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-scenario/507f8d9/ : SUCCESS in 1h 41m 54s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-py2-scenario/17886b1/ : FAILURE in 2h 10m 50s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-py2-scenario-centos-7/1b647c8/ : FAILURE in 2h 24m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-scenario-two-node/ab8f357/ : SUCCESS in 2h 03m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-py2-scenario-two-node/89b30dd/ : SUCCESS in 1h 59m 31s (non-voting)\n- octavia-grenade https://logs.opendev.org/87/674087/3/check/octavia-grenade/6b3fe9a/ : SUCCESS in 1h 21m 41s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-tls-barbican/3223a70/ : SUCCESS in 54m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-scenario-amphora-v2/da8cca2/ : SUCCESS in 1h 43m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/87/674087/3/check/octavia-v2-act-stdby-iptables-dsvm-scenario/3deccf7/ : FAILURE in 1h 21m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/87/674087/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/8529ec7/ : FAILURE in 1h 20m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/674087/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/ef5a2b4/ : FAILURE in 1h 39m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"31c781ae53730bab50c21e89a2bc7f12b698bd2d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-06 18:36:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b9acfc91493606418fff6d24a30ea43722652b17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 21:26:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/87/674087/3/check/openstack-tox-cover/3523bc5/cover/ : SUCCESS in 7m 29s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/674087/3/check/openstack-tox-lower-constraints/76feb77/ : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://logs.opendev.org/87/674087/3/check/openstack-tox-pep8/367552b/ : SUCCESS in 5m 14s\n- openstack-tox-py27 https://logs.opendev.org/87/674087/3/check/openstack-tox-py27/40d8e72/ : SUCCESS in 5m 24s\n- openstack-tox-py36 https://logs.opendev.org/87/674087/3/check/openstack-tox-py36/e782752/ : SUCCESS in 5m 00s\n- openstack-tox-py37 https://logs.opendev.org/87/674087/3/check/openstack-tox-py37/c2d6a35/ : SUCCESS in 5m 10s\n- openstack-tox-docs https://logs.opendev.org/87/674087/3/check/openstack-tox-docs/d2f1754/html/ : SUCCESS in 5m 22s\n- openstack-tox-functional https://logs.opendev.org/87/674087/3/check/openstack-tox-functional/b03dc2a/ : SUCCESS in 5m 27s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/674087/3/check/openstack-tox-functional-py36/c134da4/ : SUCCESS in 5m 43s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-noop-api/d7a9600/ : SUCCESS in 45m 11s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-noop-py2-api/e1d4de8/ : SUCCESS in 47m 35s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-scenario/f1e0705/ : SUCCESS in 1h 36m 32s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-py2-scenario/e30f361/ : SUCCESS in 1h 46m 03s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-py2-scenario-centos-7/177d5e8/ : FAILURE in 2h 09m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-scenario-two-node/b834256/ : SUCCESS in 2h 08m 36s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-py2-scenario-two-node/bd35300/ : TIMED_OUT in 2h 35m 50s (non-voting)\n- octavia-grenade https://logs.opendev.org/87/674087/3/check/octavia-grenade/79e74e5/ : SUCCESS in 1h 24m 55s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-tls-barbican/eaf1a52/ : SUCCESS in 1h 00m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-scenario-amphora-v2/5bcdb80/ : SUCCESS in 1h 41m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/87/674087/3/check/octavia-v2-act-stdby-iptables-dsvm-scenario/044f3c4/ : FAILURE in 1h 17m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/87/674087/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/757ce17/ : FAILURE in 1h 29m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/674087/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/6417115/ : FAILURE in 1h 35m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9d6232a20e89bd54728ae0f2da056d08d1c0152b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 21:26:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"90b4fdccf8451b172213fdc2a9d321d4e5baaf3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 23:38:40.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/87/674087/3/gate/openstack-tox-lower-constraints/ec96c2e/ : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://logs.opendev.org/87/674087/3/gate/openstack-tox-pep8/1d770cd/ : SUCCESS in 5m 59s\n- openstack-tox-py27 https://logs.opendev.org/87/674087/3/gate/openstack-tox-py27/dc073e2/ : SUCCESS in 4m 13s\n- openstack-tox-py36 https://logs.opendev.org/87/674087/3/gate/openstack-tox-py36/7ed4e82/ : SUCCESS in 5m 37s\n- openstack-tox-py37 https://logs.opendev.org/87/674087/3/gate/openstack-tox-py37/4d913a3/ : SUCCESS in 4m 41s\n- openstack-tox-docs https://logs.opendev.org/87/674087/3/gate/openstack-tox-docs/47472d2/html/ : SUCCESS in 6m 35s\n- openstack-tox-functional https://logs.opendev.org/87/674087/3/gate/openstack-tox-functional/cc84a49/ : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/674087/3/gate/openstack-tox-functional-py36/9e93d07/ : SUCCESS in 6m 39s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/674087/3/gate/octavia-v2-dsvm-noop-api/72c4f32/ : SUCCESS in 50m 35s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/674087/3/gate/octavia-v2-dsvm-noop-py2-api/f07b350/ : SUCCESS in 59m 57s\n- octavia-v2-dsvm-scenario finger://ze03.openstack.org/8c6142b5273c4d1cb71d45ac416c98d4 : CANCELED in 2h 09m 36s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/674087/3/gate/octavia-v2-dsvm-py2-scenario/43fb68e/ : FAILURE in 2h 09m 35s\n- octavia-grenade https://logs.opendev.org/87/674087/3/gate/octavia-grenade/153b1f8/ : SUCCESS in 1h 21m 51s","accounts_in_message":[],"_revision_number":3},{"id":"9d096daf9b59fe2b51577018dc983f5feb6a9da5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-07 00:40:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4b632cc450c6808760c085431e40caf0fc43e82f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 02:47:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/87/674087/3/check/openstack-tox-cover/2a0328a/cover/ : SUCCESS in 8m 43s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/674087/3/check/openstack-tox-lower-constraints/84538f6/ : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://logs.opendev.org/87/674087/3/check/openstack-tox-pep8/99497b8/ : SUCCESS in 5m 19s\n- openstack-tox-py27 https://logs.opendev.org/87/674087/3/check/openstack-tox-py27/324210b/ : SUCCESS in 4m 21s\n- openstack-tox-py36 https://logs.opendev.org/87/674087/3/check/openstack-tox-py36/5a3e2e7/ : SUCCESS in 4m 30s\n- openstack-tox-py37 https://logs.opendev.org/87/674087/3/check/openstack-tox-py37/cae040e/ : SUCCESS in 4m 26s\n- openstack-tox-docs https://logs.opendev.org/87/674087/3/check/openstack-tox-docs/29a0b7d/html/ : SUCCESS in 5m 42s\n- openstack-tox-functional https://logs.opendev.org/87/674087/3/check/openstack-tox-functional/54999da/ : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/674087/3/check/openstack-tox-functional-py36/0dcfa9f/ : SUCCESS in 6m 47s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-noop-api/f6ec506/ : SUCCESS in 51m 15s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-noop-py2-api/e8a9505/ : SUCCESS in 43m 53s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-scenario/3b4b0b4/ : FAILURE in 24m 26s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-py2-scenario/f36c274/ : SUCCESS in 1h 48m 23s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-py2-scenario-centos-7/b6f5c2a/ : FAILURE in 1h 59m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-scenario-two-node/b5a25e0/ : SUCCESS in 1h 18m 05s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-py2-scenario-two-node/ac69af2/ : FAILURE in 1h 24m 44s (non-voting)\n- octavia-grenade https://logs.opendev.org/87/674087/3/check/octavia-grenade/9a57340/ : SUCCESS in 1h 20m 24s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-tls-barbican/f8ad157/ : SUCCESS in 57m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-scenario-amphora-v2/eb2c5b1/ : SUCCESS in 1h 35m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/87/674087/3/check/octavia-v2-act-stdby-iptables-dsvm-scenario/01110aa/ : SUCCESS in 1h 00m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/87/674087/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/e235cc3/ : FAILURE in 1h 21m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/674087/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/9ef4bee/ : FAILURE in 1h 03m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a0686838ff51a1b9a2f209ce8aac78b35b818388","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-07 07:28:06.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6347663eb11a5336d7234a091489d1c9d9cd9549","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 10:03:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/87/674087/3/check/openstack-tox-cover/0d2ad16/cover/ : SUCCESS in 9m 03s\n- openstack-tox-lower-constraints https://logs.opendev.org/87/674087/3/check/openstack-tox-lower-constraints/2f45ab6/ : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://logs.opendev.org/87/674087/3/check/openstack-tox-pep8/884fb45/ : SUCCESS in 5m 45s\n- openstack-tox-py27 https://logs.opendev.org/87/674087/3/check/openstack-tox-py27/d5a90bb/ : SUCCESS in 5m 01s\n- openstack-tox-py36 https://logs.opendev.org/87/674087/3/check/openstack-tox-py36/2f50676/ : SUCCESS in 4m 31s\n- openstack-tox-py37 https://logs.opendev.org/87/674087/3/check/openstack-tox-py37/152150a/ : SUCCESS in 5m 50s\n- openstack-tox-docs https://logs.opendev.org/87/674087/3/check/openstack-tox-docs/672c760/html/ : SUCCESS in 6m 33s\n- openstack-tox-functional https://logs.opendev.org/87/674087/3/check/openstack-tox-functional/cd88bf1/ : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/674087/3/check/openstack-tox-functional-py36/aaf52fe/ : SUCCESS in 5m 58s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-noop-api/188efa0/ : SUCCESS in 42m 14s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-noop-py2-api/81ff671/ : SUCCESS in 54m 33s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-scenario/56d1fd5/ : SUCCESS in 1h 57m 56s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-py2-scenario/c615bbf/ : SUCCESS in 1h 28m 04s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-py2-scenario-centos-7/4c36f6a/ : SUCCESS in 2h 29m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-scenario-two-node/24cc581/ : SUCCESS in 1h 15m 48s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-py2-scenario-two-node/0690ba2/ : SUCCESS in 1h 54m 58s (non-voting)\n- octavia-grenade https://logs.opendev.org/87/674087/3/check/octavia-grenade/896b0d6/ : SUCCESS in 1h 20m 57s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-tls-barbican/75606a3/ : SUCCESS in 53m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/87/674087/3/check/octavia-v2-dsvm-scenario-amphora-v2/19f8f7c/ : SUCCESS in 1h 44m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/87/674087/3/check/octavia-v2-act-stdby-iptables-dsvm-scenario/884dede/ : FAILURE in 1h 35m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/87/674087/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/c0e3caa/ : FAILURE in 1h 25m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/87/674087/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/e3667a0/ : SUCCESS in 1h 38m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f32cc03ac002f55afddb5bac1e5cf48c3f34501b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 10:04:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f7132ec69a8afa036defae9b5b36a0cade2f575a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 11:48:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7c07c7ed38390aaf426ca9863a1248982ce73d8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 11:48:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/87/674087/3/gate/openstack-tox-lower-constraints/d5499ef/ : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://logs.opendev.org/87/674087/3/gate/openstack-tox-pep8/c7c2113/ : SUCCESS in 5m 44s\n- openstack-tox-py27 https://logs.opendev.org/87/674087/3/gate/openstack-tox-py27/a6d717f/ : SUCCESS in 4m 46s\n- openstack-tox-py36 https://logs.opendev.org/87/674087/3/gate/openstack-tox-py36/ab96cd4/ : SUCCESS in 4m 35s\n- openstack-tox-py37 https://logs.opendev.org/87/674087/3/gate/openstack-tox-py37/22923ed/ : SUCCESS in 5m 25s\n- openstack-tox-docs https://logs.opendev.org/87/674087/3/gate/openstack-tox-docs/054783c/html/ : SUCCESS in 6m 11s\n- openstack-tox-functional https://logs.opendev.org/87/674087/3/gate/openstack-tox-functional/55301c1/ : SUCCESS in 5m 27s\n- openstack-tox-functional-py36 https://logs.opendev.org/87/674087/3/gate/openstack-tox-functional-py36/86f9f47/ : SUCCESS in 7m 30s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/87/674087/3/gate/octavia-v2-dsvm-noop-api/7742fb6/ : SUCCESS in 52m 10s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/87/674087/3/gate/octavia-v2-dsvm-noop-py2-api/794ffe9/ : SUCCESS in 45m 13s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/87/674087/3/gate/octavia-v2-dsvm-scenario/9dd98aa/ : SUCCESS in 1h 41m 06s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/87/674087/3/gate/octavia-v2-dsvm-py2-scenario/4b44542/ : SUCCESS in 1h 35m 10s\n- octavia-grenade https://logs.opendev.org/87/674087/3/gate/octavia-grenade/984e3d9/ : SUCCESS in 1h 20m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"50c168e860b0947b33109dd4d94beea44400f671","revisions":{"8f84a1878174d7c6fb7b8f36ab1226d9a9c73e35":{"kind":"REWORK","_number":1,"created":"2019-08-01 16:36:28.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/674087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/674087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/674087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/674087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/674087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/674087/1"}}},"commit":{"parents":[{"commit":"be94300f5bc9812ecc3278969053197114ba8151","subject":"Merge \"Build amphora image using centos-minimal element\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be94300f5bc9812ecc3278969053197114ba8151"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-08-01 14:57:57.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-08-01 16:35:38.000000000","tz":120},"subject":"Fix wrong package names or versions for centos-minimal images","message":"Fix wrong package names or versions for centos-minimal images\n\nSince centos-minimal is used as base for centos image, dib installs\nhaproxy 1.5.x instead of haproxy 1.8.x, and dhcp client is missing\n(dhclient package).\n\nStory: 2006323\nTask: 36056\n\nChange-Id: I3be0fa18578c7c1552f24842a09e18c01e34358a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f84a1878174d7c6fb7b8f36ab1226d9a9c73e35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f84a1878174d7c6fb7b8f36ab1226d9a9c73e35"}]},"branch":"refs/heads/master"},"413a4de3db9146ea5defcfe43093e8a053073e6c":{"kind":"REWORK","_number":2,"created":"2019-08-01 16:53:35.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/674087/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/674087/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/674087/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/674087/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/674087/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/674087/2"}}},"commit":{"parents":[{"commit":"be94300f5bc9812ecc3278969053197114ba8151","subject":"Merge \"Build amphora image using centos-minimal element\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be94300f5bc9812ecc3278969053197114ba8151"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-08-01 14:57:57.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-08-01 16:52:27.000000000","tz":120},"subject":"Fix wrong package names or versions for centos-minimal images","message":"Fix wrong package names or versions for centos-minimal images\n\nSince centos-minimal is used as base for centos image, dib installs\nhaproxy 1.5.x instead of haproxy 1.8.x, and dhcp client is missing\n(dhclient package).\n\nDepends-On: https://review.opendev.org/#/c/673172/\n\nStory: 2006323\nTask: 36056\n\nChange-Id: I3be0fa18578c7c1552f24842a09e18c01e34358a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/413a4de3db9146ea5defcfe43093e8a053073e6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/413a4de3db9146ea5defcfe43093e8a053073e6c"}]},"branch":"refs/heads/master"},"50c168e860b0947b33109dd4d94beea44400f671":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-06 15:39:59.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/87/674087/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/674087/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/674087/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/674087/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/674087/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/674087/3"}}},"commit":{"parents":[{"commit":"b32064cdf2d5c70140a84e2497931941c4ee6887","subject":"Fix a unit test for new octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b32064cdf2d5c70140a84e2497931941c4ee6887"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-08-01 14:57:57.000000000","tz":120},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-06 15:39:59.000000000","tz":0},"subject":"Fix wrong package names or versions for centos-minimal images","message":"Fix wrong package names or versions for centos-minimal images\n\nSince centos-minimal is used as base for centos image, dib installs\nhaproxy 1.5.x instead of haproxy 1.8.x, and dhcp client is missing\n(dhclient package).\n\nDepends-On: https://review.opendev.org/#/c/673172/\n\nStory: 2006323\nTask: 36056\n\nChange-Id: I3be0fa18578c7c1552f24842a09e18c01e34358a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50c168e860b0947b33109dd4d94beea44400f671"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50c168e860b0947b33109dd4d94beea44400f671"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
