)]}'
{"id":"openstack%2Foctavia~674259","triplet_id":"openstack%2Foctavia~master~Iaa0f763005529cc394d3b95ce07f2be71e9b3778","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Iaa0f763005529cc394d3b95ce07f2be71e9b3778","subject":"Install missing packages in nodepool instance","status":"MERGED","created":"2019-08-02 10:43:43.000000000","updated":"2019-08-06 23:16:34.000000000","submitted":"2019-08-06 23:16:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674259-1565133394512-cf54861c","meta_rev_id":"393090aa3c85a9fe9868240abd86e85a045ce0ff","_number":674259,"virtual_id_number":674259,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-08-06 23:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-08-06 15:39:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-08-06 15:39:12.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-08-06 15:39:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-03 23:17:21.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-08-06 15:39:21.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-06 23:16:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb1e0d80447e4586cef0becc28847e36734b194b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-02 10:43:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"442577ae325b4c517b061306144a9e68cc8d7a31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 10:59:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/59/674259/1/check/openstack-tox-cover/12ab8df/cover/ : SUCCESS in 8m 30s\n- openstack-tox-lower-constraints https://logs.opendev.org/59/674259/1/check/openstack-tox-lower-constraints/04254ad/ : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://logs.opendev.org/59/674259/1/check/openstack-tox-pep8/d46ee1e/ : SUCCESS in 7m 01s\n- openstack-tox-py27 https://logs.opendev.org/59/674259/1/check/openstack-tox-py27/d7c2ae8/ : SUCCESS in 4m 57s\n- openstack-tox-py36 https://logs.opendev.org/59/674259/1/check/openstack-tox-py36/10affdc/ : SUCCESS in 6m 15s\n- openstack-tox-py37 https://logs.opendev.org/59/674259/1/check/openstack-tox-py37/cb9e2f0/ : SUCCESS in 5m 43s\n- openstack-tox-docs https://logs.opendev.org/59/674259/1/check/openstack-tox-docs/cc32ea8/html/ : SUCCESS in 5m 58s\n- publish-openstack-octavia-amphora-image-bionic https://logs.opendev.org/59/674259/1/check/publish-openstack-octavia-amphora-image-bionic/5405f77/ : SUCCESS in 10m 03s\n- publish-openstack-octavia-amphora-image-centos7 https://logs.opendev.org/59/674259/1/check/publish-openstack-octavia-amphora-image-centos7/b49dde9/ : FAILURE in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"70af4c1b92adb321877cb5986a8ff594bc742c75","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-02 11:39:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac7c00a4bcc37f174f71ed1b5529570fc81fdd29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 11:56:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/59/674259/2/check/openstack-tox-cover/efb2cf2/cover/ : SUCCESS in 8m 17s\n- openstack-tox-lower-constraints https://logs.opendev.org/59/674259/2/check/openstack-tox-lower-constraints/6a82ff0/ : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://logs.opendev.org/59/674259/2/check/openstack-tox-pep8/f9a7fd0/ : SUCCESS in 6m 33s\n- openstack-tox-py27 https://logs.opendev.org/59/674259/2/check/openstack-tox-py27/8c7cac4/ : SUCCESS in 4m 44s\n- openstack-tox-py36 https://logs.opendev.org/59/674259/2/check/openstack-tox-py36/db0215d/ : SUCCESS in 5m 34s\n- openstack-tox-py37 https://logs.opendev.org/59/674259/2/check/openstack-tox-py37/ff27427/ : SUCCESS in 4m 27s\n- openstack-tox-docs https://logs.opendev.org/59/674259/2/check/openstack-tox-docs/bfed3a2/html/ : SUCCESS in 6m 16s\n- publish-openstack-octavia-amphora-image-bionic https://logs.opendev.org/59/674259/2/check/publish-openstack-octavia-amphora-image-bionic/4093623/ : SUCCESS in 10m 20s\n- publish-openstack-octavia-amphora-image-centos7 https://logs.opendev.org/59/674259/2/check/publish-openstack-octavia-amphora-image-centos7/44afc11/ : SUCCESS in 12m 00s","accounts_in_message":[],"_revision_number":2},{"id":"215a8a67c3743714efb8d5725a8c8e7edbc2e385","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-02 12:05:23.000000000","message":"Patch Set 2:\n\n2019-08-02 11:50:59.320464 | ubuntu-bionic | 2019-08-02 11:50:59.320 | Converting image using qemu-img convert\n2019-08-02 11:52:06.936165 | ubuntu-bionic | 2019-08-02 11:52:06.935 | Image file /home/zuul/test-images/test-only-amphora-x64-haproxy-centos-7.qcow2 created...\n2019-08-02 11:52:07.132939 | ubuntu-bionic | 2019-08-02 11:52:07.132 | Build completed successfully\n2019-08-02 11:52:09.156866 | ubuntu-bionic | ok: Runtime: 0:06:06.222203","accounts_in_message":[],"_revision_number":2},{"id":"67fa1c18e3419218d13edb0769d16f3109fb0332","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-02 12:07:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a27c5adf5efbe78ced5dd0dff0483149fbfd165e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 14:51:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/59/674259/3/check/openstack-tox-cover/6444440/cover/ : SUCCESS in 8m 09s\n- openstack-tox-lower-constraints https://logs.opendev.org/59/674259/3/check/openstack-tox-lower-constraints/5e907ea/ : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://logs.opendev.org/59/674259/3/check/openstack-tox-pep8/783cd86/ : SUCCESS in 7m 29s\n- openstack-tox-py27 https://logs.opendev.org/59/674259/3/check/openstack-tox-py27/323f40a/ : SUCCESS in 7m 15s\n- openstack-tox-py36 https://logs.opendev.org/59/674259/3/check/openstack-tox-py36/c2be6ff/ : SUCCESS in 6m 48s\n- openstack-tox-py37 https://logs.opendev.org/59/674259/3/check/openstack-tox-py37/09cf58f/ : SUCCESS in 4m 42s\n- openstack-tox-docs https://logs.opendev.org/59/674259/3/check/openstack-tox-docs/9fdf74f/html/ : SUCCESS in 6m 29s\n- openstack-tox-functional https://logs.opendev.org/59/674259/3/check/openstack-tox-functional/8e24268/ : SUCCESS in 5m 28s\n- openstack-tox-functional-py36 https://logs.opendev.org/59/674259/3/check/openstack-tox-functional-py36/df6a555/ : SUCCESS in 7m 44s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-noop-api/f7a6b2f/ : SUCCESS in 44m 57s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-noop-py2-api/edb8649/ : SUCCESS in 56m 49s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-scenario/7fe0196/ : SUCCESS in 1h 35m 14s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-py2-scenario/073f3da/ : SUCCESS in 1h 54m 44s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-py2-scenario-centos-7/2613bcc/ : FAILURE in 2h 35m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-scenario-two-node/6854606/ : SUCCESS in 2h 04m 08s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-py2-scenario-two-node/9c54e57/ : SUCCESS in 1h 46m 45s (non-voting)\n- octavia-grenade https://logs.opendev.org/59/674259/3/check/octavia-grenade/6d55cb2/ : SUCCESS in 1h 22m 28s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-tls-barbican/7cd01fe/ : SUCCESS in 1h 00m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-scenario-amphora-v2/a71b551/ : SUCCESS in 1h 34m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/59/674259/3/check/octavia-v2-act-stdby-iptables-dsvm-scenario/d12c822/ : FAILURE in 1h 18m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/59/674259/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/187e5ae/ : FAILURE in 1h 21m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/674259/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/ba9cfb1/ : FAILURE in 1h 18m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ed869809085f9652eacdb80958494dbc67d59e18","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 13:37:14.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nHow did we build centos images before? O_o","accounts_in_message":[],"_revision_number":3},{"id":"545f81acd947fb59324eb312e94c0b7dbf0ea135","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-08-03 23:17:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"727196961b375c3048c1b77c4d01e615e0d5658c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 23:17:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f20d13027d79a3a834131d58999767df6dbc0aee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 23:26:03.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints finger://ze02.openstack.org/1278a1ccfcbc4b7d87496991af604996 : CANCELED in 3m 54s\n- openstack-tox-pep8 finger://ze01.openstack.org/96a47e922f4249f68144b5f71372feb8 : CANCELED in 4m 24s\n- openstack-tox-py27 finger://ze09.openstack.org/969ec20c6dfe4ddca25421f91cc765ef : CANCELED in 4m 24s\n- openstack-tox-py36 finger://ze02.openstack.org/4911f9af95e748758695d948128fc2f2 : CANCELED in 4m 05s\n- openstack-tox-py37 finger://ze05.openstack.org/4fd427c7e1cb4d97a805eac8a1d4f8fc : CANCELED in 2m 54s\n- openstack-tox-docs https://logs.opendev.org/59/674259/3/gate/openstack-tox-docs/487aca3/ : FAILURE in 4m 24s\n- openstack-tox-functional finger://ze01.openstack.org/1de091ea755b4964b39d2c1616a314b6 : CANCELED in 3m 34s\n- openstack-tox-functional-py36 finger://ze03.openstack.org/47e7662bf260443da818483cb3da25bb : CANCELED in 3m 14s\n- octavia-v2-dsvm-noop-api finger://ze03.openstack.org/e444ff25e4714a1d954c9b9e9b920017 : CANCELED in 3m 42s\n- octavia-v2-dsvm-noop-py2-api finger://ze03.openstack.org/3ff4f82e5d974339a7112fb05ad3dff6 : CANCELED in 4m 00s\n- octavia-v2-dsvm-scenario finger://ze02.openstack.org/dd035537e5df4149a988214cc73f51e9 : CANCELED in 3m 18s\n- octavia-v2-dsvm-py2-scenario finger://ze02.openstack.org/f61941c464754aa686579fcf7747fbf3 : CANCELED in 2m 51s\n- octavia-grenade finger://ze02.openstack.org/a58860435a5e4078bd2f2427172b6a4a : CANCELED in 2m 56s","accounts_in_message":[],"_revision_number":3},{"id":"0a713079f9db3c9f63aac885d5dd0cd64aba8d88","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-08-05 01:21:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"22a03309086817bf04d85201ea260dfb6c86f43d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 03:56:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/59/674259/3/check/openstack-tox-cover/f48b783/ : FAILURE in 6m 49s\n- openstack-tox-lower-constraints https://logs.opendev.org/59/674259/3/check/openstack-tox-lower-constraints/f1360ae/ : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://logs.opendev.org/59/674259/3/check/openstack-tox-pep8/6750b3a/ : SUCCESS in 4m 38s\n- openstack-tox-py27 https://logs.opendev.org/59/674259/3/check/openstack-tox-py27/e711903/ : FAILURE in 5m 55s\n- openstack-tox-py36 https://logs.opendev.org/59/674259/3/check/openstack-tox-py36/40aeb1a/ : FAILURE in 6m 41s\n- openstack-tox-py37 https://logs.opendev.org/59/674259/3/check/openstack-tox-py37/8bd8267/ : FAILURE in 6m 41s\n- openstack-tox-docs https://logs.opendev.org/59/674259/3/check/openstack-tox-docs/8877799/ : FAILURE in 4m 09s\n- openstack-tox-functional https://logs.opendev.org/59/674259/3/check/openstack-tox-functional/2f26bf7/ : SUCCESS in 5m 10s\n- openstack-tox-functional-py36 https://logs.opendev.org/59/674259/3/check/openstack-tox-functional-py36/7179a94/ : SUCCESS in 4m 20s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-noop-api/a14a3af/ : SUCCESS in 40m 02s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-noop-py2-api/210a43b/ : SUCCESS in 42m 33s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-scenario/e287074/ : SUCCESS in 2h 00m 52s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-py2-scenario/fc3c159/ : SUCCESS in 1h 27m 24s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-py2-scenario-centos-7/c1e7d98/ : FAILURE in 2h 30m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-scenario-two-node/ddddf92/ : SUCCESS in 1h 52m 33s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-py2-scenario-two-node/a93365c/ : SUCCESS in 1h 52m 44s (non-voting)\n- octavia-grenade https://logs.opendev.org/59/674259/3/check/octavia-grenade/3beb35e/ : SUCCESS in 1h 22m 17s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-tls-barbican/3b8d9e1/ : SUCCESS in 52m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-scenario-amphora-v2/88527df/ : SUCCESS in 1h 44m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/59/674259/3/check/octavia-v2-act-stdby-iptables-dsvm-scenario/26d0b77/ : FAILURE in 1h 15m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/59/674259/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/ca72f05/ : FAILURE in 36m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/674259/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/a30bf18/ : FAILURE in 1h 11m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7d14aa2b53465718e895618685f0ce83dbac95c7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-05 08:16:35.000000000","message":"Patch Set 3:\n\n(1 comment)\n\n\u003e How did we build centos images before? O_o\n\nCentOS amps were being built using the \u0027centos7\u0027 DIB element which uses the GenericCloud qcow2 images from https://cloud.centos.org/centos/7/images/. Now on centos-minimal, DIB creates a chroot, installs packages and creates the qcow2 itself.","accounts_in_message":[],"_revision_number":3},{"id":"363af916b1d6bff1b929144ca156a4b13de1ceef","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-05 08:16:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3f2d283f59663f9f5b661feb1bf45c78d32d865a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 10:53:41.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/59/674259/3/check/openstack-tox-cover/b4fd890/ : FAILURE in 7m 11s\n- openstack-tox-lower-constraints https://logs.opendev.org/59/674259/3/check/openstack-tox-lower-constraints/0442ce3/ : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://logs.opendev.org/59/674259/3/check/openstack-tox-pep8/d3a29c2/ : SUCCESS in 4m 57s\n- openstack-tox-py27 https://logs.opendev.org/59/674259/3/check/openstack-tox-py27/330c744/ : FAILURE in 6m 40s\n- openstack-tox-py36 https://logs.opendev.org/59/674259/3/check/openstack-tox-py36/e4b7afa/ : FAILURE in 6m 27s\n- openstack-tox-py37 https://logs.opendev.org/59/674259/3/check/openstack-tox-py37/4827f2a/ : FAILURE in 6m 36s\n- openstack-tox-docs https://logs.opendev.org/59/674259/3/check/openstack-tox-docs/2e3168d/html/ : SUCCESS in 4m 59s\n- openstack-tox-functional https://logs.opendev.org/59/674259/3/check/openstack-tox-functional/32db179/ : SUCCESS in 5m 12s\n- openstack-tox-functional-py36 https://logs.opendev.org/59/674259/3/check/openstack-tox-functional-py36/ebf8824/ : SUCCESS in 5m 07s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-noop-api/622f3d2/ : SUCCESS in 44m 05s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-noop-py2-api/ae58df5/ : SUCCESS in 42m 21s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-scenario/6e51ca1/ : SUCCESS in 1h 40m 05s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-py2-scenario/8d75db0/ : SUCCESS in 1h 42m 49s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-py2-scenario-centos-7/5f681b2/ : FAILURE in 2h 31m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-scenario-two-node/92e0a53/ : SUCCESS in 1h 59m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-py2-scenario-two-node/c185051/ : SUCCESS in 1h 46m 16s (non-voting)\n- octavia-grenade https://logs.opendev.org/59/674259/3/check/octavia-grenade/07b1c20/ : SUCCESS in 1h 21m 07s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-tls-barbican/d66c375/ : SUCCESS in 1h 02m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/59/674259/3/check/octavia-v2-dsvm-scenario-amphora-v2/c85dceb/ : SUCCESS in 1h 28m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/59/674259/3/check/octavia-v2-act-stdby-iptables-dsvm-scenario/5e515cb/ : FAILURE in 1h 15m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/59/674259/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/2367979/ : FAILURE in 52m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/674259/3/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/1ab34e4/ : FAILURE in 1h 07m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fe544bea2923ac13f4de81db53c440492031954c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-05 17:21:48.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"f81ac1e69536661c3cd6baf7b6154f963d9808d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 20:06:32.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/59/674259/4/check/openstack-tox-cover/1f57502/ : FAILURE in 7m 26s\n- openstack-tox-lower-constraints https://logs.opendev.org/59/674259/4/check/openstack-tox-lower-constraints/b3d87ce/ : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://logs.opendev.org/59/674259/4/check/openstack-tox-pep8/c7be0ab/ : SUCCESS in 7m 02s\n- openstack-tox-py27 https://logs.opendev.org/59/674259/4/check/openstack-tox-py27/e3c664d/ : FAILURE in 7m 50s\n- openstack-tox-py36 https://logs.opendev.org/59/674259/4/check/openstack-tox-py36/e7219b1/ : FAILURE in 6m 03s\n- openstack-tox-py37 https://logs.opendev.org/59/674259/4/check/openstack-tox-py37/01d85e0/ : FAILURE in 9m 43s\n- openstack-tox-docs https://logs.opendev.org/59/674259/4/check/openstack-tox-docs/cde0a05/html/ : SUCCESS in 6m 24s\n- openstack-tox-functional https://logs.opendev.org/59/674259/4/check/openstack-tox-functional/0148fe4/ : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 https://logs.opendev.org/59/674259/4/check/openstack-tox-functional-py36/4caa1e9/ : SUCCESS in 5m 52s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/59/674259/4/check/octavia-v2-dsvm-noop-api/31e75a8/ : SUCCESS in 54m 14s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/59/674259/4/check/octavia-v2-dsvm-noop-py2-api/48b09d0/ : SUCCESS in 38m 46s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/59/674259/4/check/octavia-v2-dsvm-scenario/559a853/ : SUCCESS in 1h 48m 23s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/59/674259/4/check/octavia-v2-dsvm-py2-scenario/50b413b/ : SUCCESS in 1h 33m 22s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/674259/4/check/octavia-v2-dsvm-py2-scenario-centos-7/fa5b375/ : FAILURE in 2h 35m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/59/674259/4/check/octavia-v2-dsvm-scenario-two-node/5af37c5/ : SUCCESS in 1h 58m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/59/674259/4/check/octavia-v2-dsvm-py2-scenario-two-node/44a0c78/ : SUCCESS in 1h 56m 31s (non-voting)\n- octavia-grenade https://logs.opendev.org/59/674259/4/check/octavia-grenade/7b4f37e/ : SUCCESS in 1h 22m 01s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/59/674259/4/check/octavia-v2-dsvm-tls-barbican/0e954a6/ : SUCCESS in 1h 02m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/59/674259/4/check/octavia-v2-dsvm-scenario-amphora-v2/c8c54c8/ : FAILURE in 2h 14m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/59/674259/4/check/octavia-v2-act-stdby-iptables-dsvm-scenario/99dcd5e/ : FAILURE in 1h 15m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/59/674259/4/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/95e8fc4/ : FAILURE in 1h 25m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/674259/4/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/454dd5c/ : FAILURE in 1h 13m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ebaa5c44c52e7c25131560db3ba14b765718903e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-06 15:39:12.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"e56a64964df70518b67e783bf4c15af9d9c0afcb","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-06 15:39:21.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e5cdd268eb2becd4f08c5b58f207e955f433d9ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 18:29:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/59/674259/5/check/openstack-tox-cover/38253e3/cover/ : SUCCESS in 7m 09s\n- openstack-tox-lower-constraints https://logs.opendev.org/59/674259/5/check/openstack-tox-lower-constraints/81c27a8/ : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://logs.opendev.org/59/674259/5/check/openstack-tox-pep8/6744b08/ : SUCCESS in 6m 59s\n- openstack-tox-py27 https://logs.opendev.org/59/674259/5/check/openstack-tox-py27/f67b209/ : SUCCESS in 4m 38s\n- openstack-tox-py36 https://logs.opendev.org/59/674259/5/check/openstack-tox-py36/73e86c0/ : SUCCESS in 4m 18s\n- openstack-tox-py37 https://logs.opendev.org/59/674259/5/check/openstack-tox-py37/c25c18c/ : SUCCESS in 4m 38s\n- openstack-tox-docs https://logs.opendev.org/59/674259/5/check/openstack-tox-docs/ea0ac84/html/ : SUCCESS in 6m 11s\n- openstack-tox-functional https://logs.opendev.org/59/674259/5/check/openstack-tox-functional/fe5b799/ : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://logs.opendev.org/59/674259/5/check/openstack-tox-functional-py36/27a4a82/ : SUCCESS in 6m 00s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-noop-api/b47553b/ : SUCCESS in 39m 17s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-noop-py2-api/a09bc04/ : SUCCESS in 50m 33s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-scenario/6c3219b/ : FAILURE in 2h 15m 47s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-py2-scenario/a0063ad/ : SUCCESS in 1h 48m 52s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-py2-scenario-centos-7/3b32b1f/ : TIMED_OUT in 2h 35m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-scenario-two-node/2ea3152/ : FAILURE in 1h 38m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-py2-scenario-two-node/1d619c5/ : SUCCESS in 1h 57m 14s (non-voting)\n- octavia-grenade https://logs.opendev.org/59/674259/5/check/octavia-grenade/8a34904/ : SUCCESS in 1h 24m 35s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-tls-barbican/dd57403/ : SUCCESS in 55m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-scenario-amphora-v2/f2b3c87/ : SUCCESS in 1h 39m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/59/674259/5/check/octavia-v2-act-stdby-iptables-dsvm-scenario/d19d7c9/ : FAILURE in 1h 06m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/59/674259/5/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/e732b6f/ : FAILURE in 1h 23m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/674259/5/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/d11718c/ : FAILURE in 1h 22m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a95246e7fb0865cd968010df7bb3d5adfca04a8f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-06 18:35:58.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"2575dc789e21e0c5de7f8442a582d957bfd48add","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 21:19:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/59/674259/5/check/openstack-tox-cover/06199e8/cover/ : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://logs.opendev.org/59/674259/5/check/openstack-tox-lower-constraints/ad839eb/ : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://logs.opendev.org/59/674259/5/check/openstack-tox-pep8/a18d18e/ : SUCCESS in 5m 16s\n- openstack-tox-py27 https://logs.opendev.org/59/674259/5/check/openstack-tox-py27/6442ac1/ : SUCCESS in 4m 24s\n- openstack-tox-py36 https://logs.opendev.org/59/674259/5/check/openstack-tox-py36/65013d1/ : SUCCESS in 3m 42s\n- openstack-tox-py37 https://logs.opendev.org/59/674259/5/check/openstack-tox-py37/50b4ffb/ : SUCCESS in 4m 58s\n- openstack-tox-docs https://logs.opendev.org/59/674259/5/check/openstack-tox-docs/60b6b6b/html/ : SUCCESS in 5m 00s\n- openstack-tox-functional https://logs.opendev.org/59/674259/5/check/openstack-tox-functional/7e074e6/ : SUCCESS in 5m 10s\n- openstack-tox-functional-py36 https://logs.opendev.org/59/674259/5/check/openstack-tox-functional-py36/3406a9e/ : SUCCESS in 4m 51s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-noop-api/3385404/ : SUCCESS in 45m 26s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-noop-py2-api/30925f1/ : SUCCESS in 49m 44s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-scenario/ef7388c/ : SUCCESS in 1h 38m 31s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-py2-scenario/14abde1/ : SUCCESS in 1h 40m 54s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-py2-scenario-centos-7/d4d77df/ : TIMED_OUT in 2h 35m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-scenario-two-node/ff01ddc/ : SUCCESS in 1h 52m 27s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-py2-scenario-two-node/5412b8f/ : SUCCESS in 1h 46m 29s (non-voting)\n- octavia-grenade https://logs.opendev.org/59/674259/5/check/octavia-grenade/0ce3058/ : SUCCESS in 1h 14m 48s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-tls-barbican/e573974/ : SUCCESS in 1h 02m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/59/674259/5/check/octavia-v2-dsvm-scenario-amphora-v2/af0c04d/ : SUCCESS in 1h 46m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/59/674259/5/check/octavia-v2-act-stdby-iptables-dsvm-scenario/e07ff59/ : FAILURE in 1h 15m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/59/674259/5/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/03354e8/ : FAILURE in 1h 17m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/59/674259/5/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/c78262d/ : FAILURE in 1h 22m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d92974de090cdb2e1df9cda8ade20db8df2c8b04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 21:19:29.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a33abd235be053a27a4da5ba67ace63f149ae2d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 23:16:34.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/59/674259/5/gate/openstack-tox-lower-constraints/3ea17b8/ : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://logs.opendev.org/59/674259/5/gate/openstack-tox-pep8/5963cad/ : SUCCESS in 7m 01s\n- openstack-tox-py27 https://logs.opendev.org/59/674259/5/gate/openstack-tox-py27/c87e48f/ : SUCCESS in 5m 46s\n- openstack-tox-py36 https://logs.opendev.org/59/674259/5/gate/openstack-tox-py36/104b7f2/ : SUCCESS in 5m 34s\n- openstack-tox-py37 https://logs.opendev.org/59/674259/5/gate/openstack-tox-py37/2f21132/ : SUCCESS in 4m 42s\n- openstack-tox-docs https://logs.opendev.org/59/674259/5/gate/openstack-tox-docs/e8f1e3a/html/ : SUCCESS in 7m 05s\n- openstack-tox-functional https://logs.opendev.org/59/674259/5/gate/openstack-tox-functional/a9f4430/ : SUCCESS in 8m 58s\n- openstack-tox-functional-py36 https://logs.opendev.org/59/674259/5/gate/openstack-tox-functional-py36/61ef80c/ : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/59/674259/5/gate/octavia-v2-dsvm-noop-api/9b6546c/ : SUCCESS in 53m 42s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/59/674259/5/gate/octavia-v2-dsvm-noop-py2-api/a26a101/ : SUCCESS in 47m 59s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/59/674259/5/gate/octavia-v2-dsvm-scenario/7b9a5b5/ : SUCCESS in 1h 45m 42s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/59/674259/5/gate/octavia-v2-dsvm-py2-scenario/3bed652/ : SUCCESS in 1h 49m 53s\n- octavia-grenade https://logs.opendev.org/59/674259/5/gate/octavia-grenade/922e1e8/ : SUCCESS in 1h 21m 15s","accounts_in_message":[],"_revision_number":5},{"id":"393090aa3c85a9fe9868240abd86e85a045ce0ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 23:16:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6e57e6fd3d046ad55d39535c2493cce997b54380","revisions":{"c50115e7caee105d5ae562d6c931f897a49e8dbe":{"kind":"REWORK","_number":1,"created":"2019-08-02 10:43:43.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/59/674259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/674259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/674259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/674259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/674259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/674259/1"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-02 10:39:47.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-02 10:39:47.000000000","tz":120},"subject":"WIP: Install missing packages in nodepool instance","message":"WIP: Install missing packages in nodepool instance\n\nThe \u0027rpm\u0027 and \u0027bc\u0027 tools do not come installed in the Ubuntu nodepool\nimage. We need \u0027rpm\u0027 to build centos-minimal amphora images and \u0027bc\u0027 for\nchecking minimum supported Ubuntu version in diskimage-create.sh\n\nThis patch also adds octavia-lib to the required projects.\n\nChange-Id: Iaa0f763005529cc394d3b95ce07f2be71e9b3778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c50115e7caee105d5ae562d6c931f897a49e8dbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c50115e7caee105d5ae562d6c931f897a49e8dbe"}]},"branch":"refs/heads/master"},"1f90d825604717057a7fdf4d8ec0a60def5d6426":{"kind":"REWORK","_number":2,"created":"2019-08-02 11:39:58.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/59/674259/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/674259/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/674259/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/674259/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/674259/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/674259/2"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-02 10:39:47.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-02 11:39:47.000000000","tz":120},"subject":"WIP: Install missing packages in nodepool instance","message":"WIP: Install missing packages in nodepool instance\n\nThe \u0027rpm\u0027 and \u0027bc\u0027 tools do not come installed in the Ubuntu nodepool\nimage. We need \u0027rpm\u0027 to build centos-minimal amphora images and \u0027bc\u0027 for\nchecking minimum supported Ubuntu version in diskimage-create.sh\n\nThis patch also adds octavia-lib to the required projects.\n\nChange-Id: Iaa0f763005529cc394d3b95ce07f2be71e9b3778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f90d825604717057a7fdf4d8ec0a60def5d6426"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f90d825604717057a7fdf4d8ec0a60def5d6426"}]},"branch":"refs/heads/master"},"22cfe365366189549165adfd4fe8d7731c737070":{"kind":"REWORK","_number":3,"created":"2019-08-02 12:07:01.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/59/674259/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/674259/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/674259/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/674259/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/674259/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/674259/3"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-02 10:39:47.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-02 12:06:15.000000000","tz":120},"subject":"Install missing packages in nodepool instance","message":"Install missing packages in nodepool instance\n\nThe \u0027rpm\u0027, \u0027yumdownloader\u0027 and \u0027bc\u0027 tools do not come installed in the\nUbuntu nodepool image. We need \u0027rpm\u0027 and \u0027yumdownloader\u0027 to build\ncentos-minimal amphora images and \u0027bc\u0027 for checking minimum supported\nUbuntu version in diskimage-create.sh\n\nThis patch also adds octavia-lib to the required projects.\n\nChange-Id: Iaa0f763005529cc394d3b95ce07f2be71e9b3778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22cfe365366189549165adfd4fe8d7731c737070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22cfe365366189549165adfd4fe8d7731c737070"}]},"branch":"refs/heads/master"},"0c30ffdb814148e7fb82817a187a25b2a0676c97":{"kind":"NO_CHANGE","_number":4,"created":"2019-08-05 17:21:48.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/59/674259/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/674259/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/674259/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/674259/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/674259/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/674259/4"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-02 10:39:47.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-05 17:20:53.000000000","tz":120},"subject":"Install missing packages in nodepool instance","message":"Install missing packages in nodepool instance\n\nThe \u0027rpm\u0027, \u0027yumdownloader\u0027 and \u0027bc\u0027 tools do not come installed in the\nUbuntu nodepool image. We need \u0027rpm\u0027 and \u0027yumdownloader\u0027 to build\ncentos-minimal amphora images and \u0027bc\u0027 for checking minimum supported\nUbuntu version in diskimage-create.sh\n\nThis patch also adds octavia-lib to the required projects.\n\nChange-Id: Iaa0f763005529cc394d3b95ce07f2be71e9b3778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c30ffdb814148e7fb82817a187a25b2a0676c97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c30ffdb814148e7fb82817a187a25b2a0676c97"}]},"branch":"refs/heads/master"},"6e57e6fd3d046ad55d39535c2493cce997b54380":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-06 15:39:12.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/59/674259/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/674259/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/674259/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/674259/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/674259/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/674259/5"}}},"commit":{"parents":[{"commit":"b32064cdf2d5c70140a84e2497931941c4ee6887","subject":"Fix a unit test for new octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b32064cdf2d5c70140a84e2497931941c4ee6887"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-02 10:39:47.000000000","tz":120},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-06 15:39:12.000000000","tz":0},"subject":"Install missing packages in nodepool instance","message":"Install missing packages in nodepool instance\n\nThe \u0027rpm\u0027, \u0027yumdownloader\u0027 and \u0027bc\u0027 tools do not come installed in the\nUbuntu nodepool image. We need \u0027rpm\u0027 and \u0027yumdownloader\u0027 to build\ncentos-minimal amphora images and \u0027bc\u0027 for checking minimum supported\nUbuntu version in diskimage-create.sh\n\nThis patch also adds octavia-lib to the required projects.\n\nChange-Id: Iaa0f763005529cc394d3b95ce07f2be71e9b3778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e57e6fd3d046ad55d39535c2493cce997b54380"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e57e6fd3d046ad55d39535c2493cce997b54380"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
