)]}'
{"id":"openstack%2Foctavia~674432","triplet_id":"openstack%2Foctavia~master~Ia986a5e1fc29badf1fe87fe51c906050bb1e2d06","project":"openstack/octavia","branch":"master","topic":"octavia_notifications","hashtags":[],"change_id":"Ia986a5e1fc29badf1fe87fe51c906050bb1e2d06","subject":"Implements notifications for octavia","status":"NEW","created":"2019-08-02 22:45:47.000000000","updated":"2019-11-25 10:43:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"be6479452e0a85a7980dc204ef75af123fc09779","_number":674432,"virtual_id_number":674432,"owner":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-09-20 06:40:29.000000000","_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"date":"2019-11-25 10:43:48.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"date":"2019-11-19 05:20:55.000000000","_account_id":31303,"name":"Yu, Yamamoto","email":"yamamoto.yu-00@jp.fujitsu.com","username":"surfinia"},{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":1,"date":"2019-08-07 18:47:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"date":"2019-08-08 18:14:40.000000000","_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31303,"name":"Yu, Yamamoto","email":"yamamoto.yu-00@jp.fujitsu.com","username":"surfinia"},{"value":1,"date":"2019-08-30 22:25:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2019-11-13 22:56:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"value":1,"date":"2019-08-30 21:31:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},{"value":1,"date":"2019-09-02 08:34:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"value":1,"date":"2019-08-30 21:27:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":31303,"name":"Yu, Yamamoto","email":"yamamoto.yu-00@jp.fujitsu.com","username":"surfinia"},{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31303,"name":"Yu, Yamamoto","email":"yamamoto.yu-00@jp.fujitsu.com","username":"surfinia"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},{"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"},{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},{"_account_id":31303,"name":"Yu, Yamamoto","email":"yamamoto.yu-00@jp.fujitsu.com","username":"surfinia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 22:45:58.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-08-02 23:16:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-03 05:50:50.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-08-07 18:47:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-30 21:27:02.000000000","updated_by":{"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"},"reviewer":{"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"},"state":"REVIEWER"},{"updated":"2019-08-30 21:31:14.000000000","updated_by":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"reviewer":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"state":"REVIEWER"},{"updated":"2019-08-30 22:25:13.000000000","updated_by":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"reviewer":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"state":"REVIEWER"},{"updated":"2019-09-02 08:34:34.000000000","updated_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2019-09-20 06:40:29.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-11-13 22:56:35.000000000","updated_by":{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},"reviewer":{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},"state":"REVIEWER"},{"updated":"2019-11-19 05:20:55.000000000","updated_by":{"_account_id":31303,"name":"Yu, Yamamoto","email":"yamamoto.yu-00@jp.fujitsu.com","username":"surfinia"},"reviewer":{"_account_id":31303,"name":"Yu, Yamamoto","email":"yamamoto.yu-00@jp.fujitsu.com","username":"surfinia"},"state":"REVIEWER"},{"updated":"2019-11-25 10:43:48.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"d888e377d48352c294837d166512db7295fb420d","author":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"date":"2019-08-02 22:45:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce955da5be05c4cf5891f79b1c987069eebe0ac1","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-08-02 22:45:58.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"a47a12e224e655802ba06c9a66955f51337f3c81","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-02 23:02:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)\n\nHi, thank you for the patch! This is something multiple people have been interested in.\nI think we need to have a specification for this that can be commented on as there are some important design decisions that should be made before implementing this.\nI did not review the full patch, but made a few initial comments.","accounts_in_message":[],"_revision_number":1},{"id":"d9f46a4e9f2b23088cf1697a463cb7e413ec8a14","author":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"date":"2019-08-02 23:12:05.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1cb1a7f5be6d718a786dd7d8e96e4c2e5ce84a73","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-02 23:12:43.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2083a6b84344212ad623f10342890801356f8a73","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-02 23:16:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f90854a3979ccd5991bca8f0967a50ccb60060c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 01:35:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/32/674432/2/check/openstack-tox-cover/112625f/cover/ : SUCCESS in 6m 33s\n- openstack-tox-lower-constraints https://logs.opendev.org/32/674432/2/check/openstack-tox-lower-constraints/8a0981f/ : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://logs.opendev.org/32/674432/2/check/openstack-tox-pep8/9050c5b/ : FAILURE in 4m 36s\n- openstack-tox-py27 https://logs.opendev.org/32/674432/2/check/openstack-tox-py27/6eb4010/ : SUCCESS in 4m 26s\n- openstack-tox-py36 https://logs.opendev.org/32/674432/2/check/openstack-tox-py36/22835ce/ : FAILURE in 4m 36s\n- openstack-tox-py37 https://logs.opendev.org/32/674432/2/check/openstack-tox-py37/27d8c6a/ : FAILURE in 4m 53s\n- openstack-tox-docs https://logs.opendev.org/32/674432/2/check/openstack-tox-docs/4f63139/html/ : SUCCESS in 5m 06s\n- openstack-tox-functional https://logs.opendev.org/32/674432/2/check/openstack-tox-functional/eaf1e1e/ : SUCCESS in 4m 21s\n- openstack-tox-functional-py36 https://logs.opendev.org/32/674432/2/check/openstack-tox-functional-py36/a604f19/ : SUCCESS in 5m 47s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/32/674432/2/check/octavia-v2-dsvm-noop-api/a1a7034/ : SUCCESS in 39m 19s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/32/674432/2/check/octavia-v2-dsvm-noop-py2-api/56ffb5d/ : SUCCESS in 45m 51s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/32/674432/2/check/octavia-v2-dsvm-scenario/a1e0c56/ : SUCCESS in 1h 28m 30s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/32/674432/2/check/octavia-v2-dsvm-py2-scenario/97e9551/ : FAILURE in 1h 36m 27s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/32/674432/2/check/octavia-v2-dsvm-py2-scenario-centos-7/eebc36b/ : FAILURE in 2h 22m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/32/674432/2/check/octavia-v2-dsvm-scenario-two-node/bd2b183/ : SUCCESS in 1h 06m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/32/674432/2/check/octavia-v2-dsvm-py2-scenario-two-node/742c95f/ : SUCCESS in 1h 05m 30s (non-voting)\n- octavia-grenade https://logs.opendev.org/32/674432/2/check/octavia-grenade/dd21507/ : SUCCESS in 1h 09m 14s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/32/674432/2/check/octavia-v2-dsvm-tls-barbican/5c2262f/ : SUCCESS in 53m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/32/674432/2/check/octavia-v2-dsvm-scenario-amphora-v2/aea2e42/ : SUCCESS in 1h 20m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/32/674432/2/check/octavia-v2-act-stdby-iptables-dsvm-scenario/d1771c6/ : FAILURE in 1h 08m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/32/674432/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/ea56859/ : FAILURE in 1h 22m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/32/674432/2/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/9c96a6c/ : FAILURE in 59m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"028693e874c84125e5f4ebe4f41216b8ddfb8495","author":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"date":"2019-08-05 19:59:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"adcd157c08d6f280306830098d0f21fe0056d791","author":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"date":"2019-08-05 20:28:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2050fdda21a5ad9c7eb0a27e777ed68e0aceb745","author":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"date":"2019-08-05 21:00:07.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"25fd84ee9c7140501fc1d848311c9ee7dddfc42b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 23:05:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/32/674432/4/check/openstack-tox-cover/220d64e/ : FAILURE in 8m 04s\n- openstack-tox-lower-constraints https://logs.opendev.org/32/674432/4/check/openstack-tox-lower-constraints/bc9a0a3/ : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://logs.opendev.org/32/674432/4/check/openstack-tox-pep8/a75f7e4/ : FAILURE in 4m 08s\n- openstack-tox-py27 https://logs.opendev.org/32/674432/4/check/openstack-tox-py27/587c425/ : FAILURE in 7m 25s\n- openstack-tox-py36 https://logs.opendev.org/32/674432/4/check/openstack-tox-py36/d982400/ : FAILURE in 6m 39s\n- openstack-tox-py37 https://logs.opendev.org/32/674432/4/check/openstack-tox-py37/e86cc42/ : FAILURE in 6m 53s\n- openstack-tox-docs https://logs.opendev.org/32/674432/4/check/openstack-tox-docs/1d8c364/html/ : SUCCESS in 5m 18s\n- openstack-tox-functional https://logs.opendev.org/32/674432/4/check/openstack-tox-functional/abd68d6/ : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://logs.opendev.org/32/674432/4/check/openstack-tox-functional-py36/8657550/ : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/32/674432/4/check/octavia-v2-dsvm-noop-api/1b848a9/ : SUCCESS in 44m 57s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/32/674432/4/check/octavia-v2-dsvm-noop-py2-api/0bf7a96/ : SUCCESS in 44m 01s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/32/674432/4/check/octavia-v2-dsvm-scenario/deb8445/ : FAILURE in 1h 34m 59s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/32/674432/4/check/octavia-v2-dsvm-py2-scenario/7ead5a1/ : SUCCESS in 2h 02m 24s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/32/674432/4/check/octavia-v2-dsvm-py2-scenario-centos-7/ddddfba/ : TIMED_OUT in 2h 35m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/32/674432/4/check/octavia-v2-dsvm-scenario-two-node/7bf58d5/ : FAILURE in 1h 34m 47s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/32/674432/4/check/octavia-v2-dsvm-py2-scenario-two-node/9968c8b/ : FAILURE in 1h 21m 16s (non-voting)\n- octavia-grenade https://logs.opendev.org/32/674432/4/check/octavia-grenade/0fe9833/ : SUCCESS in 1h 20m 02s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/32/674432/4/check/octavia-v2-dsvm-tls-barbican/d36e652/ : SUCCESS in 58m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/32/674432/4/check/octavia-v2-dsvm-scenario-amphora-v2/f0f034f/ : SUCCESS in 1h 39m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/32/674432/4/check/octavia-v2-act-stdby-iptables-dsvm-scenario/1013809/ : FAILURE in 1h 16m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/32/674432/4/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/614220d/ : FAILURE in 49m 40s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/32/674432/4/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/beeb8ff/ : FAILURE in 1h 24m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a7273963903fc9fc3c14b8510bcbffbaa8a3531a","author":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"date":"2019-08-06 16:12:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"65c6267c093777d9d5ecf68e169d033f0b27311b","author":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"date":"2019-08-06 18:26:40.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bc0d4576ec2b73bd644e55a1c9521c014efa5d68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 21:15:31.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/32/674432/6/check/openstack-tox-cover/fb580d6/cover/ : SUCCESS in 7m 03s\n- openstack-tox-lower-constraints https://logs.opendev.org/32/674432/6/check/openstack-tox-lower-constraints/2064f35/ : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://logs.opendev.org/32/674432/6/check/openstack-tox-pep8/01df4af/ : SUCCESS in 5m 28s\n- openstack-tox-py27 https://logs.opendev.org/32/674432/6/check/openstack-tox-py27/5ba07d0/ : SUCCESS in 4m 29s\n- openstack-tox-py36 https://logs.opendev.org/32/674432/6/check/openstack-tox-py36/bfddfd2/ : SUCCESS in 4m 59s\n- openstack-tox-py37 https://logs.opendev.org/32/674432/6/check/openstack-tox-py37/36de838/ : SUCCESS in 6m 23s\n- openstack-tox-docs https://logs.opendev.org/32/674432/6/check/openstack-tox-docs/7d41359/html/ : SUCCESS in 6m 41s\n- openstack-tox-functional https://logs.opendev.org/32/674432/6/check/openstack-tox-functional/9c7f353/ : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://logs.opendev.org/32/674432/6/check/openstack-tox-functional-py36/bb3d07e/ : SUCCESS in 6m 10s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/32/674432/6/check/octavia-v2-dsvm-noop-api/5de7748/ : SUCCESS in 47m 24s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/32/674432/6/check/octavia-v2-dsvm-noop-py2-api/775654d/ : SUCCESS in 44m 14s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/32/674432/6/check/octavia-v2-dsvm-scenario/c87f76a/ : SUCCESS in 1h 43m 00s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/32/674432/6/check/octavia-v2-dsvm-py2-scenario/12fb53f/ : FAILURE in 1h 53m 31s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/32/674432/6/check/octavia-v2-dsvm-py2-scenario-centos-7/0f5119f/ : TIMED_OUT in 2h 35m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/32/674432/6/check/octavia-v2-dsvm-scenario-two-node/3ffb760/ : SUCCESS in 2h 05m 51s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/32/674432/6/check/octavia-v2-dsvm-py2-scenario-two-node/a9cab25/ : SUCCESS in 1h 25m 20s (non-voting)\n- octavia-grenade https://logs.opendev.org/32/674432/6/check/octavia-grenade/6fe9204/ : SUCCESS in 1h 17m 08s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/32/674432/6/check/octavia-v2-dsvm-tls-barbican/3f49409/ : SUCCESS in 50m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/32/674432/6/check/octavia-v2-dsvm-scenario-amphora-v2/eb8954d/ : SUCCESS in 1h 31m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/32/674432/6/check/octavia-v2-act-stdby-iptables-dsvm-scenario/27bf2de/ : FAILURE in 1h 27m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/32/674432/6/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/e6a2da4/ : FAILURE in 1h 12m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/32/674432/6/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/26c6af1/ : FAILURE in 1h 17m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b35d40d2747560096eecbb393d1d34ec40ce3a0f","author":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"date":"2019-08-06 22:02:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2f95bc09759df2ada3aa3a44de7eac5e9a28017d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 00:45:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/32/674432/7/check/openstack-tox-cover/85e9768/cover/ : SUCCESS in 7m 09s\n- openstack-tox-lower-constraints https://logs.opendev.org/32/674432/7/check/openstack-tox-lower-constraints/2efeda5/ : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://logs.opendev.org/32/674432/7/check/openstack-tox-pep8/3c48222/ : SUCCESS in 5m 16s\n- openstack-tox-py27 https://logs.opendev.org/32/674432/7/check/openstack-tox-py27/8d197bf/ : SUCCESS in 4m 55s\n- openstack-tox-py36 https://logs.opendev.org/32/674432/7/check/openstack-tox-py36/d5583da/ : SUCCESS in 4m 45s\n- openstack-tox-py37 https://logs.opendev.org/32/674432/7/check/openstack-tox-py37/2cfd1b4/ : SUCCESS in 6m 28s\n- openstack-tox-docs https://logs.opendev.org/32/674432/7/check/openstack-tox-docs/4603530/html/ : SUCCESS in 6m 09s\n- openstack-tox-functional https://logs.opendev.org/32/674432/7/check/openstack-tox-functional/3aa5ab6/ : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://logs.opendev.org/32/674432/7/check/openstack-tox-functional-py36/21207dc/ : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/32/674432/7/check/octavia-v2-dsvm-noop-api/b63ae60/ : SUCCESS in 49m 09s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/32/674432/7/check/octavia-v2-dsvm-noop-py2-api/e5215ca/ : SUCCESS in 56m 32s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/32/674432/7/check/octavia-v2-dsvm-scenario/206fc9c/ : SUCCESS in 1h 37m 41s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/32/674432/7/check/octavia-v2-dsvm-py2-scenario/6be0693/ : FAILURE in 1h 53m 47s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/32/674432/7/check/octavia-v2-dsvm-py2-scenario-centos-7/93a83db/ : TIMED_OUT in 2h 36m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/32/674432/7/check/octavia-v2-dsvm-scenario-two-node/86d73cc/ : SUCCESS in 1h 51m 12s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/32/674432/7/check/octavia-v2-dsvm-py2-scenario-two-node/69f1ea5/ : FAILURE in 1h 18m 39s (non-voting)\n- octavia-grenade https://logs.opendev.org/32/674432/7/check/octavia-grenade/0553490/ : SUCCESS in 1h 18m 20s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/32/674432/7/check/octavia-v2-dsvm-tls-barbican/9ded81e/ : SUCCESS in 59m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/32/674432/7/check/octavia-v2-dsvm-scenario-amphora-v2/0487044/ : FAILURE in 1h 59m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/32/674432/7/check/octavia-v2-act-stdby-iptables-dsvm-scenario/88a2a7c/ : FAILURE in 1h 15m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/32/674432/7/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/f65431c/ : FAILURE in 1h 26m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/32/674432/7/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/c15f9a6/ : FAILURE in 1h 16m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"24fed96af5b321e75733ab72a6e82665557476d6","author":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"date":"2019-08-07 14:39:24.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"38fd2a4da58c7a76a897adab5558b932109a41f9","author":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"date":"2019-08-07 16:11:54.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"21e0ced20d204d15dc2ea885dd23144dacc7d332","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 18:47:33.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/32/674432/9/check/openstack-tox-cover/177aa04/cover/ : SUCCESS in 8m 19s\n- openstack-tox-lower-constraints https://logs.opendev.org/32/674432/9/check/openstack-tox-lower-constraints/0c917f6/ : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://logs.opendev.org/32/674432/9/check/openstack-tox-pep8/6f3911d/ : SUCCESS in 6m 25s\n- openstack-tox-py27 https://logs.opendev.org/32/674432/9/check/openstack-tox-py27/108ca9d/ : SUCCESS in 5m 10s\n- openstack-tox-py36 https://logs.opendev.org/32/674432/9/check/openstack-tox-py36/19e8db7/ : SUCCESS in 4m 45s\n- openstack-tox-py37 https://logs.opendev.org/32/674432/9/check/openstack-tox-py37/52baf03/ : SUCCESS in 5m 19s\n- openstack-tox-docs https://logs.opendev.org/32/674432/9/check/openstack-tox-docs/eaa4a00/html/ : SUCCESS in 6m 25s\n- openstack-tox-functional https://logs.opendev.org/32/674432/9/check/openstack-tox-functional/6351133/ : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://logs.opendev.org/32/674432/9/check/openstack-tox-functional-py36/c93f466/ : SUCCESS in 6m 32s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/32/674432/9/check/octavia-v2-dsvm-noop-api/44cf8dc/ : SUCCESS in 51m 04s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/32/674432/9/check/octavia-v2-dsvm-noop-py2-api/2b11b86/ : SUCCESS in 45m 52s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/32/674432/9/check/octavia-v2-dsvm-scenario/556bcc7/ : SUCCESS in 1h 53m 09s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/32/674432/9/check/octavia-v2-dsvm-py2-scenario/718107a/ : SUCCESS in 1h 44m 44s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/32/674432/9/check/octavia-v2-dsvm-py2-scenario-centos-7/3fc5bd4/ : FAILURE in 2h 28m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/32/674432/9/check/octavia-v2-dsvm-scenario-two-node/17d7169/ : SUCCESS in 2h 06m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/32/674432/9/check/octavia-v2-dsvm-py2-scenario-two-node/5d8ecd0/ : SUCCESS in 1h 21m 03s (non-voting)\n- octavia-grenade https://logs.opendev.org/32/674432/9/check/octavia-grenade/4e6bb3c/ : SUCCESS in 1h 20m 25s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/32/674432/9/check/octavia-v2-dsvm-tls-barbican/ed675a1/ : SUCCESS in 49m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/32/674432/9/check/octavia-v2-dsvm-scenario-amphora-v2/06aee0a/ : SUCCESS in 1h 14m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/32/674432/9/check/octavia-v2-act-stdby-iptables-dsvm-scenario/37743be/ : SUCCESS in 1h 09m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/32/674432/9/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/0cb172c/ : FAILURE in 1h 10m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/32/674432/9/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/81946e8/ : SUCCESS in 1h 34m 21s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"169dbc839d9b96ff535bce091fa0e92485516fba","author":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"date":"2019-08-08 18:14:40.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"766d0335988f73801258c97109bebf5ce7284738","author":{"_account_id":30047,"name":"Bruca Lock","email":"lockshi@gmail.com"},"date":"2019-08-30 21:27:02.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"04a2304c6dba99c0023e96f44193a81bd5f3fdee","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2019-08-30 21:31:14.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"92a551d0a715352125f75429848c42867a906a71","author":{"_account_id":21032,"name":"Farid Da Encarnacao","email":"farid.daencarnacao@ormuco.com","username":"haplo37"},"date":"2019-08-30 22:25:13.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"63920e3db69a6cf62cec1037c59d4c0f5420edb9","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2019-09-02 08:34:34.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"8f74b48785536bf324b409cb8e6a4d62e4d37aec","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-20 06:40:29.000000000","message":"Patch Set 9:\n\nThanks for this work! Sorry we weren\u0027t able to get it reviewed and into Train, but I promise I\u0027ll have a look once we\u0027re open again for features for next cycle.","accounts_in_message":[],"_revision_number":9},{"id":"dd0fbe3cc2a8ed65b4598fb65ba267a95f514e26","author":{"_account_id":28721,"name":"Manuel Rodriguez","email":"manu.rodri.hernan@gmail.com","username":"manu.rodriguez"},"date":"2019-11-13 22:56:35.000000000","message":"Patch Set 9: Code-Review+1\n\nIs now a good time to consider this for review? Thanks!","accounts_in_message":[],"_revision_number":9},{"id":"780ba4ffa9767a1669f2d15765603c4bc72691df","author":{"_account_id":31303,"name":"Yu, Yamamoto","email":"yamamoto.yu-00@jp.fujitsu.com","username":"surfinia"},"date":"2019-11-19 05:20:55.000000000","message":"Patch Set 9:\n\nI need this feature. But some events seem to be not supported, is that okey? (e.g. L7Rule/L7Policy, Amphora failovers)","accounts_in_message":[],"_revision_number":9},{"id":"59c8b55c43671c0b4b874bb81978d03dfadeb641","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-25 10:43:48.000000000","message":"Patch Set 9:\n\nThank you, Yamamoto. That is very good feedback. I think this work should capture the missing events you mentioned, yes.","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"98db8101b75577b89a54e8eba17e6aef65ffda91","revisions":{"337d4ba66ef3c916620e0543d6f48c4c67017e55":{"kind":"REWORK","_number":1,"created":"2019-08-02 22:45:47.000000000","uploader":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"ref":"refs/changes/32/674432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/674432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/674432/1"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-02 22:35:55.000000000","tz":-240},"committer":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-02 22:43:31.000000000","tz":-240},"subject":"Implements notifications for octavia","message":"Implements notifications for octavia\n\nWith notifications, it is possible to know the beginning and end of the creation/update/deletion of load balancers, listeners, pools, etc.\nUsers can enable notifications by setting driver in oslo_messaging_notifications to \"messaging\" in the configuration\nUsers can disable notifications by setting driver in oslo_messaging_notifications to \"noop\"\n\nChange-Id: Ia986a5e1fc29badf1fe87fe51c906050bb1e2d06\nImplements: sending octavia notifications\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/337d4ba66ef3c916620e0543d6f48c4c67017e55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/337d4ba66ef3c916620e0543d6f48c4c67017e55"}]},"branch":"refs/heads/master"},"bfd83b77cef6e531d9a665dc160d34b4c900168e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-02 23:12:05.000000000","uploader":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"ref":"refs/changes/32/674432/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/674432/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/674432/2"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-02 22:35:55.000000000","tz":-240},"committer":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-02 23:09:00.000000000","tz":-240},"subject":"Implements notifications for octavia","message":"Implements notifications for octavia\n\nWith notifications, it is possible to know the beginning and end of the creation/update/deletion of load balancers, listeners, pools, etc.\nUsers can enable notifications by setting driver in oslo_messaging_notifications to \"messaging\" in the configuration\nUsers can disable notifications by setting driver in oslo_messaging_notifications to \"noop\"\n\nChange-Id: Ia986a5e1fc29badf1fe87fe51c906050bb1e2d06\nImplements: blueprint octavia-event-notifier\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfd83b77cef6e531d9a665dc160d34b4c900168e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfd83b77cef6e531d9a665dc160d34b4c900168e"}]},"branch":"refs/heads/master"},"c18a1dd8903b942139f8730e202dad8fef766769":{"kind":"REWORK","_number":3,"created":"2019-08-05 19:59:42.000000000","uploader":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"ref":"refs/changes/32/674432/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/674432/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/674432/3"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-02 22:35:55.000000000","tz":-240},"committer":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-05 19:55:26.000000000","tz":-240},"subject":"Implements notifications for octavia","message":"Implements notifications for octavia\n\nWith notifications, it is possible to know the beginning and end of the creation/update/deletion of load balancers, listeners, pools, etc.\nUsers can enable notifications by setting driver in oslo_messaging_notifications to \"messaging\" in the configuration\nUsers can disable notifications by setting driver in oslo_messaging_notifications to \"noop\"\n\nChange-Id: Ia986a5e1fc29badf1fe87fe51c906050bb1e2d06\nStory: 2004122\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c18a1dd8903b942139f8730e202dad8fef766769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c18a1dd8903b942139f8730e202dad8fef766769"}]},"branch":"refs/heads/master"},"6fb11f6f81e28141c07c53a2ac93689ca5f0ef6a":{"kind":"REWORK","_number":4,"created":"2019-08-05 20:28:11.000000000","uploader":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"ref":"refs/changes/32/674432/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/674432/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/674432/4"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-02 22:35:55.000000000","tz":-240},"committer":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-05 20:27:08.000000000","tz":-240},"subject":"Implements notifications for octavia","message":"Implements notifications for octavia\n\nWith notifications, it is possible to know the beginning and end of the creation/update/deletion of load balancers, listeners, pools, etc.\nUsers can enable notifications by setting driver in oslo_messaging_notifications to \"messaging\" in the configuration\nUsers can disable notifications by setting driver in oslo_messaging_notifications to \"noop\"\n\nChange-Id: Ia986a5e1fc29badf1fe87fe51c906050bb1e2d06\nStory: 2004122\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6fb11f6f81e28141c07c53a2ac93689ca5f0ef6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6fb11f6f81e28141c07c53a2ac93689ca5f0ef6a"}]},"branch":"refs/heads/master"},"ed0242d3ff6854144ec52f5dfd1bf268c37f480e":{"kind":"REWORK","_number":5,"created":"2019-08-06 16:12:00.000000000","uploader":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"ref":"refs/changes/32/674432/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/674432/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/674432/5"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-02 22:35:55.000000000","tz":-240},"committer":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-06 16:03:10.000000000","tz":-240},"subject":"Implements notifications for octavia","message":"Implements notifications for octavia\n\nWith notifications, it is possible to know the beginning and end of the creation/update/deletion of load balancers, listeners, pools, etc.\nUsers can enable notifications by setting driver in oslo_messaging_notifications to \"messaging\" in the configuration\nUsers can disable notifications by setting driver in oslo_messaging_notifications to \"noop\"\n\nChange-Id: Ia986a5e1fc29badf1fe87fe51c906050bb1e2d06\nStory: 2004122\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed0242d3ff6854144ec52f5dfd1bf268c37f480e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed0242d3ff6854144ec52f5dfd1bf268c37f480e"}]},"branch":"refs/heads/master"},"23fdfeec63a31d4cc31ad035112d800b42b8f2ad":{"kind":"REWORK","_number":6,"created":"2019-08-06 18:26:40.000000000","uploader":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"ref":"refs/changes/32/674432/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/674432/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/674432/6"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-02 22:35:55.000000000","tz":-240},"committer":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-06 18:26:29.000000000","tz":-240},"subject":"Implements notifications for octavia","message":"Implements notifications for octavia\n\nWith notifications, it is possible to know the beginning and end of the creation/update/deletion of load balancers, listeners, pools, etc.\nUsers can enable notifications by setting driver in oslo_messaging_notifications to \"messaging\" in the configuration\nUsers can disable notifications by setting driver in oslo_messaging_notifications to \"noop\"\n\nChange-Id: Ia986a5e1fc29badf1fe87fe51c906050bb1e2d06\nStory: 2004122\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23fdfeec63a31d4cc31ad035112d800b42b8f2ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23fdfeec63a31d4cc31ad035112d800b42b8f2ad"}]},"branch":"refs/heads/master"},"f84e4bcf73cc02b838f02f96f7a65dedbdb3ce69":{"kind":"REWORK","_number":7,"created":"2019-08-06 22:02:24.000000000","uploader":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"ref":"refs/changes/32/674432/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/674432/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/674432/7"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-02 22:35:55.000000000","tz":-240},"committer":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-06 22:02:07.000000000","tz":-240},"subject":"Implements notifications for octavia","message":"Implements notifications for octavia\n\nWith notifications, it is possible to know the beginning and end of the creation/update/deletion of load balancers, listeners, pools, etc.\nUsers can enable notifications by setting driver in oslo_messaging_notifications to \"messaging\" in the configuration\nUsers can disable notifications by setting driver in oslo_messaging_notifications to \"noop\"\n\nChange-Id: Ia986a5e1fc29badf1fe87fe51c906050bb1e2d06\nStory: 2004122\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f84e4bcf73cc02b838f02f96f7a65dedbdb3ce69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f84e4bcf73cc02b838f02f96f7a65dedbdb3ce69"}]},"branch":"refs/heads/master"},"06a62ab2ccd357d485c8d6c6c597b97629f9a1af":{"kind":"NO_CHANGE","_number":8,"created":"2019-08-07 14:39:24.000000000","uploader":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"ref":"refs/changes/32/674432/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/674432/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/674432/8"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-02 22:35:55.000000000","tz":-240},"committer":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-07 14:38:50.000000000","tz":-240},"subject":"Implements notifications for octavia","message":"Implements notifications for octavia\n\nWith notifications, it is possible to know the beginning and end of the creation/update/deletion of load balancers, listeners, pools, etc.\nUsers can enable notifications by setting driver in oslo_messaging_notifications to \"messaging\" in the configuration\nUsers can disable notifications by setting driver in oslo_messaging_notifications to \"noop\"\n\nChange-Id: Ia986a5e1fc29badf1fe87fe51c906050bb1e2d06\nStory: 2004122\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/06a62ab2ccd357d485c8d6c6c597b97629f9a1af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/06a62ab2ccd357d485c8d6c6c597b97629f9a1af"}]},"branch":"refs/heads/master"},"98db8101b75577b89a54e8eba17e6aef65ffda91":{"kind":"NO_CHANGE","_number":9,"created":"2019-08-07 16:11:54.000000000","uploader":{"_account_id":30709,"name":"Anqi Li","email":"anqi.li@ormuco.com","username":"anqi.li426"},"ref":"refs/changes/32/674432/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/674432/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/674432/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/674432/9"}}},"commit":{"parents":[{"commit":"2111d9e264fabd0e28c1287a849bfc677ab97c79","subject":"Merge \"Add support for monitor_{address,port} in UDP members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2111d9e264fabd0e28c1287a849bfc677ab97c79"}]}],"author":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-02 22:35:55.000000000","tz":-240},"committer":{"name":"Anqi Li","email":"anqi.li@ormuco.com","date":"2019-08-07 16:11:29.000000000","tz":-240},"subject":"Implements notifications for octavia","message":"Implements notifications for octavia\n\nWith notifications, it is possible to know the beginning and end of the creation/update/deletion of load balancers, listeners, pools, etc.\nUsers can enable notifications by setting driver in oslo_messaging_notifications to \"messaging\" in the configuration\nUsers can disable notifications by setting driver in oslo_messaging_notifications to \"noop\"\n\nChange-Id: Ia986a5e1fc29badf1fe87fe51c906050bb1e2d06\nStory: 2004122\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98db8101b75577b89a54e8eba17e6aef65ffda91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98db8101b75577b89a54e8eba17e6aef65ffda91"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
