)]}'
{"id":"openstack%2Foctavia~674467","triplet_id":"openstack%2Foctavia~stable%2Fstein~I276188530a83598ed75560f02ed9d80ce9afca2f","project":"openstack/octavia","branch":"stable/stein","topic":"http_proxy_to_wsgi","hashtags":[],"change_id":"I276188530a83598ed75560f02ed9d80ce9afca2f","subject":"Add support for oslo_middleware http_proxy_to_wsgi","status":"MERGED","created":"2019-08-03 18:00:57.000000000","updated":"2019-09-17 20:56:59.000000000","submitted":"2019-09-17 20:55:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674467-1568753704601-b16e9fe7","meta_rev_id":"5f8ab1a53fbb848d03a78252b0e0eb8e87fa4607","_number":674467,"virtual_id_number":674467,"owner":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-17 20:55:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-03 21:07:08.000000000","_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-16 07:12:23.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-09-17 16:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-09-17 16:03:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2019-08-03 18:08:44.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-03 18:08:44.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-16 07:12:23.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-17 16:03:00.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-17 20:55:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b70c070b8334f9d5e9a062185b24626ad84bbea7","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-08-03 18:00:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1eadb394ebc4836e4cd555b3f7ac98d0d3c865a8","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-08-03 18:07:15.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"47e8d6cf0212adaf97a4fce363ff79e3dd9bbc33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 20:41:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/67/674467/2/check/openstack-tox-cover/0808e5c/cover/ : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints https://logs.opendev.org/67/674467/2/check/openstack-tox-lower-constraints/fc2859c/ : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://logs.opendev.org/67/674467/2/check/openstack-tox-pep8/af239e1/ : SUCCESS in 5m 04s\n- openstack-tox-py27 https://logs.opendev.org/67/674467/2/check/openstack-tox-py27/f8bf225/ : SUCCESS in 3m 40s\n- openstack-tox-py35 https://logs.opendev.org/67/674467/2/check/openstack-tox-py35/cd920ea/ : SUCCESS in 4m 09s\n- openstack-tox-py36 https://logs.opendev.org/67/674467/2/check/openstack-tox-py36/185200d/ : SUCCESS in 3m 52s\n- openstack-tox-docs https://logs.opendev.org/67/674467/2/check/openstack-tox-docs/910c906/html/ : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://logs.opendev.org/67/674467/2/check/build-openstack-releasenotes/51dd77b/html/ : SUCCESS in 4m 10s\n- neutron-lbaasv2-dsvm-api https://logs.opendev.org/67/674467/2/check/neutron-lbaasv2-dsvm-api/0b5b4ee/ : SUCCESS in 1h 12m 35s\n- neutron-lbaasv2-dsvm-py3x-api https://logs.opendev.org/67/674467/2/check/neutron-lbaasv2-dsvm-py3x-api/cee1488/ : SUCCESS in 1h 15m 38s\n- neutron-lbaasv2-dsvm-scenario https://logs.opendev.org/67/674467/2/check/neutron-lbaasv2-dsvm-scenario/1c40cda/ : SUCCESS in 1h 09m 03s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://logs.opendev.org/67/674467/2/check/neutron-lbaasv2-dsvm-py3x-scenario/cd56b68/ : FAILURE in 1h 13m 54s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/67/674467/2/check/openstack-tox-functional/a2252b2/ : SUCCESS in 6m 05s\n- openstack-tox-functional-py35 https://logs.opendev.org/67/674467/2/check/openstack-tox-functional-py35/ea82da9/ : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 https://logs.opendev.org/67/674467/2/check/openstack-tox-functional-py36/8a4f72e/ : SUCCESS in 5m 14s\n- octavia-v1-dsvm-scenario https://logs.opendev.org/67/674467/2/check/octavia-v1-dsvm-scenario/eefdba8/ : FAILURE in 55m 42s\n- octavia-v1-dsvm-py3x-scenario https://logs.opendev.org/67/674467/2/check/octavia-v1-dsvm-py3x-scenario/eb4925d/ : FAILURE in 53m 39s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-noop-api/451c8db/ : SUCCESS in 41m 28s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-noop-py2-api/116f873/ : SUCCESS in 36m 03s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-scenario/5c4f583/ : SUCCESS in 1h 26m 28s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-py2-scenario/2001aff/ : SUCCESS in 1h 51m 48s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-scenario-ubuntu-xenial/90978d9/ : SUCCESS in 1h 17m 54s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-py2-scenario-centos-7/e16ab5f/ : SUCCESS in 2h 14m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/c6d8093/ : FAILURE in 31m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-scenario-two-node/b3ae984/ : SUCCESS in 2h 08m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-py2-scenario-two-node/c0ae8d4/ : FAILURE in 1h 36m 39s (non-voting)\n- octavia-grenade https://logs.opendev.org/67/674467/2/check/octavia-grenade/0891a82/ : SUCCESS in 1h 11m 32s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-tls-barbican/50b24c5/ : SUCCESS in 47m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b7df8d587ecee0a893dcb0ef87d363ad9762de83","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-08-03 21:07:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"763492b2227477cf46eef4c4b44cc1492a483358","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 23:11:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/67/674467/2/check/openstack-tox-cover/af55790/cover/ : SUCCESS in 6m 40s\n- openstack-tox-lower-constraints https://logs.opendev.org/67/674467/2/check/openstack-tox-lower-constraints/3421c1a/ : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://logs.opendev.org/67/674467/2/check/openstack-tox-pep8/edf53a7/ : SUCCESS in 4m 24s\n- openstack-tox-py27 https://logs.opendev.org/67/674467/2/check/openstack-tox-py27/2ac3d95/ : SUCCESS in 3m 23s\n- openstack-tox-py35 https://logs.opendev.org/67/674467/2/check/openstack-tox-py35/e7b86e0/ : SUCCESS in 3m 33s\n- openstack-tox-py36 https://logs.opendev.org/67/674467/2/check/openstack-tox-py36/5f3e122/ : SUCCESS in 3m 25s\n- openstack-tox-docs https://logs.opendev.org/67/674467/2/check/openstack-tox-docs/28dd353/html/ : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://logs.opendev.org/67/674467/2/check/build-openstack-releasenotes/7edb332/html/ : SUCCESS in 3m 48s\n- neutron-lbaasv2-dsvm-api https://logs.opendev.org/67/674467/2/check/neutron-lbaasv2-dsvm-api/d242220/ : SUCCESS in 1h 18m 08s\n- neutron-lbaasv2-dsvm-py3x-api https://logs.opendev.org/67/674467/2/check/neutron-lbaasv2-dsvm-py3x-api/7a78aa7/ : SUCCESS in 1h 16m 37s\n- neutron-lbaasv2-dsvm-scenario https://logs.opendev.org/67/674467/2/check/neutron-lbaasv2-dsvm-scenario/12e3d0f/ : SUCCESS in 1h 23m 10s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://logs.opendev.org/67/674467/2/check/neutron-lbaasv2-dsvm-py3x-scenario/2e21027/ : SUCCESS in 1h 21m 27s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/67/674467/2/check/openstack-tox-functional/640c13b/ : SUCCESS in 5m 13s\n- openstack-tox-functional-py35 https://logs.opendev.org/67/674467/2/check/openstack-tox-functional-py35/e59f869/ : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://logs.opendev.org/67/674467/2/check/openstack-tox-functional-py36/94b0300/ : SUCCESS in 5m 33s\n- octavia-v1-dsvm-scenario https://logs.opendev.org/67/674467/2/check/octavia-v1-dsvm-scenario/f4aa5b7/ : SUCCESS in 1h 00m 27s\n- octavia-v1-dsvm-py3x-scenario https://logs.opendev.org/67/674467/2/check/octavia-v1-dsvm-py3x-scenario/5b0f4cb/ : SUCCESS in 1h 05m 54s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-noop-api/2f6cb62/ : SUCCESS in 45m 40s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-noop-py2-api/1142667/ : SUCCESS in 38m 22s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-scenario/4bca3a4/ : SUCCESS in 1h 31m 42s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-py2-scenario/ac61722/ : SUCCESS in 1h 39m 03s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-scenario-ubuntu-xenial/b0c79e0/ : SUCCESS in 1h 18m 39s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-py2-scenario-centos-7/8853216/ : SUCCESS in 2h 01m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-scenario-ubuntu-bionic/73ab442/ : FAILURE in 31m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-scenario-two-node/c564c2c/ : SUCCESS in 1h 15m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-py2-scenario-two-node/b73dca1/ : SUCCESS in 1h 45m 34s (non-voting)\n- octavia-grenade https://logs.opendev.org/67/674467/2/check/octavia-grenade/305d49e/ : SUCCESS in 1h 16m 09s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/67/674467/2/check/octavia-v2-dsvm-tls-barbican/c8b6889/ : SUCCESS in 58m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"92372cf423fa19d45ad2363340b54d458edec0bc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-16 07:12:23.000000000","message":"Patch Set 2: Code-Review+2\n\nWhile the patch adds a new configuration option, it uses a good default value. It also fixes an issue described in the associated story.","accounts_in_message":[],"_revision_number":2},{"id":"89e1ca8da7e2cb42bc9ab34cef498cc4835cb3dc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 16:03:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\noslo middleware is already in the requirements and the config file change has a sane default, so I think this is ok to backport as a fix.","accounts_in_message":[],"_revision_number":2},{"id":"d49759a5ab24e3836316bdd8c7a22819c1134a34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 16:03:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8114131a009ed5863860d74c70acfe51252c40b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 20:55:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0ced27977f7d02993c494ee60be5acaded8013c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 20:55:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/955b55cdef5c45d0ac8f5a59ab705b74 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1074c11319cb4fdeb4a8441f9162c2f6 : SUCCESS in 6m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b423b9fdead74fee8907afb8e591af38 : SUCCESS in 6m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/04a5c178f31d486cabbccda4e087bca1 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb66feb923c04cfa88ba795f687c7d76 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14919df514a6456c9f60f5165e476a4d : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f79a7819ae8c471ebcd612bdcb1b7f8f : SUCCESS in 9m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/91978e64065448c9aa458ac2a67c28c7 : SUCCESS in 8m 33s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c48a4f2aecf047f798c6abd11aaf7325 : SUCCESS in 9m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/635a10f87a9e4b2e824f35dcad0b4d17 : SUCCESS in 9m 07s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b59fccd1a4cc40ff81fac4bf936a0308 : SUCCESS in 1h 27m 51s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ad1a71ed955a4b62b571574e8e302728 : SUCCESS in 1h 06m 18s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c0e37873284e4629b4cb6b1625225845 : SUCCESS in 1h 27m 27s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0151c2fb4d0145088893f537b3fdf1fe : SUCCESS in 1h 14m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/43513d74fc51434c8f2f43bca19ac587 : SUCCESS in 50m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a3db1299c2bc40c89b1240e0413ac7ee : SUCCESS in 43m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1c431ecbf7fd4128a9f1194159fa2dac : SUCCESS in 1h 37m 53s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6d582f51bb9a4a37bc2be2a2fe39cde9 : SUCCESS in 1h 25m 39s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/abbeb1d724954ec3b13a8ac06a1a9cae : SUCCESS in 2h 03m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e78b59edcd554daf99ae8bdbb21ffbea : SUCCESS in 1h 25m 14s","accounts_in_message":[],"_revision_number":2},{"id":"7fbb969ed7c1352e351fb8d8ff6abad873d8f6b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 20:56:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8efd1d9972b64fbb97aff0ef77212ba0 : SUCCESS in 1m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f6b7dd46b7946c5ba9d4dbef4c28264 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6f8932e955863df5c6d1a717822664a0a8b79f97","revisions":{"c4d361fefdf897f38402b7ca9971b46390098412":{"kind":"REWORK","_number":1,"created":"2019-08-03 18:00:57.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/67/674467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/674467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/674467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/674467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/674467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/674467/1"}}},"commit":{"parents":[{"commit":"15358a71e4aeffee8a3283ed08137e3b6daab52e","subject":"Merge \"Update tox.ini for new upper constraints strategy\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15358a71e4aeffee8a3283ed08137e3b6daab52e"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-08-03 18:00:21.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-08-03 18:00:21.000000000","tz":180},"subject":"Add support for oslo_middleware http_proxy_to_wsgi","message":"Add support for oslo_middleware http_proxy_to_wsgi\n\nThis sets up the HTTPProxyToWSGI middleware in front of Octavia API. The\npurpose of this middleware is to set up the request URL correctly in\nthe case there is a proxy (For instance, a loadbalancer such as HAProxy)\nin front of Octavia API.\n\nSo, when TLS connections are terminated at the proxy, and one tries to\nget the versions from the \u0027/\u0027 resource from Octavia API, one will notice\nthat the protocol is incorrect; It will show \u0027http\u0027 instead of \u0027https\u0027.\nSo this middleware handles such cases.\n\nThe HTTPProxyToWSGI is off by default and needs to be enabled via a\nconfiguration value.\n\nIt can be enabled with the option in octavia.conf:\n[oslo_middleware]\nenable_proxy_headers_parsing\u003dTrue\n\n#Conflicts:\n#\tetc/octavia.conf\nNOTE(s10): conflict is due to I3082962841d3b645f3cbd1a6b41fc7fb28dcf7e6\nnot being in stable branches\n\nStory: 2005105\nTask: 29732\nChange-Id: I276188530a83598ed75560f02ed9d80ce9afca2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4d361fefdf897f38402b7ca9971b46390098412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4d361fefdf897f38402b7ca9971b46390098412"}]},"branch":"refs/heads/stable/stein"},"6f8932e955863df5c6d1a717822664a0a8b79f97":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-03 18:07:15.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/67/674467/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/674467/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/674467/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/674467/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/674467/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/674467/2"}}},"commit":{"parents":[{"commit":"15358a71e4aeffee8a3283ed08137e3b6daab52e","subject":"Merge \"Update tox.ini for new upper constraints strategy\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15358a71e4aeffee8a3283ed08137e3b6daab52e"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-08-03 18:00:21.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-08-03 18:06:54.000000000","tz":180},"subject":"Add support for oslo_middleware http_proxy_to_wsgi","message":"Add support for oslo_middleware http_proxy_to_wsgi\n\nThis sets up the HTTPProxyToWSGI middleware in front of Octavia API. The\npurpose of this middleware is to set up the request URL correctly in\nthe case there is a proxy (For instance, a loadbalancer such as HAProxy)\nin front of Octavia API.\n\nSo, when TLS connections are terminated at the proxy, and one tries to\nget the versions from the \u0027/\u0027 resource from Octavia API, one will notice\nthat the protocol is incorrect; It will show \u0027http\u0027 instead of \u0027https\u0027.\nSo this middleware handles such cases.\n\nThe HTTPProxyToWSGI is off by default and needs to be enabled via a\nconfiguration value.\n\nIt can be enabled with the option in octavia.conf:\n[oslo_middleware]\nenable_proxy_headers_parsing\u003dTrue\n\nConflicts:\n\tetc/octavia.conf\nNOTE(s10): conflict is due to I3082962841d3b645f3cbd1a6b41fc7fb28dcf7e6\nnot being in stable branches\n\nStory: 2005105\nTask: 29732\nChange-Id: I276188530a83598ed75560f02ed9d80ce9afca2f\n(cherry picked from commit ec83c69372d647aaeec02fea6eea29c7ee11115a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f8932e955863df5c6d1a717822664a0a8b79f97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f8932e955863df5c6d1a717822664a0a8b79f97"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
