)]}'
{"id":"openstack%2Foctavia~674911","triplet_id":"openstack%2Foctavia~master~Iaa9850d3fb8326bbf6f34fd25bc21cbd817bd73d","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Iaa9850d3fb8326bbf6f34fd25bc21cbd817bd73d","subject":"Make centos jobs experimental for now","status":"ABANDONED","created":"2019-08-06 18:47:46.000000000","updated":"2019-08-09 18:50:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"72eb0d42e292aac5f57b723a3c30eee09ab514b9","_number":674911,"virtual_id_number":674911,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"all":[{"date":"2019-08-09 18:48:01.000000000","_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"disliked":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-1,"date":"2019-08-09 18:47:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-09 18:47:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 18:49:23.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-06 20:40:27.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-08-06 20:51:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7de1a0b1f4a84e71d82683ffa240fdf2dbb930cc","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-06 18:47:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47ea2f33e24b301723095166fb09d3a71999964f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-06 18:49:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3be3724ddf543a90bbf0ead1e902e52a5491738b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-06 18:50:12.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"c5527412fdab9475a77b0350a52ba3162da9f042","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-06 20:40:27.000000000","message":"Patch Set 1: Code-Review-1\n\nThe timeout was bumped for the centos7 job just yesterday (https://review.opendev.org/#/c/673172/) and patch https://review.opendev.org/#/c/674087/ fixes two issues introduced by the move to centos-minimal DIB element.","accounts_in_message":[],"_revision_number":1},{"id":"c843de80f492d245a84e1da4233ea7bb1a364ab3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 20:51:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/11/674911/1/check/openstack-tox-cover/b67f03f/cover/ : SUCCESS in 7m 30s\n- openstack-tox-lower-constraints https://logs.opendev.org/11/674911/1/check/openstack-tox-lower-constraints/317d102/ : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://logs.opendev.org/11/674911/1/check/openstack-tox-pep8/eb1eb86/ : SUCCESS in 6m 00s\n- openstack-tox-py27 https://logs.opendev.org/11/674911/1/check/openstack-tox-py27/1cf0047/ : SUCCESS in 5m 27s\n- openstack-tox-py36 https://logs.opendev.org/11/674911/1/check/openstack-tox-py36/d19142a/ : SUCCESS in 4m 18s\n- openstack-tox-py37 https://logs.opendev.org/11/674911/1/check/openstack-tox-py37/f0a866d/ : SUCCESS in 4m 49s\n- openstack-tox-docs https://logs.opendev.org/11/674911/1/check/openstack-tox-docs/18d4a14/html/ : SUCCESS in 5m 44s\n- openstack-tox-functional https://logs.opendev.org/11/674911/1/check/openstack-tox-functional/e1674e9/ : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://logs.opendev.org/11/674911/1/check/openstack-tox-functional-py36/c4cec21/ : SUCCESS in 6m 13s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/11/674911/1/check/octavia-v2-dsvm-noop-api/ecba82a/ : SUCCESS in 53m 23s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/11/674911/1/check/octavia-v2-dsvm-noop-py2-api/f03f839/ : SUCCESS in 45m 00s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/11/674911/1/check/octavia-v2-dsvm-scenario/a30e092/ : SUCCESS in 1h 53m 01s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/11/674911/1/check/octavia-v2-dsvm-py2-scenario/6b4eb49/ : SUCCESS in 1h 41m 33s\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/11/674911/1/check/octavia-v2-dsvm-scenario-two-node/44b90a0/ : FAILURE in 1h 31m 35s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/11/674911/1/check/octavia-v2-dsvm-py2-scenario-two-node/4ffe5d9/ : FAILURE in 28m 43s (non-voting)\n- octavia-grenade https://logs.opendev.org/11/674911/1/check/octavia-grenade/3dd864f/ : SUCCESS in 1h 23m 20s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/11/674911/1/check/octavia-v2-dsvm-tls-barbican/6cfb31e/ : SUCCESS in 53m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://logs.opendev.org/11/674911/1/check/octavia-v2-dsvm-scenario-amphora-v2/f6dc876/ : SUCCESS in 1h 47m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://logs.opendev.org/11/674911/1/check/octavia-v2-act-stdby-iptables-dsvm-scenario/5233f64/ : FAILURE in 1h 17m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://logs.opendev.org/11/674911/1/check/octavia-v2-act-stdby-iptables-dsvm-py2-scenario/97909ee/ : FAILURE in 1h 01m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e886eeded111a2c8a41a1b5784cbc4631994252","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 21:34:51.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/11/674911/1/experimental/octavia-v2-dsvm-py2-scenario-centos-7/8ff7135/ : TIMED_OUT in 2h 35m 59s\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://logs.opendev.org/11/674911/1/experimental/octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7/3f316ce/ : FAILURE in 1h 17m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f1369ed32040c1d5c47251f3f7899712e99851c1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-09 18:47:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e98cbcd944c52901a05e5cce5906d43f7e692bfa","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-09 18:47:53.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"53c15063ca6e31e0e0b71da793319070dfa8eb45","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-09 18:48:01.000000000","message":"Patch Set 2:\n\nLet\u0027s see. :)","accounts_in_message":[],"_revision_number":2},{"id":"72eb0d42e292aac5f57b723a3c30eee09ab514b9","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-09 18:50:29.000000000","message":"Abandoned\n\nAlright, does look like they\u0027re passing now, from looking at some other patches that have run since. Can always restore this later if it still looks like they\u0027re a problem.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a6ee796f6cd693286f309e1e48ae2b12ca00bfbf","revisions":{"7b2e2eb6408d786cc4f2dfff07fe0cc1778f0dae":{"kind":"REWORK","_number":1,"created":"2019-08-06 18:47:46.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/674911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/674911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/674911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/674911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/674911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/674911/1"}}},"commit":{"parents":[{"commit":"b32064cdf2d5c70140a84e2497931941c4ee6887","subject":"Fix a unit test for new octavia-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b32064cdf2d5c70140a84e2497931941c4ee6887"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-06 18:46:32.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-06 18:46:32.000000000","tz":-420},"subject":"Make centos jobs experimental for now","message":"Make centos jobs experimental for now\n\nThey cannot pass, waste resources, and one of them consistently times\nout, meaning that no run can finish in less than the 2.5h max time.\n\nChange-Id: Iaa9850d3fb8326bbf6f34fd25bc21cbd817bd73d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b2e2eb6408d786cc4f2dfff07fe0cc1778f0dae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b2e2eb6408d786cc4f2dfff07fe0cc1778f0dae"}]},"branch":"refs/heads/master"},"a6ee796f6cd693286f309e1e48ae2b12ca00bfbf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-09 18:47:43.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/11/674911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/674911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/674911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/674911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/674911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/674911/2"}}},"commit":{"parents":[{"commit":"6e4da8506469de634a4ef57538c7f3ffed3cc937","subject":"Merge \"Re-enable the py36 tox environment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e4da8506469de634a4ef57538c7f3ffed3cc937"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-06 18:46:32.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-08-09 18:47:43.000000000","tz":0},"subject":"Make centos jobs experimental for now","message":"Make centos jobs experimental for now\n\nThey cannot pass, waste resources, and one of them consistently times\nout, meaning that no run can finish in less than the 2.5h max time.\n\nChange-Id: Iaa9850d3fb8326bbf6f34fd25bc21cbd817bd73d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a6ee796f6cd693286f309e1e48ae2b12ca00bfbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a6ee796f6cd693286f309e1e48ae2b12ca00bfbf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
