)]}'
{"id":"openstack%2Foctavia~675063","triplet_id":"openstack%2Foctavia~stable%2Fstein~I426255f587f36b415eb999a9eb28cf0f91de94b0","project":"openstack/octavia","branch":"stable/stein","topic":"single-process","hashtags":[],"change_id":"I426255f587f36b415eb999a9eb28cf0f91de94b0","subject":"Fix listener deletion in ACTIVE/STANDBY topology","status":"MERGED","created":"2019-08-07 08:59:38.000000000","updated":"2019-08-30 20:36:30.000000000","submitted":"2019-08-30 20:34:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675063-1567197288872-6ccbb2a0","meta_rev_id":"a7e2e090b86b48efd4183b066986e62ff61a4bda","_number":675063,"virtual_id_number":675063,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-08-30 20:34:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-08-09 12:42:01.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-08-07 13:40:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2019-08-09 12:42:01.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 13:40:08.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-09 12:42:01.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-08-30 20:34:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46f80938dcde2861d617ff3eca9d199bc197256a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-07 08:59:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e86544cf4c33b477301432991dec0709ea093999","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 11:34:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/63/675063/1/check/openstack-tox-cover/b085ede/cover/ : SUCCESS in 9m 36s\n- openstack-tox-lower-constraints https://logs.opendev.org/63/675063/1/check/openstack-tox-lower-constraints/53ed5eb/ : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://logs.opendev.org/63/675063/1/check/openstack-tox-pep8/aa9a4b2/ : SUCCESS in 8m 09s\n- openstack-tox-py27 https://logs.opendev.org/63/675063/1/check/openstack-tox-py27/c420d9d/ : SUCCESS in 3m 42s\n- openstack-tox-py35 https://logs.opendev.org/63/675063/1/check/openstack-tox-py35/abf684a/ : SUCCESS in 8m 29s\n- openstack-tox-py36 https://logs.opendev.org/63/675063/1/check/openstack-tox-py36/deb0935/ : SUCCESS in 5m 54s\n- openstack-tox-docs https://logs.opendev.org/63/675063/1/check/openstack-tox-docs/9188859/html/ : SUCCESS in 5m 06s\n- neutron-lbaasv2-dsvm-api https://logs.opendev.org/63/675063/1/check/neutron-lbaasv2-dsvm-api/a78dc2e/ : SUCCESS in 1h 18m 57s\n- neutron-lbaasv2-dsvm-py3x-api https://logs.opendev.org/63/675063/1/check/neutron-lbaasv2-dsvm-py3x-api/a39f24b/ : SUCCESS in 1h 24m 28s\n- neutron-lbaasv2-dsvm-scenario https://logs.opendev.org/63/675063/1/check/neutron-lbaasv2-dsvm-scenario/2acc44d/ : SUCCESS in 1h 20m 55s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://logs.opendev.org/63/675063/1/check/neutron-lbaasv2-dsvm-py3x-scenario/bd4591f/ : FAILURE in 1h 19m 27s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/63/675063/1/check/openstack-tox-functional/0674a23/ : SUCCESS in 6m 50s\n- openstack-tox-functional-py35 https://logs.opendev.org/63/675063/1/check/openstack-tox-functional-py35/0dd9dbe/ : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://logs.opendev.org/63/675063/1/check/openstack-tox-functional-py36/5aa6f49/ : SUCCESS in 5m 38s\n- octavia-v1-dsvm-scenario https://logs.opendev.org/63/675063/1/check/octavia-v1-dsvm-scenario/808cf20/ : SUCCESS in 1h 13m 53s\n- octavia-v1-dsvm-py3x-scenario https://logs.opendev.org/63/675063/1/check/octavia-v1-dsvm-py3x-scenario/1911d4d/ : SUCCESS in 1h 13m 00s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/63/675063/1/check/octavia-v2-dsvm-noop-api/a6c0ddd/ : SUCCESS in 40m 29s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/63/675063/1/check/octavia-v2-dsvm-noop-py2-api/7ee2140/ : SUCCESS in 45m 06s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/63/675063/1/check/octavia-v2-dsvm-scenario/b660793/ : SUCCESS in 1h 36m 29s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/63/675063/1/check/octavia-v2-dsvm-py2-scenario/60e34b0/ : SUCCESS in 1h 23m 27s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://logs.opendev.org/63/675063/1/check/octavia-v2-dsvm-scenario-ubuntu-xenial/3c1cda4/ : SUCCESS in 1h 35m 39s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/63/675063/1/check/octavia-v2-dsvm-py2-scenario-centos-7/2f29671/ : SUCCESS in 2h 23m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://logs.opendev.org/63/675063/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/3bbd627/ : FAILURE in 41m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/63/675063/1/check/octavia-v2-dsvm-scenario-two-node/64c0da8/ : SUCCESS in 1h 24m 59s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/63/675063/1/check/octavia-v2-dsvm-py2-scenario-two-node/c84fee9/ : FAILURE in 1h 49m 32s (non-voting)\n- octavia-grenade https://logs.opendev.org/63/675063/1/check/octavia-grenade/70fa76b/ : SUCCESS in 1h 00m 44s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/63/675063/1/check/octavia-v2-dsvm-tls-barbican/f950ee0/ : SUCCESS in 1h 07m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"080f913ed5217125a3f94678f49f5fcacb726584","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-07 13:40:08.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"0e43a4e10f8e26147c13df62ddb323620a83e80b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-09 12:42:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ce29eccf46d818b8a3542cdf7ea72be00cde240","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 18:28:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e473b364ed4737aa7226b3b1fa973591ac844af4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 20:34:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"557e98cdacb03fc0be7afafd6ff91404bea803f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 20:34:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4e3fc0643fd45a09c95263114bc65a5 : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3df932b78c464ae18c17952eddc25162 : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/839669f29a664d9ca2d1e763d56f8021 : SUCCESS in 6m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/315032afceab4670a1342da886b8730c : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2e0e5e38db14d7fa4bfb810c44d5ec0 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90db364222d74a508b1b042de17938f3 : SUCCESS in 6m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7eaae4bb1b804b41a4fdf59b1b26185d : SUCCESS in 6m 08s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/489205c5bdb3470381131d38c2d8d893 : SUCCESS in 7m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9cf04afd69e44a12942561b7473caa08 : SUCCESS in 6m 16s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75b5697ddd3940969101b101fd8c8523 : SUCCESS in 1h 10m 28s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/daf5d608e31541068fba388311542e56 : SUCCESS in 1h 04m 14s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ad6b4543079c467c873a5897c12371b1 : SUCCESS in 1h 25m 42s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/55b5439910b845c998602544d0108cb3 : SUCCESS in 1h 30m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/31697e32e9584c248358bb2b06590948 : SUCCESS in 45m 43s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d00f4c0a71db4e108212e704a390e5fd : SUCCESS in 48m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67d8be4bbcd34873a023d9c861cc3eb2 : SUCCESS in 1h 52m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a535a6ada3b54d17893ca3312bb9311e : SUCCESS in 1h 28m 01s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/bae59e8bce9d42ae887dfc59926037d4 : SUCCESS in 1h 56m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9d9d09a1613347649dfd0ce1d4b0b217 : SUCCESS in 1h 16m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ecff472d915f60fad6f31b6bda7fce388ee95250","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 20:36:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6678c0ecaff44532bae3135738b381d9 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0efee7efa86118560fb80144675cc98332c28b77","revisions":{"0efee7efa86118560fb80144675cc98332c28b77":{"kind":"REWORK","_number":1,"created":"2019-08-07 08:59:38.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/63/675063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/675063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/675063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/675063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/675063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/675063/1"}}},"commit":{"parents":[{"commit":"646071d85298a3f54630afb986917b5d0c05cdd8","subject":"Fix multi-listener load balancers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/646071d85298a3f54630afb986917b5d0c05cdd8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-08-02 07:12:38.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-08-07 08:58:12.000000000","tz":120},"subject":"Fix listener deletion in ACTIVE/STANDBY topology","message":"Fix listener deletion in ACTIVE/STANDBY topology\n\nWhen removing listeners, listeners are removed from the load balancer\u0027s\nlistener list just before reconfiguring each amphora.\nIn case of ACTIVE/STANDBY topology, the code is performed on both\namphorae, so the listener is removed twice from the list.\nThis commit ensures that we don\u0027t remove an already removed listener.\n\nStory: 2006329\nTask: 36065\n\nChange-Id: I426255f587f36b415eb999a9eb28cf0f91de94b0\n(cherry picked from commit b1a4758f583badf996be2fc4c405efb4cb533e5a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0efee7efa86118560fb80144675cc98332c28b77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0efee7efa86118560fb80144675cc98332c28b77"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
