)]}'
{"id":"openstack%2Foctavia~675362","triplet_id":"openstack%2Foctavia~stable%2Fstein~I75cde3ff820f085aebbdffd1e40c5ff40f16835d","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I75cde3ff820f085aebbdffd1e40c5ff40f16835d","subject":"Fixed pool and members status with UDP loadbalancers","status":"MERGED","created":"2019-08-08 14:45:00.000000000","updated":"2019-09-17 09:58:50.000000000","submitted":"2019-09-17 09:57:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675362-1568714242375-0a491f22","meta_rev_id":"6fd7e0c9faee6dbb05e14e270a59d84d1ec662e0","_number":675362,"virtual_id_number":675362,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-17 05:34:43.000000000","_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2019-09-12 16:13:10.000000000","_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-17 09:57:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-12 08:43:26.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2019-09-14 19:22:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2019-09-14 19:22:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 15:33:22.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-14 19:22:30.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-17 05:34:43.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-17 09:57:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5851fa4a5dd3498b26abc4fb89dc69d92ec0493b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-08 14:45:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cf0b28a4773713beeb88f9ac7a4846af6c1b6e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 16:50:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/62/675362/1/check/openstack-tox-cover/07acf48/cover/ : SUCCESS in 8m 46s\n- openstack-tox-lower-constraints https://logs.opendev.org/62/675362/1/check/openstack-tox-lower-constraints/84b3600/ : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://logs.opendev.org/62/675362/1/check/openstack-tox-pep8/205cbe4/ : SUCCESS in 8m 05s\n- openstack-tox-py27 https://logs.opendev.org/62/675362/1/check/openstack-tox-py27/0cb746b/ : SUCCESS in 6m 56s\n- openstack-tox-py35 https://logs.opendev.org/62/675362/1/check/openstack-tox-py35/c101137/ : SUCCESS in 6m 09s\n- openstack-tox-py36 https://logs.opendev.org/62/675362/1/check/openstack-tox-py36/5ef49b8/ : SUCCESS in 5m 40s\n- openstack-tox-docs https://logs.opendev.org/62/675362/1/check/openstack-tox-docs/5d8b70e/html/ : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://logs.opendev.org/62/675362/1/check/build-openstack-releasenotes/a2aa3bf/html/ : SUCCESS in 5m 16s\n- neutron-lbaasv2-dsvm-api https://logs.opendev.org/62/675362/1/check/neutron-lbaasv2-dsvm-api/63047c6/ : SUCCESS in 1h 22m 50s\n- neutron-lbaasv2-dsvm-py3x-api https://logs.opendev.org/62/675362/1/check/neutron-lbaasv2-dsvm-py3x-api/3894fef/ : SUCCESS in 1h 20m 33s\n- neutron-lbaasv2-dsvm-scenario https://logs.opendev.org/62/675362/1/check/neutron-lbaasv2-dsvm-scenario/33ac185/ : FAILURE in 1h 42m 39s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://logs.opendev.org/62/675362/1/check/neutron-lbaasv2-dsvm-py3x-scenario/1ed3104/ : SUCCESS in 1h 26m 31s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/62/675362/1/check/openstack-tox-functional/b52f976/ : SUCCESS in 5m 27s\n- openstack-tox-functional-py35 https://logs.opendev.org/62/675362/1/check/openstack-tox-functional-py35/d9479aa/ : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 https://logs.opendev.org/62/675362/1/check/openstack-tox-functional-py36/b891340/ : SUCCESS in 5m 41s\n- octavia-v1-dsvm-scenario https://logs.opendev.org/62/675362/1/check/octavia-v1-dsvm-scenario/7723ec7/ : SUCCESS in 1h 05m 20s\n- octavia-v1-dsvm-py3x-scenario https://logs.opendev.org/62/675362/1/check/octavia-v1-dsvm-py3x-scenario/c2c049d/ : SUCCESS in 1h 13m 47s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-noop-api/3afe8f5/ : SUCCESS in 52m 21s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-noop-py2-api/76577b9/ : SUCCESS in 48m 19s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-scenario/e650d3e/ : FAILURE in 1h 49m 15s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-py2-scenario/38acef9/ : FAILURE in 1h 47m 19s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-scenario-ubuntu-xenial/55c2333/ : FAILURE in 1h 34m 37s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-py2-scenario-centos-7/95c820d/ : SUCCESS in 2h 00m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/e94f20a/ : FAILURE in 40m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-scenario-two-node/eb1daae/ : SUCCESS in 1h 43m 01s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-py2-scenario-two-node/6649eae/ : SUCCESS in 1h 14m 55s (non-voting)\n- octavia-grenade https://logs.opendev.org/62/675362/1/check/octavia-grenade/43792a5/ : SUCCESS in 1h 30m 30s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-tls-barbican/fa383c5/ : SUCCESS in 55m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db8ead7a76f4b092b47a8207daa0f22cad699792","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-09 06:45:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fd918739a91cbe64cd50168beb8d479fd956d184","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 09:10:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/62/675362/1/check/openstack-tox-cover/28af560/cover/ : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://logs.opendev.org/62/675362/1/check/openstack-tox-lower-constraints/37a6530/ : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://logs.opendev.org/62/675362/1/check/openstack-tox-pep8/7243d07/ : SUCCESS in 5m 15s\n- openstack-tox-py27 https://logs.opendev.org/62/675362/1/check/openstack-tox-py27/ed2aa4d/ : SUCCESS in 4m 07s\n- openstack-tox-py35 https://logs.opendev.org/62/675362/1/check/openstack-tox-py35/6f89d4c/ : SUCCESS in 5m 40s\n- openstack-tox-py36 https://logs.opendev.org/62/675362/1/check/openstack-tox-py36/e5252df/ : SUCCESS in 5m 25s\n- openstack-tox-docs https://logs.opendev.org/62/675362/1/check/openstack-tox-docs/a5a5612/html/ : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://logs.opendev.org/62/675362/1/check/build-openstack-releasenotes/7b89575/html/ : SUCCESS in 4m 33s\n- neutron-lbaasv2-dsvm-api https://logs.opendev.org/62/675362/1/check/neutron-lbaasv2-dsvm-api/30a3d55/ : SUCCESS in 1h 18m 53s\n- neutron-lbaasv2-dsvm-py3x-api https://logs.opendev.org/62/675362/1/check/neutron-lbaasv2-dsvm-py3x-api/e59e46b/ : SUCCESS in 1h 12m 24s\n- neutron-lbaasv2-dsvm-scenario https://logs.opendev.org/62/675362/1/check/neutron-lbaasv2-dsvm-scenario/cf02ed7/ : FAILURE in 2h 04m 55s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://logs.opendev.org/62/675362/1/check/neutron-lbaasv2-dsvm-py3x-scenario/338bf00/ : SUCCESS in 1h 25m 54s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/62/675362/1/check/openstack-tox-functional/6f744fc/ : SUCCESS in 4m 54s\n- openstack-tox-functional-py35 https://logs.opendev.org/62/675362/1/check/openstack-tox-functional-py35/57abfe9/ : SUCCESS in 7m 35s\n- openstack-tox-functional-py36 https://logs.opendev.org/62/675362/1/check/openstack-tox-functional-py36/0daa20d/ : SUCCESS in 7m 02s\n- octavia-v1-dsvm-scenario https://logs.opendev.org/62/675362/1/check/octavia-v1-dsvm-scenario/5800694/ : SUCCESS in 1h 08m 04s\n- octavia-v1-dsvm-py3x-scenario https://logs.opendev.org/62/675362/1/check/octavia-v1-dsvm-py3x-scenario/5399975/ : SUCCESS in 51m 51s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-noop-api/3d26ecd/ : SUCCESS in 45m 04s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-noop-py2-api/cb50a84/ : SUCCESS in 46m 33s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-scenario/d6b7cc1/ : SUCCESS in 1h 23m 18s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-py2-scenario/b9ce0ee/ : SUCCESS in 1h 37m 07s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-scenario-ubuntu-xenial/35c9c1e/ : SUCCESS in 1h 45m 39s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-py2-scenario-centos-7/f23f8ff/ : SUCCESS in 2h 18m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/a61662a/ : FAILURE in 49m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-scenario-two-node/172a2b4/ : SUCCESS in 1h 14m 43s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-py2-scenario-two-node/6cbded0/ : SUCCESS in 1h 31m 08s (non-voting)\n- octavia-grenade https://logs.opendev.org/62/675362/1/check/octavia-grenade/61e932a/ : SUCCESS in 1h 20m 53s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/62/675362/1/check/octavia-v2-dsvm-tls-barbican/7966207/ : SUCCESS in 57m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb902b13018b48d5cdd3eeb43b533ee01007d028","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 15:33:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e700d237406f6598bc8cf6b9f3a834d0e7890764","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-06 06:57:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"82219862444465bcaae3d42d3e4b6ec5f460b016","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 10:27:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46d6d647f9004a35bda5d6c813d5a9b0 : FAILURE in 3m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2a14a7a903146488b7850c5e428c674 : FAILURE in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f09417eff702438e86dfaa1fb8789b92 : FAILURE in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/68063a7a43b44153bde7b94bf923817e : FAILURE in 4m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/027c914a30dc4e339279936451a915da : FAILURE in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/072c505b12af44ec9d4ff9f7607e9512 : FAILURE in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dc4b858754c46c497c09a2fb0dde9b4 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7061cd89b8d14d7bb21216ccd6963171 : SUCCESS in 4m 43s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/9b056a3c0f304785967ce9e4911bda28 : SUCCESS in 1h 17m 48s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ca8df79e4a674e4a85aba5fbe663e9c1 : SUCCESS in 1h 19m 34s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d437098f06174467b37bc4c25f5c0ad3 : FAILURE in 32m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5ab038270b3c40bf8a60d22d021e0230 : FAILURE in 26m 30s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/695c7672d3a947cd80138dd74088079f : SUCCESS in 6m 09s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/611fdfa60c234a21be81229c70fae4e2 : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7dcec9e0b1cb435faa7fcab66955f93d : SUCCESS in 6m 57s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fed4526bdd934333b59a712edf8d5dc3 : SUCCESS in 1h 09m 21s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/49d2535d18734a0094f3ef5061583792 : SUCCESS in 1h 06m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e894c4189d2d4456abf5bcaafa23260f : SUCCESS in 52m 26s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1e7f0bfc9e3241eb8186322221bc3172 : SUCCESS in 42m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec20db4c8de642f6b96c2948e654a42d : SUCCESS in 1h 31m 28s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ce152b2b93ed4e1fa13eaee55ed6ad60 : SUCCESS in 1h 47m 02s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/5fb91909ca9742788dbcf4658a22dfcd : SUCCESS in 1h 37m 15s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/265e2ccd6183455baa0d656bd4f5f7ba : FAILURE in 56m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/412562aa563b41f392a92608ae6a904d : FAILURE in 45m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/047487d081c6428dbe74f79997f0bb9c : SUCCESS in 1h 22m 43s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c213712e3e5d4201974afcf795a293a7 : SUCCESS in 2h 16m 59s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dc307b67c3e44c0abddffc8f5429644e : FAILURE in 42m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1df3d66777a341ad9b8deff15d4aa6b7 : SUCCESS in 56m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"be519cf2340b3f05dfed5ecfcc43738b4a22e69e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-12 08:43:26.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"540ed4b763649240ede666a2e850932a45ac9425","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 11:29:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9563e902790483c87f49e415ddc576b : SUCCESS in 9m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6da5735db7b64f5c9fac3b380d4269f0 : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e5418fbe8284f83a293dde9500058fc : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/905bc13f4a444739910baa60ee5cc686 : SUCCESS in 4m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/670e83ff5c6b4c16a012fbeec6b70ee6 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/134bf4d6012a43e8b4dc43907c672769 : TIMED_OUT in 41m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65cf2a4c50434db78ee21fb458be665c : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3231106ba894968ae826cc66a8f4de9 : SUCCESS in 4m 45s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/7e8c7df4cc3542d0ae2f4264b243b034 : SUCCESS in 1h 15m 29s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ef7ea92e415c446e8f940fa025725232 : SUCCESS in 1h 37m 59s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1c7dfce38d2f4099aacc020f0e89a996 : SUCCESS in 1h 28m 39s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a701915ecab44e509c720626c1987ccd : SUCCESS in 1h 30m 24s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/402036cec9db4d6f9b0652ab0f8ff045 : SUCCESS in 5m 32s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/982306d6973544af98c3ee4ffa1f6040 : TIMED_OUT in 30m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/427ef19b5f8341c1930522314d1a7eef : TIMED_OUT in 30m 46s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0fa477e4a0284fd5979cfc65dd692302 : SUCCESS in 1h 07m 00s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/dde9801ae8f040429cb74e15587464e4 : SUCCESS in 1h 09m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/813ccc3cd1d045bda2090e49028c1efc : SUCCESS in 48m 03s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9feb329e0ef5405c82ac256066800906 : SUCCESS in 49m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2fa5198b859a42979389c2af3df33a6f : SUCCESS in 1h 50m 54s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/68bfbf63d2ab47cd937b7298dbb60abe : SUCCESS in 1h 40m 52s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/44f6a3bb19444b07b865098635678d91 : SUCCESS in 1h 42m 38s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4a2878c1ed284332a681390a21657097 : SUCCESS in 2h 29m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75bea0fc6e1c47b99d8e383c2243d523 : FAILURE in 35m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f6c2c59c51d34d10851ea70d7c103fd5 : FAILURE in 58m 42s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/7c021416f9414e48a2d9301472c45718 : FAILURE in 57m 03s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/572b4f8cdb3a43de861b256e65b5c9e4 : SUCCESS in 1h 33m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2ffab6f055ae486e95c6ee5d226726c6 : SUCCESS in 1h 28m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2314efe77ead83e30c4133edec07371694c7d94c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-12 16:13:10.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f85ea9a89374429bede5e7ba725884eff2502213","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 19:31:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/582a3fe6c9d84ae3abc5de0dcc20f6c1 : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b82a6f3942044d69cb41f67f264878d : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01f73d087f604176a605ef3a9952c59c : SUCCESS in 7m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a9cf09b8236e4f2c9eddbaeb21a9d91e : SUCCESS in 4m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ec88c62af99a4df3812be57b2cfa715a : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20987a7712b846c3adbdcf57da0dbb21 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87e31bbcef5d441ea81261c925e25e26 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51c2d4afecab4063be83f454fd8ba509 : SUCCESS in 5m 20s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c8ff32ff2a9a451195cb604faaec858b : SUCCESS in 1h 25m 42s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/d59a8b3125cc4195937690e6402d7b94 : SUCCESS in 1h 19m 51s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/919864717258481eb97a8a5d31727b1f : SUCCESS in 1h 27m 43s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/00d6cba6ccf94800955e8a53d6b0d5e7 : SUCCESS in 1h 33m 48s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/29e64db3de354b6c85f01cf0294e347e : SUCCESS in 5m 25s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/58c1c0d9b6484ac1ab6519e1f68f1630 : SUCCESS in 10m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1cf97aa7e7854f1bad9979f7f484ff7c : SUCCESS in 6m 07s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c848c66858674ea49b10e1c108a57e61 : SUCCESS in 1h 07m 17s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ab5c025a480d4a39baad2b610b60c2b2 : SUCCESS in 1h 04m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/020541bd026346cfbc2b0c4661c6e058 : SUCCESS in 42m 43s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ad7b5c627e8b41819113c52e95b37b8b : SUCCESS in 53m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94914e77940e418a8eac3df734d729eb : SUCCESS in 1h 59m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/704b87fe1bed4332a10795a83f685685 : SUCCESS in 1h 25m 32s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/293831210ce9457bbd3a688d0c61dafc : SUCCESS in 1h 59m 41s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/bdc02f7af2694b6e8b1d44cff3654d68 : SUCCESS in 2h 21m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d89e1f4bc754481ab6ae7d8c28db58b : FAILURE in 32m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e32f7e8b7ba54911a59d9760b13b3534 : FAILURE in 1h 50m 23s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a81405dc74374d009765810f9c5f1c87 : SUCCESS in 1h 48m 44s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d63613a8484a41c2bb270243c8b36df9 : SUCCESS in 1h 44m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c66107bf693044539d7d6b33076a4493 : SUCCESS in 1h 08m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"25be0f73d37a6ee5e4714dd3f1d82b7313038edb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-14 19:22:30.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"1a69644d8963c1a5e9000143a314a38fe87dba4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 13:02:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"220886181f194571871ab67cd04f073c7d2bf7c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 15:24:57.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"c9e116df96751a9065dda655650e7a6fe00d8cb8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-17 05:34:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a36d6433ae27f7de3576528f0565dcc9e2707e5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 07:54:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05c45dfd994b40439f1c74e5ea0578d9 : SUCCESS in 8m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0915e8a848f4869b7bd36e36d4c2081 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/460ec87d517e4af1a20b462f32c1e447 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ec6ad63a6a341bba10dad1e4ae78fa4 : SUCCESS in 5m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5f2c627321824301adbd40d9bedb6fb4 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/603f97014cb5499a9cf76b406cc53025 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2863a04203514b8ba0fd940ac29c424d : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e2d383d6cde406ea3ee2f2c621014d0 : SUCCESS in 5m 52s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6a95fa9952ca4d068d10ec015ae26986 : SUCCESS in 1h 10m 41s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/613deac49c2348999543a04dc950bd8f : SUCCESS in 1h 17m 33s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9668879c2924925823b5b7c3a33b844 : SUCCESS in 1h 27m 06s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d09ce8d56eb8472cbcd817560ecd0b5d : SUCCESS in 1h 27m 58s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a56cc166a7cd440b896cbeba0e37f6d9 : SUCCESS in 4m 41s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/a563f50e553a42518f41827e241c6e9f : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/094ca1d1713045ea953a8da425b0b5cc : SUCCESS in 6m 00s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0cdcfd1f9b8b41719fb9a6f5662b644a : SUCCESS in 1h 09m 34s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7bf708b35b5142ad9a5ce1f1c427442a : SUCCESS in 1h 10m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/92f62d09d8004b0182a604165a1dcc7f : SUCCESS in 45m 43s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b45172711f5b43619d7b573946fc6388 : SUCCESS in 42m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7fa79c91a07c4ce4bc01d0aa1bbf1b54 : SUCCESS in 1h 29m 55s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/44399e1bf0324c24b71d7952603abae4 : SUCCESS in 1h 29m 25s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/c7f9f3a118cc4cae9fcf7557d73fc4fb : SUCCESS in 1h 58m 37s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/99f744b8b78a4b40a31858f531e138fb : SUCCESS in 2h 14m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4ca013d88fb4e5da966530d6db70c5a : FAILURE in 36m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5cfb5d1ecb1a495d96cc1c41f3232a40 : SUCCESS in 1h 51m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/27f2a8f2c57943ff8d55b1a89535bddf : SUCCESS in 1h 48m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fbb8cec720dd4e17920391139e8a66f6 : SUCCESS in 1h 29m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a3d3fddee5cf48a9b92f04afff2b74e6 : SUCCESS in 59m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"594c5bb9685adba0b1e5900d3302c73611e7b59c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 07:55:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a9439d07d4a89448798b9f1ed606b280bf831e01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 09:57:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cda05a6f0184cb99c852f6a28d68b86 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3516e3a5147e47a7b651c03a33f26e7f : SUCCESS in 7m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f880b63eaa8d44578aecd487f522f0a5 : SUCCESS in 5m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2917150813c44d838d50aba9712fd59b : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7039363a1d1140789fab2f49e1ad4317 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e3ca6ffc8bc4d7fb1bfd9255a140df8 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51068a6677e64fb1b120d938d7bf7689 : SUCCESS in 4m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8aa13dcfd4c94b7e96e178cbfc2b2935 : SUCCESS in 6m 03s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/74074ea563b047ca8686c1fb9a7e439c : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/210cc777c9dd43b487b33a895ac6bea4 : SUCCESS in 8m 45s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49fd57c00ad949d8949b16b77d8dc366 : SUCCESS in 1h 10m 12s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1e912d5ebbc443bfb2eb407493d16bf8 : SUCCESS in 1h 07m 42s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c90e1891f5cc4fe087a2ccdeae125b78 : SUCCESS in 1h 12m 25s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/c0784c939ad24a59bb16bcf66e3bf0ea : SUCCESS in 1h 30m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bcf1010495ae4ba9b37ced7f00c68c0b : SUCCESS in 50m 42s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/652989bf97204510ac3a37950f8663a2 : SUCCESS in 49m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b7d33384a24b4214b58507c9545dfeb1 : SUCCESS in 1h 36m 06s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c61f657885c94cb9ba8ea554d0a09bf4 : SUCCESS in 1h 27m 08s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/97bc0814c24e4a4c8c2dd58cbad784fd : SUCCESS in 1h 56m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/feae2554186f44e581e40c175780bcee : SUCCESS in 1h 10m 49s","accounts_in_message":[],"_revision_number":3},{"id":"81689cc68285a6610268ff71069bc2ff3fbd6d46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 09:57:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6fd7e0c9faee6dbb05e14e270a59d84d1ec662e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 09:58:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ffbf22c13ec499e94ba0b4967978ec2 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2656257c8c964bc6b4d798a664af8e3a : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bf50e3a14b72d1c1a4e881695b91ca0914bc7f70","revisions":{"edb767399c3d96425591aaebc9a86cd01411d17c":{"kind":"REWORK","_number":1,"created":"2019-08-08 14:45:00.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/62/675362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/675362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/675362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/675362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/675362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/675362/1"}}},"commit":{"parents":[{"commit":"d6bb233b96518ea09e33f6dda40325494cd1e853","subject":"Add support for monitor_{address,port} in UDP members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6bb233b96518ea09e33f6dda40325494cd1e853"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-05-20 13:53:52.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-08-08 14:41:00.000000000","tz":120},"subject":"Fixed pool and members status with UDP loadbalancers","message":"Fixed pool and members status with UDP loadbalancers\n\nThis commit fixes pool and members status when using UDP loadbalancers.\n\nStory: 2005736\nTask: 33394\n\nChange-Id: I75cde3ff820f085aebbdffd1e40c5ff40f16835d\n(cherry picked from commit 4decb6d53c728ba0e7b15c6f46a9c2b5251b4088)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/edb767399c3d96425591aaebc9a86cd01411d17c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/edb767399c3d96425591aaebc9a86cd01411d17c"}]},"branch":"refs/heads/stable/stein"},"e2b9c4740de060290099040df17a3ad915a1a278":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-06 06:57:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/62/675362/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/675362/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/675362/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/675362/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/675362/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/675362/2"}}},"commit":{"parents":[{"commit":"c3ae4c56cbcad500a579c48b4c1bcc52932b42a4","subject":"Add support for monitor_{address,port} in UDP members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3ae4c56cbcad500a579c48b4c1bcc52932b42a4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-05-20 13:53:52.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-09-06 06:56:46.000000000","tz":120},"subject":"Fixed pool and members status with UDP loadbalancers","message":"Fixed pool and members status with UDP loadbalancers\n\nThis commit fixes pool and members status when using UDP loadbalancers.\n\nStory: 2005736\nTask: 33394\n\nChange-Id: I75cde3ff820f085aebbdffd1e40c5ff40f16835d\n(cherry picked from commit 4decb6d53c728ba0e7b15c6f46a9c2b5251b4088)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2b9c4740de060290099040df17a3ad915a1a278"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2b9c4740de060290099040df17a3ad915a1a278"}]},"branch":"refs/heads/stable/stein"},"bf50e3a14b72d1c1a4e881695b91ca0914bc7f70":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-12 08:43:26.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/62/675362/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/675362/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/675362/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/675362/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/675362/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/675362/3"}}},"commit":{"parents":[{"commit":"7f683dc2c568629a3645aeaf603db24f03c3a71a","subject":"Add support for monitor_{address,port} in UDP members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f683dc2c568629a3645aeaf603db24f03c3a71a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-05-20 13:53:52.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-09-12 08:28:32.000000000","tz":120},"subject":"Fixed pool and members status with UDP loadbalancers","message":"Fixed pool and members status with UDP loadbalancers\n\nThis commit fixes pool and members status when using UDP loadbalancers.\n\nStory: 2005736\nTask: 33394\n\nChange-Id: I75cde3ff820f085aebbdffd1e40c5ff40f16835d\n(cherry picked from commit 4decb6d53c728ba0e7b15c6f46a9c2b5251b4088)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf50e3a14b72d1c1a4e881695b91ca0914bc7f70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf50e3a14b72d1c1a4e881695b91ca0914bc7f70"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
