)]}'
{"id":"openstack%2Foctavia~675364","triplet_id":"openstack%2Foctavia~stable%2Fstein~If6fb3fde9b43ac82af46eaddc48ec7a3a5b95602","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"If6fb3fde9b43ac82af46eaddc48ec7a3a5b95602","subject":"Prevent UDP LBs to use different IP protocol versions in amphora driver","status":"MERGED","created":"2019-08-08 14:45:00.000000000","updated":"2019-09-17 18:10:16.000000000","submitted":"2019-09-17 18:07:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675364-1568743656439-81340386","meta_rev_id":"82d726caaa9bd9b0c2eba24c5024c1b8d7584b1b","_number":675364,"virtual_id_number":675364,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2019-09-17 13:28:31.000000000","_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2019-09-17 18:07:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"recommended":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-09-17 05:35:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-12 08:43:26.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-09-14 19:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-14 01:16:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-09-14 19:21:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 15:32:52.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-14 01:16:50.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-09-14 19:21:50.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-17 05:35:12.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-17 18:07:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc59f6049f93cae6d991a7fb0f4ebf7d26431738","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-08 14:45:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43fb0a51deb7013c7b7ced14256be802f4ebd035","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 17:20:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/64/675364/1/check/openstack-tox-cover/6224415/cover/ : SUCCESS in 9m 54s\n- openstack-tox-lower-constraints https://logs.opendev.org/64/675364/1/check/openstack-tox-lower-constraints/5a1643b/ : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://logs.opendev.org/64/675364/1/check/openstack-tox-pep8/c5e5176/ : SUCCESS in 5m 31s\n- openstack-tox-py27 https://logs.opendev.org/64/675364/1/check/openstack-tox-py27/282fb11/ : SUCCESS in 5m 16s\n- openstack-tox-py35 https://logs.opendev.org/64/675364/1/check/openstack-tox-py35/a20ccd1/ : SUCCESS in 5m 03s\n- openstack-tox-py36 https://logs.opendev.org/64/675364/1/check/openstack-tox-py36/a0da5e2/ : SUCCESS in 6m 41s\n- openstack-tox-docs https://logs.opendev.org/64/675364/1/check/openstack-tox-docs/62c1189/html/ : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://logs.opendev.org/64/675364/1/check/build-openstack-releasenotes/719816a/html/ : SUCCESS in 6m 11s\n- neutron-lbaasv2-dsvm-api https://logs.opendev.org/64/675364/1/check/neutron-lbaasv2-dsvm-api/9977bbf/ : SUCCESS in 1h 31m 29s\n- neutron-lbaasv2-dsvm-py3x-api https://logs.opendev.org/64/675364/1/check/neutron-lbaasv2-dsvm-py3x-api/d6cacee/ : SUCCESS in 1h 19m 38s\n- neutron-lbaasv2-dsvm-scenario https://logs.opendev.org/64/675364/1/check/neutron-lbaasv2-dsvm-scenario/b75fd09/ : SUCCESS in 1h 32m 29s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://logs.opendev.org/64/675364/1/check/neutron-lbaasv2-dsvm-py3x-scenario/7a27f38/ : SUCCESS in 1h 34m 36s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/64/675364/1/check/openstack-tox-functional/23fc420/ : SUCCESS in 6m 41s\n- openstack-tox-functional-py35 https://logs.opendev.org/64/675364/1/check/openstack-tox-functional-py35/6cc66ab/ : SUCCESS in 9m 01s\n- openstack-tox-functional-py36 https://logs.opendev.org/64/675364/1/check/openstack-tox-functional-py36/28c7365/ : SUCCESS in 6m 21s\n- octavia-v1-dsvm-scenario https://logs.opendev.org/64/675364/1/check/octavia-v1-dsvm-scenario/4d95b33/ : FAILURE in 57m 35s\n- octavia-v1-dsvm-py3x-scenario https://logs.opendev.org/64/675364/1/check/octavia-v1-dsvm-py3x-scenario/ade2b58/ : SUCCESS in 1h 02m 30s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-noop-api/8985431/ : SUCCESS in 44m 04s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-noop-py2-api/1ff5015/ : SUCCESS in 47m 16s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-scenario/38618c5/ : FAILURE in 1h 44m 33s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-py2-scenario/db21c2c/ : SUCCESS in 1h 49m 23s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-scenario-ubuntu-xenial/0931a06/ : SUCCESS in 1h 49m 04s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-py2-scenario-centos-7/2e05cba/ : SUCCESS in 2h 13m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/139d892/ : FAILURE in 44m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-scenario-two-node/bddef17/ : SUCCESS in 1h 52m 04s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-py2-scenario-two-node/3d3eee1/ : FAILURE in 1h 37m 29s (non-voting)\n- octavia-grenade https://logs.opendev.org/64/675364/1/check/octavia-grenade/30aabb9/ : SUCCESS in 1h 13m 49s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-tls-barbican/83295dd/ : SUCCESS in 52m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f739982ccfb3f1c25d39dd580ce47eb25d1d9b00","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-09 06:45:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e4c94849cf1c299459a463a4531f200db663db14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 09:10:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/64/675364/1/check/openstack-tox-cover/6f30667/cover/ : SUCCESS in 8m 50s\n- openstack-tox-lower-constraints https://logs.opendev.org/64/675364/1/check/openstack-tox-lower-constraints/97cdc27/ : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://logs.opendev.org/64/675364/1/check/openstack-tox-pep8/fafdef4/ : SUCCESS in 5m 41s\n- openstack-tox-py27 https://logs.opendev.org/64/675364/1/check/openstack-tox-py27/15f314d/ : SUCCESS in 5m 24s\n- openstack-tox-py35 https://logs.opendev.org/64/675364/1/check/openstack-tox-py35/961a94d/ : SUCCESS in 4m 54s\n- openstack-tox-py36 https://logs.opendev.org/64/675364/1/check/openstack-tox-py36/ad1e3d7/ : SUCCESS in 4m 02s\n- openstack-tox-docs https://logs.opendev.org/64/675364/1/check/openstack-tox-docs/398cd0f/html/ : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://logs.opendev.org/64/675364/1/check/build-openstack-releasenotes/ac8b960/html/ : SUCCESS in 5m 47s\n- neutron-lbaasv2-dsvm-api https://logs.opendev.org/64/675364/1/check/neutron-lbaasv2-dsvm-api/e942406/ : SUCCESS in 1h 17m 25s\n- neutron-lbaasv2-dsvm-py3x-api https://logs.opendev.org/64/675364/1/check/neutron-lbaasv2-dsvm-py3x-api/dadfdb2/ : SUCCESS in 1h 27m 31s\n- neutron-lbaasv2-dsvm-scenario https://logs.opendev.org/64/675364/1/check/neutron-lbaasv2-dsvm-scenario/9796b3f/ : FAILURE in 1h 29m 13s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://logs.opendev.org/64/675364/1/check/neutron-lbaasv2-dsvm-py3x-scenario/0a3b574/ : SUCCESS in 1h 23m 36s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/64/675364/1/check/openstack-tox-functional/23382f3/ : SUCCESS in 6m 55s\n- openstack-tox-functional-py35 https://logs.opendev.org/64/675364/1/check/openstack-tox-functional-py35/2a26c8f/ : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://logs.opendev.org/64/675364/1/check/openstack-tox-functional-py36/d434e89/ : SUCCESS in 6m 04s\n- octavia-v1-dsvm-scenario https://logs.opendev.org/64/675364/1/check/octavia-v1-dsvm-scenario/27d9035/ : SUCCESS in 1h 06m 06s\n- octavia-v1-dsvm-py3x-scenario https://logs.opendev.org/64/675364/1/check/octavia-v1-dsvm-py3x-scenario/80b71cc/ : SUCCESS in 1h 01m 16s\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-noop-api/cb3bd81/ : SUCCESS in 48m 20s\n- octavia-v2-dsvm-noop-py2-api https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-noop-py2-api/6b4f02d/ : SUCCESS in 43m 58s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-scenario/43dcbeb/ : SUCCESS in 1h 30m 34s\n- octavia-v2-dsvm-py2-scenario https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-py2-scenario/5eb61b0/ : SUCCESS in 1h 46m 53s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-scenario-ubuntu-xenial/be997f7/ : SUCCESS in 1h 35m 29s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-py2-scenario-centos-7/2b79eb3/ : SUCCESS in 2h 12m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-scenario-ubuntu-bionic/99b28ff/ : FAILURE in 37m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-scenario-two-node/22cdbd1/ : SUCCESS in 1h 43m 15s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-py2-scenario-two-node/28d228f/ : SUCCESS in 1h 58m 45s (non-voting)\n- octavia-grenade https://logs.opendev.org/64/675364/1/check/octavia-grenade/5e642e6/ : SUCCESS in 1h 11m 54s\n- octavia-v2-dsvm-tls-barbican https://logs.opendev.org/64/675364/1/check/octavia-v2-dsvm-tls-barbican/1c7dad2/ : SUCCESS in 52m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba5bdbcd7bbf8e53549dd7814435bae43822768d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 15:32:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"04003d485b08056de75f1614a2b0a7fe35b608c5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-06 06:57:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"bf2aef50c06c412c8e67c030bdf55d0173feda0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 11:11:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/505b794c476a4cd4b34ad04be9f5fd27 : FAILURE in 4m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ada30de886ca4c8f8820046aa1ddf1eb : FAILURE in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92ef14d9106f46e194ac4a5ce1a43a3d : FAILURE in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c3a8e1793b042d585b45c985967e9f6 : FAILURE in 6m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/27793ad07556407ca0af1788af01a6cc : FAILURE in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ca72846d6a24fd981c1b776071b38a5 : FAILURE in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88737a8ca55f4cc889b9c4b18ccfbdfd : SUCCESS in 9m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d098f89b0253492e9d9f115cf3bdeb92 : SUCCESS in 6m 23s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/0e84e4625c2945b89a17d1416fb87b0a : SUCCESS in 1h 25m 11s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/6b4356a6341440cd9cc82748fe556372 : SUCCESS in 1h 13m 09s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d75276f3f07d415d826e070bc16ead7a : FAILURE in 38m 11s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9bfe948be4a14c6ab138d8b8e3382017 : FAILURE in 27m 16s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8b5ec7aae96249f79b288686fda61deb : SUCCESS in 7m 37s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/21f9ced77d1d47368d62519321bc4ffb : SUCCESS in 8m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/010c5b02dd6149e5ba2e4999c800cef0 : SUCCESS in 6m 19s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a40b9f0cf0094a13bc671d7b24997f2a : SUCCESS in 1h 20m 06s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9b282f54b40040fbb9ffca967e2bcbbb : SUCCESS in 1h 12m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fb7532a518db4ed5a3dedd71869689ab : SUCCESS in 49m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c3fddb9e5a5347beb1760e9f7e725ade : SUCCESS in 42m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ed2682597a0342729bdd1fbec331efb1 : SUCCESS in 1h 38m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/518fc725659c4c2d88fe4fcb3df37e28 : SUCCESS in 1h 22m 56s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/1b6b1b45df9545898ddf3b36e59e7d3d : SUCCESS in 1h 35m 24s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ea5cfb5460524017ada6acb094cd9bb0 : SUCCESS in 1h 58m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13bc3d6cd8f24a4e8f7d9493d30ec915 : FAILURE in 34m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ac089b32559a4ce98a91066a5d1eb3b2 : SUCCESS in 2h 31m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3d5a7f26adae447da483e359df8a3fb0 : SUCCESS in 1h 18m 24s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/37be4e02f7734f90935eb64bfc66a46f : FAILURE in 39m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/442060af069a49e1b8064edd10b65c1e : SUCCESS in 52m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e0c914d7a97b317dc01b5363624aaaeac8c68eae","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-12 06:45:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"742f703f2474e9fa2ad7c048af753e95ecce2291","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-12 08:43:26.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"bb8a6421c0eb970b7d7cca5a044673312a85a2e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 11:47:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/162671e25c7146db90172d009ecd3696 : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d29ba5259bea4defa7418ad90f7d4405 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36a1e231a25445c91fdd295d15db063 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/68701ba0c88b4501bfb818250427a057 : SUCCESS in 3m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/655ff39950a14af79c83fcf53881cc17 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c813ab54a91047869844ab19cbe31a97 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb7c951c8ff54f45a4e44a197af111f2 : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e52bcd6ad7b643c8b058c90c71c2aed1 : SUCCESS in 5m 54s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/e85ca77827d245ec9fa7b61caaeffe4e : FAILURE in 1h 24m 28s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/388cc5b3ad6042a8b8a3f05314ba3cd0 : SUCCESS in 1h 26m 10s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c5ced0d93d04b67a242daebf8e26cc4 : SUCCESS in 1h 33m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/839ea68b887b450eba292d2b513295a9 : SUCCESS in 1h 30m 04s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/acee06705c5c4a9b847584063a17e5f1 : FAILURE in 15m 46s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4793ebcd0b7640aebc2d4df7c4d4bbd9 : SUCCESS in 7m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/177c8f999bf64995b22b6fa9f0191829 : SUCCESS in 6m 42s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa684b306ab448b1be09208e0f8dfb1c : SUCCESS in 1h 07m 06s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/079900a55099492da3e98b47afef1227 : SUCCESS in 1h 07m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cb6969b290bb4e97aa68fbf32f7a748d : SUCCESS in 44m 13s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/245be4665b4d449dad04e5327a6550bb : SUCCESS in 49m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9bfea843b7a14706b1e8c57ff8f841aa : SUCCESS in 1h 34m 22s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/91e05d65742f450c85e358d2f24bda6e : SUCCESS in 1h 32m 46s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b16f858b13b0424f837be324d99f535b : SUCCESS in 1h 41m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c90af017b0854f94847522520c58078c : SUCCESS in 2h 25m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/306cdc96fc1f407caec5d4d28cf7d5e9 : FAILURE in 39m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/05fdaf8504b44a33b4b3a9e51d72ad15 : SUCCESS in 1h 51m 40s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/bc3f8d77ed6544bab7e201f60a7ec6b8 : FAILURE in 1h 28m 22s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7f24ee8ebbce4315b5ef2d921ffb495a : SUCCESS in 1h 18m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dc7c09da38844b0ea1af6e860943b765 : SUCCESS in 1h 05m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"12034e5e8bf149458038aac4c176ee93a86715f8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-12 16:13:01.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8cce176ce5b787a2177e847a2f7d8ec96ed0e8bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 19:10:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1761174a65943aaaf5b72c7eb58264a : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed92c5bd8fce4fe5918efc4e3a674ed3 : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca264cfe9fd64ac0aceade31c0fda134 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/017ac9bd4edd48539061428b537665f2 : SUCCESS in 4m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8ca8e3ca31bb48e79e9d1e9b9dedea80 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8653bcb1c3f49599495483b1d66c226 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c84149b14b468bb0191bc1bda7264a : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2ab3a0a3c9142deb7a0473d553a6f3e : SUCCESS in 4m 55s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/62cce4e1a27e446289dad01689a6624e : SUCCESS in 1h 19m 54s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ac69a58ac8104ae385d0daa9d6995c9a : SUCCESS in 1h 25m 26s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf2a027631c94c439eb30b5a24d799d8 : SUCCESS in 1h 45m 26s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0bbbce7f1b284dfdb0826129f4082e2e : SUCCESS in 1h 30m 11s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/455e7749336c42d6929459101ac87d3f : SUCCESS in 7m 38s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d9bfcf4fd9fe45cb9cb8d34e61f9a625 : SUCCESS in 5m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1aa0c70edc534b1a9fc536e8d46a5269 : SUCCESS in 6m 10s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ace754531155413fbb7f739889c2dc12 : SUCCESS in 1h 13m 59s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/284311de806c4d5ca893d96c1cb4930e : SUCCESS in 1h 07m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/07acdd4d4a2f4cfaa078187f4d50a3be : SUCCESS in 42m 37s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/db99e7e190514562b53efc28dbc5f09e : SUCCESS in 59m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e38c272b4d6f4ba493b8b36398c41a7d : SUCCESS in 1h 31m 00s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/bc68a65388e448f1ba54d104b6254910 : SUCCESS in 1h 52m 37s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f35e2bac9cf14cefa666871cdd8fc3af : SUCCESS in 1h 39m 51s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/22e830ef88d346f98f80054cd64dad2d : SUCCESS in 2h 13m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/788621271452492599b6b2ede81c6ba2 : FAILURE in 32m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/057c33ded63f45e5a4ffa4dc0148332a : FAILURE in 1h 50m 06s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/677456bd095e42b59da8ffdb36e91886 : SUCCESS in 1h 18m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ca0edec7f14d4717955ee8ccdc5dbb17 : SUCCESS in 1h 29m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f98cab483d9f47b69a843c1fabafb6cd : SUCCESS in 53m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9c3fb57aef29a6c872e8f2a432a021c377f264d4","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-09-14 01:16:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b3c875fe89de32b498fec33342e72a57b757d51e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-14 19:21:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"6674d33a867253b72d051e5f33f5fb831d4993d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 13:02:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"840808ec9ca1cff60c042701f3ebda3647e8d735","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 15:24:58.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"afede29fb6dc5e538bc049170b173c22aab96d37","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-17 05:35:12.000000000","message":"Patch Set 3: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c7e91954313b512aab50afdcf29c090fe24c0dca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 07:54:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cc380c79fd0470d85441bd2d6ce4bb4 : SUCCESS in 8m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b117001569b440c98e0fb05df0eaf64 : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e91a67b477c4c9daf9feb5f518c60a5 : SUCCESS in 6m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c10506263e3421384dbbb347ebfc661 : SUCCESS in 5m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e71b4428e3fa48beb2d2cf1e94a11de9 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37400893a5154a679e74466948827bc4 : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5165aa7b3f964715bd523d8cd210af15 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/471e7ba0bd494c378918d1b1bf63aba4 : SUCCESS in 10m 34s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1e0c0bae3c2740aeb1ae09af670b6802 : SUCCESS in 1h 17m 44s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f08728c7da2544048aba09dd78247228 : SUCCESS in 1h 21m 34s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6620feb6d10a46dba4034f1da864933e : FAILURE in 1h 33m 26s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8e367acdc1584c0189dca8f98e781ac4 : SUCCESS in 1h 08m 04s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/94148487b37e4fcc8863f444d85b6dd8 : SUCCESS in 12m 11s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/a50711d2cdd742a4981e8bbc97d0e779 : SUCCESS in 7m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8a4deb75eefa4dc09dbddaeb8115eeb3 : SUCCESS in 5m 54s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab949b9868d0470fa8b9122b5e03ef25 : SUCCESS in 59m 21s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7c514db1cc0645e1a7b5d593926a97df : SUCCESS in 1h 12m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3a209efcc3b945d4aab28337f456e6df : SUCCESS in 45m 17s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2926689a5c484f16a19f186ccf37a9fc : SUCCESS in 43m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf0897169e13414e8d9196aaa0e5f265 : SUCCESS in 1h 36m 01s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/11c3a5a2c72e4550b8c5a8f8608c822c : FAILURE in 1h 41m 57s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f5627a1c563148e5ba04ff20779b35c7 : SUCCESS in 1h 31m 58s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/68f132a2c9bd47a9a3c1aef14b07be5b : SUCCESS in 2h 15m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd3c779e17494a1ea19dd7d21d81489f : FAILURE in 37m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cf9beb54a81840f0b63b427dbb40044a : SUCCESS in 1h 19m 52s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ee4a6856caa044b0900268c1dd0b5fd2 : SUCCESS in 1h 57m 10s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/89686059e7b745beaf6202bb55849bec : SUCCESS in 1h 19m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/495ea8c3fc4c4379a647f53427bd70c2 : SUCCESS in 46m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"622a856f30631c3923ee0cde1a07a409b466fe8e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-17 09:54:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"66d5120c15bb31f3b81809be5b6ec4fcab68a195","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 12:38:42.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cf7ff8442d643759bba9ce7556aca23 : RETRY_LIMIT in 1m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcd6b8b94d5a425ca1fba742daae6f26 : RETRY_LIMIT in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/439b875f5ada4ed8a81dad28885dc16c : RETRY_LIMIT in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/200742d1df524f0386f9b08e9983e010 : SUCCESS in 5m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/17121047ffa14c7a963f900b83259979 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6553bf3add024ade88ade4c84adeb2ef : RETRY_LIMIT in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2efc73140a545b3af7e6f0d808125cd : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6a3cf0ca69a4567b887cd8a3d3ceee7 : SUCCESS in 5m 00s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/22c59c3fca3a403b9a91d405545bd9ff : SUCCESS in 1h 23m 56s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ac28974a59954d8f887695d86e7a61cb : SUCCESS in 1h 23m 59s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe14b27ff8ac4349826b1015d82810e5 : RETRY_LIMIT in 3m 23s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ad4b1a0126b04f1e89943f6025348301 : SUCCESS in 1h 15m 09s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7de8611ddc38463eadd9e887e41d1019 : SUCCESS in 5m 02s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/848daf2b1d954c648ab20ee82467e046 : SUCCESS in 6m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bc82b7961e2445f5b0e5a128f0636a6d : RETRY_LIMIT in 1m 40s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b03569944c954b86a0f1dfa210bdd064 : RETRY_LIMIT in 5m 00s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9c0a8e553bbc47ef825a6cfc88ffd98b : SUCCESS in 59m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e9709c0fd7964d999d85c46e5606d943 : SUCCESS in 45m 32s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/cb7e744586b84974ba5053153dfefb68 : SUCCESS in 41m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/046f77d36d134a2f923ffb14da9dbf51 : SUCCESS in 1h 45m 50s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ce914abffd0544fc865a734fa7f5729a : RETRY_LIMIT in 4m 21s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/22451872518c4ca1abd8c1ac6400486b : SUCCESS in 1h 34m 40s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/856e2ceb19d7467aa445cf5d16e19ced : SUCCESS in 2h 20m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c2213ddb6e54f28b41902fa0f1b9859 : FAILURE in 34m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/de667ee243d447afbbd8ec5759e096d3 : TIMED_OUT in 2h 34m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/769737147c9449ae8752460b058eb94d : FAILURE in 1h 34m 21s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2032711b6c3946e1b39faabf3678de73 : RETRY_LIMIT in 5m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b5e934fc7d8b4753a1844ddeeede93d4 : RETRY_LIMIT in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5912fc6b388fc58b7e5bb2932c2e5daceb068a34","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-17 13:28:31.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fb9452b8f8ad7f442c554d814e536fd038d89f75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 15:52:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23af6ac28a79497db8f72aa13ab40bc9 : SUCCESS in 7m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbf50889a62149439c395ccf78a32408 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e34911e8b6f4283a657b7ef1814bdca : SUCCESS in 6m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e1b5099fd1b4322be35db9fccf493a9 : SUCCESS in 4m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/de84f3aed3e0436bb3408150f15aba40 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0cd1294fca542ad895692daf0eea4cd : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57acec518e8d445a9a3167a5933d0470 : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c54d3b0dcd0432080f4676984453919 : SUCCESS in 5m 05s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a46ec195dc78469db79ec3fb1cda165a : SUCCESS in 1h 30m 02s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/977d9f5fddbb4b9499533d5d4991c434 : SUCCESS in 1h 25m 31s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ecf1b8ede32446ea98fe128b1da07e1 : SUCCESS in 1h 28m 18s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/80d2135e582e43fb97aa52507e15f4ac : SUCCESS in 1h 29m 12s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bbc9f374080a43c7a58e97ab84c4131e : SUCCESS in 6m 38s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/fddbb27bfd5544abbf4320321c620d26 : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/106d971fdc1641648fbb58cc01a573be : SUCCESS in 7m 41s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fff57725d0144a1e925509ce496dee85 : SUCCESS in 1h 05m 33s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d738b01fa9564d37acfeedbe7702fe25 : SUCCESS in 1h 03m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a81015794168412f9c240468fd72db9e : SUCCESS in 47m 20s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a1df470b1f5b4779a0c695606e226fd9 : SUCCESS in 47m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e8162c18857343c188c0bb9b36f9ce31 : SUCCESS in 1h 42m 56s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b1447904def043cea1eb1fae5997e535 : SUCCESS in 1h 20m 16s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/115b63d0bfb8446fa2dff2cf7f63603d : SUCCESS in 1h 47m 34s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a4aa5f9bdb7f41d2bf0b4c183b15e1de : FAILURE in 11m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/170eec211bb6497ca3c22a06eca5761c : FAILURE in 36m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/28cfb9d8802d4f2aacfd9ff1e5954c69 : SUCCESS in 1h 31m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/898e1f41db9c43e3b85fc3b3195362d9 : SUCCESS in 1h 33m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/98e571b8a1bf43138c901a74b049faa8 : SUCCESS in 1h 34m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/30be071df0b24859b4e4e14154b48a6a : SUCCESS in 1h 02m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f0e6f31571cd1ae524e7cd9087e9ade21668a0a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 15:52:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"081617d9c101fe86570669e1200fe43a9d49397f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:07:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d4d5d8f85a9493da118799551b0e095 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd683b13da25462fbd9f9b6cdfa9776f : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f17aa33a8a9b4b8cac8767081e6c2a8f : SUCCESS in 4m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9a2f51253cc642c5ad38cfdb5e61abe8 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db4aa426627546edb8d112ddbcc0eaba : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31c31a94df647b68444edcbbd5604fb : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46c108a859fd47dbaefaf4d159c8f954 : SUCCESS in 6m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/462d7b3cf4144078a96e447de6e4358c : SUCCESS in 4m 32s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/dc7c051c819945708bc171000b07e2f9 : SUCCESS in 6m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/90d419bf8db842f7a7051c0f675a3004 : SUCCESS in 5m 30s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca137d31eef547c0938f9544f30ab193 : SUCCESS in 1h 15m 22s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5075dd3ee5d948cb9b66d8d2fd6d93b6 : SUCCESS in 1h 08m 56s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/e4e4958c44c84991962d280f3dc59da8 : SUCCESS in 1h 19m 09s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/2f614446bffa4d0ebcd3c060e5e7d2c2 : SUCCESS in 1h 19m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/22e41d92da6744d9b1b803d3f67b09ca : SUCCESS in 44m 49s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/735d383776ca4888823fa79853055de3 : SUCCESS in 57m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98bc13c6be234e22810a78de4b127625 : SUCCESS in 1h 38m 14s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4f6fce3c4fb34192837be17855cc569a : SUCCESS in 1h 29m 18s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/aca2d99cbb3b4e7197805a40a306fb23 : SUCCESS in 1h 49m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1d1ab8bea7a94a76bb6452704a570f35 : SUCCESS in 1h 17m 34s","accounts_in_message":[],"_revision_number":3},{"id":"512c22ece9ee848967f4c6edf6811f616ebac6d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:07:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"82d726caaa9bd9b0c2eba24c5024c1b8d7584b1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:10:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dfa20e791dc411889965842c6f835f6 : SUCCESS in 1m 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d812989d8fc54d95a28d36cada127909 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"46ccfc66f166906bd41836093ae9f6ebe32b8776","revisions":{"0a39311f88f4a7d2832c640753277158d709cbeb":{"kind":"REWORK","_number":1,"created":"2019-08-08 14:45:00.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/64/675364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/675364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/675364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/675364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/675364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/675364/1"}}},"commit":{"parents":[{"commit":"92b1e38a20da53e37577555e51edb1a1b3f5f6f0","subject":"Fixed down server issue after reloading keepalived","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92b1e38a20da53e37577555e51edb1a1b3f5f6f0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-07-01 14:57:14.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-08-08 14:43:06.000000000","tz":120},"subject":"Prevent UDP LBs to use different IP protocol versions in amphora driver","message":"Prevent UDP LBs to use different IP protocol versions in amphora driver\n\nThe amphora doesn\u0027t support mixing IPv4 and IPv6 addresses for its\nmembers and its VIP when using UDP load balancers.\nThis commit adds a validation step in the member_create and th e\nmember_batch_update functions of the amphora driver to ensure that IP\nprotocol versions are the same in a UDP load balancer.\n\nStory: 2005876\nTask: 33689\n\nChange-Id: If6fb3fde9b43ac82af46eaddc48ec7a3a5b95602\n(cherry picked from commit 56bb1e134d670901bdfe0526495f5794a55f139a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a39311f88f4a7d2832c640753277158d709cbeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a39311f88f4a7d2832c640753277158d709cbeb"}]},"branch":"refs/heads/stable/stein"},"9fe794240bf2c3b15ed9399b2702903acf75abe3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-06 06:57:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/64/675364/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/675364/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/675364/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/675364/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/675364/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/675364/2"}}},"commit":{"parents":[{"commit":"379c61a5a290afbd2f0da81c6adc7fb3ac8a8b46","subject":"Fixed down server issue after reloading keepalived","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/379c61a5a290afbd2f0da81c6adc7fb3ac8a8b46"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-07-01 14:57:14.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-09-06 06:56:46.000000000","tz":120},"subject":"Prevent UDP LBs to use different IP protocol versions in amphora driver","message":"Prevent UDP LBs to use different IP protocol versions in amphora driver\n\nThe amphora doesn\u0027t support mixing IPv4 and IPv6 addresses for its\nmembers and its VIP when using UDP load balancers.\nThis commit adds a validation step in the member_create and th e\nmember_batch_update functions of the amphora driver to ensure that IP\nprotocol versions are the same in a UDP load balancer.\n\nStory: 2005876\nTask: 33689\n\nChange-Id: If6fb3fde9b43ac82af46eaddc48ec7a3a5b95602\n(cherry picked from commit 56bb1e134d670901bdfe0526495f5794a55f139a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fe794240bf2c3b15ed9399b2702903acf75abe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fe794240bf2c3b15ed9399b2702903acf75abe3"}]},"branch":"refs/heads/stable/stein"},"46ccfc66f166906bd41836093ae9f6ebe32b8776":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-12 08:43:26.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/64/675364/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/675364/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/675364/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/675364/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/675364/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/675364/3"}}},"commit":{"parents":[{"commit":"cf53bc6543679a33016b1dd4805e6b8c5f4eca28","subject":"Fixed down server issue after reloading keepalived","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf53bc6543679a33016b1dd4805e6b8c5f4eca28"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-07-01 14:57:14.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-09-12 08:28:32.000000000","tz":120},"subject":"Prevent UDP LBs to use different IP protocol versions in amphora driver","message":"Prevent UDP LBs to use different IP protocol versions in amphora driver\n\nThe amphora doesn\u0027t support mixing IPv4 and IPv6 addresses for its\nmembers and its VIP when using UDP load balancers.\nThis commit adds a validation step in the member_create and th e\nmember_batch_update functions of the amphora driver to ensure that IP\nprotocol versions are the same in a UDP load balancer.\n\nStory: 2005876\nTask: 33689\n\nChange-Id: If6fb3fde9b43ac82af46eaddc48ec7a3a5b95602\n(cherry picked from commit 56bb1e134d670901bdfe0526495f5794a55f139a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46ccfc66f166906bd41836093ae9f6ebe32b8776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46ccfc66f166906bd41836093ae9f6ebe32b8776"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
