)]}'
{"id":"openstack%2Foctavia~675576","triplet_id":"openstack%2Foctavia~stable%2Frocky~Iaa34db6cb1dfed98e96a585c5d105e263c7efa65","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"Iaa34db6cb1dfed98e96a585c5d105e263c7efa65","subject":"Fixed down server issue after reloading keepalived","status":"MERGED","created":"2019-08-09 10:24:20.000000000","updated":"2019-09-17 18:07:47.000000000","submitted":"2019-09-17 18:07:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675576-1568743667195-a663f06a","meta_rev_id":"9fb8e2607b9f016dda354f6d86d1d4b8fff2c59e","_number":675576,"virtual_id_number":675576,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2019-08-09 13:32:59.000000000","_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-17 18:07:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-08-26 15:32:23.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2019-09-17 15:56:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2019-09-17 15:57:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 15:32:23.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-17 15:57:55.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-17 18:07:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34296db59f5f5a671c89d7c3dafd2f2497780aa4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-09 10:24:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f70bdd438fa85bbd79314826efbce122b970ac80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 13:29:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/76/675576/1/check/openstack-tox-lower-constraints/7d3c641/ : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://logs.opendev.org/76/675576/1/check/openstack-tox-pep8/26f12e2/ : SUCCESS in 9m 27s\n- openstack-tox-py27 https://logs.opendev.org/76/675576/1/check/openstack-tox-py27/f3df548/ : SUCCESS in 4m 54s\n- openstack-tox-py35 https://logs.opendev.org/76/675576/1/check/openstack-tox-py35/7701cc6/ : SUCCESS in 7m 01s\n- openstack-tox-docs https://logs.opendev.org/76/675576/1/check/openstack-tox-docs/5d6af32/html/ : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://logs.opendev.org/76/675576/1/check/build-openstack-releasenotes/090fda3/html/ : SUCCESS in 5m 28s\n- neutron-lbaasv2-dsvm-api https://logs.opendev.org/76/675576/1/check/neutron-lbaasv2-dsvm-api/0787b07/ : SUCCESS in 1h 13m 22s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://logs.opendev.org/76/675576/1/check/neutron-lbaasv2-dsvm-py3x-api/be936fa/ : SUCCESS in 1h 18m 46s\n- neutron-lbaasv2-dsvm-scenario https://logs.opendev.org/76/675576/1/check/neutron-lbaasv2-dsvm-scenario/c693310/ : SUCCESS in 1h 34m 27s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://logs.opendev.org/76/675576/1/check/neutron-lbaasv2-dsvm-py3x-scenario/e83a84b/ : SUCCESS in 1h 37m 49s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/76/675576/1/check/openstack-tox-functional/a89314e/ : SUCCESS in 7m 22s\n- openstack-tox-functional-py35 https://logs.opendev.org/76/675576/1/check/openstack-tox-functional-py35/428f936/ : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://logs.opendev.org/76/675576/1/check/openstack-tox-functional-py36/dd28fde/ : SUCCESS in 7m 31s\n- octavia-v1-dsvm-scenario https://logs.opendev.org/76/675576/1/check/octavia-v1-dsvm-scenario/ee6169e/ : FAILURE in 53m 25s\n- octavia-v1-dsvm-py3x-scenario https://logs.opendev.org/76/675576/1/check/octavia-v1-dsvm-py3x-scenario/1fea8f4/ : SUCCESS in 57m 34s\n- octavia-v1-dsvm-scenario-multinode https://logs.opendev.org/76/675576/1/check/octavia-v1-dsvm-scenario-multinode/937a6da/ : FAILURE in 1h 37m 12s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://logs.opendev.org/76/675576/1/check/octavia-v1-dsvm-py3x-scenario-multinode/81c208e/ : FAILURE in 1h 27m 26s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://logs.opendev.org/76/675576/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/8a852ca/ : SUCCESS in 1h 09m 35s (non-voting)\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/76/675576/1/check/octavia-v2-dsvm-noop-api/93f77ea/ : SUCCESS in 53m 59s\n- octavia-v2-dsvm-noop-py35-api https://logs.opendev.org/76/675576/1/check/octavia-v2-dsvm-noop-py35-api/d1e6b32/ : SUCCESS in 53m 00s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/76/675576/1/check/octavia-v2-dsvm-scenario/3b078d7/ : SUCCESS in 1h 41m 28s\n- octavia-v2-dsvm-py35-scenario https://logs.opendev.org/76/675576/1/check/octavia-v2-dsvm-py35-scenario/9fc38c5/ : SUCCESS in 1h 39m 18s\n- octavia-grenade https://logs.opendev.org/76/675576/1/check/octavia-grenade/1679563/ : SUCCESS in 1h 33m 38s\n- openstack-tox-cover https://logs.opendev.org/76/675576/1/check/openstack-tox-cover/31034db/cover/ : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8a7e94d9d94af06e012b4fed818098603d4dd77f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-09 13:32:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"37eb4fb017e7b62345e85946463771d760ffc3c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 15:28:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/76/675576/1/check/openstack-tox-lower-constraints/f391237/ : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://logs.opendev.org/76/675576/1/check/openstack-tox-pep8/9decb33/ : SUCCESS in 7m 26s\n- openstack-tox-py27 https://logs.opendev.org/76/675576/1/check/openstack-tox-py27/8f5687f/ : SUCCESS in 5m 53s\n- openstack-tox-py35 https://logs.opendev.org/76/675576/1/check/openstack-tox-py35/5126e7b/ : SUCCESS in 5m 17s\n- openstack-tox-docs https://logs.opendev.org/76/675576/1/check/openstack-tox-docs/559df92/html/ : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://logs.opendev.org/76/675576/1/check/build-openstack-releasenotes/f4f3d99/html/ : SUCCESS in 4m 16s\n- neutron-lbaasv2-dsvm-api https://logs.opendev.org/76/675576/1/check/neutron-lbaasv2-dsvm-api/2d4e265/ : SUCCESS in 1h 18m 55s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://logs.opendev.org/76/675576/1/check/neutron-lbaasv2-dsvm-py3x-api/c3159c4/ : SUCCESS in 1h 26m 46s\n- neutron-lbaasv2-dsvm-scenario https://logs.opendev.org/76/675576/1/check/neutron-lbaasv2-dsvm-scenario/bd62ac1/ : SUCCESS in 1h 13m 24s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://logs.opendev.org/76/675576/1/check/neutron-lbaasv2-dsvm-py3x-scenario/18c5c89/ : SUCCESS in 1h 38m 01s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/76/675576/1/check/openstack-tox-functional/1a65ebf/ : SUCCESS in 5m 29s\n- openstack-tox-functional-py35 https://logs.opendev.org/76/675576/1/check/openstack-tox-functional-py35/9a5657b/ : SUCCESS in 7m 53s\n- openstack-tox-functional-py36 https://logs.opendev.org/76/675576/1/check/openstack-tox-functional-py36/e206d7e/ : SUCCESS in 5m 01s\n- octavia-v1-dsvm-scenario https://logs.opendev.org/76/675576/1/check/octavia-v1-dsvm-scenario/6b20deb/ : SUCCESS in 57m 20s\n- octavia-v1-dsvm-py3x-scenario https://logs.opendev.org/76/675576/1/check/octavia-v1-dsvm-py3x-scenario/fc166e4/ : SUCCESS in 1h 09m 44s\n- octavia-v1-dsvm-scenario-multinode https://logs.opendev.org/76/675576/1/check/octavia-v1-dsvm-scenario-multinode/5a38be4/ : FAILURE in 1h 18m 15s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://logs.opendev.org/76/675576/1/check/octavia-v1-dsvm-py3x-scenario-multinode/f80560c/ : FAILURE in 1h 24m 50s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://logs.opendev.org/76/675576/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/699ae6e/ : SUCCESS in 1h 09m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/76/675576/1/check/octavia-v2-dsvm-noop-api/bb84e89/ : SUCCESS in 41m 47s\n- octavia-v2-dsvm-noop-py35-api https://logs.opendev.org/76/675576/1/check/octavia-v2-dsvm-noop-py35-api/9f3e38a/ : SUCCESS in 41m 47s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/76/675576/1/check/octavia-v2-dsvm-scenario/1433613/ : SUCCESS in 1h 44m 31s\n- octavia-v2-dsvm-py35-scenario https://logs.opendev.org/76/675576/1/check/octavia-v2-dsvm-py35-scenario/c13f1dd/ : SUCCESS in 1h 37m 55s\n- octavia-grenade https://logs.opendev.org/76/675576/1/check/octavia-grenade/30db77f/ : SUCCESS in 1h 08m 15s\n- openstack-tox-cover https://logs.opendev.org/76/675576/1/check/openstack-tox-cover/2dfe90f/cover/ : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7e3d19ab4bb1e90c29b4396248c3c9c057b5cbfa","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 15:32:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3059d4db4663aa70d039987e918d322019f1d6ad","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 15:56:57.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"670d466930a279b3cb537c430e360926d0dcd94f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 15:57:55.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"361044871ef276551bd951294cc6ca23984e79d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 15:58:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3f9111c4f543732143249d00d76967519450ac8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:07:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c225c114cb244cebceb00d167ed10b5 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c9fd9cdd55b49908c38380fab681c3a : SUCCESS in 7m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2dc2f805ef74f9b93fdd1578d29690f : SUCCESS in 5m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f6ceeab0b4a14164b83d8cbbb76c7143 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4159635ff1948639e357e7847e87f34 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1ad30efd3d74729aeac8061977a31a1 : SUCCESS in 6m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2992496b376c4651a336251b0c10e38b : SUCCESS in 5m 32s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/447754553778411d9a7b5aa15d6134ea : SUCCESS in 7m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e50817f991734cbc97fa5ec2c0508d85 : SUCCESS in 6m 09s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/791543d1dd4f4475b6da57c02ef6e1d5 : SUCCESS in 1h 16m 35s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/64c9ea38f4454b26b49e8ceb0b2a4a1b : SUCCESS in 1h 06m 17s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/de0e3e47f4c94e05ab1692f40a70ae27 : FAILURE in 1h 30m 39s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/58387a8762444ea487f251401d13ddce : FAILURE in 1h 26m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/c585375f82f0420b845e2e5fbfe64d43 : SUCCESS in 1h 25m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72fbf22a8e994952a693b1c51bce01ae : SUCCESS in 58m 11s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/12b72b94f67040abae8269b810126781 : SUCCESS in 1h 19m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22796bca1a6442dabccd758dc6db15ee : SUCCESS in 1h 37m 03s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/39e0b84b273f429eb0ab6ed0d791b3c1 : SUCCESS in 1h 41m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5fb27233f1724931acba71c58bf2045f : SUCCESS in 1h 12m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3eb47080658e4f9d9aa4e35b58d63495 : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":1},{"id":"9fb8e2607b9f016dda354f6d86d1d4b8fff2c59e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:07:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a2491ad7e1befc1b84f8e427ca546390f61052f","revisions":{"7a2491ad7e1befc1b84f8e427ca546390f61052f":{"kind":"REWORK","_number":1,"created":"2019-08-09 10:24:20.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/76/675576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/675576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/675576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/675576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/675576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/675576/1"}}},"commit":{"parents":[{"commit":"f08e21fd6d59009465763c8049a2a82df8868d56","subject":"Fixed pool and members status with UDP loadbalancers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f08e21fd6d59009465763c8049a2a82df8868d56"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-05-23 11:55:19.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-08-09 09:27:57.000000000","tz":120},"subject":"Fixed down server issue after reloading keepalived","message":"Fixed down server issue after reloading keepalived\n\nWhen removing a UDP health monitor, keepalived is reloaded with a\nconfiguration without any checkers.\nBut if keepalived has previously detected a down server, the state of\nthe server is unchanged and it will never be added to the list of IPVS\nservers.\n\nRestarting keepalived on configuration change works around this issue.\n\nThis issue is fixed in keepalived (\u003e\u003d2.0.14):\nhttps://github.com/acassen/keepalived/issues/1163\n\nStory: 2005774\nTask: 33491\n\nChange-Id: Iaa34db6cb1dfed98e96a585c5d105e263c7efa65\n(cherry picked from commit 0a9f587015176723768a8a861265fda797c5d52e)\n(cherry picked from commit 92b1e38a20da53e37577555e51edb1a1b3f5f6f0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a2491ad7e1befc1b84f8e427ca546390f61052f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a2491ad7e1befc1b84f8e427ca546390f61052f"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
