)]}'
{"id":"openstack%2Foctavia~675577","triplet_id":"openstack%2Foctavia~stable%2Frocky~If6fb3fde9b43ac82af46eaddc48ec7a3a5b95602","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"If6fb3fde9b43ac82af46eaddc48ec7a3a5b95602","subject":"Prevent UDP LBs to use different IP protocol versions in amphora driver","status":"MERGED","created":"2019-08-09 10:24:20.000000000","updated":"2019-09-17 18:10:34.000000000","submitted":"2019-09-17 18:07:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675577-1568743671217-c3ff88a4","meta_rev_id":"5b8e07a7dc478a6ea870c52c520a3d0efc4fdded","_number":675577,"virtual_id_number":675577,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-17 18:07:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-08-26 15:34:05.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-09-17 15:57:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-09-17 15:57:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 15:34:05.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-17 15:57:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-17 18:07:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"810d6c26ea13bcc0b1a1573aba41b1fdea4b5a7b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-08-09 10:24:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69a27eea09a53180be81f6f0f1fb05365b53127a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 12:42:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/77/675577/1/check/openstack-tox-lower-constraints/2fc7028/ : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://logs.opendev.org/77/675577/1/check/openstack-tox-pep8/ed66d4e/ : SUCCESS in 7m 12s\n- openstack-tox-py27 https://logs.opendev.org/77/675577/1/check/openstack-tox-py27/dbf7622/ : SUCCESS in 5m 18s\n- openstack-tox-py35 https://logs.opendev.org/77/675577/1/check/openstack-tox-py35/29491a4/ : SUCCESS in 6m 34s\n- openstack-tox-docs https://logs.opendev.org/77/675577/1/check/openstack-tox-docs/2b9353b/html/ : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://logs.opendev.org/77/675577/1/check/build-openstack-releasenotes/22087bd/html/ : SUCCESS in 4m 45s\n- neutron-lbaasv2-dsvm-api https://logs.opendev.org/77/675577/1/check/neutron-lbaasv2-dsvm-api/a5f06a9/ : SUCCESS in 1h 20m 03s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://logs.opendev.org/77/675577/1/check/neutron-lbaasv2-dsvm-py3x-api/04933d5/ : SUCCESS in 1h 24m 52s\n- neutron-lbaasv2-dsvm-scenario https://logs.opendev.org/77/675577/1/check/neutron-lbaasv2-dsvm-scenario/5bc2713/ : SUCCESS in 1h 10m 13s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://logs.opendev.org/77/675577/1/check/neutron-lbaasv2-dsvm-py3x-scenario/61dc7dd/ : SUCCESS in 1h 50m 14s (non-voting)\n- openstack-tox-functional https://logs.opendev.org/77/675577/1/check/openstack-tox-functional/1dc659b/ : SUCCESS in 6m 39s\n- openstack-tox-functional-py35 https://logs.opendev.org/77/675577/1/check/openstack-tox-functional-py35/c8ed981/ : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://logs.opendev.org/77/675577/1/check/openstack-tox-functional-py36/56a63b1/ : SUCCESS in 7m 05s\n- octavia-v1-dsvm-scenario https://logs.opendev.org/77/675577/1/check/octavia-v1-dsvm-scenario/46fbe17/ : SUCCESS in 1h 04m 36s\n- octavia-v1-dsvm-py3x-scenario https://logs.opendev.org/77/675577/1/check/octavia-v1-dsvm-py3x-scenario/3bbb171/ : SUCCESS in 1h 10m 27s\n- octavia-v1-dsvm-scenario-multinode https://logs.opendev.org/77/675577/1/check/octavia-v1-dsvm-scenario-multinode/cf5b776/ : FAILURE in 1h 27m 47s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://logs.opendev.org/77/675577/1/check/octavia-v1-dsvm-py3x-scenario-multinode/2186b15/ : FAILURE in 1h 30m 56s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://logs.opendev.org/77/675577/1/check/octavia-v1-dsvm-scenario-kvm-centos.7/9c9d6c7/ : SUCCESS in 1h 10m 42s (non-voting)\n- octavia-v2-dsvm-noop-api https://logs.opendev.org/77/675577/1/check/octavia-v2-dsvm-noop-api/b8a082e/ : SUCCESS in 47m 13s\n- octavia-v2-dsvm-noop-py35-api https://logs.opendev.org/77/675577/1/check/octavia-v2-dsvm-noop-py35-api/1c6e5f5/ : SUCCESS in 1h 22m 57s\n- octavia-v2-dsvm-scenario https://logs.opendev.org/77/675577/1/check/octavia-v2-dsvm-scenario/eb11dde/ : SUCCESS in 1h 42m 12s\n- octavia-v2-dsvm-py35-scenario https://logs.opendev.org/77/675577/1/check/octavia-v2-dsvm-py35-scenario/b1395a8/ : SUCCESS in 1h 24m 00s\n- octavia-grenade https://logs.opendev.org/77/675577/1/check/octavia-grenade/77d8e5b/ : SUCCESS in 1h 19m 31s\n- openstack-tox-cover https://logs.opendev.org/77/675577/1/check/openstack-tox-cover/972575b/cover/ : SUCCESS in 10m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4a3c8e55e411425eae2cb09b4df597ea5fc341a4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 15:34:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e166464e43a44fa3f16dc5564d13f18ea6605ae7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 15:57:36.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"05758d532cd82e79145481b2e7a13683866ec736","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 15:57:39.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"31d63baa875a5fb113ea1fc1d1d45f4f126b5255","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 15:58:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6f6b5c4dbe5e78c33244b94e33f56ab8b19e0bcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:07:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb4c5430962c4492a94af1532855ae7e : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fea19b93bf7f4313a1b25375899e889f : SUCCESS in 7m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc6353fe721a44689db6d12e01a6b950 : SUCCESS in 5m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/84829d6fda48437c8d43a92a5416789e : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51cb0cfc95c0474f96649ce54cd89f06 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61df8bbb0d19467183f24229f1848947 : SUCCESS in 5m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/feb25f28ace64f99822ff1a7288a6365 : SUCCESS in 5m 41s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/3ebf1410cfbf45aeb9b94e00e082f53f : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/544274a4ee7548aa8fed0bd7f5391fd0 : SUCCESS in 5m 36s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4526b4de3bb84ba4bef7f784c16bf7b7 : SUCCESS in 1h 02m 39s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0b1920f0b4b74b06aa1139ddac6284c1 : SUCCESS in 1h 01m 56s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/e8a2102c25cb4e0d854817443cd2256f : FAILURE in 1h 11m 33s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/f57ca02b796443c9ab82c14a7f5e5284 : FAILURE in 1h 14m 01s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/3070ae206e82425eb980cf31adb204b1 : SUCCESS in 1h 11m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc5a8fe771b24650ab13dabb72643f6d : SUCCESS in 47m 40s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/32bdad644b4c4b08938062521ee34420 : SUCCESS in 42m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac7868a827d74a73b2a7d144cf3c0ffc : SUCCESS in 1h 44m 43s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/369de820366942d589ad563cd374ca08 : SUCCESS in 1h 52m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3339d399558548a1b3b8b1c94b507d47 : SUCCESS in 1h 16m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0edc24db1394748ba238627af84895c : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":1},{"id":"43c831f0dde9be493eead67ead5793b5b2061173","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:07:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5b8e07a7dc478a6ea870c52c520a3d0efc4fdded","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:10:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c84710c1f8bd43cd93f8e3bc15e2a04a : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/588d9aa103be4df3aa308c9c7afad1d1 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14169cbcfc1451b2f87464cb23e512216f72c340","revisions":{"14169cbcfc1451b2f87464cb23e512216f72c340":{"kind":"REWORK","_number":1,"created":"2019-08-09 10:24:20.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/77/675577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/675577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/675577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/675577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/675577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/675577/1"}}},"commit":{"parents":[{"commit":"7a2491ad7e1befc1b84f8e427ca546390f61052f","subject":"Fixed down server issue after reloading keepalived","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a2491ad7e1befc1b84f8e427ca546390f61052f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-07-01 14:57:14.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-08-09 09:28:06.000000000","tz":120},"subject":"Prevent UDP LBs to use different IP protocol versions in amphora driver","message":"Prevent UDP LBs to use different IP protocol versions in amphora driver\n\nThe amphora doesn\u0027t support mixing IPv4 and IPv6 addresses for its\nmembers and its VIP when using UDP load balancers.\nThis commit adds a validation step in the member_create and th e\nmember_batch_update functions of the amphora driver to ensure that IP\nprotocol versions are the same in a UDP load balancer.\n\nStory: 2005876\nTask: 33689\n\nChange-Id: If6fb3fde9b43ac82af46eaddc48ec7a3a5b95602\n(cherry picked from commit 56bb1e134d670901bdfe0526495f5794a55f139a)\n(cherry picked from commit 0a39311f88f4a7d2832c640753277158d709cbeb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/14169cbcfc1451b2f87464cb23e512216f72c340"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/14169cbcfc1451b2f87464cb23e512216f72c340"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
