)]}'
{"id":"openstack%2Foctavia~676013","triplet_id":"openstack%2Foctavia~master~Ifc9ce78755b33b6ad88b1a0c4f100c27f35bd77e","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ifc9ce78755b33b6ad88b1a0c4f100c27f35bd77e","subject":"Set Libvirt hw_rng for devstack","status":"MERGED","created":"2019-08-12 19:44:36.000000000","updated":"2019-08-15 11:07:24.000000000","submitted":"2019-08-15 11:07:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676013-1565867244936-d21f4b70","meta_rev_id":"af8b011049420d832a71198adf62f5d8b43bfd64","_number":676013,"virtual_id_number":676013,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":2,"date":"2019-08-15 11:07:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-08-13 17:47:58.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-08-15 02:34:12.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2019-08-14 03:14:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-08-15 02:34:12.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 17:47:58.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-08-14 03:14:13.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-08-15 02:34:12.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-08-15 11:07:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12900a7e3eecd9bf67933278e4a6e8b619a946b8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-12 19:44:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8f30efae60587d16a30a6d9627c707e030792fa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-12 19:44:45.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"8be6369d637e8e8322cdb15b813ca8c6821756ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 22:13:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62014bed206d4da69fddcc9b08c7b5bf : SUCCESS in 8m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c4e0a04ebfb4634bfa8e4c8d80d6721 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbe6c5494e264e4da59061de60412258 : SUCCESS in 7m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54cb2ceed5054efaa847380790e56925 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59856256ce9c4d268ca3efbc48834d8e : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd4dd6e4baf84053921a112f617aa9d4 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc757bca4d684cfe95734f5b57d6c259 : SUCCESS in 7m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a8c71a69d3574d28bf82ba29f44b77a5 : SUCCESS in 5m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5a707fda39d0411db5f4ccd60a29e1d8 : SUCCESS in 5m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/15b614bf387641e6b850403ecedba370 : SUCCESS in 53m 53s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0396e74ccadc4228b96730504d3da769 : SUCCESS in 49m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1266e8715f44ecfbd6757cc8ccce864 : SUCCESS in 1h 30m 28s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3052186b3cab422999cd6f1d5f1fee8e : SUCCESS in 1h 41m 01s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/668b56ed850a46c6a55e317cf2d90109 : SUCCESS in 2h 20m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/287f3254bb3f4d509d803eedd862d399 : SUCCESS in 1h 20m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/95f6beba5b904a68b1e478a9824087bc : SUCCESS in 1h 19m 39s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a5e55bff791e4aaaaffe77678dcb6b37 : SUCCESS in 1h 29m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7d3a94e2aa9d4dc682df534d7f6569a0 : SUCCESS in 53m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8281cf2c83324fba841dcbac34c012fb : SUCCESS in 1h 38m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/df2c6c7b94834155b119830cb91af9d6 : SUCCESS in 1h 05m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f0e6fa83e24a409989e2cd1c8d418285 : FAILURE in 1h 03m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/bf5d1ed998614538bf6aa71981c310db : SUCCESS in 1h 18m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0187d635e1715c0e7f417278bff6c17284a4f75","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-13 15:44:20.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"1fa62238d87aba69f7113c35a3a84ddb7dace391","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-13 16:03:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"10d058d0f949ed7fac17d583813d2a1929bb1a49","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-13 17:47:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"560103707cd372370a0940a104671c3f77bf2d73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 18:09:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/216a6be992c6417a8866d44c6a8bb3a3 : SUCCESS in 6m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf1a91cc11564bad9e154073dac432f3 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1de48158d464242a7782657029af8fa : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ca6c0b724964904a25eb09f494c94c1 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/861b6c72db5848a2afcae1282acd0fac : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e06a73f638114a1a9c4263f9c37eb489 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3460575f214241afaa24948ba386f812 : SUCCESS in 6m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b681bb97af944de08d1d2ff790770558 : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/59a46fbcd8dd4413a044cab2790e270e : SUCCESS in 7m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7b243717f9b94a1ca97288cd7a0e9c9a : SUCCESS in 45m 34s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b95453c5dc7d418d8b6046eeb5e59100 : SUCCESS in 45m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9466e026c6ed460bbe6a027ca5ead2e9 : SUCCESS in 1h 43m 09s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/58639ba1691249bba7582f9959085547 : SUCCESS in 1h 47m 00s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6bfccd900fd94c0f98bf0b7471ba446c : SUCCESS in 2h 00m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2c483321c6b49bcabff4c64742a1d92 : FAILURE in 1h 32m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/cc87c40a708944c0a525a90f39f589e9 : FAILURE in 1h 25m 50s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/30c97119622647d58404682d209f7e99 : SUCCESS in 1h 19m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8fabb7674b294b39842f37bcef7e5759 : SUCCESS in 51m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7f6bd3905ac34d36868dca2f86944b3e : SUCCESS in 1h 34m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ea0efc7502464337a6a5c5fa53646d47 : SUCCESS in 1h 09m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/491ec4ec98d44809a8ae84e401455ec0 : FAILURE in 1h 08m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/05a550c0070549878c00d33d6d5fd478 : SUCCESS in 1h 24m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"96237f8de5df79832eee19aa27c6e60d67c38ccd","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-08-14 03:14:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fe95de30eeba6ddc2ac43b5d8d90c2511511d714","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-08-15 02:34:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a89837625f6df5dc6f891e1bae8fb856c4861feb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 09:05:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"edecb295c6ab7f791b16f7d0cf650fbb146b849d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 11:07:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"af8b011049420d832a71198adf62f5d8b43bfd64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 11:07:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38ab7b87a783473eb8906508d9f6eefe : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bb93410651546c4a6e4c4686fe6e989 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/553d7a38180e40f4888954e2f3633dc1 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9746de7f8d5942c590327ba1cc91fb96 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0092e6891cdf4727bf380d340b6ee6b0 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce8775e9bd734478a44e85639b9a8248 : SUCCESS in 5m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b7114e1c878146e49b6289c964443ed0 : SUCCESS in 6m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/31ad23b0ad3349d59a88b9e51838b1d8 : SUCCESS in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/11f42b54ea224ad9bdc61797b016e45b : SUCCESS in 49m 50s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c7e0a5e41d4c4deb89e3f22667b5e9b8 : SUCCESS in 45m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca48e3a149e44adea4e8d1bcdf616cd3 : SUCCESS in 1h 57m 55s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6f07732490fe4d969160c568a10c2b09 : SUCCESS in 1h 41m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/38e7a5f458054d41a774e097e8a14f00 : SUCCESS in 1h 16m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"37b50c1fef1ae5b181f701c8d8930c0d5a2d9458","revisions":{"e46e33f55400970f53a1dad9e84fa9a06877f596":{"kind":"REWORK","_number":1,"created":"2019-08-12 19:44:36.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/13/676013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/13/676013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/13/676013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/13/676013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/13/676013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/13/676013/1"}}},"commit":{"parents":[{"commit":"6e4da8506469de634a4ef57538c7f3ffed3cc937","subject":"Merge \"Re-enable the py36 tox environment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e4da8506469de634a4ef57538c7f3ffed3cc937"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-12 19:43:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-12 19:44:35.000000000","tz":-420},"subject":"Set Libvirt hw_rng for devstack","message":"Set Libvirt hw_rng for devstack\n\nDepends-On: https://review.opendev.org/675745\nChange-Id: Ifc9ce78755b33b6ad88b1a0c4f100c27f35bd77e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e46e33f55400970f53a1dad9e84fa9a06877f596"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e46e33f55400970f53a1dad9e84fa9a06877f596"}]},"branch":"refs/heads/master"},"37b50c1fef1ae5b181f701c8d8930c0d5a2d9458":{"kind":"REWORK","_number":2,"created":"2019-08-13 16:03:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/13/676013/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/13/676013/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/13/676013/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/13/676013/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/13/676013/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/13/676013/2"}}},"commit":{"parents":[{"commit":"6e4da8506469de634a4ef57538c7f3ffed3cc937","subject":"Merge \"Re-enable the py36 tox environment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e4da8506469de634a4ef57538c7f3ffed3cc937"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-12 19:43:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-13 15:59:19.000000000","tz":-420},"subject":"Set Libvirt hw_rng for devstack","message":"Set Libvirt hw_rng for devstack\n\nThis patch makes sure that the amphora service VMs have an rng\ndevice available to them that by default sources from\n/dev/urandom on the compute host.\n\nDepends-On: https://review.opendev.org/675745\nChange-Id: Ifc9ce78755b33b6ad88b1a0c4f100c27f35bd77e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37b50c1fef1ae5b181f701c8d8930c0d5a2d9458"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37b50c1fef1ae5b181f701c8d8930c0d5a2d9458"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
