)]}'
{"id":"openstack%2Foctavia~678180","triplet_id":"openstack%2Foctavia~stable%2Fstein~I4230feae5b6f0778b738bddbc69c5a0cd3e6930c","project":"openstack/octavia","branch":"stable/stein","topic":"re-add-FailoverPreparationForAmphora","hashtags":[],"change_id":"I4230feae5b6f0778b738bddbc69c5a0cd3e6930c","subject":"worker: Re-add FailoverPreparationForAmphora","status":"MERGED","created":"2019-08-23 08:33:15.000000000","updated":"2019-08-26 03:39:25.000000000","submitted":"2019-08-26 03:37:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678180-1566790629102-6ccbb2a0","meta_rev_id":"1ab5e5786377601ee3fe37e27d83b1a6e321d3e5","_number":678180,"virtual_id_number":678180,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-08-26 03:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},{"value":2,"date":"2019-08-26 01:34:19.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-08-23 16:01:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},{"value":1,"date":"2019-08-26 01:34:19.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2019-08-23 08:33:15.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 08:33:15.000000000","updated_by":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"reviewer":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"state":"REVIEWER"},{"updated":"2019-08-23 16:01:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-26 01:34:19.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-08-26 03:37:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"75a5221e1c78ffc775db103c738aeb7a0bc85038","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-23 08:33:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"375b70e30e8a1df5161db54808bdb5464e7cade6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-23 08:33:58.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 4494c62149a9ce562c5e7e87233119af0a781ada","accounts_in_message":[],"_revision_number":1},{"id":"e67b451d34391cedfe40d02a8797c6e07c1a58c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 10:48:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2feab6a7602641518f5bbd6647a69ec2 : SUCCESS in 9m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bf1789d932549cb8821d368c7ec56ba : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48b73455dc2743d9b3c65b31e7a86a42 : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ee909dad47d4360b49bf85e873660e6 : SUCCESS in 5m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b9fd40281198478ea06dc1f5c4cd89b1 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cad9ea7127904dafac8af052e44ca45f : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e56277f9359e422d9a98ff2154aae754 : SUCCESS in 9m 37s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a045ec8a7d2a4ec797678fe4d74fd56a : SUCCESS in 2h 01m 01s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/4a330d38074c49e6aae9d9b227d53494 : SUCCESS in 1h 20m 06s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9787d2e6a5f24da2a9b2738cb1760110 : SUCCESS in 1h 28m 01s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/89c872a5c75741b392ab42797fae2ee3 : SUCCESS in 1h 31m 27s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/be057247e21149bb86e76b2aeb21d031 : SUCCESS in 6m 12s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/cd67c2d990774d7cbf6eb36ba4112d94 : SUCCESS in 6m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4f4b491ffd884efa91aada30ecab7266 : SUCCESS in 6m 52s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e40b9e9390df48b2bffb062ce8488b26 : SUCCESS in 1h 13m 19s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/efaa47c588214c3597be1d01114cac8e : SUCCESS in 1h 04m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/20279d6e09cb40a2a78334876f2985f3 : SUCCESS in 46m 08s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a63747171df846a0810750679e934f08 : SUCCESS in 48m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0638a33b2314cf4a4b260a416285dca : SUCCESS in 1h 38m 18s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3c972ee613ff4d8ca10be8cebee3ef44 : SUCCESS in 1h 57m 19s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b2e9df17c690431a81fd414eeccb88e9 : SUCCESS in 1h 34m 29s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/386f5b91c3e24898a1debc211cd25d7e : FAILURE in 2h 06m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53d57c7ab9244775bede97ac0b3178a8 : FAILURE in 53m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b63cbd9963d545c7bbb277f95710e9dc : SUCCESS in 1h 44m 34s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/330ee88c2df24c54aaaab340d4e95bd1 : SUCCESS in 1h 45m 30s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7076f92721d448cfb7a42456c698674d : SUCCESS in 1h 12m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c436cb8bca454af0bfe68330a92af201 : SUCCESS in 59m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dbf3813eeddb74eb2631d908269ee397ad6393d8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-23 16:01:25.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"80a8f2b34272d855a9f39bb6008b8d7c2b05c02a","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-08-26 01:34:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a0cdcaaf057c5bb4d305308c95ce62ec0da5a636","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 01:34:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd9826ee577e5d49a7f7386c6e2416582017cac6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 03:37:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a9393d906d44631b986ea0cb6c6b13e : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1180602f9fa74b1f84175efffd15bc8d : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3191da44e34d4cc9866e876edcba321b : SUCCESS in 5m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d135843764be414d89ece62696fea7e6 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7816051b048d4e3a82f630a4b3aa4f83 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51660079f7374a84a004052fe56cd986 : SUCCESS in 19m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a3c16fe043d34c5d9774aff845f6a9ac : SUCCESS in 7m 02s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/a4813ab7bb6b47e8be3007c94467adbc : SUCCESS in 7m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0557eb82e555485d9b69400ba614f579 : SUCCESS in 6m 10s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c6b7bd2825047138b334e05675be26b : SUCCESS in 46m 16s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/776d5062c24a4d95b5fbce760d7ec301 : SUCCESS in 58m 04s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6a8390ca603e4c8b8e18d022a8029d9f : SUCCESS in 1h 10m 40s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ad8fcf6651bd4c09b33f41ad8b85e788 : SUCCESS in 1h 11m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b1dbc66951bf447fb354780612bd80b2 : SUCCESS in 37m 04s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ca56b4cda97e4ee290c41158468e346e : SUCCESS in 39m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ed197f045204e06ba29bf3424b40947 : SUCCESS in 1h 19m 12s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/451495284eca4d63b2359fe5f948278d : SUCCESS in 1h 19m 22s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/aeaa7cb1c7f546e2962e2c7e2c595dd1 : SUCCESS in 1h 39m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d62091d8b4274cf98b1b374d686c99fe : SUCCESS in 1h 05m 38s","accounts_in_message":[],"_revision_number":1},{"id":"3daba9b7387028709c3d1d08b76e0d1863a321d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 03:37:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c5cb38501236d07e6b50f4d81d2426d1d0031036","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 03:39:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deb244d52e1743c2b51f26dd4906ace6 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b7fe7f80c4f66096dc5b625807199463f3e6c64","revisions":{"4b7fe7f80c4f66096dc5b625807199463f3e6c64":{"kind":"REWORK","_number":1,"created":"2019-08-23 08:33:15.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/80/678180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/678180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/678180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/678180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/678180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/678180/1"}}},"commit":{"parents":[{"commit":"15358a71e4aeffee8a3283ed08137e3b6daab52e","subject":"Merge \"Update tox.ini for new upper constraints strategy\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15358a71e4aeffee8a3283ed08137e3b6daab52e"}]}],"author":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-01-16 12:44:34.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-23 08:32:29.000000000","tz":120},"subject":"worker: Re-add FailoverPreparationForAmphora","message":"worker: Re-add FailoverPreparationForAmphora\n\nThis tasks removes the dns_name option from ports, so that they can be\nplugged into a new amphora, which has a different name. This was lost in\nchange I04d267bd3cdedca11f0350c5255086233cba14ec.\n\nStory: 2006205\nTask: 35791\n\nConflicts:\n    octavia/controller/worker/v1/flows/amphora_flows.py\n\nChange-Id: I4230feae5b6f0778b738bddbc69c5a0cd3e6930c\nSigned-off-by: Sven Wegener \u003csven.wegener@inovex.de\u003e\n(cherry picked from commit c4c7387f8904075b961612b5a08aff47c51c081c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b7fe7f80c4f66096dc5b625807199463f3e6c64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b7fe7f80c4f66096dc5b625807199463f3e6c64"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
