)]}'
{"id":"openstack%2Foctavia~678337","triplet_id":"openstack%2Foctavia~master~I35f18b0fa8d9f49f1dfeac5eadcedae5beb49cc8","project":"openstack/octavia","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"I35f18b0fa8d9f49f1dfeac5eadcedae5beb49cc8","subject":"DNM: Test PDF building","status":"ABANDONED","created":"2019-08-24 07:36:46.000000000","updated":"2019-08-25 15:27:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"66f31b2b804640a77f95f97107beb07770fe0efd","_number":678337,"virtual_id_number":678337,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-08-25 06:31:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-25 06:31:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31fdfd12d08f2d29c57279206505cc0cd95ccd1c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-24 07:36:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fe9ce3823c7594386e47d5d4c9d9330bcb6d81a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-24 07:40:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6dbe2717b1b26202de2916e3dbb818c149bc2454","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 07:42:18.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"96dabb46899abfe85c817fb267dcea8be97f693a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-24 07:42:49.000000000","message":"Patch Set 2: Workflow-1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"06c95083c89e9c87ab1a725bef82d8bfd0b6947e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 08:09:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cb6b7fddfa44c96a8a808e84af8f873 : POST_FAILURE in 14m 17s\n- octavia-test-multiple-pdfs https://zuul.opendev.org/t/openstack/build/9cdadc607f554a848e5007904ee1dbaa : POST_FAILURE in 12m 53s\n- octavia-test-no-pdfs https://zuul.opendev.org/t/openstack/build/8b6386fe546f4f4c9b7e7e560cb6414d : POST_FAILURE in 13m 17s","accounts_in_message":[],"_revision_number":2},{"id":"9d29c26ffb4bb4f4fa5f9d975a6322a3b8c794fd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-24 08:37:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e6fe50283924cb947fd0cc53e57fe45a5c300587","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 09:03:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6be2a5ef1d7414dab3fbbc3654ebcec : POST_FAILURE in 14m 01s\n- octavia-test-multiple-pdfs https://zuul.opendev.org/t/openstack/build/a7a65606679c4c4ab44ebe46326b7a69 : POST_FAILURE in 15m 09s\n- octavia-test-no-pdfs https://zuul.opendev.org/t/openstack/build/ff09c8e2e3514372980ca555c5c5d033 : POST_FAILURE in 11m 21s","accounts_in_message":[],"_revision_number":2},{"id":"f6227300117842a890a2e2e93eaa00ba0fc0fda3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-24 09:32:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a9cfa88e40b465e8a60ecd174452a38d1de28a5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 09:43:14.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"fb6c70bcb3fb98f5d527f815422676a7417a197c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-24 09:43:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5272f472dcf61efbc48ed91475dd78fd7759887a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 09:57:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88cc458327074827b8ff42d7069d7125 : SUCCESS in 10m 54s\n- octavia-test-multiple-pdfs https://zuul.opendev.org/t/openstack/build/001a23e7f83d4049832bc55ad399acdf : SUCCESS in 11m 19s\n- octavia-test-no-pdfs https://zuul.opendev.org/t/openstack/build/9eb69d7c604948de81d2e60540818e89 : SUCCESS in 11m 20s","accounts_in_message":[],"_revision_number":2},{"id":"ceef7a49404854a213388905812c6ae9dc9bed7c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-25 05:47:54.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9d9c7a4040e058d8cacc7161e3f9b37e5e91c138","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 06:01:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de762558299443839c71d5b6e49d434e : SUCCESS in 11m 37s\n- octavia-test-multiple-pdfs https://zuul.opendev.org/t/openstack/build/a6258dad5c3d4f3eac602df3dca663a4 : SUCCESS in 11m 50s\n- octavia-test-no-pdfs https://zuul.opendev.org/t/openstack/build/324b35fea5b54abcb37533732fc2fdbe : SUCCESS in 11m 09s","accounts_in_message":[],"_revision_number":3},{"id":"fd48035ea7ec137563d3aae4b623f7d5d81081ad","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-25 06:01:59.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"210465a90ea2020df38184574eb507c1b62cfc15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 06:31:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9276847feda4b278cabc05343bd0e8b : SUCCESS in 13m 48s\n- octavia-test-multiple-pdfs https://zuul.opendev.org/t/openstack/build/1e8168a803eb4674b48036150e0efd03 : SUCCESS in 18m 52s\n- octavia-test-no-pdfs https://zuul.opendev.org/t/openstack/build/471f061553f048c78aff0de5ca1343df : SUCCESS in 17m 34s","accounts_in_message":[],"_revision_number":4},{"id":"2ae4015660c482bce23e1ccb06ba8f39003f574f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-25 15:27:00.000000000","message":"Abandoned\n\nworks nicely IMHO","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4aa0779665bc08dfd24578483ec56ae3eb9e18dd","revisions":{"251587fff4349398be59766b5f666f0a31ac7352":{"kind":"REWORK","_number":1,"created":"2019-08-24 07:36:46.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/37/678337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/678337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/678337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/678337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/678337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/678337/1"}}},"commit":{"parents":[{"commit":"ce2377be0d8ba5b142baa8ca5f5a31b81e97a474","subject":"WIP: Generate PDF documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce2377be0d8ba5b142baa8ca5f5a31b81e97a474"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-24 07:35:47.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-24 07:35:47.000000000","tz":120},"subject":"DNM: Test PDF building","message":"DNM: Test PDF building\n\nDepends-On: https://review.opendev.org/678160\nChange-Id: I35f18b0fa8d9f49f1dfeac5eadcedae5beb49cc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/251587fff4349398be59766b5f666f0a31ac7352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/251587fff4349398be59766b5f666f0a31ac7352"}]},"branch":"refs/heads/master"},"a9e2798954c76934e66783e5a26b17ec8127d8b1":{"kind":"REWORK","_number":2,"created":"2019-08-24 07:40:24.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/37/678337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/678337/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/678337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/678337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/678337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/678337/2"}}},"commit":{"parents":[{"commit":"ce2377be0d8ba5b142baa8ca5f5a31b81e97a474","subject":"WIP: Generate PDF documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce2377be0d8ba5b142baa8ca5f5a31b81e97a474"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-24 07:35:47.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-24 07:39:38.000000000","tz":120},"subject":"DNM: Test PDF building","message":"DNM: Test PDF building\n\nUses different scenarios to test https://review.opendev.org/678336\n\nDepends-On: https://review.opendev.org/678336\nDepends-On: https://review.opendev.org/678160\nChange-Id: I35f18b0fa8d9f49f1dfeac5eadcedae5beb49cc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9e2798954c76934e66783e5a26b17ec8127d8b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9e2798954c76934e66783e5a26b17ec8127d8b1"}]},"branch":"refs/heads/master"},"fd588083904d8ae812f7c69af46fc541355da472":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-08-25 05:47:54.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/37/678337/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/678337/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/678337/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/678337/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/678337/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/678337/3"}}},"commit":{"parents":[{"commit":"ce2377be0d8ba5b142baa8ca5f5a31b81e97a474","subject":"WIP: Generate PDF documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce2377be0d8ba5b142baa8ca5f5a31b81e97a474"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-24 07:35:47.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-08-25 05:47:50.000000000","tz":0},"subject":"DNM: Test PDF building","message":"DNM: Test PDF building\n\nUses different scenarios to test https://review.opendev.org/678336\n\nDepends-On: https://review.opendev.org/678336\nChange-Id: I35f18b0fa8d9f49f1dfeac5eadcedae5beb49cc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd588083904d8ae812f7c69af46fc541355da472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd588083904d8ae812f7c69af46fc541355da472"}]},"branch":"refs/heads/master"},"4aa0779665bc08dfd24578483ec56ae3eb9e18dd":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-25 06:01:59.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/37/678337/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/678337/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/678337/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/678337/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/678337/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/678337/4"}}},"commit":{"parents":[{"commit":"50efe7941bfe5424b2fd4bc23321f0e9f32831e8","subject":"WIP: Generate PDF documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50efe7941bfe5424b2fd4bc23321f0e9f32831e8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-24 07:35:47.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-08-25 06:01:59.000000000","tz":0},"subject":"DNM: Test PDF building","message":"DNM: Test PDF building\n\nUses different scenarios to test https://review.opendev.org/678336\n\nDepends-On: https://review.opendev.org/678336\nChange-Id: I35f18b0fa8d9f49f1dfeac5eadcedae5beb49cc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4aa0779665bc08dfd24578483ec56ae3eb9e18dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4aa0779665bc08dfd24578483ec56ae3eb9e18dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
