)]}'
{"id":"openstack%2Foctavia~678460","triplet_id":"openstack%2Foctavia~master~Ic414f34f376db8bba6e5c1b53341b9a88d8934dd","project":"openstack/octavia","branch":"master","topic":"modify-diskimage-elements-for-suse-support","hashtags":[],"change_id":"Ic414f34f376db8bba6e5c1b53341b9a88d8934dd","subject":"(WIP): Modify the diskimage elements to support SUSE packages","status":"ABANDONED","created":"2019-08-26 04:59:24.000000000","updated":"2024-05-07 16:01:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"61854eae1362b440b2f5493a9bb21ddbccbe5a47","_number":678460,"virtual_id_number":678460,"owner":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-15 23:27:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-05-15 21:27:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 21:27:44.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-05-15 23:27:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7513a220622c744e3ac6463c7bf362de459f0391","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-08-26 04:59:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40f0c696d8caa32910e04203a958836f0d446dec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 06:19:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d364da6ecb0946658a2cb98a732ece7c : SUCCESS in 7m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb659d4dd28f4520a2aa03f15f5bdaad : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93abfb8a8c754da2846d723beb36e1da : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e80e34b58d8047809032df3ab2401ed3 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6971a2b0e1b4ce88aa76bca27fc7f4c : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/935a2769d64e48b7941604fa0d076866 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c52350f2886442dbaf8f7f246b13114b : SUCCESS in 7m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d7bd490f4fea412293a50004880adf16 : SUCCESS in 4m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c23c1229699a45799e90b77116ea5937 : SUCCESS in 6m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7bde3a16da0a4a2da8acbe74f773ce04 : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8ff956a5770e43a997f091cb530f91f0 : SUCCESS in 5m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1242c51d60fb4547a4d4122ee41ae9f1 : SUCCESS in 37m 45s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/69071617577449f4916cd251500f22d7 : SUCCESS in 43m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/304f0a6408744e1480e5d9d52dc5fa19 : FAILURE in 28m 15s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c6b41bc0c8b1432b8270f6016be7b035 : FAILURE in 30m 48s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1e705aa94bfe45f28f1d4b848af3a813 : FAILURE in 32m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a42a9a63789e4c4e898c10c698e8b924 : FAILURE in 32m 43s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/306ef7e99a03464db484f5766af951af : FAILURE in 30m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/995d99f06e354067982d9275be92994f : SUCCESS in 1h 18m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d6670022f0be4c0a8fbafe6da8c1d26c : FAILURE in 31m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4879623dbd1a45f699c6830cf1c124c4 : FAILURE in 28m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b778ae71121496191c14cd03e527933 : FAILURE in 27m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1dbfdd88c5c545f7966f35ee95523f0c : FAILURE in 28m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/83310017ea94490ba7583f2a30af7f7d : FAILURE in 36m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"20626572eb57602ce04c80be5c1b053785671f80","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-08-26 19:04:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"70582e44ba795c103130fdef4196f3afe88fb7c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 20:37:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc085396d1fb4dcaa1b8788322512b05 : SUCCESS in 8m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d93ac625ce9741ce87a718fd65bf2122 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d69bd6a072b48ddaf8c3fa6bc727d52 : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/81d674a7e5de4c30a8aea031501ec0c8 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/956687292af645198644c1e10208ac2e : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0acdfcbfc12244158817f39a26537745 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb24543812244c80b5ac5336c6c097e9 : SUCCESS in 8m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fb96552f9dd446d090b293520b4616da : SUCCESS in 5m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/679995e6f64c433fa8c58ddfec0d2c7a : SUCCESS in 6m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/de73372a424347da9c51719f21df2a0d : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/00b9ad3267fb4be3be6601506258d861 : SUCCESS in 6m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a65d805c05464bcb8c5e9575a2dc43f6 : SUCCESS in 43m 43s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e3d1269ceb1f43f7994383fba2c09609 : SUCCESS in 48m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb7cf5f43f784adf8f36ca1601e8f43b : FAILURE in 32m 57s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fc0e8f66a2894eb393a8681cc2b1b6c2 : FAILURE in 31m 25s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fe5d7546922a41b8aecd91c22f4a3f50 : FAILURE in 37m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/efffad2361d242808808bfbe376337b3 : FAILURE in 34m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1045a5af2c624bf3b2250f49a7e1b5ef : FAILURE in 36m 10s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bff40fd4222b48378febc661c1396afc : SUCCESS in 1h 27m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b4146596875541c8887d65303b871ed2 : FAILURE in 34m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/73f2d40ef07c41a99b9bda5097848fa3 : FAILURE in 31m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b65866d96df48c894136225dd92704a : FAILURE in 32m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a681e2869c164bed973f494a6eeb4a5d : FAILURE in 31m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e3712fd5dd274417905059b9b86072af : FAILURE in 31m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74f898b3e5cee0cd9852d00b5afd92367eb0673c","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-08-26 21:06:18.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"8b3ef0ad6c9c166e3c874137bb29fa5fe9159312","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 22:28:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d31fcd03307e4409ad1934d70f1ef554 : SUCCESS in 7m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e183fe7d1044c77aae10c11370efdfb : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36205331de31444982ef1bbeb4d515bc : SUCCESS in 7m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/90000ec8aac5472abed0b9a1de4e05b2 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4b8434539df42f591bf40ab270d5f90 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c1d31e92df14c5baf97a03aeca9a76b : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/157edd852ab14346831688d7f9cc0a7c : SUCCESS in 7m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/707cb6f0563a4583b98b9ad0fe0460da : SUCCESS in 7m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d685d65263384b9187b2a9721516223c : SUCCESS in 7m 10s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1b54b3472ae84007b287c7eaef4e5692 : SUCCESS in 6m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3c77656aab6f428884d9d09b5d291f73 : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3dc2bd5ef7de423ebcdf888eb9b6acc2 : SUCCESS in 49m 35s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/aca2c7189f8841a494ab6d6344a4ed55 : SUCCESS in 47m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/391c1838a5024ab6817ab4ecafd223be : FAILURE in 30m 43s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4a0a667682744c26991be8e6db60e0a8 : FAILURE in 43m 29s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/af42dcdfab2d4160b2b41ea922f432a0 : FAILURE in 44m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1cb291c87dd84fd2bf5a89ae676b6084 : FAILURE in 46m 34s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/04ad5d29e8ad47b0a5df6db9cb7ed163 : FAILURE in 38m 11s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/18cb60dcfabb47bf9c3c11c3a9319d18 : SUCCESS in 1h 21m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aae1b8e3cd8d42bc8dd8947fac5f5522 : FAILURE in 35m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/983b6c7a11314c2b910b10d799e2d636 : FAILURE in 34m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf63832e9df2496da7469f64768d847f : FAILURE in 34m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9614c9e38246484aa814d4f13bab24d9 : FAILURE in 31m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/767c2575a6eb4cbead748295d262d2c8 : FAILURE in 39m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8fb4ad2ac52c774ad7f595519277635e4d3d8235","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-08-27 15:54:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0ecdfe5ba91ace290e040caea5ba369897e906e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 18:38:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21e47d9c286c4199957c3fcad48f56b9 : SUCCESS in 8m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40f3a9a043b14a5fb7a218b69072b783 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14cfdaff65e442949a7fa4d7d7db621e : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f38463bf481545ad91f514c01c65b2bf : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0091d186def4a56a867658f78304726 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/168ae64c90b4464b9d16e01e3ad09a55 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f1294852154514b31a5a8e34f1336f : SUCCESS in 9m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a7189e7f44ad425392446129e4253bf4 : SUCCESS in 5m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d99d07a539a74e58a6b3e759e0182488 : SUCCESS in 8m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3d35c891af97428c95a9d073ac1d30f8 : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/427d2e8eb6b64342ac419f1df68cc3d6 : SUCCESS in 6m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/345e3c3cfacc48409bc9f4c275c78b99 : SUCCESS in 48m 05s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3a7d58a364e7419db3e3b57684746075 : SUCCESS in 40m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d974ba4b89e54eea97d4e2a0e5f06b82 : SUCCESS in 1h 54m 15s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/50dcd91247cc462db877ac5427301c96 : SUCCESS in 1h 40m 11s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e083cf0c3f2f4e83975cd03ab9534589 : TIMED_OUT in 2h 36m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2ab534f8c7a04989b9501cbbf1826ce7 : SUCCESS in 2h 12m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/70070bb06691406fa90487311d566020 : FAILURE in 1h 30m 53s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/49ab905b896c4c9fb9ff6866357e33ca : SUCCESS in 1h 15m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/25bbeb6c55724c8fb9051558711e1c1c : SUCCESS in 1h 01m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5f98198f2f1e4c8ba61d1e5399f6e432 : SUCCESS in 1h 44m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1b06100e9f084e3baf16eb0fe3c2f2f0 : SUCCESS in 1h 13m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1ff7114ff2df463e89319db18ed85879 : SUCCESS in 1h 18m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/56acb0bfc1dc4ecd8a38436c55f2f0e6 : FAILURE in 34m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9b23b824f8f754a6368a53180334fa9352226fd9","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2020-05-15 21:19:46.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"dbd55abeb16195cfbfc794b68a8367459ae6f2cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 23:27:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c9ab33ce7334b04ad625bfdf19e8dd5 : SUCCESS in 9m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5081aee727748859498236c6792f2cf : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0932fa72837a496c8fea12ba433e2a6b : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/812084d5c3f649388f294d2a45c96194 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8379f27c1e1b4fc8956ed93b0b375b6a : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/addb24ac259b480cb56247b4e5cc0322 : SUCCESS in 12m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e6f8a419e05e4abe9d015edb2709c95e : SUCCESS in 5m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/16e705c0b1da494b8be8ce6871127465 : SUCCESS in 10m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/93249c8cb74444e89712ed61a877343a : SUCCESS in 9m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/70586b57d0de4b51947d1f58898d4b8b : SUCCESS in 46m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/61d33c150b9b45458c785dc350a58cfe : SUCCESS in 1h 49m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7535ce28b94a4a339e2740ce66639c2b : SUCCESS in 1h 03m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8604b10217074b0c8c6b1e60aea92544 : SUCCESS in 55m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8b2ec5448a84db7861014e61eaba74a : FAILURE in 57m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4b8ae914fceb42ac98a990975d4d2131 : SUCCESS in 1h 43m 36s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/76aa46bc77f743858646315de387f742 : SUCCESS in 2h 01m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3a746efb93d24e6589c5d4eaa05758ad : SUCCESS in 1h 27m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2e0f9da562ee4223a99015867728591a : SUCCESS in 1h 46m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"61854eae1362b440b2f5493a9bb21ddbccbe5a47","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-07 16:01:17.000000000","message":"Abandoned\n\nThis patch has not been touched for four years. Abandoning","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"47f2f3d7f1fb7a7ac25c679bb490fdb9ae45aca8","revisions":{"61ef9cd6633d2a1bcabd725c5c44aee59066b558":{"kind":"REWORK","_number":1,"created":"2019-08-26 04:59:24.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/60/678460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/678460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/678460/1"}}},"commit":{"parents":[{"commit":"6e89fe783c3bd93e114f90447505f3c32b6c2a0a","subject":"(WIP):Enable devstack octavia plugin to support SUSE distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e89fe783c3bd93e114f90447505f3c32b6c2a0a"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-08-26 04:55:48.000000000","tz":0},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-08-26 04:59:06.000000000","tz":0},"subject":"(WIP): Modify the diskimage elements to support SUSE packages","message":"(WIP): Modify the diskimage elements to support SUSE packages\n\nModify the elements folder to support SUSE packages.\n\nChange-Id: Ic414f34f376db8bba6e5c1b53341b9a88d8934dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61ef9cd6633d2a1bcabd725c5c44aee59066b558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61ef9cd6633d2a1bcabd725c5c44aee59066b558"}]},"branch":"refs/heads/master"},"9d666570df14ec4a0ac7ce9830996f287ac9cc70":{"kind":"REWORK","_number":2,"created":"2019-08-26 19:04:02.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/60/678460/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/678460/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678460/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678460/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678460/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/678460/2"}}},"commit":{"parents":[{"commit":"6e89fe783c3bd93e114f90447505f3c32b6c2a0a","subject":"(WIP):Enable devstack octavia plugin to support SUSE distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e89fe783c3bd93e114f90447505f3c32b6c2a0a"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-08-26 04:55:48.000000000","tz":0},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-08-26 19:03:55.000000000","tz":0},"subject":"(WIP): Modify the diskimage elements to support SUSE packages","message":"(WIP): Modify the diskimage elements to support SUSE packages\n\nModify the elements folder to support SUSE packages.\n\nChange-Id: Ic414f34f376db8bba6e5c1b53341b9a88d8934dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d666570df14ec4a0ac7ce9830996f287ac9cc70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d666570df14ec4a0ac7ce9830996f287ac9cc70"}]},"branch":"refs/heads/master"},"2224134b4ff509168e995543a8a3f3459110642c":{"kind":"REWORK","_number":3,"created":"2019-08-26 21:06:18.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/60/678460/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/678460/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678460/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678460/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678460/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/678460/3"}}},"commit":{"parents":[{"commit":"6e89fe783c3bd93e114f90447505f3c32b6c2a0a","subject":"(WIP):Enable devstack octavia plugin to support SUSE distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e89fe783c3bd93e114f90447505f3c32b6c2a0a"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-08-26 04:55:48.000000000","tz":0},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-08-26 21:06:10.000000000","tz":0},"subject":"(WIP): Modify the diskimage elements to support SUSE packages","message":"(WIP): Modify the diskimage elements to support SUSE packages\n\nModify the elements folder to support SUSE packages.\n\nChange-Id: Ic414f34f376db8bba6e5c1b53341b9a88d8934dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2224134b4ff509168e995543a8a3f3459110642c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2224134b4ff509168e995543a8a3f3459110642c"}]},"branch":"refs/heads/master"},"a28823c854fa3acbe0124254ee4db75a8ccc588e":{"kind":"REWORK","_number":4,"created":"2019-08-27 15:54:01.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/60/678460/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/678460/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678460/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678460/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678460/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/678460/4"}}},"commit":{"parents":[{"commit":"6e89fe783c3bd93e114f90447505f3c32b6c2a0a","subject":"(WIP):Enable devstack octavia plugin to support SUSE distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e89fe783c3bd93e114f90447505f3c32b6c2a0a"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-08-26 04:55:48.000000000","tz":0},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-08-27 15:43:24.000000000","tz":0},"subject":"(WIP): Modify the diskimage elements to support SUSE packages","message":"(WIP): Modify the diskimage elements to support SUSE packages\n\nModify the elements folder to support SUSE packages.\n\nChange-Id: Ic414f34f376db8bba6e5c1b53341b9a88d8934dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a28823c854fa3acbe0124254ee4db75a8ccc588e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a28823c854fa3acbe0124254ee4db75a8ccc588e"}]},"branch":"refs/heads/master"},"47f2f3d7f1fb7a7ac25c679bb490fdb9ae45aca8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-05-15 21:19:46.000000000","uploader":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"ref":"refs/changes/60/678460/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/678460/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678460/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678460/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678460/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/678460/5"}}},"commit":{"parents":[{"commit":"c1f491e45454fa99e125dbd60b7ddc79868c715c","subject":"(WIP):Enable devstack octavia plugin to support SUSE distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1f491e45454fa99e125dbd60b7ddc79868c715c"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-08-26 04:55:48.000000000","tz":0},"committer":{"name":"Keith Berger","email":"keith.berger@suse.com","date":"2020-05-15 21:19:46.000000000","tz":0},"subject":"(WIP): Modify the diskimage elements to support SUSE packages","message":"(WIP): Modify the diskimage elements to support SUSE packages\n\nModify the elements folder to support SUSE packages.\n\nChange-Id: Ic414f34f376db8bba6e5c1b53341b9a88d8934dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47f2f3d7f1fb7a7ac25c679bb490fdb9ae45aca8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47f2f3d7f1fb7a7ac25c679bb490fdb9ae45aca8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
