)]}'
{"id":"openstack%2Foctavia~678525","triplet_id":"openstack%2Foctavia~master~Icddd2873abeb56a389a35356995df6dde70872b2","project":"openstack/octavia","branch":"master","topic":"patch-jinja2","hashtags":[],"change_id":"Icddd2873abeb56a389a35356995df6dde70872b2","subject":"Fix template that generates vrrp check script","status":"MERGED","created":"2019-08-26 11:51:45.000000000","updated":"2019-09-03 11:13:45.000000000","submitted":"2019-08-30 15:50:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678525-1567180233478-5a467fae","meta_rev_id":"29cf412e5a695db20d6a169eb307bf43b0d3bde4","_number":678525,"virtual_id_number":678525,"owner":{"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-26 11:51:56.000000000","_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-08-30 15:50:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":2,"date":"2019-08-27 00:22:35.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2019-08-26 14:29:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":2,"date":"2019-08-26 15:07:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":1,"date":"2019-08-27 00:22:35.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2019-08-26 14:29:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 11:51:56.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-08-26 14:29:33.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"},{"updated":"2019-08-26 15:07:22.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-27 00:22:35.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-08-30 15:50:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ef5fbb230d3fb7731e7299e8ad9672cc9fcfd10e","author":{"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"},"date":"2019-08-26 11:51:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d4713f180cbd34f6a7bfb29ceb48609ea1c29ec","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-08-26 11:51:56.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"a12b8b84be08d5adeed95e5ffd810ce52f2a99ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 14:22:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5941b7c7b63a4c33a9e8cb4e5ecdcbbe : SUCCESS in 7m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f18e1c4e4a1b4270baa744e4b1b0441f : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3942a869d9f54011b1b3622647cc7111 : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ad3d4d77941443f8614d94cf9836589 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34ba1915a4314e32b9b722cafd02e2f7 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bd77bf309374bc49d42a8e9b978c38f : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08c2cfdc6a894f34bb3b1619552dfcb6 : SUCCESS in 6m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/adc4e2469a124a9885e8bc383f8974cf : SUCCESS in 5m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f5ac0ce890c24ead92851295efaa3fd3 : SUCCESS in 7m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b17a89bcdedd42b994cd7a180d3720b1 : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/af28997cca4e49279e429c99edd53880 : SUCCESS in 5m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/633701a05f4d4f9381242e27a4699765 : SUCCESS in 43m 53s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9f80e3b6c1b047db8a84e9a23c85c81f : SUCCESS in 54m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/617eee880bd54da594cf4785d3e064b8 : SUCCESS in 2h 01m 28s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e32a65476e45451d9891c64532056ba6 : SUCCESS in 1h 44m 17s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2c91d7c7826c4b95aa2f029dda99ecba : SUCCESS in 2h 24m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b5d9e25745fa470d85f614d66ba3be28 : SUCCESS in 1h 54m 01s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff8470b4870c481ebaf4f1b50e4a2338 : SUCCESS in 2h 22m 31s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b0571181cb094f3896215b2c5949f53a : SUCCESS in 1h 21m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1feab3daae174a459c233bf45c25a63c : SUCCESS in 57m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/96ee2746eee44c52bccc14c05046999d : SUCCESS in 1h 40m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6305236573e3411391258a044eee3e54 : SUCCESS in 1h 16m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0176f1df84db46c9ab72e894cef7413e : SUCCESS in 1h 18m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/43e2a6b3fc7742e1a49820fd188a7d20 : FAILURE in 1h 17m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af291d339edc82581f766e0e5a4ffffcf8296bf5","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2019-08-26 14:29:33.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1\n\nThis indeed fixes the VRRP check script for me.","accounts_in_message":[],"_revision_number":1},{"id":"c1b5d9a2aa8f9fd52452dbea815b4b47ca3a4566","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-26 15:07:22.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"c2de5e833ceab1e33050ac3375ef6060486e7834","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-08-27 00:22:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54bce58159d775037a9227c69234c2364ab7f56e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 00:22:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f3e908788ed73124e0dbb2515002f7543a10132","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 00:51:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffb66e149e39400aa8dfc72edaed5dd0 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/271e55e418964835a1e9abda50be49e8 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/639c070af804427897ee083594268ea2 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/755a2ea671a54d39809c5ab80361322c : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/713f6ed7ca4c47909f8dcb3d77af2be7 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/186dd3a28fea49f3a9c1837b7d776e8e : SUCCESS in 7m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a48fa40cb3194a8b9f1b2810e2ccc7a6 : SUCCESS in 5m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f3cfd3da510c4d38b3e44ea108e37595 : SUCCESS in 6m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bb7a7d3195584ead884866422200f2f1 : CANCELED\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3b21aaef5ef44c6e9c52e27395a0400d : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22a205b77d8541179eada972df2724d3 : CANCELED\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0c7b36346d824ca481f452b8efdc99cd : FAILURE in 26m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/66dedd64627542bb85799f9a2ca5d40d : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"90d84a2617e90b5c27c96448369254011ce05105","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-08-27 02:22:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f325a8ebcdeae6bac9d92b48c1420b5ee72afdf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 04:44:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cc3e20e929a4200a60e18c9e1ce4f29 : SUCCESS in 7m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/685afeaf2ac94c7d8e5ab5c3fb69620b : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9590770644ac4471afa1868df5c51281 : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/22fba5a1f0c64b60b82990ef0107fdb2 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/884cda48ee484112b4610c961502fd37 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff8e58e62747469cab78437970681c6e : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c547aca841946769c6ae0172ff7663c : SUCCESS in 7m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6aa2892a856246b987dd77f813e6cdcb : SUCCESS in 4m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b48ced547a294e1facbedb6841377964 : SUCCESS in 5m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d4b4a92ffa024a8abe210b483b7be4fc : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9b838b74614f476fb7e248e6170936e6 : SUCCESS in 6m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/30585a3e2c7a4ca7a262e56aaff99ae5 : SUCCESS in 45m 45s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7f6be0119b6148b39c285751a1d38ae5 : SUCCESS in 40m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79f966a4939945829b9506543a99a77e : SUCCESS in 1h 36m 32s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b5f15983e5014cb4836a605d2066fca5 : FAILURE in 1h 44m 55s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d767dd8cc11845e0bb4260ff854664c3 : SUCCESS in 2h 20m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cae65079494b4b11899b951b251ecb8d : FAILURE in 1h 31m 15s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4416d3bb9ec04e099f7853c8cd73f403 : SUCCESS in 1h 20m 15s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6291c589872f4fa392eaf037818df9bc : SUCCESS in 1h 27m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4a1524b2e5ce401db96b2a7f3ca9ccc1 : SUCCESS in 48m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ffed637006d148ba962072171a8fbe6f : FAILURE in 1h 52m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7491eb3a3814fb2aa9a67f6e2c1fc54 : FAILURE in 1h 12m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/12a508c7aeed475e99732c7275d178b1 : SUCCESS in 1h 16m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/006e77325f3a4ab3b8f500eb2695cd67 : SUCCESS in 1h 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c1989972c0f100566d44fcff60ae4f046f65363","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-08-28 08:27:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ba8303d61b6ce96715a74ef5f24f1cce57410fc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 11:04:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e103fa0f752478ab6b1d0f2ab7bcede : SUCCESS in 8m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e38f7ed6b631487c90d7b0e5dce55201 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bbc60b442ab4dc4bbdd7f87bcd88409 : SUCCESS in 6m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/87be46f35d5d4876aa89abcd78d907e1 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/842437c835ba4644b66d06e1598d99da : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d88fded84b704b04b1198800029c8e70 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d894d28a97b542eaa765cf52b8639c39 : SUCCESS in 7m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/eaff300d69354d758f3dd63116046a50 : SUCCESS in 6m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2e2f13666b064151a2e99c694c04a669 : SUCCESS in 7m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ba964198fbac4b3486c8653a88ca8ed6 : SUCCESS in 5m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/159a77712a47490393dc265875970e64 : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e52e5e02745542578000a58d4034b763 : SUCCESS in 49m 14s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5879c667d94741438fc9dedb8f064f9c : SUCCESS in 47m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/747722de05f541f6b33c688b9c4a7561 : SUCCESS in 2h 00m 48s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ca57001a4a014c9ebba32afdb44573c8 : SUCCESS in 2h 01m 01s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d2c18668f1664e34882e9690a58f2cc9 : SUCCESS in 2h 29m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ad91975a8e3340bea0b68df4e9949459 : SUCCESS in 1h 41m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2320a69d0c94094b8730c4a85ff618d : FAILURE in 1h 35m 59s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4440c3779f5c41e7b891feb40d54185d : SUCCESS in 1h 23m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4b5c0bc434e04c89aa071620fe81bdd7 : SUCCESS in 55m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0dfe266197d3410194787b64a9c68ced : FAILURE in 2h 23m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a099cd2ac8b74d64bdb61e36b788b292 : SUCCESS in 1h 12m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6b71e04467444887bcc6a49935190116 : SUCCESS in 1h 21m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3b6c86d2a030472a9d6f816bb5065c66 : FAILURE in 1h 19m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d07793134f2e2d1860b2f59a090393f58659216","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 11:04:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d5897115eaa7f5afb8be3b3dfb03718538bcba5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 13:00:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d36f77c81f1449c7b925c296f72d58f2 : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d2a78f679c74c65931ba70d84747b94 : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4dcb46c7d3194f128d6e2113288a6bc9 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68697d7385794b0b9ec4cb9476e9b011 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6eb9bec2619a44caa924de3b895ba483 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a5e4c0dee1c4d03a0f90568bdc9ccdf : SUCCESS in 7m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/231fc93aa55e45faa4e47b6fc3b25d47 : SUCCESS in 5m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/392ee06bb72a43798e212f90024557f0 : SUCCESS in 5m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/84f0dfbe43324b38a4bef2df2ed9129f : SUCCESS in 49m 17s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/173decf118394a6fa0204cfffd71b040 : SUCCESS in 48m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d4c6fa69551844159bb0b091c0cb1ce4 : FAILURE in 1h 53m 42s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/57776f819e9c41b8bdf7f2ed8c902ea5 : SUCCESS in 1h 50m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/44655aa73d2d49fb9215276ad577d6b2 : SUCCESS in 1h 08m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e2984353463e90e20359fb86edffa7ca74a2f236","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-29 22:54:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2a5208d6cd9cc7c6023865aafbcf3668cb712b53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 01:52:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07561dcf1f974d0db986c26a30d67e19 : SUCCESS in 11m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3aa0747c2444bbf82bd3a0fffcf0a55 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9afdc3d032a240b08c04b25318c770ec : SUCCESS in 14m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f1ce42fc78b94ce6a3f597ed6b7c9aff : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b88b02dd66d44aaf9e2f573e8de92609 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ff366dfbd55410b94d2f6c5f837fa99 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b190249a98fb4bc59f18a1342a3c9143 : SUCCESS in 9m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e09864f02010486eb60eb93cc6d94397 : SUCCESS in 5m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1c2319a0a5464b3caa7e7d1db50690a5 : SUCCESS in 6m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5cf5fed4225247c2b9ef91b336e32867 : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1a78ecf337874e4cbfed2e608ca8b851 : SUCCESS in 13m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff7e49bf98e14c0fb7928cdcfd112e34 : SUCCESS in 43m 35s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/805b8a179f534efd9f182be034f41284 : SUCCESS in 47m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58b73cea667140d4a8bd73e5318af7ed : SUCCESS in 2h 12m 22s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a4d1f79cb6f644069764e63a9da80a36 : SUCCESS in 2h 29m 03s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2583f8a76f5c4403a8020984c0eb53c8 : TIMED_OUT in 2h 36m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/245565b1a96c411e84417d127e9f9b12 : SUCCESS in 2h 06m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3da5d2556a314d538d23fa3ad5dbf437 : FAILURE in 2h 33m 57s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5874ca8956964adeb3ec95fac6601217 : SUCCESS in 1h 29m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0f4e499d55584897b1181f57b151c56c : SUCCESS in 56m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d697b9f377c24b739db3564449ee8993 : SUCCESS in 1h 38m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f734fa9a75b34a9a8e4e6bc4274057a2 : SUCCESS in 1h 15m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1f85e5b8f56f42649254465495c08b01 : FAILURE in 1h 18m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/348b985aa10443ffaa478f3818146696 : FAILURE in 1h 25m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5cee9d9a350b53c867559b4ee958d91fc6ab85e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 01:53:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bbd5301f2d9213e611a1501ef9e1798322494412","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 02:40:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/149e0e6292b1401486d6d78bccfb9477 : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/997786cded0d48c8a267497b4587d9ed : SUCCESS in 7m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3309ea13fa6d42d797b83e8372891e4e : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53f0e644cf4d42ac8e6cd83525706144 : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91e3044be9e54734bbe3e28de02d381a : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53e57fb4fcc24724a40ad073a6573b5f : SUCCESS in 8m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a0a26446ad1d482084efdb3b004ee9bd : SUCCESS in 8m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/daba12b69f7c4c8f925dfdb60a32f96e : SUCCESS in 5m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/47c14b5d11d449a88caba592dfcea900 : SUCCESS in 46m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/80683f72176448569b19680224ff3e36 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bfca4bb4742a49b59e3f8450b840390c : CANCELED\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6af8409faba642a68ed6893f2a1618d7 : FAILURE in 41m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4303b7708abb47ed97c705b71aed4e45 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"a310b6ad2c4e66db4dcc1c8ffaf0a65eccc02fd8","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2019-08-30 11:01:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f402a7afba94a7538c9d1f5dac5757df88df4ba3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 13:40:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/605e299008bc4598ac8ab37739a86fca : SUCCESS in 7m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f294b699c54b4e79a64130572b3043a1 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeb523d542274b55b6a1a20caba37e7f : SUCCESS in 6m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f03d30011c941c6b8a2e526da2c5184 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/800ecb9a273944e2b023f785df84bdb6 : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b0fe3c6deff4863acfad5262e8dff66 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/144146d71c004a0fae64c668f0e61e07 : SUCCESS in 8m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f33b1f34567c4c4a9eced87f7831a03f : SUCCESS in 5m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f90c0d69c02e4b5a9814409481b3cd23 : SUCCESS in 5m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a1dba2604f9e4e3380392dc64c17e0db : SUCCESS in 7m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d218b6e2ea5f4bbc958c90108db0f406 : SUCCESS in 6m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7cccac57031643b8a7c1f97899269984 : SUCCESS in 48m 39s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9717d3642c45461594d5e612638c701f : SUCCESS in 47m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b51d394e00c84399bf31ad25e267ae7c : SUCCESS in 1h 50m 31s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6b48a38528b04133960da1efd305747a : SUCCESS in 1h 42m 16s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/39002a49195a475d9b8096927a934252 : SUCCESS in 2h 28m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/889381ec058b46bcbeb615314528880e : FAILURE in 1h 49m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/af0c5cfaea9444a4acced28cdee51ca8 : SUCCESS in 2h 12m 43s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f9dec9f5679f4a2bb8c34b9712280d55 : SUCCESS in 1h 19m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/12cb645d7b584b8fb052a8579c0446ce : SUCCESS in 1h 00m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/08c51e03463b409db6c738c8d64164a0 : SUCCESS in 1h 31m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bdccdae3efd54431aad4d27591857a9c : SUCCESS in 1h 16m 40s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/89278ad4809f4182bc036ab208cf0f65 : FAILURE in 28m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/35acfb811daa48f4981567ae01f620bd : SUCCESS in 1h 36m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"215d07209e47fb83dd6827fe9dcf14cabbbfa48b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 13:40:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10fd08168542c4b43c3cc755729afc3e16fcec6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 15:50:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27517eef944046da93b250d9aa13f959 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae5979b646204805b415cc9b27a5e23c : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00ec95093e3945ea8b730568092c3df9 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a08d774f9fc4a1f82eaa939028f2b99 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5bceefacd04d4954a4496f2936a40775 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da202024e4e34d83b8a5045b74063e0b : SUCCESS in 10m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0d6be8e71f9a45c5909da2013920f4e7 : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/805601ad76084cdc843f38ac2cb472a8 : SUCCESS in 9m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5379c45016d64a7c8c0f71abbefd272c : SUCCESS in 49m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/18e55cd90c444ff6adbcdd04eb98d56b : SUCCESS in 51m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3dfa50bc9cf4445abfce9b40856656cf : SUCCESS in 2h 01m 55s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/43c0048f7cf949aeb55bee85f179c9d5 : SUCCESS in 1h 57m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f401500dce194aaab906a893fa3a896e : SUCCESS in 1h 24m 23s","accounts_in_message":[],"_revision_number":1},{"id":"9cf7ca79e17126a601f49ae93da868285e510965","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 15:50:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"828330abff3bd83bc3bb413cdbb30128d51a162c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 15:52:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7f744a74550428c80ca81b1a88ed7aa : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c46e13b1ce74e5506299c8cb822c4b193b2a81ef","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-09-03 11:13:45.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 896b1cfde1761e78688e9216cb05331baddb1eea","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"905499162b33fbac1fd604688d21b73f559ab70d","revisions":{"905499162b33fbac1fd604688d21b73f559ab70d":{"kind":"REWORK","_number":1,"created":"2019-08-26 11:51:45.000000000","uploader":{"_account_id":30881,"name":"Rene Luria","email":"rene.luria@infomaniak.com","username":"reneluria"},"ref":"refs/changes/25/678525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/678525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/678525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/678525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/678525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/678525/1"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Rene Luria","email":"rene.luria@infomaniak.com","date":"2019-08-26 11:50:42.000000000","tz":120},"committer":{"name":"Rene Luria","email":"rene.luria@infomaniak.com","date":"2019-08-26 11:50:42.000000000","tz":120},"subject":"Fix template that generates vrrp check script","message":"Fix template that generates vrrp check script\n\nCorrect the inline comment to not include an empty new line at the start\nof generated /var/lib/octavia/vrrp/check_script.sh that leads to this\nkind of error:\n\n\u003e  Aug 26 11:49:32 amphora-12184e15-1ec3-4d80-98a7-c7d1ddb6716f\n\u003e Keepalived_vrrp[15265]: Error exec-ing command\n\u003e \u0027/var/lib/octavia/vrrp/check_script.sh\u0027, error 8: Exec format error\n\nChange-Id: Icddd2873abeb56a389a35356995df6dde70872b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/905499162b33fbac1fd604688d21b73f559ab70d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/905499162b33fbac1fd604688d21b73f559ab70d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
