)]}'
{"id":"openstack%2Foctavia~678534","triplet_id":"openstack%2Foctavia~stable%2Fstein~I728a9aa82986edef8128907a48c9df87d119fa8f","project":"openstack/octavia","branch":"stable/stein","topic":"fix/rollback-when-able-stable/stein","hashtags":[],"change_id":"I728a9aa82986edef8128907a48c9df87d119fa8f","subject":"only rollback DB when we have a connection to the DB","status":"MERGED","created":"2019-08-26 12:49:48.000000000","updated":"2019-09-04 01:38:49.000000000","submitted":"2019-09-04 01:36:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678534-1567561009542-4fe53e00","meta_rev_id":"8d8311547701d9ed05a85a60244f90d4cc6e33ec","_number":678534,"virtual_id_number":678534,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-04 01:36:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2019-08-30 06:52:07.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},{"value":2,"date":"2019-08-29 22:55:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":1,"date":"2019-08-30 06:52:07.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 12:49:48.000000000","updated_by":{"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},"reviewer":{"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},"state":"REVIEWER"},{"updated":"2019-08-29 22:55:45.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-30 06:52:07.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-09-04 01:36:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84eefa396569230e13f2f648ad8697f8f38815e4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 12:49:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9f059b1ea327c96420ec5b2bf6968276481d2981","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 12:49:59.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 7a3d6b0e79a3fd5895c8f13a0ff524fc781a45a1","accounts_in_message":[],"_revision_number":1},{"id":"6452b8aff250aaf1c7ce966cb68b7fc2237e8fcf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 15:02:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2e918c5fb79b602cbe30de7c86dcb42a69ba0be0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 19:00:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12bbb2344eac43f5ac9d4b7378e261ba : SUCCESS in 9m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/642be1adab094e4b97c649eae3219510 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5900d427ea247219f0502b9051aed1b : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59555da6b23d4e328268f0137e764497 : SUCCESS in 5m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/601eb24e6dd94b5184b156d12daa0e21 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d463de0c6a0b43c8b58b940d5499ed38 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50bd334c48374c18aada2686e244af5e : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49ae2415319f4088b7624c8304b340f2 : SUCCESS in 5m 30s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/32c706f7f097424899bd30da5bedd71b : SUCCESS in 1h 15m 46s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b1f36cae95384df5890ab0e4a61efbf6 : SUCCESS in 1h 22m 49s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4f3815657224a4bb03e2979998ac072 : SUCCESS in 1h 32m 12s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ac5ee6dc75334c9192c20f863a436ea6 : SUCCESS in 1h 34m 15s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/232c7a5fbd51474b8f81b5848a7d066e : SUCCESS in 6m 35s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e543b8085d33419f959abb22985c58a5 : SUCCESS in 9m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5059c338c8b24a9780308feec65154a7 : SUCCESS in 8m 14s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9212a3af33c4c348aeefc1359b67059 : SUCCESS in 1h 03m 24s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/dc32e732df5a4cc2ae0600448d2170d4 : SUCCESS in 1h 06m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a653cf5aed4a4f55971de8d49677f26f : SUCCESS in 42m 52s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e660715fc9654954b186c5d6e1b9afb7 : SUCCESS in 49m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8fc08f4072b43c99581f108a5e866d2 : SUCCESS in 1h 30m 29s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0c78fd5df807469a92f4d99fefbe7477 : SUCCESS in 1h 32m 08s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a0a6c66d6bcf43409286df572cef55a1 : SUCCESS in 1h 41m 42s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c30a3385a8424f35b6e01e21fcf26e34 : SUCCESS in 2h 25m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b196f2930041432e8462c431dad02b70 : FAILURE in 33m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f6107a38d7774162930b004078f27b9c : SUCCESS in 1h 54m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/20e99ff1cd714d3e954e41b32cdce70f : SUCCESS in 1h 47m 31s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ffe3c1ef012342f0b9cee4714526a2db : SUCCESS in 1h 19m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/53cb25c9d2a540848b76172160cfdaa5 : SUCCESS in 53m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57d2f9116da6664273ed8d129d1318e981d4419b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-29 22:55:45.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"7b281d2b0fbcb4c6e573dbd2db54969aca4470b9","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-08-30 06:52:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"808819ccc792f25b21d4ff7a93caa741987928f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 06:52:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"97dc7f3dc5528a9d915d9f3fcde5036830704273","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 08:34:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5bc375a5cc648d284a35c16d431e7d1 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1c24afdc60849d78498af5f95836dbe : SUCCESS in 6m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a91cd5add9749b989d6df6df8c68ac4 : SUCCESS in 6m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/079a6778f8b34d19818210d20170f293 : SUCCESS in 10m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c731fc122f3b431198bc4a8e59e08508 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61d3ad33b60846f986e4bb88634e42b9 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/671d321a78cb4101b309680e231e81cd : SUCCESS in 7m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d3bc391b324b45e19e03435bac8335c5 : SUCCESS in 10m 28s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/9c3f164c39dd4c66bafb969fde4b90a3 : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c9794c58346c44419ab38e0757803de1 : SUCCESS in 7m 28s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aa8272dc6de5496c81382c05c6519545 : SUCCESS in 1h 04m 08s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/cae58e2df5784e6baa8fdc658ec8bfe4 : SUCCESS in 1h 12m 42s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/5ff95f0ba3b14fc491846430c63aa22a : SUCCESS in 1h 24m 15s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/3efd134ca0054472a43ffd42cf587221 : SUCCESS in 1h 13m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8a91fb1b39f1458c90300be4c1c7f6d6 : SUCCESS in 42m 07s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a4bf24039355418eaf7d492c0d0e9d68 : SUCCESS in 46m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67c79cc566734cfd96190ec9ce19d5cf : SUCCESS in 1h 35m 24s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a423d3e3f1c84264a56d31bb1f8d4ce2 : FAILURE in 1h 34m 47s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/7ed69fdf40864ba5b1316b1957c779a7 : SUCCESS in 1h 31m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a6ca73e910504993b5cddda5517f9221 : SUCCESS in 1h 16m 33s","accounts_in_message":[],"_revision_number":2},{"id":"3ecba497a352a5526c5bd20d6ac74c97d4ceca9a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-03 21:22:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"29f49229908835459bdc73f205eec1f2ce5e5cb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 23:48:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d85a15eebaa423996ca2a5af1a467a1 : SUCCESS in 8m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fe0b717e2ae47aebb98ef817b10ba1a : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e3be48537ac4ac0969d2f65c89a6a3a : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a76ceec3d614f95ad4c43da7f583d44 : SUCCESS in 6m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4765e8c512b341ac9b9eca67c88b9684 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea4099219baa4474a4d0c39b6fdad508 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9609388632d74581bb038ba8a0bf877c : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1671b39be1134fe0b997873a03969160 : SUCCESS in 6m 36s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/2621a9a206a94b479e972d822c6d3e37 : SUCCESS in 1h 22m 06s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/7c4c618dc55241efbeeae31234e9f233 : SUCCESS in 1h 19m 41s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d55ea8a36ea74ef69799d9c29477f05d : SUCCESS in 1h 33m 10s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7e94b8fc412b46dbb7a06a79e3ad47f0 : SUCCESS in 1h 40m 44s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b5ef8db14d354ff29d470c61320d2f94 : SUCCESS in 6m 30s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/3f5fa0faea614adfbbea0fa9cbcad22b : SUCCESS in 8m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5b753aaf1c3b480db38cef13d8dbae7e : SUCCESS in 6m 55s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0bd9d5bdb7484339a8b267d12f937d78 : SUCCESS in 1h 17m 48s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/6e523dc77ccb418a847a2d45912f3618 : SUCCESS in 1h 09m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4672239253094353b5bb3fc272874ccb : SUCCESS in 52m 50s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/fa6d76516f73494bb27c4e5baf54502b : SUCCESS in 42m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e96374f6416349dc97d6d12f3d561ef6 : SUCCESS in 1h 34m 15s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/36c537e2738241e397e0d3612ad5846e : SUCCESS in 1h 40m 19s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/e745326da8d84cd2b45720215414780c : SUCCESS in 1h 54m 14s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9470dceb280246c8a16f93690a9380d0 : SUCCESS in 2h 20m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2f9adaba71340448240169c0cb9cc76 : FAILURE in 35m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0ac76fdd102b43ceb1212c7240c438e5 : SUCCESS in 1h 58m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/712f3cc2fb3840ca853624577d5b4d0d : SUCCESS in 1h 48m 07s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/016a7ecf5d3d4ebabc5bb2ffa81bec5d : SUCCESS in 1h 16m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5bdafcee0c684b50ba3c17db6e6d874c : SUCCESS in 50m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c6517318d5dc45f62c2c6a5f0d91fc7fa12a9c3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 23:48:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e92e81bde331c44ac12c9be1ad6200bc5276c7e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 01:36:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3575cd7fb6a6400781c7ba2e302f7bab : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e97c11217ffd4c399ff7f6b89ed7a897 : SUCCESS in 6m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1674cd0044914780bba682aa64e837c1 : SUCCESS in 5m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/51ba319b8c654af7b7e703a36754e79e : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/134fabd3a8984bb59d0ecf99653b4d78 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23208fdfbd554dcabb9ba9aa13aa0ff4 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6cf8c8473be45a79ec1a0d784b7c8be : SUCCESS in 5m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f7824e95a4414729a8a5bc49c8db66d1 : SUCCESS in 6m 43s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b5a13bd66d3444a483444557837b8dc8 : SUCCESS in 8m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a315f22ccb540d4901cd63c79ce71a9 : SUCCESS in 6m 35s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/677b528e538e4a3a93cfbdb26923a435 : SUCCESS in 1h 09m 10s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/27e56295a99144f9a9b76b3e8654a271 : SUCCESS in 1h 05m 07s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/e7cc8e0e185e482392a9735e692eb226 : SUCCESS in 1h 18m 53s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/4eb1fc2d87ed4dad8e83a1a6ea5ce7db : SUCCESS in 1h 22m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cee6e552d2624fc4b7372846defe9b32 : SUCCESS in 44m 17s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/cea62a306dcc497c989e8c975264a3b0 : SUCCESS in 44m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15eef24c0e6a4d5eba5f74da627e08a0 : SUCCESS in 1h 35m 23s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f1fbd23615aa40c48b30d89d2c0f960b : SUCCESS in 1h 34m 09s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/add1d0e88a8f41f6be733f70279a4f8d : SUCCESS in 1h 32m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3aeb36f9d2434ceb9f07d700189bd34a : SUCCESS in 1h 18m 24s","accounts_in_message":[],"_revision_number":2},{"id":"18704e8d3f7e375534c2112da79661daf0117374","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 01:36:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fa005808085d7af6eca8ba538228946f29d51a2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 01:38:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aea6768e3274e108c7b0b81f4ae44ae : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"40462b2b9f62e9734d3363873f6fe8dbd0c42bb2","revisions":{"15a40d31b7dbd1660826d7d078b94cca6f3dbd09":{"kind":"REWORK","_number":1,"created":"2019-08-26 12:49:48.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/34/678534/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/678534/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/678534/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/678534/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/678534/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/678534/1"}}},"commit":{"parents":[{"commit":"4b7fe7f80c4f66096dc5b625807199463f3e6c64","subject":"worker: Re-add FailoverPreparationForAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b7fe7f80c4f66096dc5b625807199463f3e6c64"}]}],"author":{"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","date":"2019-06-27 19:28:52.000000000","tz":-240},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-26 12:49:48.000000000","tz":0},"subject":"only rollback DB when we have a connection to the DB","message":"only rollback DB when we have a connection to the DB\n\nStory: 2006062\nTask: 34775\n\nChange-Id: I728a9aa82986edef8128907a48c9df87d119fa8f\n(cherry picked from commit feaa1ce02170cd2ff3eddfe28b1e61a6b07e15a6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15a40d31b7dbd1660826d7d078b94cca6f3dbd09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15a40d31b7dbd1660826d7d078b94cca6f3dbd09"}]},"branch":"refs/heads/stable/stein"},"40462b2b9f62e9734d3363873f6fe8dbd0c42bb2":{"kind":"REWORK","_number":2,"created":"2019-08-26 15:02:59.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/34/678534/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/678534/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/678534/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/678534/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/678534/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/678534/2"}}},"commit":{"parents":[{"commit":"4b7fe7f80c4f66096dc5b625807199463f3e6c64","subject":"worker: Re-add FailoverPreparationForAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b7fe7f80c4f66096dc5b625807199463f3e6c64"}]}],"author":{"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","date":"2019-06-27 19:28:52.000000000","tz":-240},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-26 15:02:50.000000000","tz":120},"subject":"only rollback DB when we have a connection to the DB","message":"only rollback DB when we have a connection to the DB\n\nStory: 2006062\nTask: 34775\n\nChange-Id: I728a9aa82986edef8128907a48c9df87d119fa8f\n(cherry picked from commit feaa1ce02170cd2ff3eddfe28b1e61a6b07e15a6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40462b2b9f62e9734d3363873f6fe8dbd0c42bb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40462b2b9f62e9734d3363873f6fe8dbd0c42bb2"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
