)]}'
{"id":"openstack%2Foctavia~678536","triplet_id":"openstack%2Foctavia~stable%2Fqueens~I728a9aa82986edef8128907a48c9df87d119fa8f","project":"openstack/octavia","branch":"stable/queens","topic":"fix/rollback-when-able-stable/stein-stable/rocky-stable/queens","hashtags":[],"change_id":"I728a9aa82986edef8128907a48c9df87d119fa8f","subject":"only rollback DB when we have a connection to the DB","status":"MERGED","created":"2019-08-26 12:50:10.000000000","updated":"2019-09-04 01:25:46.000000000","submitted":"2019-09-04 01:24:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678536-1567560254994-fd577826","meta_rev_id":"031b92846504620c242a89b782844a9e1d9b7ace","_number":678536,"virtual_id_number":678536,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},{"value":2,"date":"2019-09-04 01:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2019-08-30 06:52:50.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-29 22:58:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":1,"date":"2019-08-30 06:52:50.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 12:50:10.000000000","updated_by":{"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},"reviewer":{"_account_id":30587,"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","username":"squarebracket"},"state":"REVIEWER"},{"updated":"2019-08-29 22:58:15.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-30 06:52:50.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-09-04 01:24:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2819e9f001cf2600c056b4bc439e1cb51fa18dca","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 12:50:10.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"6435029e27f1ddb3fb15cbb3028e797209a39136","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 14:42:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be75779c88674ba291e6f1517e14170b : SUCCESS in 14m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4668a32d08854d92aa67a8d8d21686bd : FAILURE in 5m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f7a46ea0c68d49afa6e0379d4b38ec88 : FAILURE in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b3a815612cb42899370450de5eb8392 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4c10efefb8b4528b39045dcef28326e : SUCCESS in 5m 21s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/f89caf446efc41328f21f5fa4624a5a5 : SUCCESS in 1h 18m 01s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ee549d17fe8a4f25b7c6ec259ac75942 : SUCCESS in 1h 26m 30s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21a60c68e1034b598e296f9a2cc1ff7e : SUCCESS in 1h 25m 56s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1e376b6ebcd8474aac78f7e19acc2e07 : SUCCESS in 1h 22m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e05dff71eef64578b21e12570145b81a : FAILURE in 7m 34s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2ff13019301c4a1996f6c4089c8d98db : SUCCESS in 6m 14s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/441595a505444e2691a538c39fe91159 : SUCCESS in 11m 42s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aabac0d9136c45ed8fc69a56b429622d : SUCCESS in 1h 07m 09s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1e92ab3bc4a549658cbeb74c0f45c2e5 : SUCCESS in 1h 04m 22s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/e59e5c0f8dae4cd79deaddd2cd5d7834 : SUCCESS in 1h 22m 24s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/28f9c0aef2624f78b7f95ae84c85f75f : FAILURE in 13m 33s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/9002634f29b74179a4c7cac16ee01060 : SUCCESS in 1h 15m 30s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/58d281ba65814ce0af4e29350f2d6432 : SUCCESS in 43m 59s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/d3c5ff4ae13247238ac95ab5ec467feb : SUCCESS in 51m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6cb655b57a1d41c1ba4dc27f2b9dc1b7 : SUCCESS in 1h 42m 12s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/eaf292055634443b95000a026f7df9db : FAILURE in 1h 27m 05s","accounts_in_message":[],"_revision_number":1},{"id":"967c31f65c09a0c07fa7cdd4583ac6fdf01ffc25","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 15:03:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ff1d69ad300779a4b0489fa0413aa2d1a4c8a5a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 18:33:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebc56bd80684453c9daa7e60f19cc20a : SUCCESS in 11m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aafa098654604d51ae5cac9f310b0385 : SUCCESS in 7m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a9e544db2fab4a3b9dffea51bb9524b8 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abedf59b0992498abc909507ead32454 : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2aa553def8984f1487753bdbde93489e : SUCCESS in 5m 26s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/b101276c732649a5805caf7f69a5b474 : SUCCESS in 1h 18m 33s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0e8092f84d254aabb59aa47927edae31 : SUCCESS in 1h 22m 02s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5561eaed068a4e2e9b862cfdf4c7f262 : SUCCESS in 1h 26m 21s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/455210c4834740a795d90461bc03ad01 : SUCCESS in 1h 21m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f11fcdc0fe4443ce9690a8eadc3aa279 : SUCCESS in 9m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b3a4ba95c4ae4915b3e6209b7cb5ce65 : SUCCESS in 8m 41s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/495e4da9aaba4b0caf7f520c57b3f8e2 : SUCCESS in 7m 35s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f738d9177f6749ae82232f09593dfc0d : SUCCESS in 1h 02m 15s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/662536375bba451fa45f005d40b74484 : SUCCESS in 1h 00m 13s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/da215940af2c4cf78ef73d6237003590 : SUCCESS in 1h 14m 09s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/561f71f6d087418b90ebf72222f05eb0 : SUCCESS in 1h 18m 39s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/0c8f20c4397441e78894875e7f2ae62a : SUCCESS in 1h 06m 40s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/20a0aa6ed7bc4e27b687ab86b26c922b : SUCCESS in 47m 55s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/41f6970e2a5949159035fe3a119f876f : SUCCESS in 46m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/924f110e34f44897971b995410f5ff17 : SUCCESS in 1h 38m 58s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/dd9be0cd7eb6424cbbcf2c6bb12f32b5 : SUCCESS in 1h 35m 39s","accounts_in_message":[],"_revision_number":2},{"id":"5ad8a1a175b61ec3f93effe1d96b01c736fd1765","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-29 22:58:15.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"95f3686b760e3dff478d649f2898451cc511c5ac","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-08-30 06:52:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"44795dfbcd1ecee4e878b27168fca24cef2768af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 06:53:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"85c002c3ef1ef528024d5bdfe3f856c18df6b280","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 10:51:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e74802a779a64eb89cdc4a2d0e902992 : SUCCESS in 9m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/481725a969ca43a2a7fad357487c7508 : SUCCESS in 5m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ab387cd4fd3748a5a7861a3db65c8e72 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ae1b6a3aabf4b1baa7e37e03ab62856 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1400875197d740d892280c501d784b12 : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56b7b9f5728e437d937e6753383f77c4 : SUCCESS in 9m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8c95bb8ed047441483a0ffd4cd95c7fa : SUCCESS in 6m 40s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f86d18d1f910462d8c00b50a602cf97d : SUCCESS in 7m 38s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/07838ef9493948ca837c33ec9778a5c5 : SUCCESS in 1h 14m 27s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/696199efdc2043f291065704c643525b : SUCCESS in 1h 20m 54s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ae9a9b6527442eb89b56e8bda3edc8b : SUCCESS in 1h 04m 02s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5799cb38f0ae4021be37ecdac4585a15 : SUCCESS in 1h 00m 32s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/0b661fdffdda4ea699d4738704e672e9 : SUCCESS in 1h 12m 25s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/2e29becad436415ebecd06ef25d94092 : SUCCESS in 1h 20m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/59dddcf1deab4bf681e03d7feff4d220 : SUCCESS in 48m 16s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/fd834afeca57446abf0e19a3dafbc638 : SUCCESS in 47m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d0ab6fe916b4983820904d6d8de5fee : POST_FAILURE in 1h 44m 48s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/9e101c3ed48b41b3810a6e8627406425 : SUCCESS in 1h 37m 53s","accounts_in_message":[],"_revision_number":2},{"id":"e8715fce0a05be32a8e984ebb38cf33edcf91ae8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-03 21:26:56.000000000","message":"Patch Set 2:\n\nrecheck\nPOST_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"cbf285ef9f615241168c2de697f3f378962fc46b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 23:12:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cd264ad25c6406da1ea01041cf4dac9 : SUCCESS in 9m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f53d088c36564f7dbfae44d7c61bc068 : SUCCESS in 6m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/930bb821c21c42d290a37fefef78742c : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e3cd2a39524477dbc5dae0432fc002a : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09296379f1b046228dd5f561eccf9942 : SUCCESS in 5m 47s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/e6a1088e8c244526ba01858da2daa9bf : SUCCESS in 1h 26m 08s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/06d34c2604234bd0ac9b5c6ccaf77a1b : SUCCESS in 1h 23m 10s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a298f7ee3d74f7c8da013d5087669e8 : SUCCESS in 1h 23m 03s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d1bc6cef9e7c429089573b6393f791de : SUCCESS in 1h 29m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/146d4ceb862f478c9e04695c93ee54ca : SUCCESS in 9m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/22ff1006860e4ff09077a7555b85d70b : SUCCESS in 6m 45s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/44ca632da1c34032beb73a41ee0ef434 : SUCCESS in 7m 16s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73212e81567c48e1874981b5cf211b4e : SUCCESS in 58m 00s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/30f31ccee26d4e42aee6d5fb25e8b5b8 : SUCCESS in 1h 11m 32s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/7bc758ebc7ce430dbe5640e8b906a39b : SUCCESS in 1h 34m 54s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/cfad3eafac33467e91f4379654431e51 : SUCCESS in 1h 20m 47s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/20a64837edec4757a3a003f94743591b : SUCCESS in 1h 05m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8c6d9d5455114df9b699f4153f38e9ad : SUCCESS in 48m 01s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/2e8281fda6414817a33e057322d70bfe : SUCCESS in 47m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6d453897d194599a674a10b7965e641 : SUCCESS in 1h 38m 48s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/8b4bcc1e51b048078d9b4cd0d01b40b4 : SUCCESS in 1h 27m 58s","accounts_in_message":[],"_revision_number":2},{"id":"5150f2d2ec3a650f44c6b58629876c757f2270f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 23:12:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"32181021a8d7896701dfce6301712e5a7b8ca602","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 01:24:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"73c36874bfc45fc7d7a87ae839d040914b769c2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 01:24:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/900cc0601bc249c68aadb84dfddf861b : SUCCESS in 9m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73885a94db5f494ba2a7c5a59054bffc : SUCCESS in 8m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/13d4fa52d5f24d0780f5fc508e5a69c3 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a8f4b23ffc34c4ebc0cc7e3aed0540a : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a75f08a3b56492fba0e1012a1e2ff26 : SUCCESS in 5m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc4f92d5c64a4cf3a3a8840d22eafb77 : SUCCESS in 8m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/86902929c5d04b01b1d52523060aeb7e : SUCCESS in 6m 48s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/29bffaade7fc4ca09d140948aa3173b0 : SUCCESS in 7m 22s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/2c944c4f39b84d07b6071e091a4e5943 : SUCCESS in 1h 13m 51s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/e3172f56626d4585b5b76ba90b246993 : SUCCESS in 1h 30m 32s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95fe346adeb84f288df5f2e9bff27a08 : SUCCESS in 1h 09m 15s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d5196c02e7eb443d8a9c0f7e0cf59244 : SUCCESS in 49m 04s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/6808a964533a4c3bbc7d4682ca56e7a9 : SUCCESS in 59m 44s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/4451a5d6eef54ae1ad40bf32eb4b592e : SUCCESS in 1h 07m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b6c206578a414e6bbf6412c6282599ec : SUCCESS in 48m 52s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/73b39b9b585e4334a10dfe01e220d4dc : SUCCESS in 47m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74fc3e4b905340b1bf04eba448f8cc0d : SUCCESS in 1h 33m 42s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/89665f0bc974427b82e6924b5be86560 : SUCCESS in 1h 34m 13s","accounts_in_message":[],"_revision_number":2},{"id":"1391aba065e00fafd02c6f769c7b98cba434fb8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 01:25:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c997a35f17a74751bf6b865d0ebf6474 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d43c4f42ccf304d0dcbde35b97beb42fe434944a","revisions":{"08571186fffa758a707ba34530e7a25f5c82915b":{"kind":"REWORK","_number":1,"created":"2019-08-26 12:50:10.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/678536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/678536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/678536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/678536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/678536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/678536/1"}}},"commit":{"parents":[{"commit":"37aad5db40d688c6c718e8a7f892b61b0924143a","subject":"worker: Re-add FailoverPreparationForAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37aad5db40d688c6c718e8a7f892b61b0924143a"}]}],"author":{"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","date":"2019-06-27 19:28:52.000000000","tz":-240},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-26 12:50:10.000000000","tz":0},"subject":"only rollback DB when we have a connection to the DB","message":"only rollback DB when we have a connection to the DB\n\nStory: 2006062\nTask: 34775\n\nChange-Id: I728a9aa82986edef8128907a48c9df87d119fa8f\n(cherry picked from commit feaa1ce02170cd2ff3eddfe28b1e61a6b07e15a6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/08571186fffa758a707ba34530e7a25f5c82915b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/08571186fffa758a707ba34530e7a25f5c82915b"}]},"branch":"refs/heads/stable/queens"},"d43c4f42ccf304d0dcbde35b97beb42fe434944a":{"kind":"REWORK","_number":2,"created":"2019-08-26 15:03:53.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/36/678536/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/678536/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/678536/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/678536/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/678536/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/678536/2"}}},"commit":{"parents":[{"commit":"37aad5db40d688c6c718e8a7f892b61b0924143a","subject":"worker: Re-add FailoverPreparationForAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37aad5db40d688c6c718e8a7f892b61b0924143a"}]}],"author":{"name":"Chuck Wilson","email":"chuck.wilson+ubuntu@gmail.com","date":"2019-06-27 19:28:52.000000000","tz":-240},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-26 15:03:33.000000000","tz":120},"subject":"only rollback DB when we have a connection to the DB","message":"only rollback DB when we have a connection to the DB\n\nStory: 2006062\nTask: 34775\n\nChange-Id: I728a9aa82986edef8128907a48c9df87d119fa8f\n(cherry picked from commit feaa1ce02170cd2ff3eddfe28b1e61a6b07e15a6)\n(cherry picked from commit 3e1d0abece2b73c3472d83a55182542a9b6869ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d43c4f42ccf304d0dcbde35b97beb42fe434944a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d43c4f42ccf304d0dcbde35b97beb42fe434944a"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
