)]}'
{"id":"openstack%2Foctavia~678537","triplet_id":"openstack%2Foctavia~stable%2Fstein~I9f1b6e58004d7c2024f8d60d50ac998240abb6ee","project":"openstack/octavia","branch":"stable/stein","topic":"task/35814-stable/stein","hashtags":[],"change_id":"I9f1b6e58004d7c2024f8d60d50ac998240abb6ee","subject":"elements: add arch property for ``open-vm-tools``","status":"MERGED","created":"2019-08-26 12:50:56.000000000","updated":"2019-09-18 08:02:35.000000000","submitted":"2019-09-18 08:00:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678537-1568793640606-354c625b","meta_rev_id":"dd4e5730c7a8e274725198657314bdde378fb77e","_number":678537,"virtual_id_number":678537,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-18 08:00:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-18 05:37:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":2,"date":"2019-09-17 15:55:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-09-18 05:37:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2019-08-26 12:50:56.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 12:50:56.000000000","updated_by":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"reviewer":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"state":"REVIEWER"},{"updated":"2019-09-17 15:55:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-18 05:37:05.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-18 08:00:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d019dfab2f7a19ce3826f4eefcc35dc3d42a4ec4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 12:50:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5dc42e78b4505bacce8a3bdce15c60baa8b776e6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 12:51:04.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 4074dbd919f6e25096371e40caa0830d1ab19e6f","accounts_in_message":[],"_revision_number":1},{"id":"89fe028b3cb4f37917488b55939e746747b0cb82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 15:18:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c04a166c9e52455eb76967c07db7fc50 : SUCCESS in 8m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f3600bf09944034be2aa95721080bea : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40b684116d524a12bc336059555f7040 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ffb54e9976aa42a19d2f9fbaffa9495d : SUCCESS in 5m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6b851ed94ccc46e1938aef3226519bf7 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5f7699326b64cc6b6954b7735684cac : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/001106ec9010489e9dd68a91c004c94e : SUCCESS in 9m 08s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/9e668ad5498e461f9c025770512787f0 : SUCCESS in 1h 16m 49s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/e36b97f2b05c428cb066011d97185dab : SUCCESS in 1h 20m 14s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c33633b4339a4fbb9787edbeecbe9700 : SUCCESS in 1h 34m 20s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/141c443a88c346e1ab66a3b36ffdb3fd : SUCCESS in 1h 32m 00s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/69ea5ee2f98b4646a88760de5950f8ae : SUCCESS in 7m 42s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7a35809d9d8642d895dee5634c8eb53a : SUCCESS in 8m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8000431f2d564de89dc549c29fbfa373 : SUCCESS in 6m 52s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99c015376df4481f9d5ddcf79dbbc058 : SUCCESS in 1h 04m 22s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/853bc8c860df4d63bb8609be45b2fd84 : SUCCESS in 1h 03m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/20c79b3eef424ee89ab37239f34c0901 : SUCCESS in 45m 56s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c995ccf8887a4dfba93d90e128244833 : SUCCESS in 46m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7eb1a3fb07ab49a3a6e38c362a27954a : SUCCESS in 1h 31m 55s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d5477b4744a849b892210591b726174e : SUCCESS in 1h 32m 48s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b20e594a9d8c4837a8d3fc0b7730a88e : SUCCESS in 1h 37m 54s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/583c48dcc5694dd299cd1b8c76a8f1d5 : SUCCESS in 2h 16m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec9f354312474b79b9b713bdfcd2ecfb : FAILURE in 38m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b55004d27a7b4defbd096ee6dc826a75 : SUCCESS in 1h 17m 12s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/47bd8dd511264b03b5274324ac3d5599 : FAILURE in 1h 53m 14s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b777bdb59fd54a8cabfecc6263b65d18 : SUCCESS in 1h 14m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f846f31669154e68bd4726afe246aac2 : SUCCESS in 1h 09m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03db8a99e306313ae2b5e8e848ba87ade7e320d9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 15:55:36.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"e5874fc4cc86200c1ac753695399e887b1281c66","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-18 05:37:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a794cf12ca6fa51ee80c7e858c103247ac1b1300","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 05:37:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d71ac6e079823399c8488afbc271b54675791290","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:00:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3ec421f15a988fb40399f01d6a1e8cfb03f6369a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:00:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87312a2a7c1f427da7a459903dd8fc69 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/486b5c4ae3304d00a1a1068fdb0d81c8 : SUCCESS in 7m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/333302553ce24cc7817ae22bcd0c70a7 : SUCCESS in 6m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dca65aa31a23413fb3518bca07e411bb : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52e1397a3a1643098fea8f52fb7bf11e : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc1deaf73eab4edfa558c552186694c1 : SUCCESS in 9m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f96e84478a86406493e638f160374ea5 : SUCCESS in 7m 41s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/afbdc9ef7efc454b835179b54952f5d5 : SUCCESS in 8m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/90126a4174bd42ffbe725d4c344ca798 : SUCCESS in 6m 55s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d306eb1bb6454ad98515f13e80e56adc : SUCCESS in 1h 15m 50s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4af1846749d54342aa611e46f7e9c8e9 : SUCCESS in 1h 07m 59s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/655dc116e107437588d0c03c5fe7cf99 : SUCCESS in 1h 26m 14s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/58f35070a36c4c5a95d9cd895cbf6782 : SUCCESS in 1h 22m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/994c5d4a6b66426fbea53fc7b7dbb981 : SUCCESS in 53m 14s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4503d8b0ead3489381ce60caffa9f6a0 : SUCCESS in 52m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cfc46f4e57924d0996566e4143770bcd : SUCCESS in 1h 44m 56s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7bde40876efa4bf697f46c348a708b09 : SUCCESS in 1h 37m 02s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/41ac501b79a140abbee9860f393a6b3b : SUCCESS in 1h 38m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/797feb45dd6347c0b82aad27aa5b25ec : SUCCESS in 1h 19m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ac39c71c6750c0184500a18203ad41bc95fd1120","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:02:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c711387adc2b4c80a5628667b8db7012 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0571e0e4a0cbfd206fc410625794bd1a3f12f195","revisions":{"0571e0e4a0cbfd206fc410625794bd1a3f12f195":{"kind":"REWORK","_number":1,"created":"2019-08-26 12:50:56.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/37/678537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/678537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/678537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/678537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/678537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/678537/1"}}},"commit":{"parents":[{"commit":"4b7fe7f80c4f66096dc5b625807199463f3e6c64","subject":"worker: Re-add FailoverPreparationForAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b7fe7f80c4f66096dc5b625807199463f3e6c64"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2019-07-15 06:01:54.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-26 12:50:56.000000000","tz":0},"subject":"elements: add arch property for ``open-vm-tools``","message":"elements: add arch property for ``open-vm-tools``\n\nThe ``open-vm-tools`` package only exists for x86-platforms.\n\nTask: 35814\nStory: 2006219\n\nChange-Id: I9f1b6e58004d7c2024f8d60d50ac998240abb6ee\n(cherry picked from commit 5114a9a4f9d037d8ffb75342713fc27ff692235f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0571e0e4a0cbfd206fc410625794bd1a3f12f195"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0571e0e4a0cbfd206fc410625794bd1a3f12f195"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
