)]}'
{"id":"openstack%2Foctavia~678538","triplet_id":"openstack%2Foctavia~stable%2Frocky~I9f1b6e58004d7c2024f8d60d50ac998240abb6ee","project":"openstack/octavia","branch":"stable/rocky","topic":"task/35814-stable/stein-stable/rocky","hashtags":[],"change_id":"I9f1b6e58004d7c2024f8d60d50ac998240abb6ee","subject":"elements: add arch property for ``open-vm-tools``","status":"MERGED","created":"2019-08-26 12:51:04.000000000","updated":"2019-09-18 08:09:24.000000000","submitted":"2019-09-18 08:06:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678538-1568794013853-d6d561dc","meta_rev_id":"a1f9c094e58d6d078f505778a59d0f531d32590a","_number":678538,"virtual_id_number":678538,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-18 08:06:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-18 05:37:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":2,"date":"2019-09-17 15:55:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-09-18 05:37:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2019-08-26 12:51:04.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 12:51:04.000000000","updated_by":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"reviewer":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"state":"REVIEWER"},{"updated":"2019-09-17 15:55:22.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-18 05:37:15.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-18 08:06:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63d5de23223c8c321e4f1977a8938b13c9a97589","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 12:51:04.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"8db37497d1e54d906567bab131b4edc02044c096","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 14:57:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ab1b323e9414a66b59031fbbedeca50 : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ba8d95dee494f5488b60c9d02ee5d8a : SUCCESS in 10m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a9b8c02ec96f408e932891fe86b6fe6f : SUCCESS in 6m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/895ca2d7777a4ea5928a2692fdc722c6 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cd6b037e4a242ef8e1d76f023fa00c8 : SUCCESS in 8m 16s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/027f5bd480404943ace9f55d1a2fb37a : SUCCESS in 1h 22m 28s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/a5d8cbf5b59c455bb2f27b325db9a198 : SUCCESS in 1h 20m 23s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/85d2f0f1f1ea44fca450704963781b28 : SUCCESS in 1h 26m 28s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c09c8dff3444454bbca390631b64e671 : SUCCESS in 1h 29m 09s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5fae506ee28943e58f749c98c7836e88 : SUCCESS in 5m 49s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f375ba13f34a4550a6188b719ea102fc : SUCCESS in 7m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3c1995f12bb4874a75362e1975c30a4 : SUCCESS in 5m 51s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f922a87fc68c47c1a6fedd52cee42ae3 : SUCCESS in 1h 06m 16s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/23fe79e41699488e95590de782de77d8 : SUCCESS in 1h 05m 20s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/945703efe3c84f2682a5fec2d4ff366b : FAILURE in 1h 06m 16s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/7a5191a96862456d920481f966b2d360 : FAILURE in 1h 29m 48s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/086bac3b51aa4e59acc937054b5c3529 : SUCCESS in 1h 06m 59s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d3f67da1d874425ab6f437beec17cfc3 : SUCCESS in 44m 20s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/ad781bc1d87d4b9dacead3414e75abb1 : SUCCESS in 56m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75f691750d3c40948d69a601159ab9e1 : SUCCESS in 1h 31m 43s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/ee96c88cd02b4f548eb8e06226e83a19 : SUCCESS in 1h 26m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4842bc11ca0548dd887adaff31e4cf2c : SUCCESS in 1h 16m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3689ee33e1741d18dc046f74b901614 : SUCCESS in 10m 08s","accounts_in_message":[],"_revision_number":1},{"id":"eea2f00be9d7bb615ad31e6ae83354731ccd8006","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 15:55:22.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"04fa95017dc69d066ba1e8b4512b966b3f430e74","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-18 05:37:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"345e885f96643efc025bfb38a04d6281cf728dd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 05:37:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8619c3a9514d032837693a1f915d8a4ccd4172a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:06:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"da4b0b9bf5848cf2eaf43078a8b575a617d270a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:06:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0a251d755f64f91841d1db86d3c2553 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5342c95c04b5401393a5074d1174a207 : SUCCESS in 6m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3178d175cfae4b138ed6c946a6e83591 : SUCCESS in 5m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1f915a14d6d94e68829924c6030fa8e0 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ef5baafb0a44beda2334c143131350b : SUCCESS in 8m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1c6d895cd7da4b35bda2535a86b95473 : SUCCESS in 6m 27s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/81153ca18ed14e7a9386820427a45d6a : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d719447bcf674c1e826388fffceeae1a : SUCCESS in 7m 28s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/107752e9230d431cb6de49d92442915a : SUCCESS in 1h 02m 09s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/08454b3b61ad4cbfa2f0c9d5596163ec : SUCCESS in 1h 14m 13s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/e9986e36e78f42efaef82e20f440a8da : FAILURE in 1h 20m 31s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/bdbfab9d3b6547f58bc8c38b23077187 : FAILURE in 1h 32m 19s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/7a4e82b5666c4b90b3cb2c86b1cc36f8 : SUCCESS in 1h 24m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/db41e0c2d40b46a8bd38f8e5fe7314e5 : SUCCESS in 50m 18s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/4ff2f52a05294d9e80c7bff0f5877ceb : SUCCESS in 48m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e8404a5a15f4704a9169974bd13d985 : SUCCESS in 1h 49m 13s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/473286bb9b754b7d9c500cc4f5870df4 : SUCCESS in 1h 41m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8fbe404a0065462e803974abcccda8d6 : SUCCESS in 1h 11m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3314ae82e9af4a8885e9e412a37c7f69 : SUCCESS in 9m 30s","accounts_in_message":[],"_revision_number":1},{"id":"04cc45916d6ab5afc14b14819088f4d27c3a6498","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:09:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24a01029669644968cd7ecf11a1eaed5 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4074dbd919f6e25096371e40caa0830d1ab19e6f","revisions":{"4074dbd919f6e25096371e40caa0830d1ab19e6f":{"kind":"REWORK","_number":1,"created":"2019-08-26 12:51:04.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/38/678538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/678538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/678538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/678538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/678538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/678538/1"}}},"commit":{"parents":[{"commit":"4494c62149a9ce562c5e7e87233119af0a781ada","subject":"worker: Re-add FailoverPreparationForAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4494c62149a9ce562c5e7e87233119af0a781ada"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2019-07-15 06:01:54.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-26 12:51:04.000000000","tz":0},"subject":"elements: add arch property for ``open-vm-tools``","message":"elements: add arch property for ``open-vm-tools``\n\nThe ``open-vm-tools`` package only exists for x86-platforms.\n\nTask: 35814\nStory: 2006219\n\nChange-Id: I9f1b6e58004d7c2024f8d60d50ac998240abb6ee\n(cherry picked from commit 5114a9a4f9d037d8ffb75342713fc27ff692235f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4074dbd919f6e25096371e40caa0830d1ab19e6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4074dbd919f6e25096371e40caa0830d1ab19e6f"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
