)]}'
{"id":"openstack%2Foctavia~678548","triplet_id":"openstack%2Foctavia~stable%2Frocky~I7793d7a9113b23ac88e7c53d5dc292a70b9453b5","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"I7793d7a9113b23ac88e7c53d5dc292a70b9453b5","subject":"Add warning log if auth_strategy is not keystone","status":"MERGED","created":"2019-08-26 13:50:54.000000000","updated":"2019-09-13 16:25:48.000000000","submitted":"2019-09-13 16:24:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678548-1568391841068-e46e57b4","meta_rev_id":"f537cbe9e3fdceb652dd827fbfcf766fffaf84b5","_number":678548,"virtual_id_number":678548,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"date":"2019-09-13 12:04:11.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-13 16:24:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2019-09-05 06:43:47.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-09-03 21:21:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":1,"date":"2019-09-05 06:43:47.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 21:21:40.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-05 06:43:47.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-09-13 16:24:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b63690f217e81c276a12ed2c2b3c4593b589f2dd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 13:50:54.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"c7254c0167f1f8fc42ed2aca478cb02addde43e5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 13:51:01.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit e7686135d5c83d589b5820ab47957e7d3b940494","accounts_in_message":[],"_revision_number":1},{"id":"b5206825de168e7594fb33368d6afdc0f710ab6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:08:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50c3af95bbad4cae882c1e579b0db07b : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c3d7daf350d40eeaa07f82bed3959d7 : SUCCESS in 7m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c22df02a82f44b06ba35487fc0a74a3e : SUCCESS in 5m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d3f63a38b02b460495f0eb2ee0a0f77b : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03f9ec65d0e045c088691b34e15eda23 : SUCCESS in 6m 52s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a58abfc99c964748bcc78be521e241c5 : SUCCESS in 1h 11m 17s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f5727a948b294867b15f18be46768803 : SUCCESS in 1h 20m 19s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43647d86a6784f8b811e3e36a382da8c : FAILURE in 1h 19m 24s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/805b69d6243346af966890605e1242ca : SUCCESS in 1h 25m 45s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f63cbe029ced456d89ec4b5f6b8b5383 : SUCCESS in 6m 13s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b293112daeb04ed9b23b96b22b1d1fe1 : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2f7fab7fcd0e460a8448be12d3abdf2e : SUCCESS in 7m 24s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28679653000f4c3c94cb006e7e0eee55 : SUCCESS in 58m 54s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/02e0d95f8ecd4dc0b45377025453893c : SUCCESS in 1h 04m 33s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/80da7e1e3f644660840ce187b0d7efd9 : FAILURE in 1h 21m 01s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/57fb486b8427436086ee5bdc4ea7f772 : FAILURE in 1h 27m 24s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/3243a86fc83540e1bec0a9ec0af330a1 : SUCCESS in 1h 12m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d7df8389dafd401c9f2f6cfa6c9c828f : SUCCESS in 49m 27s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/d1ef0a8832f64af1a05c3f9dd5f93303 : SUCCESS in 49m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ff29d5b27d84cb7b173297e748a065a : SUCCESS in 1h 39m 51s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/d941c34db5504a1ea482dde004195a03 : SUCCESS in 1h 28m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/070a2eab8a9f4e0488d4857dd144939d : SUCCESS in 1h 20m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88a256fde7304b119c3c37e4a5aa1141 : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"05f11bd28c7c465661a88cdb5652049aa4474858","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-29 23:02:06.000000000","message":"Patch Set 1:\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"d9f621f262745dd112324358a7d99430133e5462","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-03 21:21:40.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"56e4ca21f0bdda735458786175c293874ee1cf88","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-09-05 06:43:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"704b047024662ad6c6be8466aebba4b020522d36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 06:44:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57da0d8d3ed64518e8999a645b8f4d80c646f410","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 11:02:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31486d87c3234c1f9f053c1388d39356 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ef16f96cf32421e80f6a0677ff6567b : SUCCESS in 8m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1df2e7be7b784ca8af40c7bf64a83210 : SUCCESS in 5m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/44625355d1e0478496a3b94594d8d3cf : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/448bb85a3b9f47758d28c372cee5554c : SUCCESS in 8m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7ed2023c20394b51841a7150d44987a5 : SUCCESS in 6m 30s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/94f496d38b754fb5b31859934f88096e : SUCCESS in 8m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/17231389a2824d638d919c2e426865bb : SUCCESS in 6m 33s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1880e546e2e14187ab444f5657013da5 : SUCCESS in 1h 09m 11s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d7ccf4fedc724a3ebfcafda45a913183 : SUCCESS in 1h 08m 11s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/62ffd154dbb049f5a4bb29d54885f96c : FAILURE in 1h 39m 45s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/e71bccc6650e409090422c22aa0a8014 : FAILURE in 1h 45m 56s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/169d571619c240d09253f6581e03a35d : SUCCESS in 1h 22m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bee3c17c46dc450e98e3694bfe8c5117 : SUCCESS in 48m 42s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/340dd3c518764bdb85a98d7b336503fc : SUCCESS in 52m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74c4de2c8671446f89471a03415789fb : SUCCESS in 1h 43m 18s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/a9617f7520ce43b1b37ea02ecdcb3a07 : SUCCESS in 1h 43m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2fab7fc370864a4396cebd49b6fd6ba6 : FAILURE in 35m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3a756aa7cb3411083423a001f0f1059 : SUCCESS in 9m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7d4d29e99db9d4c04806a737dea23fb531b68fc7","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-09-09 02:54:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a79c6f81b32af63c52dbb8c90bfa61114746c78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 04:54:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/212047769b764e3fa7bd5827fd3cc774 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9029b4cc47f4643a129861cdab59665 : SUCCESS in 6m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a4051c05bcf41e3b42b4494de0d4f89 : SUCCESS in 4m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cdc0380946164c4f859e58d6dddcc776 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0557a8c520f4e03b5e14890d9966d42 : SUCCESS in 7m 15s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/145e2ba1a5b04806ba6cebb6c47af81d : SUCCESS in 1h 18m 59s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/5a0c4891ee3947a7ad750629c9fa09c5 : SUCCESS in 1h 17m 57s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9bab60dafa6b4045b529ad54dfcc26f8 : SUCCESS in 1h 21m 44s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/83c9c246481d43978c0f0b9d06677932 : FAILURE in 1h 15m 50s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a46f5207ff9b4a8f8a1082979a4784da : SUCCESS in 4m 37s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/39786bd9be6d4512b80f467ca42b8809 : SUCCESS in 5m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c52d00656179497bb509737d51b2cc2b : SUCCESS in 5m 25s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8962e3d8390a429e9033ad1f76e6c010 : SUCCESS in 1h 06m 28s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c895efb4ebcb4b4d84d4b99fbf5942c2 : SUCCESS in 1h 02m 33s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/630661830ac54fcc844f41ab93495b95 : FAILURE in 1h 18m 33s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/cb36ec1ac9d44d32ae24dfe1d65b9236 : FAILURE in 1h 18m 25s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/2aa6fabdf79042a5812f8417e4120d8d : SUCCESS in 1h 08m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c84f40481fea4dd491c941e3b27754f4 : SUCCESS in 52m 39s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/377883896f56422fa8b5e37e552f72b8 : SUCCESS in 43m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eae382e5ae554ee9bd7001b3861121ed : FAILURE in 1h 43m 37s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/79e5a3dc60344c8d9ca1b8aa58b015f5 : SUCCESS in 1h 28m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0b6670c953024e8aa454f599a28638d3 : SUCCESS in 1h 10m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02a585aed60441878e88573274aadda3 : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":1},{"id":"164713ead41fe78d9601e80d2a62340a67b5ab2b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-13 12:04:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fbd6383caeafdd22a92ecb07b8d41222941b7fd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 14:13:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d41a4415fecd4616be33f79f441bccd9 : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56a89b9142894e12bdd0492d95933813 : SUCCESS in 6m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d1745407cff4681b0cc5c15d449f286 : SUCCESS in 5m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fe2118d0bb41457d8d56047c9564f757 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c343f483834a22894d29bd6eec6ac5 : SUCCESS in 6m 16s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/7b95768258584176ae282833f718f1f5 : SUCCESS in 1h 17m 03s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/d08ec6aa86834ecea98807f13069f2b0 : SUCCESS in 1h 21m 06s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f9285f79a154e93b924bc342724a271 : SUCCESS in 1h 30m 24s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4a5091dca1f14e10973993fd84bbaf7d : SUCCESS in 1h 39m 54s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/562c85a4e9254f0784277176ee1f9cf1 : SUCCESS in 6m 10s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/25c20e7323e9479ea998bb75eef964f0 : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/44e1a7c04b7e40d1b0e4a0aa7ef71107 : SUCCESS in 6m 09s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4062c1f3aac45a399cb684150d9fe82 : SUCCESS in 59m 55s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0e7a353ba0da4303b438eb74dd7809cb : SUCCESS in 1h 02m 33s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/cbee8b325dc7407eaac3a88cc584f64e : FAILURE in 1h 31m 46s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/cec53ad8a21c45a6a3b8c2311e56e059 : FAILURE in 1h 26m 15s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/cfcc93aaf29a4179aee1207b19849091 : SUCCESS in 1h 07m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4d0fd68ea3414286b605397525aa2709 : SUCCESS in 35m 27s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/53698221add34da1a04f2b7d8c4258b4 : SUCCESS in 46m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9ed57b07c92406896af5a47011151bf : SUCCESS in 1h 45m 16s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/fac5de23212d458c9422b2fe87e1001c : SUCCESS in 1h 51m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8da2eb31f24c4b4ea28f31dff4540f54 : SUCCESS in 1h 11m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afc4803a01db4d96a99c8f658673b40c : SUCCESS in 8m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f067adf97e31d152487c1c9fd7490996775672f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 14:13:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a82a4734e23e3325b9831cd57cc3c28a2b61afb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 16:24:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9b02d3ca42f44598297f6c8578b38b5 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/062ff33935934033b7ca45ba2d4d7414 : SUCCESS in 7m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d62a3bf0c4d749f48b9314ffdefd4db0 : SUCCESS in 5m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2b1522e393284be999be23bff8071068 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65eacb153f69431abf6622ceb6441217 : SUCCESS in 6m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c79e87d9f9234b08b9e52571c9f1e349 : SUCCESS in 5m 49s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/49a1eca318ef4087a48594fe1e942c38 : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/11aae1f948ca4c699e3bf9a63cfae09a : SUCCESS in 5m 12s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e43ad9252774c799accfff9473b89e8 : SUCCESS in 1h 13m 10s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/53ff8795b1da47a5a76b0116fa9536b4 : SUCCESS in 1h 14m 33s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/3f87b21411f44874838828beb8df630e : FAILURE in 1h 31m 55s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/63d7437f4e43404fb950fa6a9c2c974f : FAILURE in 46m 07s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/099600bf0cba4991b60f2486944d3dbb : SUCCESS in 1h 15m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ba58eeed16c84629b582cd0e197aa08e : SUCCESS in 52m 04s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/5bc59e40472d4fe8b3edd754d05a067f : SUCCESS in 42m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e1a9a6ee8f449a9ad126e94b2c54d72 : SUCCESS in 2h 02m 10s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/8a2fd41e279a4484803685a62a44ef7f : SUCCESS in 1h 42m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/db4a3cefcda640149cc02e54de310241 : SUCCESS in 1h 18m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71cc3ccfe6964ca58abf29162da9b4d7 : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f69a0cac10467723487d024e026c0659dd1d27ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 16:24:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f537cbe9e3fdceb652dd827fbfcf766fffaf84b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 16:25:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f0f67f5eb2c440280714c644bed0210 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f58b07fd8caf7c7dea18b483d342b2b0e057b12e","revisions":{"f58b07fd8caf7c7dea18b483d342b2b0e057b12e":{"kind":"REWORK","_number":1,"created":"2019-08-26 13:50:54.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/48/678548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/678548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/678548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/678548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/678548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/678548/1"}}},"commit":{"parents":[{"commit":"4494c62149a9ce562c5e7e87233119af0a781ada","subject":"worker: Re-add FailoverPreparationForAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4494c62149a9ce562c5e7e87233119af0a781ada"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-11 16:06:23.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-26 13:50:54.000000000","tz":0},"subject":"Add warning log if auth_strategy is not keystone","message":"Add warning log if auth_strategy is not keystone\n\nA user came to the IRC channel with CLI errors:\n\"Client-side error: Validation failure: Missing project ID in\nrequest where one is required.\"\n\nThe root cause was the [api_settings] auth_strategy was set to\n\"noauth\" instead of \"keystone\".\n\nThis patch adds a warning log message to the API process that\nwarns users that typically the auth_strategy should be set to\nkeystone.\nIt also points the user to have an administrator check the keystone\nsettings in the octavia.conf.\n\nChange-Id: I7793d7a9113b23ac88e7c53d5dc292a70b9453b5\n(cherry picked from commit 48371c98ee49807b5c2b73a442d649bad40b73ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f58b07fd8caf7c7dea18b483d342b2b0e057b12e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f58b07fd8caf7c7dea18b483d342b2b0e057b12e"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
