)]}'
{"id":"openstack%2Foctavia~678550","triplet_id":"openstack%2Foctavia~stable%2Fstein~I4e8007ff98f585ce6622f7128ecdd53cdfa926a7","project":"openstack/octavia","branch":"stable/stein","topic":"api-none-stable/stein","hashtags":[],"change_id":"I4e8007ff98f585ce6622f7128ecdd53cdfa926a7","subject":"Fix l7rule API handling of None updates","status":"MERGED","created":"2019-08-26 13:51:26.000000000","updated":"2019-09-18 04:03:17.000000000","submitted":"2019-09-18 04:01:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678550-1568779265290-a663f06a","meta_rev_id":"ab09a08038c603ae2faab1f4933f9701874336b7","_number":678550,"virtual_id_number":678550,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-18 04:01:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2019-09-18 02:09:00.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-09-17 16:04:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":1,"date":"2019-09-18 02:09:00.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 16:04:41.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-18 02:09:00.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-09-18 04:01:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e423c0dcb311b988fd8fa97e6548f5ad1a242473","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 13:51:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7ad058297dbcbc842dfc5696d8f25a3dffaa402d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 13:51:35.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 2234394dcea1411ddc6c6c4509be66cec95c1aef","accounts_in_message":[],"_revision_number":1},{"id":"e19bf5dfcb09d1f2cbf3fb00790837c3df58fa92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:20:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4db786ae95948dda28e930b3be50a27 : SUCCESS in 8m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b92b875d78f4a42926b0a3ded08367a : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19638f5724de4decb4ceb3f122a73bff : FAILURE in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ec2fca8088b41c08342a24f9b786f26 : SUCCESS in 4m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/57044cb7cb974d7caf633476ca739866 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b32c1e1c13a9496d81b3f9441cd3b6f3 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87b7c597d8024436b89385459c40f92c : SUCCESS in 7m 28s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d0d77d6efcb94ab69e0b27d3d837a2eb : SUCCESS in 1h 27m 55s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/bacf84bab84242e2b1d27296bf95d14d : SUCCESS in 1h 21m 59s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/85fdc1b441ea4a7ba632733a4659a5da : SUCCESS in 1h 29m 25s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/58a699509c314d11b8c8138b0c4803b1 : SUCCESS in 1h 26m 45s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9ac233a2563f4ce1833a39d6635ec55b : SUCCESS in 6m 53s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/bb9eb273cde6457d8b49139e448f4a8e : SUCCESS in 8m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/18543a84a7124d80862fa68905551e6f : SUCCESS in 8m 43s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff3a495289c04515814356654c23a2c8 : SUCCESS in 1h 03m 55s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/080d2c083f37480ea7561e5eb8412555 : SUCCESS in 1h 05m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/65955bd2ecd14f1fabff7c7effd7b22e : SUCCESS in 49m 56s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/475a4c12b3a84722b411ddbae2f32d8c : SUCCESS in 45m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7051e307e7842ebad09b8c584d411c7 : SUCCESS in 1h 35m 38s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1b3b6dc26c054af6aff5a0f474fe1b90 : SUCCESS in 1h 39m 01s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b48ad4e57a134fcf86381b18ad07a96a : SUCCESS in 1h 41m 17s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7f9ff1a699374bc3a8a4a5e480fa221c : SUCCESS in 2h 13m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7260878cf1f149fe9341e811164d4790 : FAILURE in 38m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2c4a198f3d014d539cb3b69ba44148c4 : SUCCESS in 1h 47m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a038495a8fe24cf0a0540f588d7d8065 : SUCCESS in 1h 46m 02s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a8f374d0f1f34d41858cdadbcb11d7e7 : SUCCESS in 1h 14m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5ffafcd206fb4fb4b63e74f4c5dd3642 : SUCCESS in 53m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b863f56c2cc73f2d6d1007c2e1f8ea54e151d728","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-16 07:57:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ebd421f20d3d9e706e8d71aed1a809d757f3273","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 10:40:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5c0bf4abeff428aa3c1c67c11197f3e : SUCCESS in 7m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0bc54f03c1747748be7508d91a67cc7 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b7a71ef9d2e42158d7bac51de1d02bb : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2bdfd5e6e4494ae7bde8cb3385f79b3e : SUCCESS in 3m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9a0a7511cb504f3abcf8c35916425cad : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c57bc6000bd45fe899e3b4716a5471b : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d3e17fde8e4b0db49e832d69cc3f5c : SUCCESS in 7m 39s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/0633180f12824fb381cb05593b67c8a6 : SUCCESS in 1h 22m 27s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/26c7af8e0bf545189da574536cca04bc : SUCCESS in 1h 22m 57s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c3de77801fb5446682085e3dcbf931c7 : SUCCESS in 1h 28m 38s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ce2f344d8d3445bc99fd1e76f21cd8ea : SUCCESS in 1h 33m 22s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4d962b302fe84c1d979d020503e9628d : SUCCESS in 5m 21s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/6e1dae0ae3484a408067bba8b2a76a16 : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5394a312c0eb4d25816ac927399d64d7 : SUCCESS in 7m 07s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d6a15c1b2d240b9958c5fbcade1848a : SUCCESS in 1h 10m 29s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/93dba8536916465696d0b979ad964351 : SUCCESS in 1h 07m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f531d808c9944daab30f790a1da294fa : SUCCESS in 48m 54s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0c97777d3459458f955fe8173ac45b1d : SUCCESS in 44m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c42399263c548b8845977ba54a7d968 : SUCCESS in 1h 33m 59s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/772d16f280e345cd8241e5734da72b9c : SUCCESS in 1h 53m 11s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/3069740d053e4d7d898abae0fcd2244e : SUCCESS in 1h 48m 25s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5ba7527b917444b699a732b2a71dbea2 : SUCCESS in 2h 30m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/150703dc02fa49eba171687dc5dbbf91 : FAILURE in 32m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b7431f78eba4449c896a714a03a26cc4 : FAILURE in 1h 30m 19s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c4dbd35d80f34f1db905447a916cb8a8 : SUCCESS in 2h 04m 36s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f1f04c76aac947b2b10c4587b0d322ad : SUCCESS in 1h 06m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/45818247332c4c399c480b6ca2c5011f : SUCCESS in 48m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c803be96ceedb54c169d1993f9c1821cf110875d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 16:04:41.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"0d23506b6ebd19f57852bdc08268d82a735235ac","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-09-18 02:09:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"28d405886fff15dbefb57646b64deef9b501ef97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 02:09:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3c248c105a7e0f813c70d538a44687f19155418f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 04:01:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ce80974aea4646db1d5028e9ad82d54f6518856b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 04:01:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c52be141e6494bf6b858f459ca549803 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/850cbea21cb44000a75d4157cb1f4ff0 : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/092580baa84740de842d6feef0dfbfba : SUCCESS in 4m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/059dbab8e66247b7bb08f44a8218f2d5 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/468f942ca6a04f12898bafba14eb6f27 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/034c862b021d433c90c87daddc4fa537 : SUCCESS in 7m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6de1a323b7d9420b87b74a4c56ff2d4c : SUCCESS in 5m 14s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c1cfd54086764b6dae07f25ff4cf6be4 : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8fdb72ce69f04325bd4e8ae66267d752 : SUCCESS in 5m 00s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a378e1783714737a3818d503ce24ea2 : SUCCESS in 1h 05m 12s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/cb812aba764d459b9d7b9afdb3dfd016 : SUCCESS in 1h 06m 54s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/3eb0d41fc4c84fff983babe1b2846857 : SUCCESS in 1h 11m 23s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/91f36298c1e942d69fcf5a011f136a20 : SUCCESS in 1h 23m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c676fbad149b41fcafabed44584d64f0 : SUCCESS in 39m 16s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/641229d5cb0e485db5b8182013c09e56 : SUCCESS in 40m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45db59173b9a4fae922fdaffe867377e : SUCCESS in 1h 34m 03s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/81f2f98ba813494cb698a4aefe5f351c : SUCCESS in 1h 24m 49s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/9e682d85a2fd44349c58c3027f51e4ca : SUCCESS in 1h 45m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/62d3080bd71b41fbbc65a6c4b083b9ab : SUCCESS in 1h 16m 30s","accounts_in_message":[],"_revision_number":2},{"id":"2fa2e835662ebe273bd2e5dc4c691b2cc06da5d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 04:03:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f3de793afe84e8896dac49add568028 : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"55b04f279bb2587b8e78667f386f582b347f4e03","revisions":{"b74c4be9cb233745919117aaedff6b1051c9c420":{"kind":"REWORK","_number":1,"created":"2019-08-26 13:51:26.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/50/678550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/50/678550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/678550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/678550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/678550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/50/678550/1"}}},"commit":{"parents":[{"commit":"4b7fe7f80c4f66096dc5b625807199463f3e6c64","subject":"worker: Re-add FailoverPreparationForAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b7fe7f80c4f66096dc5b625807199463f3e6c64"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-31 20:09:42.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-26 13:51:26.000000000","tz":0},"subject":"Fix l7rule API handling of None updates","message":"Fix l7rule API handling of None updates\n\nThe current l7rule API does not properly handle\nclearing/reseting values on update.\n\nThis patch corrects this to appropriately handle None/null updates\nto the l7rule parameters.\n\nStory: 2005374\nTask: 36023\n\nChange-Id: I4e8007ff98f585ce6622f7128ecdd53cdfa926a7\n(cherry picked from commit e4e7b4a880c19054b63cd295cb19adee527517e5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b74c4be9cb233745919117aaedff6b1051c9c420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b74c4be9cb233745919117aaedff6b1051c9c420"}]},"branch":"refs/heads/stable/stein"},"55b04f279bb2587b8e78667f386f582b347f4e03":{"kind":"REWORK","_number":2,"created":"2019-09-16 07:57:55.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/50/678550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/50/678550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/678550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/678550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/678550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/50/678550/2"}}},"commit":{"parents":[{"commit":"a680a19e18a6853e2a1362a83a6d628ab0fd601b","subject":"Merge \"Add failover logging to show the amphora details.\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a680a19e18a6853e2a1362a83a6d628ab0fd601b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-31 20:09:42.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-16 07:57:42.000000000","tz":120},"subject":"Fix l7rule API handling of None updates","message":"Fix l7rule API handling of None updates\n\nThe current l7rule API does not properly handle\nclearing/reseting values on update.\n\nThis patch corrects this to appropriately handle None/null updates\nto the l7rule parameters.\n\nStory: 2005374\nTask: 36023\n\nChange-Id: I4e8007ff98f585ce6622f7128ecdd53cdfa926a7\n(cherry picked from commit e4e7b4a880c19054b63cd295cb19adee527517e5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55b04f279bb2587b8e78667f386f582b347f4e03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55b04f279bb2587b8e78667f386f582b347f4e03"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
