)]}'
{"id":"openstack%2Foctavia~678557","triplet_id":"openstack%2Foctavia~stable%2Fqueens~I4e8007ff98f585ce6622f7128ecdd53cdfa926a7","project":"openstack/octavia","branch":"stable/queens","hashtags":[],"change_id":"I4e8007ff98f585ce6622f7128ecdd53cdfa926a7","subject":"Fix l7rule API handling of None updates","status":"MERGED","created":"2019-08-26 14:05:43.000000000","updated":"2019-09-21 00:57:28.000000000","submitted":"2019-09-21 00:55:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678557-1569027358551-88704861","meta_rev_id":"6f7ab3b616e092bbeb0dde10d08250501e152a0c","_number":678557,"virtual_id_number":678557,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2019-09-20 14:03:53.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-09-21 00:55:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-20 21:22:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-20 21:21:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-09-20 21:22:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 21:21:19.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-20 21:22:38.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-21 00:55:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0057b1578a3526c854a0d58dbc5db8af9511650","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 14:05:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6976dd570a83ea07788386bc4bec789e5683f334","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:25:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f603b5392e7a40fcb0534e46f6857eb7 : FAILURE in 6m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21ca0042dc52437cb0719847e84668d5 : SUCCESS in 6m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d6bef50e3c9046a8993b845f1d8e9147 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4435583de92249899e96008305b32ed3 : SUCCESS in 7m 53s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/cb698ac8ef8340f7a0166e31035320c2 : SUCCESS in 1h 25m 12s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f13f08ac0164449a88de981e81590866 : SUCCESS in 1h 26m 42s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2904354d1edd44619ebf434623abe3c3 : FAILURE in 1h 14m 21s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e68bb6de1954478cb46095d55c22f2dd : SUCCESS in 1h 09m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7b65a8324d44b1fb6a086bb9770e383 : FAILURE in 9m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/eb923ba6fca24ba6986b489456882697 : FAILURE in 8m 38s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d90773c0dbb94f758dc675e02ee33274 : FAILURE in 8m 51s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/70c1fd3b17774baf9e7d4c8ff12cb177 : SUCCESS in 1h 02m 30s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7fd46487b4f043628bc38f0339eb9295 : SUCCESS in 1h 04m 44s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/6660b910c404406681ca3941b4873186 : SUCCESS in 1h 27m 04s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/f12da61088764eb6ba77bdc3fda47ef2 : SUCCESS in 1h 16m 22s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/0505ed2b0e884292a12178eef182a069 : SUCCESS in 1h 08m 40s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/47d5ff801cf64f909c9b806ad0c9196f : SUCCESS in 50m 55s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/97f1dda80330435887d2d717d6740b4a : SUCCESS in 45m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bc366317cbe8427a9438976b1ba6fc32 : SUCCESS in 1h 28m 51s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/6e0d1d6c57fc434ab2ec0a134bad3e18 : SUCCESS in 1h 40m 10s","accounts_in_message":[],"_revision_number":1},{"id":"935fc72646c5e7fccc1e8794fead3b51c7a5445b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-16 08:06:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8df99e1200b2b4ca50649f904cd95e880106add5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 10:25:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de53c4df643f4bff8cb8827a524fe5d2 : SUCCESS in 8m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a0924ad95ac24ee6a4283a40afd9b5c4 : SUCCESS in 5m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8586ed507ae84f18a752735d8ae57861 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d40630898a974b798a6c4c875e8fbce3 : SUCCESS in 6m 32s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/e44a7401d6e1432bbfbe3438d196c6ab : SUCCESS in 1h 26m 14s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/13dd231e3b23426fa86a2126fd27c6fa : SUCCESS in 1h 21m 29s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7862c565f382431a9d1de77d492af7f3 : SUCCESS in 1h 31m 09s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f8100ca818d442d7b64a4cae227fb6bd : SUCCESS in 1h 36m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/289fabda1709447093f6cf4d5a9aeb01 : FAILURE in 7m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dd86384b526b4d1989911e21ba2601fa : FAILURE in 5m 37s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f07518377472486287b93a9de7794058 : FAILURE in 7m 21s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bb05546309284823bab3a3917ee40110 : SUCCESS in 53m 03s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e8a737a61eeb4628a4f43338a0379fed : SUCCESS in 1h 22m 35s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/cf828874914a4727999146cf6bd1bb4c : SUCCESS in 1h 07m 41s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/0cbe99e6d74e4f03b015703538ee3d9b : SUCCESS in 1h 11m 16s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/c964344f378f4449b6e1f6ee61920f00 : SUCCESS in 1h 10m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee5717623a464dddbf6f3c3bf06214df : SUCCESS in 42m 12s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/c82b4c368ddf4e4b9f2b168dac08b0c9 : SUCCESS in 47m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/731e4bb14d6c4ee5943cbf70e44f0c4a : SUCCESS in 1h 32m 28s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/efc328cb01974e7a9d1eeaf39f99aa5e : SUCCESS in 1h 35m 44s","accounts_in_message":[],"_revision_number":2},{"id":"058c8362d283e6fbf989a67e05fdc23d4c4d7c34","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-17 15:03:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a79958c3d9dfc99670063ffb72eb955eeb771dfe","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 16:05:33.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"f943125d470538b6f02befb7a2d1691a78d02226","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 17:29:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/770025cdd4024420bcbcbce91dd625d2 : SUCCESS in 9m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/862e24cc55094c0b9333f8f56f5b95de : SUCCESS in 6m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f26d6b151954495683b6cac7084ebb25 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b70db78155ab41c786c7859292b7aebe : SUCCESS in 7m 24s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c6fb32b4700e4af8a6d3edc832f3392c : SUCCESS in 1h 26m 53s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/d4bb367111a24a47a399412b66d650d8 : SUCCESS in 1h 17m 10s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ecfa4ee0188d41cdb00f8164f5a4ca60 : SUCCESS in 50m 51s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7cb8925844d541ab8b8bf121a1cd1c40 : SUCCESS in 1h 23m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b56987d67fdc4f0b9e419172b8349672 : FAILURE in 8m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ce8c125ac39a42d4abba7cd445e93cc3 : FAILURE in 5m 25s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/44a841013acf423b856b5c1f0b7bd114 : FAILURE in 7m 31s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a299b3eac964b109b97cf4d8535c450 : SUCCESS in 1h 15m 05s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2237ad6a7a6e450b90d88a0f656c6d7a : SUCCESS in 1h 04m 45s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/edf2265e2b15485fb5b5747bfb681278 : SUCCESS in 1h 09m 55s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/1c4372885a1e457fb55d447c187fd284 : SUCCESS in 1h 15m 06s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/fdd0f7a467544f3c868b0b0f00e9217b : SUCCESS in 1h 12m 52s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4a1bc989e5364652a465f26eeda57956 : SUCCESS in 53m 29s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/7eb1b7802d9d4c47863f03289be4dc10 : SUCCESS in 50m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/14a13dd0fdfc42da8a25eb99b64bbd60 : SUCCESS in 1h 56m 42s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/da380906c2684acc80e6e038cd4d5054 : SUCCESS in 1h 38m 15s","accounts_in_message":[],"_revision_number":3},{"id":"2cccf22d253eb3493aba878d145b2c5713208e10","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-18 05:42:15.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"3a3824bc07f42cf705deb0520a5af9a556d17dc2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-20 10:32:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"44b5b188c20b21458e5f261e021bac4a5a639524","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 12:24:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a11dccb76a2f450e991f315de094efab : SUCCESS in 9m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0316ebd07d5245ba8b2d375b7e93654f : SUCCESS in 5m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5db827597b8646c3b86556e3d23ecd23 : SUCCESS in 11m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/537df92d8dab42dbb86f9541703488ae : SUCCESS in 6m 54s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d66de4e3c828448ebd2bb6c8c668b46e : SUCCESS in 1h 22m 12s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/7b86ba76f43744ca8f6da8382227a4ba : SUCCESS in 1h 19m 09s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/37c8c844d8b84390acda419562d79a3d : SUCCESS in 1h 24m 33s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/caf4c24c290e406fb735601323303e5e : SUCCESS in 1h 23m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/143665732dbd4d5b9d6b664ca580270d : SUCCESS in 8m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4398e9bcb4024b03a14853b082ce3b82 : SUCCESS in 6m 34s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/67dee70649ec479fb7cb44a5f7128c81 : SUCCESS in 5m 44s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e77a730d66b4dacb3778bdd7b78e01b : SUCCESS in 1h 01m 00s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/97f0b69c91524d418e7a5297359ae406 : SUCCESS in 1h 03m 36s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/7c985ea9547540d39267cdac30b6de93 : SUCCESS in 1h 16m 59s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/c8387ea5e67a480cab0fbd29dfa8a050 : SUCCESS in 1h 10m 15s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/022d6123ff58459183e4a268f6a35c5a : SUCCESS in 1h 34m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cf94e065720341099b76337ba182e87a : SUCCESS in 58m 37s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/afbfff90d9794362a653fde0e1ecb336 : SUCCESS in 59m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a14991b3374487584d48b3fc3e3948d : FAILURE in 1h 25m 50s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/a521cd120e254b5b833159e0c216bd12 : SUCCESS in 1h 31m 00s","accounts_in_message":[],"_revision_number":4},{"id":"151ad5f24352a830c0638c7fc84c9eb9518dd482","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-20 14:03:53.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"205282c844ecc7b9c547b8e8df11ea51a1361d23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 15:57:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed6fba36efb34041a43fdbb43f478493 : SUCCESS in 9m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ef379ea9a424b89970bffdd39213a46 : SUCCESS in 5m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2d5093ac3d0d4d62ad21c7266ac1121c : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6855a09baa744808b95a0abf379b1790 : SUCCESS in 6m 58s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/7afcba6705fa47f881d4a213c7b8015f : SUCCESS in 1h 19m 16s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0c715e3d3d2d47f18fe30d8336213e11 : SUCCESS in 1h 20m 44s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8215913acc44bc593269ebe186e17e9 : SUCCESS in 1h 22m 31s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/be5e521b3c2b4af88ea9e760714fce51 : SUCCESS in 1h 29m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a7b9b9780d84572b0322f9ab43ff93f : SUCCESS in 10m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/08fc75f848e440d59647a00cea72f0eb : SUCCESS in 6m 20s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/5f5f044c585c4c8abffc38ecc1b33526 : SUCCESS in 6m 06s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f0b58789c3045fcb8efc96eab7f27cd : SUCCESS in 1h 02m 52s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c53859bd652145e386833556ec3b2d0d : SUCCESS in 1h 00m 01s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/cf4d8fd65ee943c9b51aa6b59d4552a1 : SUCCESS in 1h 15m 13s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/7775526779a5411faab03854f0c3e0a1 : SUCCESS in 1h 11m 22s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/0613909ae79c4cbaa18f7cb426531c86 : SUCCESS in 1h 24m 52s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/429279ac5d854389aab51ff552eb2b46 : SUCCESS in 1h 05m 35s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/19adeb018de64b87bafc2cf8e16a862b : SUCCESS in 46m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4e234cc643f244f88d23da9369d72608 : SUCCESS in 1h 45m 41s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/10ecbce9a11a464097275c0db2bc01e4 : SUCCESS in 1h 41m 11s","accounts_in_message":[],"_revision_number":4},{"id":"fbd3f3ae9d1e4ec4acf90fa5449fe9af6dd74a47","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-20 21:21:19.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"58e0641571f0e391335bf296d0862a5cebf35e94","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-20 21:22:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"02004b09f5f6f640224cfbd85e02f7f34eb77320","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 21:22:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4d8487025f5f5aa932b1bd1051514ecaaa8c6a8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 00:55:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"645716137bd81bb7d29e7948bb5e20f6edba0cc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 00:55:58.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a953ccf6c4584ea18f56b05272fbe4af : SUCCESS in 8m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1cafe8332374e94a20acba292be3bdf : SUCCESS in 5m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/acbb0ca23ea246c4b2b54ec4e04dc1be : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c96b62e3d1584c58a126ba6d4ca5bb38 : SUCCESS in 5m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42e953ce3277483da5c39b88f4836178 : SUCCESS in 10m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c498902020af41e18389b1bd10ce39c1 : SUCCESS in 6m 17s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/814254c324154e848f46fab96ff8b2be : SUCCESS in 7m 10s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/48b6c978f24f4912b631f8517eda2448 : SUCCESS in 1h 16m 51s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/fee8f274f8b945339cb46c5dce952bce : SUCCESS in 1h 21m 35s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/219a8216b36f48dca8c485929cbec2ed : SUCCESS in 1h 07m 04s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c2d095c7395243a88a4bb975959115d1 : SUCCESS in 1h 19m 52s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/8c52698ace334d3b804b8fd60d500024 : SUCCESS in 1h 07m 40s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/d16e88ab4e1847a38b43822c4b910af5 : SUCCESS in 1h 19m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1cee40dc7e914910b6a1d2adc3e58ca9 : SUCCESS in 48m 41s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/a955988186cb48a39198b61a92d8d20a : SUCCESS in 51m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b9bb14d8528425bb238eb385a45bc72 : SUCCESS in 1h 26m 32s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/c7d15ebd7e5340da8ad8938434a13845 : SUCCESS in 1h 31m 50s","accounts_in_message":[],"_revision_number":4},{"id":"6f7ab3b616e092bbeb0dde10d08250501e152a0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 00:57:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0660cfa9701b494a96050e859cdf9786 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"431d9c9b95f7a7ac8b15315ddd39131f065d69db","revisions":{"e2e3356240360039c72929622d46756373387ca7":{"kind":"REWORK","_number":1,"created":"2019-08-26 14:05:43.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/57/678557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/678557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/678557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/678557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/678557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/678557/1"}}},"commit":{"parents":[{"commit":"37aad5db40d688c6c718e8a7f892b61b0924143a","subject":"worker: Re-add FailoverPreparationForAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37aad5db40d688c6c718e8a7f892b61b0924143a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-31 20:09:42.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-26 14:05:12.000000000","tz":120},"subject":"Fix l7rule API handling of None updates","message":"Fix l7rule API handling of None updates\n\nThe current l7rule API does not properly handle\nclearing/reseting values on update.\n\nThis patch corrects this to appropriately handle None/null updates\nto the l7rule parameters.\n\nConflicts:\n    octavia/api/v2/controllers/l7rule.py\n\nStory: 2005374\nTask: 36023\n\nChange-Id: I4e8007ff98f585ce6622f7128ecdd53cdfa926a7\n(cherry picked from commit e4e7b4a880c19054b63cd295cb19adee527517e5)\n(cherry picked from commit 2234394dcea1411ddc6c6c4509be66cec95c1aef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2e3356240360039c72929622d46756373387ca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2e3356240360039c72929622d46756373387ca7"}]},"branch":"refs/heads/stable/queens"},"181f7b018226dc6ae492edc00d6a1bad7395cae2":{"kind":"REWORK","_number":2,"created":"2019-09-16 08:06:34.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/57/678557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/678557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/678557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/678557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/678557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/678557/2"}}},"commit":{"parents":[{"commit":"f71539ab0eb9e02e24837325af6a93c5e396692d","subject":"Merge \"Add failover logging to show the amphora details.\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f71539ab0eb9e02e24837325af6a93c5e396692d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-31 20:09:42.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-16 08:06:11.000000000","tz":120},"subject":"Fix l7rule API handling of None updates","message":"Fix l7rule API handling of None updates\n\nThe current l7rule API does not properly handle\nclearing/reseting values on update.\n\nThis patch corrects this to appropriately handle None/null updates\nto the l7rule parameters.\n\nConflicts:\n    octavia/api/v2/controllers/l7rule.py\n\nStory: 2005374\nTask: 36023\n\nChange-Id: I4e8007ff98f585ce6622f7128ecdd53cdfa926a7\n(cherry picked from commit e4e7b4a880c19054b63cd295cb19adee527517e5)\n(cherry picked from commit 55b04f279bb2587b8e78667f386f582b347f4e03)\n(cherry picked from commit b8f68c2da9f8678419b2e4787a19a234ef020b23)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/181f7b018226dc6ae492edc00d6a1bad7395cae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/181f7b018226dc6ae492edc00d6a1bad7395cae2"}]},"branch":"refs/heads/stable/queens"},"52c45d6e493b031080ddcb8a9f5adc999e09a170":{"kind":"REWORK","_number":3,"created":"2019-09-17 15:03:23.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/57/678557/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/678557/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/678557/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/678557/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/678557/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/678557/3"}}},"commit":{"parents":[{"commit":"f71539ab0eb9e02e24837325af6a93c5e396692d","subject":"Merge \"Add failover logging to show the amphora details.\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f71539ab0eb9e02e24837325af6a93c5e396692d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-31 20:09:42.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-17 15:03:13.000000000","tz":120},"subject":"Fix l7rule API handling of None updates","message":"Fix l7rule API handling of None updates\n\nThe current l7rule API does not properly handle\nclearing/reseting values on update.\n\nThis patch corrects this to appropriately handle None/null updates\nto the l7rule parameters.\n\nConflicts:\n    octavia/tests/functional/api/v2/test_l7rule.py\n\nStory: 2005374\nTask: 36023\n\nChange-Id: I4e8007ff98f585ce6622f7128ecdd53cdfa926a7\n(cherry picked from commit e4e7b4a880c19054b63cd295cb19adee527517e5)\n(cherry picked from commit 55b04f279bb2587b8e78667f386f582b347f4e03)\n(cherry picked from commit e0913562de2a5f23d1dade40d36679736dc25d9d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/52c45d6e493b031080ddcb8a9f5adc999e09a170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/52c45d6e493b031080ddcb8a9f5adc999e09a170"}]},"branch":"refs/heads/stable/queens"},"431d9c9b95f7a7ac8b15315ddd39131f065d69db":{"kind":"REWORK","_number":4,"created":"2019-09-20 10:32:53.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/57/678557/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/678557/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/678557/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/678557/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/678557/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/678557/4"}}},"commit":{"parents":[{"commit":"189938ee635b3c597b15e978f7a59a2e157047ea","subject":"Merge \"Work around strptime threading issue\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/189938ee635b3c597b15e978f7a59a2e157047ea"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-31 20:09:42.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-20 10:27:43.000000000","tz":120},"subject":"Fix l7rule API handling of None updates","message":"Fix l7rule API handling of None updates\n\nThe current l7rule API does not properly handle\nclearing/reseting values on update.\n\nThis patch corrects this to appropriately handle None/null updates\nto the l7rule parameters.\n\nConflicts:\n    octavia/tests/functional/api/v2/test_l7rule.py\n\nStory: 2005374\nTask: 36023\n\nChange-Id: I4e8007ff98f585ce6622f7128ecdd53cdfa926a7\n(cherry picked from commit e4e7b4a880c19054b63cd295cb19adee527517e5)\n(cherry picked from commit 55b04f279bb2587b8e78667f386f582b347f4e03)\n(cherry picked from commit e0913562de2a5f23d1dade40d36679736dc25d9d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/431d9c9b95f7a7ac8b15315ddd39131f065d69db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/431d9c9b95f7a7ac8b15315ddd39131f065d69db"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
