)]}'
{"id":"openstack%2Foctavia~678560","triplet_id":"openstack%2Foctavia~stable%2Frocky~Iae839b6e59c43bc4be1e12702b3e2197b038b4c2","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"Iae839b6e59c43bc4be1e12702b3e2197b038b4c2","subject":"Use the infra pypi mirror for DIB","status":"MERGED","created":"2019-08-26 14:08:53.000000000","updated":"2019-09-05 14:14:47.000000000","submitted":"2019-09-05 04:25:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678560-1567657544813-99383969","meta_rev_id":"54b6351e9811fefcfee2a316f291faf6b665cf31","_number":678560,"virtual_id_number":678560,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-05 04:25:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2019-09-05 02:24:58.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-09-03 21:20:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":1,"date":"2019-09-05 02:24:58.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 21:20:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-05 02:24:58.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-09-05 04:25:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80a9fe79a016e379a31544ebfbf60e52785706d7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 14:08:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d6073f048c7c7f19bf2487d4adae720960e6f27","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 14:09:04.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 9ae2f61b7331eae4d946914290e5b6a53896f01c","accounts_in_message":[],"_revision_number":1},{"id":"c5653fa587828b9af5c5bb1918b8ab82459d9008","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 17:09:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a36106607f443989dc07a4f17eda1a1 : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb6e11a3ade44acca1b7d2ae8bf7e81d : SUCCESS in 8m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5737ad3d8eac4f32a73231932ccd6688 : SUCCESS in 5m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/63fca6d58c0b44cd971f353a4cad8f5d : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d7c9eea231342af8ef1521d2851232f : SUCCESS in 7m 20s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/dd66b482cffb4eda8d4263ba7cec6ae5 : SUCCESS in 1h 13m 37s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/3e8ab04c132d48b2a8d9a024e73dc9be : SUCCESS in 1h 17m 10s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c4473b8a6f74dcf8c42a5260cb3500e : FAILURE in 44m 52s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/353f7062076c4ebaa36aa85c55065daf : FAILURE in 49m 28s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5a08cce3a2e144b29eb542fb6a3c1b0c : SUCCESS in 6m 01s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b2d674678b63447eb0524573d86db973 : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a036a77ed7f0499baa51d114b635c405 : SUCCESS in 6m 50s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c721989b835a4cae8cf930b3e23e7494 : SUCCESS in 1h 06m 19s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ae7cb79019394b83a5503d1a0f807398 : SUCCESS in 1h 02m 41s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/0d480704b83f4780bad42dc720c6a39d : FAILURE in 1h 28m 54s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/6b6343ca0bf743d4a93de25c5bb08d37 : FAILURE in 1h 32m 27s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/ada35d7c8bfa4526969a789919ef6d1d : SUCCESS in 1h 08m 51s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aa97f0ef995947ffaad83e6342fcd062 : SUCCESS in 44m 29s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/192013c9d97c42a7995489d806da5432 : SUCCESS in 46m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa88f97947fb44bdb1468589f732e04c : SUCCESS in 1h 21m 37s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/dcc139b125a74ea4ab34b5b42bedd8f2 : SUCCESS in 1h 37m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fb6c489a651f4c7ea33e33dc04636e54 : SUCCESS in 1h 14m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aacb6dc254be49c39f6ce40e44663e8e : SUCCESS in 8m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0bd7c9216417421ab1421208d34005cbad57dee9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-29 22:59:59.000000000","message":"Patch Set 1:\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"d669653ca3a34a191695d22737b05bfa0f96e4a2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-03 21:20:44.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"a844b3b1487786df9823e067ce8f4a8b3f457531","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-09-05 02:24:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"64ec0a1fd2bf159479dd5945f30df31f39f26db5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 02:25:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78c38c596b7b73aa20c750fe8551dd5dbe322da0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 04:25:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ae5d63c4bb5a071edf68af9be96c18a78486d497","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 04:25:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5240bde9cd6479aad2bc86140a3b994 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/372d619f68a9444eabb7a1caacb3d7a1 : SUCCESS in 8m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a28d9c2bda594e0e9096dde1e1469c1a : SUCCESS in 6m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/932309dfaa80477b9f64b06eb5ae4206 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d38253f75fa41f98b9e1c9a0f08cf36 : SUCCESS in 8m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ccfb5de2955648eba8492cc80026b604 : SUCCESS in 5m 44s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/5a9d75518fa541699cd5e351fee02d7f : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c278503588d24ca9a0c9c9cbacbac0eb : SUCCESS in 6m 07s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c503d61a30bf4d50b5d8950d62df6de2 : SUCCESS in 1h 04m 22s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/00cfbf0eb36e4423a9d4e6c889fd7c7b : SUCCESS in 1h 06m 32s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/2e5427ec2963455bb13cee3f2295420b : FAILURE in 1h 42m 04s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/76bc26c6e5d54932b9bcdaae9c396b2e : FAILURE in 1h 32m 17s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/1211e8c149f24d0abaef719f4bc33302 : SUCCESS in 1h 16m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/227d5b94de7942768ebcee96ec6a1eca : SUCCESS in 47m 04s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/46c4e0ad9e944c0eb7f29f4f03042af9 : SUCCESS in 47m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6e33f57a4f143ab8d8385dd9b7574b8 : SUCCESS in 1h 40m 13s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/c48364dbc5ca4e2e81755e9ca3ea35f1 : SUCCESS in 1h 38m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3dc4d8ded3414bf5a29fe74ddde3e349 : SUCCESS in 1h 08m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb9303b298e04145958c900999e2c6e9 : SUCCESS in 8m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9d331eb54c9f4d47c0316fc058a242b7a1e8ac83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 04:28:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d424deffdd5f43f0a2bf7bffdc72005f : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":1},{"id":"54b6351e9811fefcfee2a316f291faf6b665cf31","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-05 14:14:47.000000000","message":"Created a revert of this change as I001068059b02c909e3b7b105f2f168e5bfed8291","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7486abcc34c04f1846eca27bf180f36b408f471f","revisions":{"7486abcc34c04f1846eca27bf180f36b408f471f":{"kind":"REWORK","_number":1,"created":"2019-08-26 14:08:53.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/60/678560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/678560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/678560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/678560/1"}}},"commit":{"parents":[{"commit":"4494c62149a9ce562c5e7e87233119af0a781ada","subject":"worker: Re-add FailoverPreparationForAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4494c62149a9ce562c5e7e87233119af0a781ada"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-08 16:11:20.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-26 14:08:40.000000000","tz":120},"subject":"Use the infra pypi mirror for DIB","message":"Use the infra pypi mirror for DIB\n\nThis patch changes the devstack plugin to use the infra pypi\nmirror if one is declared in the environment.\n\nDepends-On: https://review.opendev.org/675468\n\nChange-Id: Iae839b6e59c43bc4be1e12702b3e2197b038b4c2\n(cherry picked from commit 1a26ba4a054017c285053e217e71e7f816f8df9e)\n(cherry picked from commit 1e43a3071badf10e7cb4132c71673a1ae11d7c4c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7486abcc34c04f1846eca27bf180f36b408f471f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7486abcc34c04f1846eca27bf180f36b408f471f"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
