)]}'
{"id":"openstack%2Foctavia~678562","triplet_id":"openstack%2Foctavia~stable%2Fqueens~Iae839b6e59c43bc4be1e12702b3e2197b038b4c2","project":"openstack/octavia","branch":"stable/queens","hashtags":[],"change_id":"Iae839b6e59c43bc4be1e12702b3e2197b038b4c2","subject":"Use the infra pypi mirror for DIB","status":"MERGED","created":"2019-08-26 14:09:04.000000000","updated":"2019-09-05 14:15:06.000000000","submitted":"2019-09-05 04:33:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678562-1567658014604-233c4fa7","meta_rev_id":"587f4c3f92a7b4b9ddf67e2430e92cc993fb0682","_number":678562,"virtual_id_number":678562,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-05 04:33:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2019-09-05 02:25:13.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-09-03 21:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":1,"date":"2019-09-05 02:25:13.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 21:18:29.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-05 02:25:13.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-09-05 04:33:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"456c1eb1a38e5a45031498ab70fa9af8acc21965","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 14:09:04.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"b245a4772c8d4abed60423f197cc9ab0c18457be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 17:01:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87c6e44f1daa4c568d3181f2cd6b0b56 : SUCCESS in 9m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d4a121bde1164f239dc0252fe52263ad : SUCCESS in 4m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6237c2254708490e8f63b34df6423926 : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b525f7dac9420099d011439a5dae9b : SUCCESS in 9m 21s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/fa260ee8ec1f4544a3363dbbc54abc79 : SUCCESS in 1h 18m 35s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f424666999804ffbb74a66309e7d6078 : SUCCESS in 1h 19m 34s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f32abecc6ed3473f9b0a7a11772de4c1 : FAILURE in 36m 26s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/752e3478594b4690bf330a20ceb9b10e : FAILURE in 28m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c91eb7fa846f401bac006fd38a9a8747 : SUCCESS in 13m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9c46c46189dd477e83be307b4cb7c195 : SUCCESS in 6m 12s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/2e445d33a35545da8f979316fef9b0da : SUCCESS in 8m 30s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5bbff2838b084afda8623f5a2fd08a6c : SUCCESS in 1h 11m 02s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d5f741d3bd594d60947846b0a6316f9c : SUCCESS in 58m 31s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/2dfef6aa09c14139a7e8cd9f6335af02 : SUCCESS in 1h 13m 54s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/e082245ddf6043438c4140621774dd91 : SUCCESS in 1h 09m 03s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/c4de4f6cf1a24e3a856294fab8e93866 : SUCCESS in 45m 19s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/51f48db81e4748208941271515d42013 : SUCCESS in 40m 00s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/ae83e7d16c5e4a17b459c2f0a4cacf4e : SUCCESS in 54m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afba05b8f03b402da5f3d6a5c8af80d5 : SUCCESS in 1h 37m 27s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/7892652dbf4a4d5c9f7205b09c923014 : SUCCESS in 1h 21m 15s","accounts_in_message":[],"_revision_number":1},{"id":"35775e8eae48f61adb84dc5c6ef214f37e2e0b76","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-29 23:00:15.000000000","message":"Patch Set 1:\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"e2b3d992036777378d86424898fc07de03d46148","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-03 21:18:29.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"75ba6ebb43efc7db5b34ec541f96af4ed56339da","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-09-05 02:25:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"79e49b20d033b178436490a9d521ba93ef610085","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 02:25:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8cc57eb2cd5ba2ddb3d4d1f7813afdf5f16d0d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 04:33:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2a48dad0183e2dcd1532da459c85d09cc3d93050","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 04:33:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8519a21744a1433591a66c9f292709d7 : SUCCESS in 9m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7562746d2d204bfc8ee1d5095cd190f2 : SUCCESS in 5m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4e3f4397a3014a938b079df7a733249c : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28bb2fa3fbf545aea4c3ddabd6ab2e7c : SUCCESS in 8m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6167432071d8419c8d86b2138592c0fe : SUCCESS in 9m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/650e7f159cf247a9a5b796cbcef791b0 : SUCCESS in 6m 15s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/fc6be149a3d142e3a656629f7d0b4886 : SUCCESS in 8m 19s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1f3f744fc69b4e78a9693cfe35ee593d : SUCCESS in 1h 22m 14s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/365f023c17714cc9ac3cfde3b16e890f : SUCCESS in 1h 25m 03s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18647a50648547478e03419d205fa5d1 : SUCCESS in 1h 10m 00s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1eee90da427e41249200cad2aef5bf76 : SUCCESS in 1h 25m 19s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/980314cdd34a4507a275c086e45e8a59 : SUCCESS in 1h 06m 56s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/164035137ff844c587149577e1d35758 : SUCCESS in 1h 56m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/78b59d66a84c404991707f39949d50ee : SUCCESS in 46m 37s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/5f53b8dd654242f6a17f8f900af738c2 : SUCCESS in 55m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce2f01c3891d4fa49e35387ad3d6910e : SUCCESS in 1h 37m 29s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/85e9bfa60511438e8da7133456185bd6 : SUCCESS in 1h 18m 37s","accounts_in_message":[],"_revision_number":1},{"id":"56ae4bb4f49fb2d6d6bb49b7ac04f0820d282eb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 04:35:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02808003accb440cb9af66431b3884a0 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1},{"id":"587f4c3f92a7b4b9ddf67e2430e92cc993fb0682","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-05 14:15:06.000000000","message":"Created a revert of this change as I2220577e5121fda53dc5d0ef3e2cc841c21ebfbe","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ae2f61b7331eae4d946914290e5b6a53896f01c","revisions":{"9ae2f61b7331eae4d946914290e5b6a53896f01c":{"kind":"REWORK","_number":1,"created":"2019-08-26 14:09:04.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/62/678562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/678562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/678562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/678562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/678562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/678562/1"}}},"commit":{"parents":[{"commit":"37aad5db40d688c6c718e8a7f892b61b0924143a","subject":"worker: Re-add FailoverPreparationForAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37aad5db40d688c6c718e8a7f892b61b0924143a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-08 16:11:20.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-26 14:09:04.000000000","tz":0},"subject":"Use the infra pypi mirror for DIB","message":"Use the infra pypi mirror for DIB\n\nThis patch changes the devstack plugin to use the infra pypi\nmirror if one is declared in the environment.\n\nDepends-On: https://review.opendev.org/675468\n\nChange-Id: Iae839b6e59c43bc4be1e12702b3e2197b038b4c2\n(cherry picked from commit 1a26ba4a054017c285053e217e71e7f816f8df9e)\n(cherry picked from commit 1e43a3071badf10e7cb4132c71673a1ae11d7c4c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ae2f61b7331eae4d946914290e5b6a53896f01c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ae2f61b7331eae4d946914290e5b6a53896f01c"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
