)]}'
{"id":"openstack%2Foctavia~678563","triplet_id":"openstack%2Foctavia~stable%2Fstein~I4380605bdb346eee824b2ce05fa25175c4cc3034","project":"openstack/octavia","branch":"stable/stein","topic":"lib-get-stable/stein","hashtags":[],"change_id":"I4380605bdb346eee824b2ce05fa25175c4cc3034","subject":"Fix L7 repository create methods","status":"MERGED","created":"2019-08-26 14:09:32.000000000","updated":"2019-08-30 12:31:42.000000000","submitted":"2019-08-30 12:29:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678563-1567168191289-26e757ed","meta_rev_id":"dac60cd6ef4243035a23b16c45543c4617baf966","_number":678563,"virtual_id_number":678563,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-08-30 12:29:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2019-08-30 06:53:16.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-08-29 22:58:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":1,"date":"2019-08-30 06:53:16.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-29 22:58:45.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-08-30 06:53:16.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-08-30 12:29:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bddcea604260eb9cdd42defd611b1d50e059999a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 14:09:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e3f8836f6fc2a31a9de9d9272185b9b616e1fd92","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 14:09:57.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 0f8eabab9c90a6084fc6016ebc3bceccc0e9c703","accounts_in_message":[],"_revision_number":1},{"id":"1dc2ccb5ca626e00c863439523a86edde53b1fee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 18:26:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0abd91d5feb64d69a88b29dd82663c78 : SUCCESS in 8m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30de4076bdc949d0ba81f8494aa2e89a : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/514817f055f94c4990bfcaca17161f55 : SUCCESS in 6m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6b431dd01554d00b8f6d292797b9284 : SUCCESS in 6m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ec52c18a768540a09978371a8f3f2ffe : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3f404cc74184bedb5b0e1886f2cf8e5 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57f00d4c29874d6f92fe1d19318f0f27 : SUCCESS in 7m 48s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/4217fc3b766a4e65a1e0c2f233c6ecbe : SUCCESS in 1h 21m 13s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/1a16203e37084081a215fca823e7296f : SUCCESS in 1h 19m 28s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/571f0ef17985426f8ac229b2339bfdb6 : SUCCESS in 1h 27m 55s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/aa8c6ca95e894a0f86ddb1478e3f743f : SUCCESS in 1h 28m 07s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0e95ec67df804a49aca08f39d63189a9 : SUCCESS in 7m 17s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/6aba6b52e4db441eac3c55f03b4c40c5 : SUCCESS in 8m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/38add8ebdb3f46529581c2c66879d35e : SUCCESS in 8m 24s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0a41565015c94e3897eb728a2cf33fca : SUCCESS in 1h 03m 25s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/47088d7471c7432b8c1cfbff29c2fa92 : SUCCESS in 57m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/999f1f303bbe4c15a575d938d1926250 : SUCCESS in 45m 55s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/65ab2f0d085d4ca4bfe72d64cb14d6ff : SUCCESS in 48m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8208cfb70274d7f943f80c5c19496e6 : SUCCESS in 1h 33m 47s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/671c791a6eca4f6e9bca75942bbce46f : SUCCESS in 1h 29m 37s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/122619282bcd4ca09e5b2d6a8d4dca8c : SUCCESS in 1h 38m 44s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/117c71ddba544f9289c173fdd7fb0e61 : SUCCESS in 2h 12m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d694be728ca14eafb9919dfa2b715d5c : FAILURE in 36m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/26a026658fdc4544848070d332f2b516 : SUCCESS in 1h 16m 16s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/19a29d22898b4d1dae63335bee489aef : FAILURE in 1h 41m 22s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e99f6a8c34b74622a8007723e312ad58 : SUCCESS in 1h 20m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/35449bec777a43069b779253b40e5a17 : SUCCESS in 1h 01m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62d98c570bc570777f16166e3168ea96d7a0adf7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-29 22:58:45.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"afe7cd711114918903234c37da02a16045cd5d73","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-08-30 06:53:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2be032f6e42e93ff1656c364fdad23a9ed1548c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 06:53:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e47a5e7b0f3d3df38896d5a214a99cbf7e0e0f75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 12:29:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9cd611ea689634af3fa73f0c642bd299fe3aa5a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 12:29:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbe1826ca5d3407ab17d99044d7bac4b : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5c2ba05ddce431dabf849345fd3f737 : SUCCESS in 6m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2df418131c344664a2603f4e01e516ac : SUCCESS in 5m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b857a4e9fa5e4815a2c12da448e7f856 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5917a51035a442d80e8d28fd9144986 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/383f509855784704bde0d36caa2e3644 : SUCCESS in 6m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/64359855b9f048de90eddcd03e8fbb9e : SUCCESS in 6m 37s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/fea479ab78594bea9b941863c4153d8b : SUCCESS in 9m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/48c81cdab8044323a9d7e23cd4c5f45c : SUCCESS in 6m 56s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18dd3a8400d54966b68f5b6dbe9e37de : SUCCESS in 1h 10m 49s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b499e94cfb644e959b04ff88490ffd02 : SUCCESS in 1h 08m 47s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1783f8be3df14699b9719d070d6da317 : SUCCESS in 1h 25m 22s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/42dd12cf2a094b7bac6de2087dde7315 : SUCCESS in 1h 27m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/060dd9ce6d7542a48ce81d58e34a1640 : SUCCESS in 45m 53s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9331808074d3404ca4389c0ef5542bab : SUCCESS in 42m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/97b4999fe24e4fa7906c64d389cc6a1b : SUCCESS in 1h 46m 52s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/05dd3702f7ed41b1a6898b1b77fa4509 : SUCCESS in 1h 51m 48s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a2b8dfc14cd141af8957c1ad7679f47b : SUCCESS in 1h 50m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f759c8dfba454e52828c9c911c7320de : SUCCESS in 1h 28m 47s","accounts_in_message":[],"_revision_number":1},{"id":"27e88d6e8e3957fa1b4d370897a7f2a86a898ee0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 12:31:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa3b0a26d9a47168a6a5b9dad65df13 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bdd0d44e99bbe6e95c50a0a9dc978a19ee56a759","revisions":{"bdd0d44e99bbe6e95c50a0a9dc978a19ee56a759":{"kind":"REWORK","_number":1,"created":"2019-08-26 14:09:32.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/63/678563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/678563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/678563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/678563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/678563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/678563/1"}}},"commit":{"parents":[{"commit":"4b7fe7f80c4f66096dc5b625807199463f3e6c64","subject":"worker: Re-add FailoverPreparationForAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b7fe7f80c4f66096dc5b625807199463f3e6c64"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-27 22:41:57.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-26 14:09:32.000000000","tz":0},"subject":"Fix L7 repository create methods","message":"Fix L7 repository create methods\n\nSQLAlchemy can get confused if the parent database objects are not\nlinked when creating new records using the repository.\nFor example, when creating an L7 policy, even if you specify the\nlistener_id, sqlalchemy will drop the ID from the final SQL INSERT\nparameter.\nThis patch corrects the L7 policy and rule create methods to have\nthe required parent objects linked.\n\nStory: 2006305\nTask: 36021\n\nChange-Id: I4380605bdb346eee824b2ce05fa25175c4cc3034\n(cherry picked from commit 36afa82d0c9c61aa74ea922fcc769125afebd962)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bdd0d44e99bbe6e95c50a0a9dc978a19ee56a759"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bdd0d44e99bbe6e95c50a0a9dc978a19ee56a759"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
