)]}'
{"id":"openstack%2Foctavia~678565","triplet_id":"openstack%2Foctavia~stable%2Fqueens~I4380605bdb346eee824b2ce05fa25175c4cc3034","project":"openstack/octavia","branch":"stable/queens","topic":"lib-get-stable/stein-stable/rocky-stable/queens","hashtags":[],"change_id":"I4380605bdb346eee824b2ce05fa25175c4cc3034","subject":"Fix L7 repository create methods","status":"MERGED","created":"2019-08-26 14:10:05.000000000","updated":"2019-09-18 07:22:40.000000000","submitted":"2019-09-18 07:21:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678565-1568791279925-354c625b","meta_rev_id":"7bcdf891d33ad54dfe16d88cc6fc38da594c2d17","_number":678565,"virtual_id_number":678565,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2019-09-11 12:59:00.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-18 07:21:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-18 05:36:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-09-17 16:00:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-09-18 05:36:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 16:00:03.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-18 05:36:46.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-18 07:21:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1060520ef0fa358cc7da64825d4a61eb8779e6d2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-26 14:10:05.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"6e4edb4419e69e79ee8a03e493e4094cd796757d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 17:55:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a4cbebae550467f80cfe7b8809d96d3 : SUCCESS in 11m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/71914a991fd542018f9d6370c6b2ecd1 : SUCCESS in 7m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ac9a898a6b9b45ccb94f4d33757bb0b0 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1a7eb4d8a9d475bb559a4d62963d548 : SUCCESS in 8m 24s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/f9261e33a9814f2296fbd37473687d17 : SUCCESS in 1h 25m 02s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/d885ef0d286842bda0db5c3dfa69ccf3 : SUCCESS in 1h 19m 15s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec95578f1ca04e0d8b6e8583d15f8084 : FAILURE in 1h 15m 28s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ed36c66a85aa4090bf9a5062b6364dee : FAILURE in 1h 16m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1db1675d2eea480995a21a3843d5b5f9 : SUCCESS in 11m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1efaab3e11134cde93d1099081f593f6 : SUCCESS in 8m 29s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/737d7a30ffe14e5e8f62a7425826254d : SUCCESS in 8m 19s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4ac71a240764b1e93c94e02e67f2896 : SUCCESS in 1h 03m 54s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/6e092884b64444ec99035c831c947952 : SUCCESS in 59m 34s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/e4b70a84aea3404cbe73fafc3ee56066 : SUCCESS in 1h 08m 01s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/6051f0736cad4672bbbe04958069840d : SUCCESS in 1h 02m 40s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/beb72bb750744ab6a9be09a39aad4e97 : FAILURE in 49m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/655fbf6ad5824d25ab5f9739d976a403 : SUCCESS in 46m 29s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/a8d3505e77044668b6bb3b1b4e20badb : SUCCESS in 47m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63cec67bc12949f092a319647c547cf3 : SUCCESS in 1h 30m 52s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/1b4e5aa8a00446e3984deb6178dad400 : FAILURE in 1h 36m 19s","accounts_in_message":[],"_revision_number":1},{"id":"4b7cbaa284806e4956664fc621b687c73df5d9f8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-08-28 07:13:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"564b907fb2ea4106e4ff4d82f7ee4a85911ed863","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 09:18:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/404adc8c07b54c6ea737323b6c30953b : SUCCESS in 10m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3c45fc646c44f158d974648d8e438a1 : SUCCESS in 6m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0fd30c96daa64899aa2fe1e1e1b8385a : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a25cdc773704de6a263d8775a4d4095 : SUCCESS in 8m 32s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/bb98cd604d314a0a86b2bf6226671ebc : SUCCESS in 1h 18m 36s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ccb642ba84b841eab1cdb264695d8d7d : SUCCESS in 1h 19m 41s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a2fb864d10824b32b4ca5c8e3309e530 : SUCCESS in 1h 22m 09s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b7be758850374f4e89ac8b0b86167598 : SUCCESS in 1h 22m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb608754e11040daa245c6467bd5be48 : SUCCESS in 10m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a28467570e704e6ab265a6d163b909d6 : SUCCESS in 6m 08s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d3b82a18db154db4a998530f1fb0f4ed : SUCCESS in 6m 13s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4a438a87ba14b3ba156a8be8a2abfdb : SUCCESS in 1h 00m 11s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1eef2d577a544e0cbe3c2d8ccf473f22 : SUCCESS in 1h 04m 11s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/c3cb386e25be4761b803425e04613f5e : SUCCESS in 1h 17m 51s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/5603315fa4a547e88b4294a1a67c3680 : SUCCESS in 1h 11m 18s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/21bee3aec30d45c8976e12bfd1323d3c : SUCCESS in 53m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2e389b4455a74300b395889c149a5c1a : SUCCESS in 47m 29s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/7a1edc3f1d534c6ea5e91ae7a3642771 : SUCCESS in 49m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f73e01218464436dbb429452a5e087cd : SUCCESS in 1h 35m 46s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/9307a51fa76642b7ba3709e4fa25dfab : FAILURE in 1h 29m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5f4e3e1890ccbafe1c7c1c940086c5b029e773b5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-11 12:59:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"402c02cf3d66f723585822dd3e85c5c127dc5685","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 14:40:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd03a8d8b044463a797c36a5021c554 : SUCCESS in 10m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0fe07e00a53346e99b14438534209ad0 : SUCCESS in 5m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6514a29ffb8245e78807bea8601b2c5c : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9157577341de424c92526e9dc338111d : SUCCESS in 6m 44s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/5c42e2e0db4b450f88316631013d7a45 : SUCCESS in 1h 16m 41s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/2e0d49f1719a4445b9e8a478c23ebc8e : SUCCESS in 1h 25m 16s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d2e1d79fdb74ed9906c0cdfeefe89c4 : SUCCESS in 1h 25m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/423210f536324183bec539938ce71573 : SUCCESS in 1h 24m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e55c65d431d849949074bf24d8e7a5ba : SUCCESS in 9m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b48bb15d3e6b44069bfa09176b608050 : SUCCESS in 6m 55s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/63aa0c29b7b14801862380a559c0ec19 : SUCCESS in 6m 41s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3a544bac94ed4e33b49856ab497be9f2 : SUCCESS in 1h 10m 33s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f7b2946e80d749569c5efa593319b01b : SUCCESS in 57m 10s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/00a703c764c647dda5cdd91e580e8f38 : SUCCESS in 1h 11m 14s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/4634d9cdc819430296ef8e3d8d25e55d : SUCCESS in 1h 13m 09s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/88185efba8564169b74a7897df2a2d3b : SUCCESS in 1h 14m 57s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fe922616d19a4775a843f170483867af : SUCCESS in 43m 06s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/dd8c84f94f29430a81c7809b10bc0ec6 : SUCCESS in 44m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0ecbee63b9fc4dd5a3bb1b3a8642d0ee : SUCCESS in 1h 31m 25s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/3f4c3e0361bf4ea78dadc8f962e1bdb3 : SUCCESS in 1h 31m 26s","accounts_in_message":[],"_revision_number":1},{"id":"920e095327c2dfc77f1c77b0a2707b528eeba446","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-17 16:00:03.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"4380581beb483e11398900b11197b3b6996edf28","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-18 05:36:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"430690b60350036d63641f1c809dff7abdd35948","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 05:36:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"04abdc1ec014ebf86eecdea7f11b0ad2a9f2587a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 07:21:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2649512a11664e6db11be2a1cdfee6a0 : SUCCESS in 10m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f4e4a6c4e3b3485ebaaeaa956a9bb7b9 : SUCCESS in 4m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/98ee12be4fd047ca81c798e94f5350a5 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2af75dddae154c2da7f9e36b4b0580e4 : SUCCESS in 6m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64eda017c1db4b04918916333669183b : SUCCESS in 9m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/13f6b267c0504072964efcddf651f7a4 : SUCCESS in 5m 42s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d7e7785eec014c6c9e1ce91b3e462c14 : SUCCESS in 6m 29s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/56ccb23d4a5f4a20b600c778314df506 : SUCCESS in 1h 15m 38s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/d3f881fd479d4f0ca730580b29ea0789 : SUCCESS in 1h 19m 12s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82398a4bb6be421bbb09c8f05ade9c6b : SUCCESS in 1h 02m 30s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/420832f7ea2c47b99820c11d8766f377 : SUCCESS in 1h 01m 18s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/f5bf30a1e3434d8382d78423410ddea7 : SUCCESS in 1h 15m 22s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/54f032461bc9423a9650ed5380badd3c : SUCCESS in 1h 14m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7d54ce934d294aab8f5bf7cb816b48d3 : SUCCESS in 1h 04m 20s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/75ea2a33ec014d0b97cc571e803f44f4 : SUCCESS in 49m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d64210ae8b24770ab6be2f33508a4a0 : SUCCESS in 1h 32m 49s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/62651bd8f18546b4853893ffdb2e42f4 : SUCCESS in 1h 31m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e8034e3210d7ce15d292ae426644ed5c17dd1e76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 07:21:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b67233cf8963d6caaf65e0ed22c3e4cf427392fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 07:22:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff189c3acd648cc996004d5329708c1 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66d71b015009e2eeb1a71f1211e7ebe27f624a74","revisions":{"66d71b015009e2eeb1a71f1211e7ebe27f624a74":{"kind":"REWORK","_number":1,"created":"2019-08-26 14:10:05.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/65/678565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/678565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/678565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/678565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/678565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/678565/1"}}},"commit":{"parents":[{"commit":"37aad5db40d688c6c718e8a7f892b61b0924143a","subject":"worker: Re-add FailoverPreparationForAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37aad5db40d688c6c718e8a7f892b61b0924143a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-27 22:41:57.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-08-26 14:10:05.000000000","tz":0},"subject":"Fix L7 repository create methods","message":"Fix L7 repository create methods\n\nSQLAlchemy can get confused if the parent database objects are not\nlinked when creating new records using the repository.\nFor example, when creating an L7 policy, even if you specify the\nlistener_id, sqlalchemy will drop the ID from the final SQL INSERT\nparameter.\nThis patch corrects the L7 policy and rule create methods to have\nthe required parent objects linked.\n\nStory: 2006305\nTask: 36021\n\nChange-Id: I4380605bdb346eee824b2ce05fa25175c4cc3034\n(cherry picked from commit 36afa82d0c9c61aa74ea922fcc769125afebd962)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66d71b015009e2eeb1a71f1211e7ebe27f624a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66d71b015009e2eeb1a71f1211e7ebe27f624a74"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
