)]}'
{"id":"openstack%2Foctavia~678923","triplet_id":"openstack%2Foctavia~master~I3ec135766c9a1ddb7ac6655c0ee1ccb1e78ead5c","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I3ec135766c9a1ddb7ac6655c0ee1ccb1e78ead5c","subject":"Use dual intermediate CAs for devstack","status":"MERGED","created":"2019-08-27 17:54:40.000000000","updated":"2019-09-18 08:59:37.000000000","submitted":"2019-09-18 08:59:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678923-1568797177647-1830412a","meta_rev_id":"4faa7f4e6a7a25d21d43441dbf083ec92191f37b","_number":678923,"virtual_id_number":678923,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-09-18 08:59:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-18 05:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-13 16:07:42.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-09-18 05:54:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"disliked":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":-1,"date":"2019-09-18 05:54:29.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":-1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-13 16:07:42.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-18 05:54:29.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-18 08:59:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b93f303ed86444fc646f54a54b77d003624b83cf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-27 17:54:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77f5f0ddf99622edc1156b82cb167964d6645cdc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-27 17:54:48.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"fa34295311bae37cd48f2b69311fd3d7c6e7026b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-27 18:07:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"02aab8a378ed5b7b23017c10b2d60db9318745ed","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-27 18:08:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"283b409f41446d0bd408f3ec537abd376c6377d5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-27 18:08:52.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"9519c7e2e71ce42d79f314160c45b1683f69ad3d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-27 18:46:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"32de52ef8ad452248d48a9967572be51d9852cc2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-27 19:49:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a03791664701d49192fe7362bf4f94547515f75a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 22:41:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/476be8b3800c400983d1bdd097a4b717 : SUCCESS in 8m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/691e511874a14c019940edc4cb785d25 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cefc4bff000431e81613311e561abe1 : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e371429a1d304c3e9186b1ccf9ff64d2 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b71fbcc14ca4c42940034e3f6cc17be : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2afc79b5fce240bf8b385b503232588a : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78f3e50dde6142e38f23d025a926eb9d : SUCCESS in 9m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ba5321acf9bb4c7cac8ee316d2267656 : SUCCESS in 5m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d2f0e85e21a44d5b83d61353dbaf429b : SUCCESS in 6m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dfdb5951f69f4ec5b34a4dec5e667da4 : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/49f96815c5584e82b3117d14ed04c397 : SUCCESS in 7m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/71e8a7a39e2f48e6874e48a0cc622e27 : SUCCESS in 49m 31s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c3a218ef01824b579527af2820c7ef00 : SUCCESS in 45m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/29b92d6c32334f6f966bdef7d15743bc : SUCCESS in 1h 27m 38s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/01f0bd7238a94248abdace3fac187371 : SUCCESS in 2h 02m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5cf7fd0809234d23bfcfefca06b0f016 : SUCCESS in 2h 04m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/27e05e049bc4409391e88b91f5820944 : FAILURE in 1h 10m 34s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2cf0ef7f2a9402badceebb6d8b98b6a : FAILURE in 1h 04m 47s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c213fd3da4934d048749fbd4db8c612e : SUCCESS in 1h 11m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3954754cf2ca4c4f9d7247026d4dfdde : SUCCESS in 1h 04m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6066cd607ed74a30b9eaab2b112df37e : SUCCESS in 1h 29m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/70977f08b3c54644abdef5cc10d13637 : SUCCESS in 1h 08m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/58077561ee074b59828b6955d0a7d494 : SUCCESS in 1h 11m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1a72ac3755ad435cad5ea113a4a82e3d : SUCCESS in 1h 19m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"36d19f1ec4a6f2d2e04d8ed7e41576334958dd11","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-27 23:13:21.000000000","message":"Patch Set 5: Workflow-1\n\nNeed to fix multi-node","accounts_in_message":[],"_revision_number":5},{"id":"74b539c5d833e448559fc997d3962ea031eda058","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-27 23:28:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a14dcea480558d537ec7988cbc2dddf6f431d976","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 01:41:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3d2ceb81e2c48d6a16675e57c047400 : SUCCESS in 8m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70197be1125747adbdc08ee205dc1ff8 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c35191be6d93422291757a944dafe922 : SUCCESS in 6m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/810b8d31b7c748b9b1ab14fc70e30352 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e694e97a63464c56b72bf67870d0e5d9 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d70b9ad6d9b04942b92dc5acf2381ab2 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc7fe988492a436081a45a9264968938 : SUCCESS in 7m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/53d9e43b2ca446e7b3c3bdec75d9371c : SUCCESS in 6m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/091b691c053f4d02b0996f0cf527d836 : SUCCESS in 6m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dc24ba9d4e334328bf4b4dc64cd49612 : SUCCESS in 8m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/71b8984b2e664c50983070609e6cd3ee : SUCCESS in 5m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/77822abf614548579da25edd22d29442 : SUCCESS in 53m 41s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8583885fb1ce4bed846613ce28dedb6f : SUCCESS in 57m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de4712a590e34eaab7a7634d56ca84e4 : SUCCESS in 1h 43m 45s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/85dd5e86b07f464680a422dc0225a5bb : SUCCESS in 1h 55m 43s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/638bf05f8cac4a24956a2eec8adcf3f5 : SUCCESS in 1h 55m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/551d646372ba4449b435f7b7e458bb0c : SUCCESS in 2h 05m 56s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/fdf3b32ee2de4a1fb312b82c9f9a60d7 : SUCCESS in 1h 51m 39s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e9cdb9b93f264106bccbc721f16aa11d : SUCCESS in 1h 22m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ab7202a725094743b6692a0fca67b636 : SUCCESS in 55m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/522dc48a102f4216883c99bf72c845f6 : SUCCESS in 1h 36m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7d541e3c355474989e8605e5c76cf02 : SUCCESS in 1h 13m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ecacf7a1bbf2481798ca341d9cabd0c0 : SUCCESS in 1h 27m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/94272d28b56e42dba3ef26d7fd05db2d : SUCCESS in 1h 20m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"58abe48a1ea3df2744aed56df3a732aa9fd7ba8b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-13 16:07:42.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"53b6d4a60dcb1439e6194bad6652e22d913d7522","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-18 05:54:29.000000000","message":"Patch Set 6: Backport-Candidate-1 Code-Review+2 Workflow+1\n\nI wonder if we could have made the single and dual CA scripts just be a single script with an arg or something? but this is fine for now. It\u0027s a good proof. no need to backport tho.","accounts_in_message":[],"_revision_number":6},{"id":"dc3194016c68355022875966f507dd6e75f0cb32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 05:54:46.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"28f992dc2098d8b7beae834cb6e0216265c40e9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:59:37.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e34e1ba6174a4a31acd70bb85b494f10 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b93709bb12fc4b37aba6578b1b6564cb : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8f38baa3136745ff97fc738a7c2809a8 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ece08fc2592144f2ad054ef819919f66 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5cabc0cb271c4ee498770ebc90dba117 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1101b156cfa2474198e2ef108a3fab38 : SUCCESS in 7m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f7fb371c9e5643d59df32cc20349b143 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b876d06eee64255a5652286b8a8161d : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/51dfa5788d864e96a1a150329be15679 : SUCCESS in 44m 45s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8faaa83f89ce47b39f6729e2db15c286 : SUCCESS in 48m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f0909a35edf41f2a38a91fcad713e1a : SUCCESS in 1h 45m 04s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2772fb97367a48c780d52bb582997186 : SUCCESS in 1h 43m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5d9a1ae57a0f4d9cbf8bf96f8973869b : SUCCESS in 1h 18m 35s","accounts_in_message":[],"_revision_number":6},{"id":"4faa7f4e6a7a25d21d43441dbf083ec92191f37b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:59:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"950faea6e99433c162f460a8ec38b2688fdd5e99","revisions":{"e65aa0a39adaf784dfd859ef9e5b069b00d0608f":{"kind":"REWORK","_number":1,"created":"2019-08-27 17:54:40.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/23/678923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/678923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/678923/1"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-27 17:51:52.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-27 17:54:39.000000000","tz":-420},"subject":"Use dual intermediate CAs for devstack","message":"Use dual intermediate CAs for devstack\n\nThis patch updates the devstack plugin to use a dual certificate\nauthority (CA) with intermediate CAs for the Octavia controller\ndeployment.\nThis is a more realistic deployment model for testing.\nNote: This change uses weak security to save gate resources. Please\nrefer to the Octavia Certificate Configuration Guide for production\ninstructions.\n\nChange-Id: I3ec135766c9a1ddb7ac6655c0ee1ccb1e78ead5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e65aa0a39adaf784dfd859ef9e5b069b00d0608f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e65aa0a39adaf784dfd859ef9e5b069b00d0608f"}]},"branch":"refs/heads/master"},"76dbc425aa1e49ee99f998c21986d1205320b79a":{"kind":"REWORK","_number":2,"created":"2019-08-27 18:07:46.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/23/678923/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/678923/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/678923/2"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-27 17:51:52.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-27 18:07:21.000000000","tz":-420},"subject":"Use dual intermediate CAs for devstack","message":"Use dual intermediate CAs for devstack\n\nThis patch updates the devstack plugin to use a dual Certificate\nAuthority (CA) with intermediate CAs for the Octavia controller\ndeployment.\nThis is a more realistic deployment model for testing.\nNote: This change uses weak security to save gate resources. Please\nrefer to the Octavia Certificate Configuration Guide for production\ninstructions.\n\nChange-Id: I3ec135766c9a1ddb7ac6655c0ee1ccb1e78ead5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76dbc425aa1e49ee99f998c21986d1205320b79a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76dbc425aa1e49ee99f998c21986d1205320b79a"}]},"branch":"refs/heads/master"},"ba6984c31c19ddbc033870d327a29223197e920e":{"kind":"REWORK","_number":3,"created":"2019-08-27 18:08:29.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/23/678923/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/678923/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/678923/3"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-27 17:51:52.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-27 18:08:23.000000000","tz":-420},"subject":"Use dual intermediate CAs for devstack","message":"Use dual intermediate CAs for devstack\n\nThis patch updates the devstack plugin to use a dual Certificate\nAuthority (CA) with intermediate CAs for the Octavia controller\ndeployment.\nThis is a more realistic deployment model for testing.\nNote: This change uses weak security to save gate resources. Please\nrefer to the Octavia Certificate Configuration Guide for production\ninstructions.\n\nChange-Id: I3ec135766c9a1ddb7ac6655c0ee1ccb1e78ead5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba6984c31c19ddbc033870d327a29223197e920e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba6984c31c19ddbc033870d327a29223197e920e"}]},"branch":"refs/heads/master"},"2c2f4fd3ccdbc970245e15322eddd5fc9fb61427":{"kind":"REWORK","_number":4,"created":"2019-08-27 18:46:04.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/23/678923/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/678923/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/678923/4"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-27 17:51:52.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-27 18:45:58.000000000","tz":-420},"subject":"Use dual intermediate CAs for devstack","message":"Use dual intermediate CAs for devstack\n\nThis patch updates the devstack plugin to use a dual Certificate\nAuthority (CA) with intermediate CAs for the Octavia controller\ndeployment.\nThis is a more realistic deployment model for testing.\nNote: This change uses weak security to save gate resources. Please\nrefer to the Octavia Certificate Configuration Guide for production\ninstructions.\n\nChange-Id: I3ec135766c9a1ddb7ac6655c0ee1ccb1e78ead5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c2f4fd3ccdbc970245e15322eddd5fc9fb61427"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c2f4fd3ccdbc970245e15322eddd5fc9fb61427"}]},"branch":"refs/heads/master"},"198daf308de8bfa13b5376c4980966e7ae7904ca":{"kind":"REWORK","_number":5,"created":"2019-08-27 19:49:40.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/23/678923/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/678923/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/678923/5"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-27 17:51:52.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-27 19:49:34.000000000","tz":-420},"subject":"Use dual intermediate CAs for devstack","message":"Use dual intermediate CAs for devstack\n\nThis patch updates the devstack plugin to use a dual Certificate\nAuthority (CA) with intermediate CAs for the Octavia controller\ndeployment.\nThis is a more realistic deployment model for testing.\nNote: This change uses weak security to save gate resources. Please\nrefer to the Octavia Certificate Configuration Guide for production\ninstructions.\n\nChange-Id: I3ec135766c9a1ddb7ac6655c0ee1ccb1e78ead5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198daf308de8bfa13b5376c4980966e7ae7904ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198daf308de8bfa13b5376c4980966e7ae7904ca"}]},"branch":"refs/heads/master"},"950faea6e99433c162f460a8ec38b2688fdd5e99":{"kind":"REWORK","_number":6,"created":"2019-08-27 23:28:32.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/23/678923/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/678923/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/678923/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/678923/6"}}},"commit":{"parents":[{"commit":"9ec9859bbf3361363bb91dee057e4ed2ea7c8212","subject":"Merge \"worker: Re-add FailoverPreparationForAmphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ec9859bbf3361363bb91dee057e4ed2ea7c8212"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-27 17:51:52.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-27 23:28:12.000000000","tz":-420},"subject":"Use dual intermediate CAs for devstack","message":"Use dual intermediate CAs for devstack\n\nThis patch updates the devstack plugin to use a dual Certificate\nAuthority (CA) with intermediate CAs for the Octavia controller\ndeployment.\nThis is a more realistic deployment model for testing.\nNote: This change uses weak security to save gate resources. Please\nrefer to the Octavia Certificate Configuration Guide for production\ninstructions.\n\nChange-Id: I3ec135766c9a1ddb7ac6655c0ee1ccb1e78ead5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/950faea6e99433c162f460a8ec38b2688fdd5e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/950faea6e99433c162f460a8ec38b2688fdd5e99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
