)]}'
{"id":"openstack%2Foctavia~679153","triplet_id":"openstack%2Foctavia~master~Ieab13c1152fed64da7390891b315b5e67513ce3e","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ieab13c1152fed64da7390891b315b5e67513ce3e","subject":"Fix base (VRRP) port abandoned on revert","status":"MERGED","created":"2019-08-28 22:34:17.000000000","updated":"2019-09-18 08:45:33.000000000","submitted":"2019-09-18 08:15:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679153-1568794525314-7b169225","meta_rev_id":"d8651e8dd0cb208779fffa98ad12b3771452a7f1","_number":679153,"virtual_id_number":679153,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-18 08:15:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"recommended":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2019-09-03 08:27:59.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2019-09-18 05:38:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-13 11:56:19.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-09-14 09:17:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-14 01:07:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2019-09-18 05:38:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-13 11:56:19.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-09-04 16:09:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 08:27:59.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2019-09-13 11:56:19.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-14 01:07:45.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-09-14 09:17:05.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-09-18 05:38:45.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-18 08:15:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea2c3ab83eaaddbb7c137e2b6a92cf68495ade48","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-28 22:34:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7afc1cee829b25cf75d7475b52c96c5d2eca593e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-28 22:34:37.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"87f64e8afcafe2c01d4e6e80b47ef74e507d8d04","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-08-28 23:23:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c3889904a3d45df86f192e57037803172d227503","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 01:47:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0062414a73bd4699a31f6d568cd1d0aa : SUCCESS in 8m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/911557cedaf54d1f974249bada263c6e : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f155b42d88e47478eaa8104a29d797b : SUCCESS in 6m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56e33a36a04a4ac0bb1f401803348a87 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3850757a231424f82452336c4cbac03 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/056e3d1103f64c6980bf19958037f68b : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/689e144807d7457d9c2d43a4244da99e : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b59830f81b58400c86aea5af16e852dd : SUCCESS in 5m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/75d4a0446b5f4934b0b25aefe5baf2f9 : SUCCESS in 6m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6d699bbc6376467a959e2882eba123b3 : SUCCESS in 6m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/804ec23c648d401b9ac5662f0d3baa75 : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/27df940970ef46dea1f3bf09f9cab913 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4c22bbacfbc8438a9cc1816f059fe379 : SUCCESS in 45m 06s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a2cd3d0a72944a00b6429719c8ca5554 : SUCCESS in 51m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/871b758c30f140f0a40d3bc90ce2a594 : SUCCESS in 1h 35m 53s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1dd973772d4a44dda30fdef0eab9a63e : SUCCESS in 1h 37m 48s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2b14e0d13b844c3a8ffd7e3071de5ee4 : SUCCESS in 2h 08m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/deee31fdbda44375a11b0576b98f882d : SUCCESS in 1h 26m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/2d0c2b777a734fd1b045364feed62132 : FAILURE in 1h 49m 27s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3588105df927429abb22f130fadcc549 : SUCCESS in 1h 21m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3fe772e75eea4113bc1293713bc4af2d : SUCCESS in 54m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5bdfad273f814217b42a9468bafb6022 : SUCCESS in 1h 36m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/996661e8db224cf1ba98ff6999fceaf3 : SUCCESS in 1h 05m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fb4849aa919e4d048f6d95fc75268194 : SUCCESS in 1h 04m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/119fa987a0f84394876c7a6dd30e6408 : SUCCESS in 1h 27m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"adf8eb878a9f29ee4da0bc7cb6f3d0c42abb9c54","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-09-03 08:27:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e7816cff2a922ab6bb1fdcb71712f016159d966f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-04 16:09:06.000000000","message":"Patch Set 2: Backport-Candidate+2","accounts_in_message":[],"_revision_number":2},{"id":"0a0d95ef425c1ec2500942bb4e751f553f1eceb3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-13 11:56:19.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b956d8e302d3cd8b12debdfd202e76e82c0ad5f5","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-09-14 01:07:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"08726b97a996101d9c44fbf0ac720e1292e7f050","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-09-14 09:17:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"298a45b844bd573f17d955540a9034ffac76f7a0","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-18 05:38:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLooks like a good bugfix to me.","accounts_in_message":[],"_revision_number":2},{"id":"572a6630dbfccb09865af4de027c18218c087789","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 05:39:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"be9377ea40f54fb3e76c4fd20e65e4d0f2b34d31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:15:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"89880611b1de46e74ba49db9c867b59622e63b35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:15:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e909532c9fa4cc7ba18a67333504183 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/320a3231615346f4be31ae72d476c963 : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9d7678c6b09c46aba3b4b832c27ec2be : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cf526af8df8461a9b9f67f4893da280 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45c340b3104f4d03b87280dc5427902b : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccbe98e1d94c41c391b39bf39123b2b5 : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f998cbdfe3b4ce78ddd38033f6c63f1 : SUCCESS in 6m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/da1c83f62dd040e3b21e0357d67b516f : SUCCESS in 7m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/74e01bf761cd483d8a9551ff0614d21b : SUCCESS in 7m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/365bb9c2fb1b40e48c4cfcd0dcc8921d : SUCCESS in 44m 10s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/987f18e4de144038840d32a60e8429d8 : SUCCESS in 50m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf45e9d837a64c74b8cb83ed425e16f1 : SUCCESS in 1h 59m 12s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/28da7694305c4e9c9ea3ad5605922e17 : SUCCESS in 1h 54m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d6d16436016b4b3cab9c21c54e2c79ed : SUCCESS in 1h 38m 25s","accounts_in_message":[],"_revision_number":2},{"id":"a5580673f3d3343ae5c7f04e31df02101aea792c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:17:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4753ea1d0c634ef0af1ea24890115dbb : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/443d98201e9344cc9d34464510b4e4ad : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2},{"id":"83f2067c3cc6d1483f55256fd6acccd0dbc88f6f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-09-18 08:23:52.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 701a900164689aa88549232961f91c7d9c38c03a","accounts_in_message":[],"_revision_number":2},{"id":"d8651e8dd0cb208779fffa98ad12b3771452a7f1","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-09-18 08:45:33.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit bd47d3554012fab6dae4dd34c2008458c93fce26","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f20dc158441faad8422f7836dc7d0af50dcbcb0a","revisions":{"2dfb2776d66ab1761974b602aab568fd88cae814":{"kind":"REWORK","_number":1,"created":"2019-08-28 22:34:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/53/679153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/679153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/679153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/679153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/679153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/679153/1"}}},"commit":{"parents":[{"commit":"60bd083ce640981e6ccd00cbff76c32ff73cf829","subject":"Merge \"Force DIB Python version for py2 in diskimage-create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60bd083ce640981e6ccd00cbff76c32ff73cf829"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-28 22:31:07.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-28 22:34:16.000000000","tz":-420},"subject":"Fix base (VRRP) port abandoned on revert","message":"Fix base (VRRP) port abandoned on revert\n\nIf the base (VRRP) port fails to attach to the amphora instance, the\nport would not be deleted as part of the revert cleanup.\nThis patch splits the two plug VIP phases and attempts to clean up the\nbase (VRRP) port, that was created in the first phase, should the port\nattach fail.\n\nChange-Id: Ieab13c1152fed64da7390891b315b5e67513ce3e\nStory: 2006468\nTask: 36399\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2dfb2776d66ab1761974b602aab568fd88cae814"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2dfb2776d66ab1761974b602aab568fd88cae814"}]},"branch":"refs/heads/master"},"f20dc158441faad8422f7836dc7d0af50dcbcb0a":{"kind":"REWORK","_number":2,"created":"2019-08-28 23:23:29.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/53/679153/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/679153/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/679153/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/679153/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/679153/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/679153/2"}}},"commit":{"parents":[{"commit":"60bd083ce640981e6ccd00cbff76c32ff73cf829","subject":"Merge \"Force DIB Python version for py2 in diskimage-create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60bd083ce640981e6ccd00cbff76c32ff73cf829"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-28 22:31:07.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-08-28 23:23:23.000000000","tz":-420},"subject":"Fix base (VRRP) port abandoned on revert","message":"Fix base (VRRP) port abandoned on revert\n\nIf the base (VRRP) port fails to attach to the amphora instance, the\nport would not be deleted as part of the revert cleanup.\nThis patch splits the two plug VIP phases and attempts to clean up the\nbase (VRRP) port, that was created in the first phase, should the port\nattach fail.\n\nChange-Id: Ieab13c1152fed64da7390891b315b5e67513ce3e\nStory: 2006468\nTask: 36399\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f20dc158441faad8422f7836dc7d0af50dcbcb0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f20dc158441faad8422f7836dc7d0af50dcbcb0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
