)]}'
{"id":"openstack%2Foctavia~679768","triplet_id":"openstack%2Foctavia~stable%2Fstein~I9bdd6f2913c0da2b0772b4b694309ef993289360","project":"openstack/octavia","branch":"stable/stein","topic":"bug/2005947","hashtags":[],"change_id":"I9bdd6f2913c0da2b0772b4b694309ef993289360","subject":"Add failover logging to show the amphora details.","status":"MERGED","created":"2019-09-03 12:05:44.000000000","updated":"2019-09-11 18:20:46.000000000","submitted":"2019-09-11 18:18:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679768-1568225912583-b56adfd4","meta_rev_id":"443c741a479647a74e98afec59af73a804f2f3a9","_number":679768,"virtual_id_number":679768,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9057,"name":"Noboru Iwamatsu","email":"rockpine@gmail.com","username":"r0ckpine"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2019-09-11 13:24:15.000000000","_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2019-09-11 18:18:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":9057,"name":"Noboru Iwamatsu","email":"rockpine@gmail.com","username":"r0ckpine"},{"value":2,"date":"2019-09-05 06:44:16.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-09-03 21:25:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":9057,"name":"Noboru Iwamatsu","email":"rockpine@gmail.com","username":"r0ckpine"},{"value":1,"date":"2019-09-05 06:44:16.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2019-09-03 12:05:44.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9057,"name":"Noboru Iwamatsu","email":"rockpine@gmail.com","username":"r0ckpine"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":9057,"name":"Noboru Iwamatsu","email":"rockpine@gmail.com","username":"r0ckpine"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 12:05:44.000000000","updated_by":{"_account_id":9057,"name":"Noboru Iwamatsu","email":"rockpine@gmail.com","username":"r0ckpine"},"reviewer":{"_account_id":9057,"name":"Noboru Iwamatsu","email":"rockpine@gmail.com","username":"r0ckpine"},"state":"REVIEWER"},{"updated":"2019-09-03 21:25:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-05 06:44:16.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-09-11 18:18:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6bd649e1e4dd8a50ca73902120b45b651cd1a291","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-03 12:05:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a8f1cc956d61c271b91b507a34d04725fb6a347","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 14:44:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56a5962b5ed74370901607125ce1e852 : SUCCESS in 9m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a08b9f0f67604fd49c75212a5be4a859 : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bba781322a841b89b167979cf72d4d7 : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d05e003cd6a9482d8ab99943e060b649 : SUCCESS in 4m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/30ef6c58304848e4bddef7a88bab6cd5 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2b98a1042a742bf824e95911b15c3a3 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0260b53eb49b4fea913f7c8dab5de12a : SUCCESS in 7m 17s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d63419bd0c2a449a861556341c4c6328 : SUCCESS in 1h 18m 11s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/90fa5e11c59341879587f18ab7a97ac8 : SUCCESS in 1h 15m 07s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f7b8e244dc7444590371b52b8385d73 : SUCCESS in 1h 05m 35s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/357be1ed6bb941aa98c5d230e480becf : SUCCESS in 1h 31m 44s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3f5c9e3d212d4aeea8a7e932be348285 : SUCCESS in 7m 12s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c16f45c44767416a9c7ef8ed5d3c8c91 : SUCCESS in 7m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5eb254d274154d6a87dc149ba1aa7567 : SUCCESS in 6m 59s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5dd3fca277aa40d2b4244f7d92ac839f : SUCCESS in 1h 10m 13s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/77298711f1e54f73b5cf4d42956232a0 : SUCCESS in 56m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/99c88d8b4fb2484cbc56dd3264f96e28 : SUCCESS in 52m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f3965f2967e741fa96f9bd3a0d6a6691 : SUCCESS in 48m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e88f7e0daff34e35b2b2e857ce33d14c : SUCCESS in 1h 44m 44s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/019536313c6647a1ba1869a53eb3b5fd : SUCCESS in 1h 52m 41s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/8837fb46e9f0435b9b7dd4a5238e95c6 : SUCCESS in 1h 46m 23s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4450a9ee74de48368b60f5df3d3b2384 : SUCCESS in 2h 32m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c0a7de158b24fbda69360be574b21ea : FAILURE in 33m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9244d135e5594db0bd920606bf145945 : SUCCESS in 1h 29m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/9eaf78730ddf4a4ba17b9b19cd26507c : SUCCESS in 1h 49m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/44c9148c18a54db68bffaabd5ceb53e0 : SUCCESS in 1h 15m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2e27040d85034ff6993c4306fc770b40 : SUCCESS in 56m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77ed4716133726a4404078bd746d40c7e4f17cec","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-03 21:25:38.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"23b8fc708ccadf8b323fd5a336f2582eb1d5d000","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-09-05 06:44:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c1329c12d26791bf01dc1efc08e4f2a3943c02b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 06:44:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cad66e4ebd2a034ccfa7aa0afe0bbcc05fcdae2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 11:36:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43cf97e6bad847e68f3478901804ed3c : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72b44ba276564887a118b6b074f3be92 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/55dace95b24345d990c3773fe9e47a87 : SUCCESS in 4m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/483c3d89c6f14f70984de4b8eae949fe : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6546ffa4b07d4ce78a6117cd7b08d9b9 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/833a06d72a4f412780c4df4a2c4241be : SUCCESS in 8m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b851d850e9cf4196a8a37da1c379f4ec : SUCCESS in 6m 20s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/a556e711af804f079d63e8ce86be9077 : SUCCESS in 8m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/57887bafe3ef466185e39be842e7fcac : SUCCESS in 7m 04s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/839a2a114e084fc58e01812c73bfd9c1 : SUCCESS in 1h 10m 42s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/44df4b71467448f7a9bade08b288a67d : SUCCESS in 1h 04m 54s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/65c8b6ba58794a52b7f4a28b665f8009 : SUCCESS in 1h 12m 16s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/73e6054d28cf4dfa9493acecce7dd205 : SUCCESS in 1h 20m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/034c300d9ec44e6fa3b8557df2a55e41 : SUCCESS in 43m 08s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c7d36def244a4457bd02c43a271521fa : SUCCESS in 44m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b0e84a1232c4bf7b56463cc98085c40 : SUCCESS in 1h 44m 32s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ef7c2daf55d1481985100549378afaf3 : SUCCESS in 1h 43m 22s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/1b9f6ec8f8334036a34d52b503ccb312 : SUCCESS in 1h 45m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1381c17fb876495eb5431134f9ebf8aa : FAILURE in 36m 55s","accounts_in_message":[],"_revision_number":1},{"id":"bb07175dcfc3af6bc1b212cbfd5ba7d42e845d1a","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-09-09 02:54:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"94c2ef241df30cd87d664a7f78c1444516e664bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 05:11:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49dd2a78235b4312be8832aebbebf802 : SUCCESS in 8m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36fa4ac2720441ce846723dee8dcb02f : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4d2cc3bf1df4e6cbf40cfedab530e2f : SUCCESS in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/99ffca43aa4b4c41ab6aaad63bdd6651 : SUCCESS in 5m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e45ec0e4fcd74714ba9bc652efa29d02 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aceaa3d2bf224826a796587db8d8659a : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc3dd526bc014ad086f381bf064d0555 : SUCCESS in 7m 52s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/dbffb8cbd5cb4184a723023e156d5f2b : SUCCESS in 1h 23m 33s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/913b24675da34f0a8035dd51ebf32511 : SUCCESS in 1h 32m 12s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d98a8b0c07e145cdbef18c5dab6d0afa : SUCCESS in 1h 28m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/10b44beab325475f928c33643c28151c : SUCCESS in 1h 04m 54s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/efacf88ac46141a0af9426c9227e9471 : SUCCESS in 4m 54s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/2f6fc164d8d04aa89c17134649c0d0b5 : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0f7910e6f2184b06be868bde08fb69d8 : SUCCESS in 6m 39s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/66a7c3e7aefa4737854880f5206db89e : SUCCESS in 59m 35s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/24551957ff7c400baf7eeb261f3cc8e7 : SUCCESS in 58m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a244d881765845b2841e755770cd3ae6 : SUCCESS in 42m 10s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/bea58ce9cde64a2e89ead625a225f846 : SUCCESS in 43m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e3bdddaa990483aa819c6e0903b46bd : SUCCESS in 1h 28m 45s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fe60bddd46e14d8e83d652567d0553f0 : SUCCESS in 1h 39m 52s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/c35d5afd716e46c68335b805ab398352 : SUCCESS in 1h 34m 27s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/87f6fa8bcbbe4a9d9d5bbb24eb8dfbb6 : SUCCESS in 2h 14m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b230d0010ca8406cb8645f29724f9e9d : FAILURE in 31m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/134e3707ed0b4ecead557ad9d8d26311 : FAILURE in 1h 29m 47s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/fca2648e229746c88b89ae9686532037 : SUCCESS in 1h 52m 04s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/345e91313c4541cd88b8aabc43b4f7fc : FAILURE in 1h 17m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b1920555fdaf45518025e44a57393b0e : SUCCESS in 47m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f1273d77c24b97faeb9e2a62be946e884832ee1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-11 08:04:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4c30fc4bd77dad670b5b1cb23e86679485750128","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 10:42:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/086fcf20cf2c4ecf8b0b745ad19f4fd6 : SUCCESS in 8m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27189bdc3f734a07b7ab64b3d4f4fc58 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd06273684aa4138a8a8a4231e93787f : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/431d98d39b9c431cae76c2ae22a7e1e4 : SUCCESS in 3m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7eadfdef6e604e6c872c853009e6ac5d : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdb7f28dc1304e6593733d7c2899fef1 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0fdcf0c2a604560bd9c2ad540f314b0 : SUCCESS in 6m 50s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/7791d8da18824b10b3c2c06a2bb4b2cf : SUCCESS in 1h 15m 35s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0089b1116f714264ae4615da17a55991 : SUCCESS in 1h 19m 10s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25a8f1dc9297431dbbf6337497b23316 : SUCCESS in 1h 28m 55s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/df2df403402c430c88da2e987bc4cb17 : SUCCESS in 1h 20m 20s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e6c31d21edc94e9b94e31e48a683e4cd : SUCCESS in 5m 40s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ad044217b7624c3e93c1474aff36680c : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/46679dd8a6674ac98906a406a34ac6c4 : SUCCESS in 7m 37s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b7133bbde0e5417f849f7f7d138534ce : SUCCESS in 1h 03m 49s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5845dde75d87467c87f1056b589858ba : SUCCESS in 1h 00m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/512605bd84704876b6f8973bed77fb07 : SUCCESS in 45m 38s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b50f7e8a5f80400b9c10f9f2c7b50c39 : SUCCESS in 48m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01c2542701854861a5bd81ca1609c4c0 : SUCCESS in 1h 39m 39s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9fabdc6e3ce1447b91f5f989bf764a27 : SUCCESS in 1h 32m 14s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/c4a238b2d39148a186881cca5b2d70dc : SUCCESS in 1h 45m 27s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/817750e496d04027bf5b4153b41f5028 : SUCCESS in 2h 23m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d48179db67148b8894c4f88aef60a33 : FAILURE in 35m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9e3a5163377b473eb7bdec4736de092b : SUCCESS in 1h 12m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ec08a097a48549218bbcff43d6606fce : SUCCESS in 1h 34m 19s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d6fb34c8e7f942b7b03b46f08b28671a : SUCCESS in 1h 20m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/091fd82966904949a8b04caca2b36df9 : SUCCESS in 47m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0cf768517613425684a05c551cffda8e447d1b50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 10:43:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"968162f564a4da624579be82159625ec76f0a675","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 12:47:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ecce65d7199491fb741c3e7d0b45174 : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/621acd4d751f489b87f150b65771fb8b : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7cf59a2ceba84187b434f58f3169ed6b : SUCCESS in 4m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1c4563530e974f90a93b30755b72d67c : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce89b24773264623b795969113aae697 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8e1305ec5fb4ad4a3d5163477ec24aa : SUCCESS in 6m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dc638d9758bf4c728697b5fae1881b6f : SUCCESS in 6m 08s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/00328f59215f4082a877483aef5f5075 : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/269f074109874b18b73b8a1ef3c67f70 : SUCCESS in 8m 07s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58d58731d0f146aa9f9af604ff02df9c : SUCCESS in 1h 20m 51s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/760698b11c7a413b9a599b95d4d9d226 : SUCCESS in 1h 11m 57s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1f9b6e81a1c3497d900ff224b3ffc89e : SUCCESS in 1h 21m 40s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f00505acae724f43911c6eeddcb71d26 : SUCCESS in 1h 24m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f4a4ce64b68e43468b03776ce9ff7320 : SUCCESS in 46m 28s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/57d1c98a58cb422e889009ffcb3c5799 : SUCCESS in 41m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59af2552eb734815924b3564cef96dac : SUCCESS in 1h 45m 53s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a7c70748db0840ebb4a7735be13bf30f : SUCCESS in 1h 42m 40s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/ee29fe13abdd4a27816a4a89eadec189 : SUCCESS in 1h 56m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/822f1fb16a904608829835f03477bacb : FAILURE in 57m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c38951a196c4dfca76c1abcd6efcaae77c1bb264","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-11 13:24:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e08721836ad3258b419f26a956d15155ea525717","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 16:14:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5acf247d238e46cfa25a910abadf13bf : SUCCESS in 8m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b15e0aef8bd24cc9a8de88e5eb86178f : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89b36e8fd42841b78f0270c5db320aa6 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/68f8a934ea5e44558f9ac0da933f468b : SUCCESS in 4m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4a06ceeed5104c3487a9dc4126abe9b2 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6065218d03a34212b1803c6fad929547 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9e674eda4ef44f3bbc8879ed86a2ccb : SUCCESS in 6m 05s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1a1c9edd383c4942bc51347fc9a49c04 : SUCCESS in 1h 23m 32s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/c486f3ede3e4448eac3b4958282db019 : SUCCESS in 1h 25m 54s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67a02ad5f9784c41a55c7cd5f5b7458a : SUCCESS in 1h 32m 47s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8d6f496aed2c45dab3ccfbb00ab0adb5 : SUCCESS in 1h 03m 57s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/32aa6fb401c7473a8b50b73f7588fac9 : SUCCESS in 6m 05s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0f25ee3e4e8f4bc3a367edcc1008229a : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8c303a157ccf44f5a80328bfcd4246f5 : SUCCESS in 5m 35s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22fccd8df7f34894a977ba23539cf2a7 : SUCCESS in 1h 10m 55s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/dfac7e4391764230bb5b6c4f1c001cfe : SUCCESS in 1h 08m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0fdebecff8de4480b8142ebf9919e334 : SUCCESS in 39m 25s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9f2c4a20c64342d7a856fee3afc42bc9 : SUCCESS in 40m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7eb4c2c04b704d4cb64f93284b44f94d : SUCCESS in 1h 47m 34s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8784e1dcd3964e2cb9bdf703bee2fd59 : SUCCESS in 1h 49m 18s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d4dab2cd92c14d278927f3dd07aeb4cf : SUCCESS in 1h 48m 57s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/40c80dd996264f159348554bc8360ea0 : TIMED_OUT in 2h 35m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa23bcdd6372446c9156b7f2439716e3 : FAILURE in 37m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/519aec39b1524a28a1c49bd91898b0ed : FAILURE in 1h 55m 34s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/efddc849531a4f1689ac539283a3f78b : FAILURE in 1h 43m 03s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b19c3a833c3344ad850b540dd6ec0705 : SUCCESS in 1h 21m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d2c1c721fcd042d383cb08a26d240c78 : SUCCESS in 48m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da0d813a5bef948219be915a1db0fabdca054684","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 16:14:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"993be48d3ee7fec1010f4c12c18c0489ba1b8b71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 18:18:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36400e0296d54d14af6e25aa66727a89 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70f90131c6ba4a33b4c1bd71fb33d291 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/afd17d9ec46047d2b0a002ea36859106 : SUCCESS in 5m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/05b4701ca6914f39a817d3d6443c01b3 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34a14d931da7498fb642040d3d8ae2a2 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cecd7ba734242809cc14cb473612424 : SUCCESS in 7m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/92782f3066534e10a4e0a7ad0777dada : SUCCESS in 7m 10s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/26bdb04983c646fd8578b39241b9d9fd : SUCCESS in 9m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/69c4c441bb644a16ad151c983f45f506 : SUCCESS in 6m 44s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a69b58a3979f4629b910963d3d489f39 : SUCCESS in 1h 10m 42s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/916cb4bad74243d5a4e43de7c09c2dfc : SUCCESS in 1h 13m 35s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d0800f8b6b14491dba90cebe8a5eeae3 : SUCCESS in 1h 21m 35s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ed9321ba85e941c1a5d074e4f8895326 : SUCCESS in 1h 23m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5be38aee404b401a852f15e6b1424c35 : SUCCESS in 59m 05s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b11062dfd078442d929185ebce5fc99d : SUCCESS in 1h 00m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/edae9cccd15141e9acf0efeb9fa5ee63 : SUCCESS in 1h 34m 53s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3948b2845ee64fe3850a8341f4e89d71 : SUCCESS in 1h 38m 34s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f06423e06b63483993f8fb121c251498 : SUCCESS in 1h 37m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7d493491a3c14dbfb00a2443451bc954 : SUCCESS in 1h 16m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6ae4992666ca7bd314ceccdd081d2ce7bffb3d3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 18:18:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e709fb26d72b15e9f661b578555c37f97f92e401","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 18:20:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96e87f612c7f43fc973174f49d52bddd : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f412d8521c79d9361c86dc1a1d74a119b1942b7c","revisions":{"f412d8521c79d9361c86dc1a1d74a119b1942b7c":{"kind":"REWORK","_number":1,"created":"2019-09-03 12:05:44.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/68/679768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/679768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/679768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/679768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/679768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/679768/1"}}},"commit":{"parents":[{"commit":"49990fd89fefb852b8d26415431818833902d3a6","subject":"Merge \"Fix listener deletion in ACTIVE/STANDBY topology\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49990fd89fefb852b8d26415431818833902d3a6"}]}],"author":{"name":"Noboru Iwamatsu","email":"n_iwamatsu@jp.fujitsu.com","date":"2019-06-25 09:24:48.000000000","tz":540},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-09-03 12:02:06.000000000","tz":120},"subject":"Add failover logging to show the amphora details.","message":"Add failover logging to show the amphora details.\n\nWhen failover occurs, Octavia health-manager doesn\u0027t log any\ninformation about the loadbalancer.  Due to this, operators can\u0027t\nspecify which loadbalancer got failed and which user was affected\nby the failover.\n\nThis commit fix this issue by logging not only amphora_id but also\nthe linking information (e.g. loadbalancer_id, compute_id, ..)\nwhen performing/completed the failover.\n\nChange-Id: I9bdd6f2913c0da2b0772b4b694309ef993289360\nTask: 34312\nStory: 2005947\n(cherry picked from commit 93e06a9239a0f4b2bcdcdca9e13e800e75b1f237)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f412d8521c79d9361c86dc1a1d74a119b1942b7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f412d8521c79d9361c86dc1a1d74a119b1942b7c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
