)]}'
{"id":"openstack%2Foctavia~679770","triplet_id":"openstack%2Foctavia~stable%2Fqueens~I9bdd6f2913c0da2b0772b4b694309ef993289360","project":"openstack/octavia","branch":"stable/queens","topic":"bug/2005947","hashtags":[],"change_id":"I9bdd6f2913c0da2b0772b4b694309ef993289360","subject":"Add failover logging to show the amphora details.","status":"MERGED","created":"2019-09-03 12:07:26.000000000","updated":"2019-09-13 21:01:40.000000000","submitted":"2019-09-13 21:00:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679770-1568408412883-65f41938","meta_rev_id":"89f1a091d66b875e7d1df6b8ecea58e843a5ceb8","_number":679770,"virtual_id_number":679770,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9057,"name":"Noboru Iwamatsu","email":"rockpine@gmail.com","username":"r0ckpine"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"date":"2019-09-11 12:58:44.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2019-09-13 17:11:31.000000000","_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2019-09-13 21:00:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":9057,"name":"Noboru Iwamatsu","email":"rockpine@gmail.com","username":"r0ckpine"},{"value":2,"date":"2019-09-05 06:43:36.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-09-03 21:26:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":9057,"name":"Noboru Iwamatsu","email":"rockpine@gmail.com","username":"r0ckpine"},{"value":1,"date":"2019-09-05 06:43:36.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2019-09-03 12:07:26.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9057,"name":"Noboru Iwamatsu","email":"rockpine@gmail.com","username":"r0ckpine"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":9057,"name":"Noboru Iwamatsu","email":"rockpine@gmail.com","username":"r0ckpine"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 12:07:26.000000000","updated_by":{"_account_id":9057,"name":"Noboru Iwamatsu","email":"rockpine@gmail.com","username":"r0ckpine"},"reviewer":{"_account_id":9057,"name":"Noboru Iwamatsu","email":"rockpine@gmail.com","username":"r0ckpine"},"state":"REVIEWER"},{"updated":"2019-09-03 21:26:22.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-09-05 06:43:36.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-09-11 12:58:44.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-13 21:00:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b3f13488741b951dd4ffc54431cfeaba1ff0368","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-03 12:07:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7544fb7bbfc92e4ef1b9470058dfc441ddc71de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 13:51:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f63075c99ca5463fbc068875e8217cd1 : SUCCESS in 10m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6885550dfbc64d05a0bcc27b6b84f3aa : SUCCESS in 5m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6a57c8a57e814852897922960e08a240 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d2a682ffbe746b0a798a0b364a0df05 : SUCCESS in 6m 37s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/192499a7213e41cbb0c2a0e639e07782 : SUCCESS in 1h 24m 08s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/33219ed148e04e608c06d2f5910db073 : SUCCESS in 1h 20m 07s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7554999fe594970a6f89a63132aeb7e : SUCCESS in 1h 15m 57s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b1a1f7be56254dbe8ecd0cb38f4761f3 : SUCCESS in 1h 36m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23ff993d4a1340989c8f6ca83e901740 : SUCCESS in 10m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5627a450257248f2a6d03b1192fa2201 : SUCCESS in 6m 30s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/651d06b2f4554d948b7496a950812f38 : SUCCESS in 7m 08s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fdfee8b2c21d42c591f94ad1d732b9aa : SUCCESS in 1h 10m 00s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b51e201805c44ed6a2611d736597b1c5 : SUCCESS in 1h 07m 22s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/51a9851a8e474bd18f7ea2bbccb06809 : SUCCESS in 1h 23m 07s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/dca31516da6046059b5ea9529a6d1716 : SUCCESS in 1h 10m 34s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/c4c279610fbf4f029e26188ae3f55996 : SUCCESS in 1h 05m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df06a808e5e4432a9d65d431a27137fe : SUCCESS in 49m 28s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/50faa0c647c0480bbc6002e10d01d361 : SUCCESS in 45m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b3af34ff6e34fb1a44c29b52e855862 : SUCCESS in 1h 33m 31s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/cf8d5b9a074a4e8f96e260a70544c804 : SUCCESS in 1h 38m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a615232ab69cfbb84ec44fefe34d40847cd834f0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-03 21:26:22.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"4b8f9b0712b7fcec4f85a03245b7dbca1f79962e","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-09-05 06:43:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f95e551002145651ee61092ac93c6ef196592e2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 06:43:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27fdcd3b0acd6216b63eaa03d58df259d1969455","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 09:10:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4412f816be64663b5e067ef1d894daa : SUCCESS in 9m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5b620ceb3d1a470d8105413456c70b47 : SUCCESS in 6m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/58af1b61f87f4f469f517faa21e02a05 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/245692dc297c4c5ab9b949916113816c : SUCCESS in 7m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/015c3fcb4ba745d88103bbd733b9d92e : SUCCESS in 10m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d621bba193bb456ab8790a951f50a145 : SUCCESS in 7m 25s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c3a9720a3a0546e4870b6b8dff1f7036 : SUCCESS in 6m 28s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/3e40f5b9fed54ecab386a496e40a0f6f : SUCCESS in 1h 14m 22s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/73a455b12ad749d0a04f239207cebffd : SUCCESS in 1h 23m 06s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c433cdfb9cd4ad6a733447563b13f42 : SUCCESS in 1h 03m 49s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/429afa61558b47c2b5caf83c9773ab7b : SUCCESS in 1h 07m 22s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/a8cc3472ba854352bbc3e24f81a9457e : SUCCESS in 1h 20m 25s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/20b9785cb3bf4af2813fdf4cc5fddfb7 : SUCCESS in 1h 28m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da89a1d891fb4af486ccf62ec09581de : SUCCESS in 1h 00m 32s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/dc30ff278b454ce9be2bcc15dad121b0 : SUCCESS in 48m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd1c90dba7d34b27b2371b940deabc28 : FAILURE in 1h 42m 59s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/a212afc1f2ba4f5f92d18c54e88a5c08 : FAILURE in 1h 40m 52s","accounts_in_message":[],"_revision_number":1},{"id":"114c590d786216066297cec20f7df7c03a004f71","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-09-08 23:57:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"23ea64aa9157c37198750bc558cc6a20050d586d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 01:48:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b852ddb8725844418df3fd84ecde5647 : SUCCESS in 8m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b6cadbc14ca4921b44bc2936005a335 : SUCCESS in 5m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5831e483fc144e64a2f502c5dbd81d3c : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd119359f76d4c54a7644be2745fedf5 : SUCCESS in 6m 29s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a2f04f7b69af4d7980204ecb68fac68f : SUCCESS in 1h 20m 48s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/56344e1535c34edd9f878e1518ed978e : SUCCESS in 1h 18m 29s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7efe75162c54bbfa98be57bccf6eaef : SUCCESS in 1h 13m 24s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ba614e379d83448d8118daf0d5f9683b : SUCCESS in 1h 16m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d48d14bd7404669bfd6526ca4c4d1ee : SUCCESS in 8m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dd92f24dc1974408bb9418d94fdac866 : SUCCESS in 6m 43s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/047cde73d2a24e3e8475aed7b5d5e85d : SUCCESS in 6m 23s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a7c6d1d9c224443aa1e9f1962ce3402b : SUCCESS in 51m 27s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/deb1556c7d30469992006ffa1b2c6a34 : SUCCESS in 48m 00s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/82c0d898cfe343a284554f10c93b43f9 : SUCCESS in 1h 14m 25s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/e62a67ff7004460a89f4ae3c5802716e : SUCCESS in 1h 03m 04s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/c053562c3b974528b2febaec612cef0e : SUCCESS in 59m 52s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/edd453f63b94491b867490decc69036f : SUCCESS in 41m 01s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/7a179766ef6347adafa1bf7d7140188e : SUCCESS in 45m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ef2e0ee1333496c8b720e5f0da72265 : SUCCESS in 1h 13m 20s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/6c5e53f641b14f4b8c5b2885588431fa : SUCCESS in 1h 27m 25s","accounts_in_message":[],"_revision_number":1},{"id":"7240b8dce16ca807121a41c300f8cafe0f02d8fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 01:48:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"07ae934f5a43e5d507b7c5054117987ffbb5208f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 03:42:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e54f9a22c04a403e991673a9b9b854f2 : SUCCESS in 8m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f64c9f055f5640b3b3ce974b2f268e13 : SUCCESS in 4m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7057a62b94db4d2dae3f6c6baf830d0d : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/194f24d99a274c9dbd100fd315ccbf10 : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/822e3a8563af4356babe7094614123b1 : SUCCESS in 7m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ace7aa0e7327466f84dad63548bc370a : SUCCESS in 5m 27s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/21790cd7f4cb4b30896b1e9b0d31bde2 : SUCCESS in 6m 06s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6209b8b839e74d149073c86df1fccae7 : SUCCESS in 1h 16m 39s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/30864beb3aa44c18bc6be3de57b1b412 : SUCCESS in 1h 15m 53s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8766bc772e148bdb7a3a922fbc8060d : SUCCESS in 1h 05m 52s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9d5dff7e1d4241ff8ccd7244f2f6d0b1 : SUCCESS in 53m 40s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/60d1053ab5df44a598781b3934d88a58 : SUCCESS in 1h 09m 09s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/963fbac1bc0741249b2b543660c997f4 : SUCCESS in 1h 07m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6eaadd4585624dd4b446ad1a3591189d : SUCCESS in 43m 31s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/87a9e9b056f54ed7a1ccf5f9cdfe185e : SUCCESS in 42m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d8b0be755c44a36b5d203ef247e77ef : SUCCESS in 1h 33m 12s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/d2c09931e82d44d7b07e242c3ed8646a : FAILURE in 1h 34m 07s","accounts_in_message":[],"_revision_number":1},{"id":"85f0eda5634affc34c9b9dbf00baf5077dceb587","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-11 08:04:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"34562efb26b21049328178792499eace185bb101","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 10:04:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17f8c8af5e4246c092b4eedb3fb5083a : SUCCESS in 8m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c41ea4b845474316aa1d44327c11ee92 : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4c50898c20aa4f3a89b2ddc3c9ee2e31 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57483d36146640838a9091d9d7b20acb : SUCCESS in 7m 56s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/10eca8c88b444cd5b901e1f0ee09abcd : SUCCESS in 1h 14m 12s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/287e632de1f84af1b58671278c972937 : SUCCESS in 1h 23m 58s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0bac842f66864466bb61abfab5db8d13 : FAILURE in 59m 19s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9e8afea18817464a96a66f195f272637 : SUCCESS in 1h 37m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51d887134700420b8863d945ba66a558 : SUCCESS in 10m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/60b3296a609949979ffc81f63259d8a7 : SUCCESS in 6m 42s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/eae47937f871483dbba4596d6f833dd7 : SUCCESS in 6m 42s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8dc99cc373c94bc8bc93e6605a72adb2 : SUCCESS in 1h 08m 08s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/bbd4055c75784fcf93748c9c6448125d : SUCCESS in 1h 00m 44s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/95339dbdb8fd4251913f8a9d7c61d5f5 : SUCCESS in 1h 11m 33s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/49a8ed184ed049fe9b60bacac20f53f2 : SUCCESS in 1h 04m 12s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/d42eb8f1e10740edb8db5552cb6d4b70 : SUCCESS in 1h 15m 04s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bf2e80d40e434303a012f1e3c6e652f1 : SUCCESS in 44m 06s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/25acbd651a5e4b0984ebeae0d4027a2a : SUCCESS in 45m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce2cc8c542aa48718e5a67500235af21 : SUCCESS in 1h 29m 08s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/2ab712952292471f9135cae6e413e43d : FAILURE in 1h 42m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b0e6d08a115c3869ad5364c6f8737552ac41e12a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-11 12:58:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"269ab5016da6a4e92d7dd77e0ebba05022c4047d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 14:47:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18802299202b471082d823d37d5039d3 : SUCCESS in 8m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66b520c260d04d3480fd3ecf4f281b70 : SUCCESS in 4m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0e2fc87437724bcab3024fd6a000691b : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a48d4a57de14eab8839484421a47d07 : SUCCESS in 8m 53s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ed9e868860e241d0850b1d57fa752109 : SUCCESS in 1h 20m 23s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/56ff2ca09237489ea22038681e3e451d : SUCCESS in 1h 22m 48s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f7a981a879d44a9bf979f50bb941a2c : SUCCESS in 1h 25m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/af0695ae92a8497aa716376ce2c2e54a : SUCCESS in 1h 41m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a27ab5560f1f4bfe86bdddf7d91f0967 : SUCCESS in 9m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/be7aeb55d38e40a38fec087f26c7f440 : SUCCESS in 7m 36s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/837b20769f844bedb94a85797c08752e : SUCCESS in 6m 54s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e837b8d06b9e423aa01ede7756aad77b : SUCCESS in 50m 39s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c0c485a49fc949a599aeaa3577ac853f : SUCCESS in 44m 24s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/e94fb2a4f52244b5b935c00fe00aac67 : SUCCESS in 53m 28s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/bc3142a12d074bf1981fa7ea16d6c9a9 : SUCCESS in 1h 17m 06s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/c9f69d1216aa4131a1405219f7163dd9 : SUCCESS in 55m 51s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/57a29ffedabb4a19a6581cacd3c7a71a : SUCCESS in 48m 40s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/e270250ea2764675b76ba9741529645f : SUCCESS in 53m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b930b829e85a4e83bf170d72b279eb8d : SUCCESS in 1h 23m 12s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/959614a9a7b3425a82701888362c51da : SUCCESS in 1h 43m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b5c89a5cd1fac65837fbbe419ad1136c3ca7c12f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 14:47:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"153c69c67368846e802c72eed2b5ea7e0b4b1962","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 16:46:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aee36375a50430386e3e14bdf2674f6 : SUCCESS in 10m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3cb5e7585737483390dcb8db0a026fdd : SUCCESS in 8m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ef36e57724c34cebb28cc7c491fb7f12 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/913f7207d4c84b83abea58501be5c8b2 : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ca9b3ce30d043dab02c290d2240befe : SUCCESS in 10m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/36b09c349e794d3cb4dce2096b0c6a48 : SUCCESS in 6m 48s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d5f9b90ed83545529743bc67d1a07495 : SUCCESS in 5m 49s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d9d15b0d1ea3483bab8b594d3b69ea08 : SUCCESS in 1h 20m 13s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0d045d4591884502b3feabc3a1b32151 : SUCCESS in 1h 21m 18s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/034d2fd0e9b846d2b03a1226e1b4fe7f : SUCCESS in 58m 46s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5abb16488ad94e18a715cd86f5a89e9a : SUCCESS in 1h 02m 09s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/b0ee3a33c7614015a2af5f06c54123d7 : SUCCESS in 1h 44m 28s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/f252f173f1b14d09b1bb3f722fa5c42a : SUCCESS in 1h 09m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8d0eddcd9000487c87601bbc2d289e3a : SUCCESS in 58m 46s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/ae3598e5053a404b8e0552eff1696910 : SUCCESS in 38m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2927ab05df034ed792a0e42c2945d0d9 : SUCCESS in 1h 55m 50s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/a397c003275a4720822bf12ff8333b7b : FAILURE in 1h 31m 12s","accounts_in_message":[],"_revision_number":1},{"id":"63245904060d238206dad59a0252c29b901c532a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-12 06:40:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d167191b9d36593d5cf36da42c433451f87efeb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 08:47:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51720420eeb44a23833c5f8bd365798d : SUCCESS in 9m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ea676db0939440485b1a9dd04c2158d : SUCCESS in 5m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/90943fea23c849349ef5801ba248de3e : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ec0aaca12964e2a9efcc912804e0d29 : SUCCESS in 8m 59s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/efc9cbe977eb446b963601661c6ece38 : SUCCESS in 1h 35m 10s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/46296c10533340ff98174b73e406412e : SUCCESS in 1h 33m 31s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b871b95033cc4664858b560f196a4778 : SUCCESS in 1h 46m 01s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a3dae97e1e6f40f387a72f3f732c2c0c : SUCCESS in 1h 21m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c13a0f29801149e59ec46aeef340a5c6 : SUCCESS in 9m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cba275b4b4bf49d19237905ba8e4078a : SUCCESS in 7m 04s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/1a6e698342db4d9a9d4253042515aacd : SUCCESS in 8m 30s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e332887b93742849940fe710a9024e7 : SUCCESS in 1h 03m 01s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1c1b0befd9984b819e0f4289a0918252 : SUCCESS in 1h 03m 44s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/ce6bce8f923547b2874f10996d5ff3fc : SUCCESS in 1h 04m 44s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/0d56522027114bb48351d7f68cbba3a5 : FAILURE in 1h 06m 59s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/4b2aabc8af1d4035b29c6fa84bcae48f : FAILURE in 1h 12m 14s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/89a98c12ae03466c9ea9ae4ef8c65485 : SUCCESS in 56m 59s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/aac02f44c76f457e8ac78e9474fdda25 : SUCCESS in 50m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d15ea4c47106452e9a6d8f6910a06789 : SUCCESS in 1h 58m 50s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/bd03ffeb67f5475a915889615dfaeea1 : SUCCESS in 1h 32m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a6cdd4afa734149de6f2dfcbad406f7667de243c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-12 09:11:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cb9d0cdd177df2b35530ba77440a44f3dae28cee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 11:18:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a884ecb6286c4bf8b735e534799def6b : SUCCESS in 8m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f79429b602cf4b7690e92630c83b3d37 : TIMED_OUT in 41m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b26df3e353274459ab442f643a6902e8 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ca836dff01244d2a503a8eb6a45e035 : SUCCESS in 7m 56s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6e10091994eb45b3a56c6de1f2c8a01f : SUCCESS in 1h 19m 31s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/68c42aaa62284bbb95ec1d18b36917b2 : SUCCESS in 1h 27m 05s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2700d4e8d72428f8ddfb4c698662275 : SUCCESS in 1h 26m 19s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0f1b2b4324494541a9c13a8e1f20b44d : SUCCESS in 1h 36m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a482e66ff7b493c9c11a27243ae347e : SUCCESS in 8m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/740b81e04ad64974bf050335d6030162 : SUCCESS in 5m 56s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d53ed025d7fc40448809bd020c254ea1 : SUCCESS in 6m 39s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d366f468427845e0ba21b6c82e13eb0b : SUCCESS in 1h 04m 35s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/140e2fc012544287a756a24e77f9368c : SUCCESS in 1h 00m 50s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/3da815bd92914816b3c32256da2181ab : SUCCESS in 1h 07m 22s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/7309ec1bf98e47abb6ae73dc4fbe61f0 : SUCCESS in 1h 18m 25s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/846d840fd8984ea39840c7ef11a5272a : FAILURE in 1h 00m 52s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f50681e2b8134ebc9cc442662cd9a758 : FAILURE in 44m 07s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/50324ef3538a45a9b8982b59b424df5f : SUCCESS in 46m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e53f4b1062ac4d359f317071cf14ce2b : SUCCESS in 1h 31m 12s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/f5e2f4ac94e643349f4a5fbed7e0effd : SUCCESS in 1h 33m 44s","accounts_in_message":[],"_revision_number":1},{"id":"fdaa2364425cb133509249f51c64addc740e0dd1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-12 11:51:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"661eaa035ba572dda0f6570d186ffca41ec3d96d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 15:30:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48518316ec66497da5f0be5d7441137e : SUCCESS in 9m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/776d8a3e06894568aa3eba49f413dcbf : SUCCESS in 8m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8a719b14d8684977ad86dfb525e77c05 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/523a0d4a64f147deb478aa916d4915bd : SUCCESS in 6m 33s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/4bed99a4fd914ed3a54900ea188f35c2 : SUCCESS in 1h 37m 01s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ddcb356e536e4255b9a16747c493e1e4 : SUCCESS in 1h 11m 43s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/025f75320cfb4d48b52e73796a1be2f7 : SUCCESS in 1h 21m 38s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/316ab897239f42909ab65718a004b83b : SUCCESS in 1h 20m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdfaeb5281f84dbab688d5a97c4b8096 : SUCCESS in 12m 34s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/59044ce36e624d35a5ca27792c4cdacf : SUCCESS in 6m 45s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/dc7060000def4ade9e8caff731f3e26d : SUCCESS in 8m 16s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b14c73844444268b35c5c3a55d32313 : SUCCESS in 1h 25m 40s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/12699ec87aca46d29791e4a3d2d4f050 : SUCCESS in 1h 14m 48s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/c40d971e66664608a406d1c589b6713f : SUCCESS in 1h 26m 16s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/180d6efbcfcd48e79ced0ab717b62a95 : SUCCESS in 1h 14m 13s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/fbfadbbcb0e04deebefd54d7e954ca2a : SUCCESS in 1h 03m 13s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/46cd458a17694856b2ad927e55494b11 : SUCCESS in 45m 37s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/776b508f84f94e1ab6d808abd14be5d3 : SUCCESS in 46m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7d22fead55a48c9b93522f2006fe515 : SUCCESS in 1h 33m 30s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/7f03439c64d8488393ce6046fe699231 : FAILURE in 1h 31m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9fb437a8401d0c5a1a81e20fa6dfe512c026043c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-12 15:59:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"25bfaea60056c08129294e9ad013d8f12e623b52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 18:00:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48ff77cbc63c40409b4e2cc9396e1248 : SUCCESS in 8m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9425d8eeca274dcab9c1660f213e4543 : SUCCESS in 4m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1dc2c1f5d42a4e469af24d504c5ab159 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ae47a579e2c4b8bb9682a2e4f3a3299 : SUCCESS in 8m 04s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/e682a997152b47ceb12764aec79516f2 : SUCCESS in 1h 22m 52s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/29438e7d620349f6b0db949a36780a64 : SUCCESS in 1h 16m 29s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/389b36cf79c941228519dcd6a280d20b : SUCCESS in 1h 35m 59s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4ef82628096c47589777cd94607ffd2b : SUCCESS in 1h 28m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/724f1b6b18b442a18b1c9d8c169bbfa1 : SUCCESS in 10m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9a9da31bb0014c799bd2be4f6dca893f : SUCCESS in 8m 20s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7e25d7b0791a4fbf97432fb885a01339 : SUCCESS in 5m 48s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ee86014039447199b071ff965a14939 : SUCCESS in 1h 01m 04s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/dfcbd9742fa94a4e9598ec7963ec5c7a : SUCCESS in 53m 41s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/6e5c401eca234e3e8a5687fafb079f32 : SUCCESS in 1h 37m 57s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/012977d1a0954132a072e86b27b05560 : SUCCESS in 1h 37m 16s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/231560ea35a541b9933b34b0f7b9eb0d : SUCCESS in 1h 04m 06s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fb105dd80794454fa4a5d1701f22ef5e : SUCCESS in 49m 14s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/28dca2bb38b5415688086aa12fd822c5 : SUCCESS in 53m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e95055cb025d4fecb9d4f2d9f2832c2a : FAILURE in 1h 35m 24s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/d7d97e99407c453ab70e2bdce449c2f8 : SUCCESS in 1h 31m 26s","accounts_in_message":[],"_revision_number":1},{"id":"551a49ec6fb4aee6e19448be52471dc27910ea63","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-12 19:15:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"933c1c8d97939a69d1c1c5435f0b7ede244a7699","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 21:06:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/312cb7de3e594649a0247ff8dac73110 : SUCCESS in 9m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e66d06ace7e9463b995147cb54e9607a : SUCCESS in 4m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9927fac3202a46109ce4b79c79ef018d : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68e7291f00834f7a975391f10e9cbac8 : SUCCESS in 8m 49s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ef8df339482f437faec2d5fc31d5a289 : SUCCESS in 1h 21m 39s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/3506c4e7189c4679a5f3dde6cfe66aa2 : SUCCESS in 1h 25m 03s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5cffba2084de462d8bff5bcb30acc61e : SUCCESS in 1h 22m 58s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0038b53a6cd2401eb823b84ac5d1a65e : SUCCESS in 1h 43m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2073b0af8cce4ee6bddf665bc34881c7 : SUCCESS in 11m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/78a7d2fb311541ad8c89ea21ade80c13 : SUCCESS in 6m 50s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/560781a1206a4233b710b711f116d1e5 : SUCCESS in 5m 03s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/268758bdfea54bbea6e93f46178bfe80 : SUCCESS in 1h 00m 16s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/831a3c9398a146f6b48a002749d6e3db : SUCCESS in 1h 05m 24s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/da157cbc995b47f8996cc71826fd1575 : SUCCESS in 1h 12m 33s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/eb02cc9c648846278933be4da4f06d9d : SUCCESS in 1h 17m 59s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/9a2e05aaa3884f7fb21712a30f4e0f0b : SUCCESS in 1h 06m 02s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/81a643cc8676449eac91fdb3e6d14919 : SUCCESS in 1h 06m 09s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/4a0700092c7a4c63948a8ee31e9213f4 : SUCCESS in 1h 08m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7d1dad6174447759bf92c8e85f05c5a : SUCCESS in 1h 44m 36s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/f0fd4aa7ab4f4509a76a927d06731b0d : SUCCESS in 1h 30m 53s","accounts_in_message":[],"_revision_number":1},{"id":"09637f2fbf2cbab62baf5eafecbd13bad43c76fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 21:07:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2a6dde09bd02eb530f7c773d20e6400d643ad58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 22:45:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81fa036eb41449c4bab86fefca614645 : SUCCESS in 11m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08625dad162a4e5fa191efec1778902b : SUCCESS in 4m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fdf6d20161a84fa9a594cc09004e37b5 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f97bffba5024047b3ef1f670c5d5748 : SUCCESS in 7m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a1a2238a88945b2a72e89fc7fa21a5b : SUCCESS in 9m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c583413756164d13829dc701fadcf9b9 : SUCCESS in 5m 36s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/37f468c9d8084643ad284afbb6ac5d44 : SUCCESS in 6m 06s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a6d88b7ba3b7471cbf7ea04e2cfda914 : SUCCESS in 1h 33m 51s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/3f93baabd71e40c1926b53d8da8f387b : SUCCESS in 1h 24m 50s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b4d45adeb774e8195ef5b1d354cfbc1 : SUCCESS in 1h 12m 08s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9ddb2c1a336a49b1b53b0190ff22474a : SUCCESS in 1h 04m 48s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/454d7c06c28d4d45927a9f22d71662c6 : SUCCESS in 1h 13m 02s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/0d84b77a2ee54884844b9e6cff30953e : SUCCESS in 1h 05m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/038ea0e7844e4f2a82c44dbab6810795 : SUCCESS in 43m 31s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/07c2633236e6447bae0df18592849366 : SUCCESS in 55m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bffa4508462f4be5bc23acc9055f5689 : FAILURE in 43m 19s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/7d964f13b6a7407a9ebaaa4a6528f3fe : SUCCESS in 1h 35m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a3da05a1975e2bf7bd1dfef8bf7d3dce03e85b47","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-13 06:36:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6b69ef8e1281f8956076111c8df830c2e21e78d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 08:42:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9378343efda44e9996b56ad716c8da44 : SUCCESS in 9m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/68409a6b74df4d30acdf0b306369da43 : SUCCESS in 4m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7ab8e150aba44fd1a1078ef3d0adf261 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c103b76aab8146a9905fb1473cc259ea : SUCCESS in 7m 26s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/2cc98388149b4082b20f35fece8b201e : SUCCESS in 1h 40m 57s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/36257b6f47b242c3a5a1fc9f75934094 : SUCCESS in 1h 26m 43s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7e5a4cbf8ae44f6cba8260a59c403058 : SUCCESS in 1h 30m 06s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c83c27809ab34fcea6f99ee91a6ce305 : SUCCESS in 1h 45m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f66769915df4337b741c3a112cef87e : SUCCESS in 7m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/091844c62d1842d3beb9a3f8d1172e59 : SUCCESS in 5m 25s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4c1d6f007e47414aa792ad17bf1402e6 : SUCCESS in 6m 22s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cbc2d39da7114149b6a12f2f5863c0e8 : SUCCESS in 1h 04m 42s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b2752f25ef1d49e5b4ee2c5ccbff4f69 : SUCCESS in 58m 36s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/eee3c810e3ca4bc1b212752ec12375ec : SUCCESS in 1h 18m 04s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/19b7a0c6646248b6ab8a4a0dc8477e16 : SUCCESS in 1h 13m 03s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/4e888ce5cf3b411bb1d5cfee869a0e92 : SUCCESS in 1h 03m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d4ef9d25c5f44902938749267c91952d : SUCCESS in 43m 32s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/ec5af33cf865482ca2f487237efb1b28 : SUCCESS in 48m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/454b719fcd2147f8857a9c65d1352125 : SUCCESS in 1h 26m 40s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/1842ef31684d4df6a5f684c34339ae3e : SUCCESS in 1h 37m 58s","accounts_in_message":[],"_revision_number":1},{"id":"4c9102aea60623ac9d75a838c8ec4880ec8a4a3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 08:43:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea7befba2276c3747dcdcc6d2a3f747945cdd62e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 10:26:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/386ed8e725484f8c900689fec860e043 : SUCCESS in 9m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d7562f8c4e6453cbc3c200bb6b7ba69 : SUCCESS in 6m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4a881a26015442ecaf3521cf0e4c1f47 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c851c8b328674a73991c18c3f7a92c46 : SUCCESS in 6m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffab699493e6487f967225e51381d731 : SUCCESS in 8m 10s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9c1b6ccdeafc4462a4bea8814ef18b00 : SUCCESS in 6m 09s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/cd5022d9f3fc4a1ca7ee10762c0f15fa : SUCCESS in 6m 57s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6fd821c347564961bc099ad48e7139cc : SUCCESS in 1h 19m 12s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/93157c7270b845c18b97dcee5a939953 : SUCCESS in 1h 21m 57s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e0c14b159cb42aca74e33dbbf2b85fb : SUCCESS in 1h 02m 28s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/56b4ae4643b247e9bc88f792511e6cec : SUCCESS in 47m 40s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/5968a3eab64545e2a8d3f3841487b489 : SUCCESS in 1h 13m 01s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/babedf5ea6b149419c7ab017797221be : FAILURE in 13m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bdd40e66046e4b87bc0442f4528c6dbd : SUCCESS in 41m 31s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/eaa1ccbcd78b4836b8933db8483eb142 : SUCCESS in 50m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e774216b8dbf44c48294d05b79649233 : SUCCESS in 1h 20m 16s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/9f5c6b66765d44458c067ca0f423a85d : FAILURE in 1h 30m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ed344f3b10b75ba605a76a4c488262f6937bc6a3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-13 12:23:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f0c43d89e62506a1dff5943db5f5708907913c00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 14:21:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0a51435423f419198ee1c5e72f47e62 : SUCCESS in 9m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/373de33d767141878ea73d40c750921d : SUCCESS in 4m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1a0ef3191e4e4a408e1552061fef2a2a : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73312670404f4af0b965f90fd75449a9 : SUCCESS in 6m 43s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a3f6eaea491b477db40fcb9e6d897804 : SUCCESS in 1h 21m 09s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/c6742eebfff8446c9ee8d1fd4b2556eb : SUCCESS in 1h 18m 37s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b32601a5a5b94bca84e298205a000214 : SUCCESS in 1h 16m 51s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a01fd3036747492cac690316b0b83ddf : SUCCESS in 1h 24m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2776b06835af4750ae78c693e4fb4b13 : SUCCESS in 8m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/671ccbeb017c4a36bce3920ae4a09384 : SUCCESS in 5m 54s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/becfbdc3d9224d668af8a03361c2520c : SUCCESS in 5m 56s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a40dc3ae5d74aa092cfb2580e0c450a : SUCCESS in 1h 05m 30s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f6dd9ba84c9b4ef782372dba388cec0a : SUCCESS in 59m 51s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/a802bd62a02540f0ac2af274b38cf576 : SUCCESS in 1h 07m 40s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/e683b89bebec41cd82936822683b88ae : SUCCESS in 1h 10m 58s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/7917aab570564aad9860a53d2169b144 : SUCCESS in 1h 11m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a4f07177a51b49b99bb4e20aed3d2721 : SUCCESS in 44m 01s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/48e24d9493ac4c36a896e8a72614231a : SUCCESS in 44m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/402a22b4715042848ede1e1c84d088db : FAILURE in 1h 38m 28s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/e1d26a8b7eee47ae90e49362eb09218d : SUCCESS in 1h 51m 02s","accounts_in_message":[],"_revision_number":1},{"id":"4b4c6855484b6a1272fc0f84555b26ab182127d7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-13 15:15:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e0095dacaa8178b13fda7c6115c236c1a23c5c6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 17:07:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e6d8ccf77b448b4b0cbcc538af5c3c7 : SUCCESS in 9m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3333b32bcdb4380bbf2e5ade3e84c13 : SUCCESS in 6m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9aef0c591512476b9a183b1b01153b54 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95d7fd60062942a08c691ab014e0bd00 : SUCCESS in 8m 19s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1466f01b0bdd47f2af514b8d90958e42 : SUCCESS in 1h 30m 58s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0bbe56ab7a124b1f9aa9c07737c3c20d : FAILURE in 12m 45s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f1e3ea27e5c4d7884d52cbaa86b9b89 : SUCCESS in 1h 22m 41s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8a59a5cbafef408bbecb6f8891e19077 : FAILURE in 10m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b270dbbf989f464ba2bec8c2af3e94bb : SUCCESS in 11m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/365e5c82120d4bb081013b1106ed0489 : SUCCESS in 8m 12s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/9e2deea5704c49f3829e16b253a09318 : SUCCESS in 8m 20s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0ffd23982fd740a497db5afa38be0878 : FAILURE in 11m 50s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3f21897499f54cee8aadea23c6c569b9 : FAILURE in 13m 46s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/31b209bd35b04365a32af1fac58018d1 : SUCCESS in 1h 13m 44s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/c29101d6e6c344eea36913705abb370b : SUCCESS in 1h 06m 25s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/a4c49a6b37cb4190b38b26c5a4646384 : SUCCESS in 1h 09m 04s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2908b32178a04eb6a23411e2081be1cc : SUCCESS in 54m 24s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/cc8e1dc59a3742d2add9f7a493d4a431 : SUCCESS in 56m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48b41ae91050479c8187366f50e51659 : FAILURE in 1h 47m 13s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/46b8f51a89764c19b73a98f4c8e58ab4 : FAILURE in 8m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3d4a3c2a3c0bfdeabee92691dfd8b8d2ba276174","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-09-13 17:11:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cbd7d9c28da8024fb1a74838d2b03dbf9411d6f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 19:11:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c8be1ee53e242818548bd83d9217bd5 : SUCCESS in 13m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e39b7bc217bf492b83e5fe2f7abaacaa : SUCCESS in 5m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0e4dee4ab8c14794b813531e559552e4 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad241f38ca294959933b299f14dc21ab : SUCCESS in 7m 56s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/dcf1836893754970a54569403c3f3495 : SUCCESS in 1h 08m 45s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/38e576a54be94effa465d1171d2f502f : SUCCESS in 1h 20m 10s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/40ed1d13437c48c9bc0e3d73caf075ba : SUCCESS in 1h 14m 19s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2563283b12f84c0c836bab02fbf85861 : SUCCESS in 1h 19m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b55834b591d413bb5931cac1a3572d0 : SUCCESS in 10m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6207a591918348c48cd53b1f9678f9c9 : SUCCESS in 10m 04s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c638e6d36ca249c2aca3114f7c8e5c19 : SUCCESS in 5m 34s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84ed30632ae84dc3a97d451ed5bda901 : SUCCESS in 56m 29s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e8288e2acb6b4de5b93196241bceed6c : SUCCESS in 1h 11m 31s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/0248d7c5c74d4501a4eff83584141fe4 : SUCCESS in 1h 14m 53s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/48c7bf77c7084416abb8bb44d2ea5b2b : SUCCESS in 1h 02m 58s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/41009ecabb5d4dbaaa65965c7369a52e : SUCCESS in 1h 35m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a7369c6ff4764eb5adece79fd88881c4 : SUCCESS in 44m 03s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/db1bc6512b8e4657a0ed134cf475c91c : SUCCESS in 49m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5109484f1b7e4accbda0fedd40bd6e1f : SUCCESS in 1h 37m 00s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/0c89f7c6960e44dd98f2e798b436d60a : SUCCESS in 1h 48m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a58562d985975d81c49053f0d16185297d76e4ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 19:11:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2060fba265a37fe1915dfd56625be04ecc85d016","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 21:00:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ead324bb688547afb96cf5be205b38b2 : SUCCESS in 8m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/20cc27baa9a24773a0c6fe8cc468bbff : SUCCESS in 6m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ded28807e3b8411492a6f68eebd1af15 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abc92dcdf9304ea189473878f389fc67 : SUCCESS in 8m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7ac0538fdd24a41813df34f5682cd4d : SUCCESS in 8m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9f29995241d5431b952192421db82651 : SUCCESS in 5m 59s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/bc955e05e2e24c008b3fd423a831f0e2 : SUCCESS in 6m 52s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ca7a64d09bcb434f959a345b1747aadc : SUCCESS in 1h 31m 18s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b48755055fdd46149c87cf078b13ee89 : SUCCESS in 1h 41m 05s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5ca2b5b186c45f1a5b941079c192115 : SUCCESS in 1h 10m 21s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0d82c9369e184924830730cdd59a223a : SUCCESS in 1h 10m 08s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/b179f69be91c48d1848679c3b542289c : SUCCESS in 1h 24m 22s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/37d1dcb8164346b785cd02d082901e94 : SUCCESS in 1h 18m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0db5d1144f134dc9a3c209be7f25949e : SUCCESS in 41m 08s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/746a8bb1c4244274b09ba29b90e4ae7d : SUCCESS in 41m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1b480155be0849d4bc52681324a5dd7f : SUCCESS in 1h 44m 41s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/9ee25cf8f291415ba925f53bafa7fc9a : SUCCESS in 1h 30m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7689a1c323e0ec05033872152b82de2d5991d583","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 21:00:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e2d68fa45a1da439c3507334ce7349497df7b0f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 21:01:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f4070f44b6496aab2ef763d9b85d59 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2959d88b7149aeb3f5e9c3dcdcae91913d47cdfb","revisions":{"2959d88b7149aeb3f5e9c3dcdcae91913d47cdfb":{"kind":"REWORK","_number":1,"created":"2019-09-03 12:07:26.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/70/679770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/679770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/679770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/679770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/679770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/679770/1"}}},"commit":{"parents":[{"commit":"37aad5db40d688c6c718e8a7f892b61b0924143a","subject":"worker: Re-add FailoverPreparationForAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37aad5db40d688c6c718e8a7f892b61b0924143a"}]}],"author":{"name":"Noboru Iwamatsu","email":"n_iwamatsu@jp.fujitsu.com","date":"2019-06-25 09:24:48.000000000","tz":540},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-09-03 12:07:15.000000000","tz":120},"subject":"Add failover logging to show the amphora details.","message":"Add failover logging to show the amphora details.\n\nWhen failover occurs, Octavia health-manager doesn\u0027t log any\ninformation about the loadbalancer.  Due to this, operators can\u0027t\nspecify which loadbalancer got failed and which user was affected\nby the failover.\n\nThis commit fix this issue by logging not only amphora_id but also\nthe linking information (e.g. loadbalancer_id, compute_id, ..)\nwhen performing/completed the failover.\n\nChange-Id: I9bdd6f2913c0da2b0772b4b694309ef993289360\nTask: 34312\nStory: 2005947\n(cherry picked from commit 93e06a9239a0f4b2bcdcdca9e13e800e75b1f237)\n(cherry picked from commit f412d8521c79d9361c86dc1a1d74a119b1942b7c)\n(cherry picked from commit 7722133fb2fa63a5ae486c379ddb08262ad5f996)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2959d88b7149aeb3f5e9c3dcdcae91913d47cdfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2959d88b7149aeb3f5e9c3dcdcae91913d47cdfb"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
