)]}'
{"id":"openstack%2Foctavia~680422","triplet_id":"openstack%2Foctavia~stable%2Frocky~I001068059b02c909e3b7b105f2f168e5bfed8291","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"I001068059b02c909e3b7b105f2f168e5bfed8291","subject":"Revert \"Use the infra pypi mirror for DIB\"","status":"MERGED","created":"2019-09-05 14:14:47.000000000","updated":"2019-09-06 20:22:04.000000000","submitted":"2019-09-06 20:20:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680422-1567801223920-b79a56ac","meta_rev_id":"0635770e840927ac55981a653a3b35606df268aa","_number":680422,"virtual_id_number":680422,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"date":"2019-09-06 15:38:39.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-09-06 20:20:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-05 16:16:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-09-05 15:08:29.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-09-05 16:16:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2019-09-05 14:14:47.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-05 14:14:47.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-05 15:08:29.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-09-05 16:16:13.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-06 20:20:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8315851e010e107e530016f71f1e6aa99bf2abc0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-05 14:14:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43dad7cbfead9fad3676486b933b4944a541886b","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-09-05 15:08:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c43d28379868fd7167a8657fdc4983e2ad280f54","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-05 16:16:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1c06abad0d74b089d8b45a3dabaf2a68858ea2b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 20:49:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/482863a20cd84540b7bbc25a3d2eea4f : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/699f98c23c1843c1829c973d14a7970b : SUCCESS in 7m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e5e996497bbc44d2a602f565f818a1ff : SUCCESS in 7m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/386499cb14ad4ca190adc1d3733ccef9 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9029d0607e446edbffde7cb097f5ef4 : SUCCESS in 7m 11s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6c6422b93e0d440ba6f5f6173c67b5da : SUCCESS in 1h 13m 19s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/c31e82f4a88040b88e520d6635353a40 : SUCCESS in 1h 18m 15s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/97a514f58b00435a97aa2231ebb0a1fd : SUCCESS in 1h 35m 42s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/490638d9bf204c1d81d83a1e4b23c5a4 : SUCCESS in 1h 38m 06s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2ce171dc6f8642da9ee2a5cc3522a724 : SUCCESS in 6m 26s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/78c80979702f462e865f47f04098c9c1 : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7a20c21695a44a9b9f13875a732e19d5 : SUCCESS in 7m 29s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0841465748514e17a696401a9e2cc261 : SUCCESS in 1h 04m 43s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c3fc7e5e71ae4558a5476fa0222b3cc7 : SUCCESS in 1h 09m 21s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/b59b0037f302496881c61a184ce7692f : FAILURE in 2h 10m 38s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/2800d7fc7670421fbeed9283413f3e4d : FAILURE in 1h 25m 56s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/8ae3e9c5210a4302aa1763f80980df52 : SUCCESS in 1h 04m 03s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/26ff6dac410a4aaba1f48a8a4ad23ff5 : SUCCESS in 1h 07m 09s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/82e3b3ba32c047839ced866a834f3798 : SUCCESS in 1h 09m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/55289ad610b64c97a976c831bc01d7c1 : SUCCESS in 1h 44m 33s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/7a864dedb2f546e0ac4278552ede78da : TIMED_OUT in 2h 36m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5157fbb83be04e69a6da4ebc32f9ad12 : FAILURE in 33m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0c2fd2fcc604dc18ea598b8668cf380 : SUCCESS in 10m 12s","accounts_in_message":[],"_revision_number":1},{"id":"4988c5a073a9da5c4072aaf5cb11cccbdbc1b88e","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-09-06 13:18:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1f675a31c550ce63f6737f4ee7a28e73f88b12d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 15:35:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d06c0460037447b09d4700a2eb6ec0d2 : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb85aba4f77b49aa9b48f27690b2d893 : SUCCESS in 6m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e158c0c9789456eb6a9a696227d7958 : SUCCESS in 5m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4b1ce49084764b36b9f38e8d0b1ae9a5 : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a821398058344fc0805bb20337656eea : SUCCESS in 5m 47s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/eaf41cbffc3748629106f30740801658 : SUCCESS in 1h 23m 15s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/9c3c37d7750340a9b27afb06809b1180 : SUCCESS in 1h 27m 51s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c1614a14d564db88808c2c7a1bebc9a : SUCCESS in 1h 34m 22s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d7cc180ca6834de7a8d4888aa47b0ad6 : FAILURE in 1h 37m 21s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8dda85d8b3034d9d8778a2b7c30454df : SUCCESS in 6m 23s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/38e3d4a137844d7a85c92f2f7fb9010d : SUCCESS in 9m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3c4d952ae8d24b1ba78f92ebfcf3b2a4 : SUCCESS in 6m 02s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be94a7489bbc482fa90056c67159d130 : SUCCESS in 1h 23m 56s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7162009f480b46d98091c5ec5fd27e8c : SUCCESS in 1h 18m 22s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/52ee645d15714cbeb8fee4ffe16daffe : FAILURE in 1h 39m 27s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/cd1e0e6fb12a4a9194e13190a75ada0e : FAILURE in 1h 38m 13s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/ba968c2ccd124aa280dfbd2169cf4868 : SUCCESS in 1h 13m 42s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a4447498a24c48e5b4a82fd99bc84986 : SUCCESS in 52m 24s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/a5e97e98414345229219b18d37d05949 : SUCCESS in 50m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68594c90a3ea452bb910a6617cdb533e : SUCCESS in 1h 45m 20s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/5165012c33f54d29ac37101c7254bf10 : SUCCESS in 1h 48m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a7efbca0567e449cada51b24913b9282 : FAILURE in 2h 07m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0eaf452fb8e5467d954313c70fef9f7d : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":1},{"id":"2edff38456a8078b13b195da9af81242924e860c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-06 15:38:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6a6b3502576e7f0017f883011a85416ccf844dd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 17:43:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89d84096f1f548d4b0f05aefe5846e15 : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2d1276b5cfb4cacbc570f3f7119542c : SUCCESS in 7m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7389dfc84da24fd79007de6d6ef8fe47 : SUCCESS in 5m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/999bbb8a54c94406a2dd5d71f9bb076c : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c62eeb68ca4cba9ced2ab1fd8e9483 : SUCCESS in 8m 06s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a83468293a814f67844775cc491a793a : SUCCESS in 1h 14m 35s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/6992e5f50814459082f33df1e58ce590 : SUCCESS in 1h 21m 28s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26646c01863b4185b2680a70cdf37359 : SUCCESS in 1h 03m 33s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/987019b3bac247a78d018a8477359bda : SUCCESS in 1h 40m 15s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9fabe17827464916b7f19fa2c7bd1428 : SUCCESS in 5m 00s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/1a0e72b6183046f78085b15a471af9bd : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/45e5381c23b84488ace26b41e98bb207 : SUCCESS in 5m 47s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21428bf52afc46fa8e3ffafd8240e463 : SUCCESS in 48m 32s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3291dd55984b40bd98ec6184b6cb82fb : SUCCESS in 1h 03m 55s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/d5b47b6772e34cbdbf7fadd1cd69c80a : FAILURE in 1h 37m 53s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/827ee1986fd1439c80a0f9c923fd488c : FAILURE in 1h 39m 46s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/eec433fe60bf4b3391f1883ddccc11ba : SUCCESS in 1h 25m 10s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/903fc0a4341242d2bb4870a7d4a07651 : SUCCESS in 45m 57s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/fc1269003a4a4ee0922ede35d0f0b9b7 : SUCCESS in 58m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67a5e674d58947e0b3d95cf74248a66f : SUCCESS in 1h 55m 02s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/aa0c2d5d497f489fa03b1e5f9e13537d : SUCCESS in 1h 38m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3d2e78b27b1241c7822f76195879278b : SUCCESS in 1h 25m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/035a904763324857bb9e7d2a5110188b : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e28c470f59e490f6e33f01b8773c5a6cb6eae9be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 17:44:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eb39bf92be99b6743bff97e239af97759e013e10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 20:20:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c5ab74d6521446d87a815258af69ceb : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da8bf181be9b45f4b501aaa581c6370f : SUCCESS in 7m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/398e353cc620452fb6c82d5b3dd7eb7f : SUCCESS in 4m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b8b53541cfd34fda9346457b49a20d95 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdcfdc1abbe44ac597659bab222e2d0e : SUCCESS in 6m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ec44a528350946898fa29b088260d416 : SUCCESS in 4m 38s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b0b1b6e48a474160ae05b00f33c1835e : SUCCESS in 5m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ed72b007d1e348d1bc1bb73e1270be02 : SUCCESS in 5m 14s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da2226683c074b9682508c099301a436 : SUCCESS in 1h 05m 38s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/6c0df605223d4c6a80582e4c870fd1c7 : SUCCESS in 1h 25m 06s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/43f298d11315499e8968cbfe7f40a677 : FAILURE in 1h 23m 52s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/720a90d6e68c411dbf6b15cfd917d30b : FAILURE in 1h 46m 30s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/af37ecbbd82f4f18aac13c2c63fe3ebd : SUCCESS in 1h 26m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3ea619fdf9c449a28eac349b0189c319 : SUCCESS in 47m 01s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/8f8e6f883f1b46dbb387ff328542476b : SUCCESS in 54m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53e815d4edd54679b4216b2a58db1ae0 : SUCCESS in 2h 30m 10s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/bd2a6b3778394466be07526805daa44b : SUCCESS in 1h 44m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/02484c2d3f0c4a72bd9b3c03c92d63c8 : SUCCESS in 1h 38m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a3d49bcd84f48abbf51ac25437eff75 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ed506199dfd880e41b2993e6ec0d5ccbfaf914a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 20:20:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0635770e840927ac55981a653a3b35606df268aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 20:22:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5181387eac84178833be7076218462a : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e3a69774c885698c7bb01835e01d5a847770789","revisions":{"3e3a69774c885698c7bb01835e01d5a847770789":{"kind":"REWORK","_number":1,"created":"2019-09-05 14:14:47.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/22/680422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/680422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/680422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/680422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/680422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/680422/1"}}},"commit":{"parents":[{"commit":"7486abcc34c04f1846eca27bf180f36b408f471f","subject":"Use the infra pypi mirror for DIB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7486abcc34c04f1846eca27bf180f36b408f471f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-05 14:14:47.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-05 14:14:47.000000000","tz":0},"subject":"Revert \"Use the infra pypi mirror for DIB\"","message":"Revert \"Use the infra pypi mirror for DIB\"\n\ndiskimage-builder is broken in the version in stable/rocky upperconstraints.txt.\n\nThis reverts commit 7486abcc34c04f1846eca27bf180f36b408f471f.\n\nChange-Id: I001068059b02c909e3b7b105f2f168e5bfed8291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3e3a69774c885698c7bb01835e01d5a847770789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3e3a69774c885698c7bb01835e01d5a847770789"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
