)]}'
{"id":"openstack%2Foctavia~680833","triplet_id":"openstack%2Foctavia~master~If0ff2a855ad5b9d9ef3742ad0596c97a6dbf81ed","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"If0ff2a855ad5b9d9ef3742ad0596c97a6dbf81ed","subject":"Bump diskimage-builder minimum to 2.24.0","status":"MERGED","created":"2019-09-07 19:32:56.000000000","updated":"2019-10-09 16:13:30.000000000","submitted":"2019-10-09 01:32:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680833-1570584769825-fe29816b","meta_rev_id":"ade5154f0c6f68c742c54bd856f9b365548d2486","_number":680833,"virtual_id_number":680833,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 01:32:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-10-08 23:18:33.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-10-08 23:49:10.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-10-08 23:49:10.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2019-10-08 23:18:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-08 23:18:33.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-08 23:49:10.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-10-09 01:32:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c9fbb318548c43f2eb43e467dce303c709d4bc6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-07 19:32:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bfefb942864b8d0c3e88711c3ae61853aef07b7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-07 19:35:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d86045e66839f5b7abea82c079d2bd473ea563c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 23:52:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5a742e0387b5463c86c0f039f2185e3c : SUCCESS in 2m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0327767c3df34429b0323b434f5c6067 : SUCCESS in 6m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5c037c5b6a34fda9d89fbd18fb7229f : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ded83f25d2174afbbc7dae2b0b98dcd2 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/53d88ae89996460eb5fe5bf3f8085222 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f70fe9ffb0e44ddb9d2b01ac6493404a : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/850c0bc2fb6f4ace95e0e736e59dd974 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01905a49b4a844899f3b1e9f23297168 : SUCCESS in 6m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/57e1b1a40e8e4de696e1ef074e86412d : SUCCESS in 4m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/845dad9e01d149d89a3b60096b41926b : SUCCESS in 5m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/24a79796504e48828891d3fd919a3425 : SUCCESS in 4m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d2cf407c6f544175b85c5adcae471ce3 : SUCCESS in 5m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f187a17a96e1447883b6a2af0792d026 : SUCCESS in 42m 26s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5dd060b6876f4d23b29bf53b43753ff9 : SUCCESS in 42m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1a353415ebc54af2990ac28ad825d606 : SUCCESS in 1h 50m 07s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/aa0389b1c00f44149f639ef24667a501 : RETRY_LIMIT in 55m 35s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f843a273526748d9a35ff41dd8967ea4 : SUCCESS in 2h 18m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fc633a2dd8644873ad4b63af41d46358 : FAILURE in 1h 55m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/07f63722bdd54277a812875135e9d3e9 : FAILURE in 1h 30m 10s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2fed4740cc964461b9faab968800f1c2 : SUCCESS in 1h 18m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/95c256176c9f41309501f5dfbefe61cd : SUCCESS in 47m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/29ba8dba3d1d43caa6f8cdf84d6b8cdf : SUCCESS in 1h 37m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b7df1c24fe6e4190a94990fc5051e5f8 : FAILURE in 1h 32m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/256cf8f1fe2642928b0ed75b49e9fba4 : FAILURE in 1h 13m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7560f59e0d9f4a75adcea5376c797f92 : SUCCESS in 1h 21m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1dd292ababfdbeccec9c3122f9863809db774ebe","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-10 22:20:42.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"fa25a983955349310df265729325cf19df9bf298","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 01:15:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/46496044bf09432d8af556156817c313 : SUCCESS in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3d57e3d088e43ebab71207c147aeec5 : SUCCESS in 7m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a19c250cb1054834be1e917fed06208b : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0117de494be4a35a4c5b7916c8b326e : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2397e55c8a4844a4b1aeac030fc25bcd : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fc10f60ce01434494eb8d1b7b6f054d : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/580b86e3ba8840419bcb4bcfc9d9d551 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/917b6441cc2447999e2088f66ccd8bc4 : SUCCESS in 8m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bea215ca3bc248808feec263835b179d : SUCCESS in 6m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bc149d8891404751a0d19b975ac19360 : SUCCESS in 6m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ac9b971d5841424a9ce65bf10b16d530 : SUCCESS in 7m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/352a6d2f846648ae837ee754e87cb2a4 : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/70565e95afe34cf6bb368593929de114 : SUCCESS in 47m 25s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9166c6d8abdb4bab98324d14a8b89954 : SUCCESS in 1h 11m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99856d4b1dc24a35ad8d431c5975a935 : SUCCESS in 1h 45m 49s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/711cb55a0bcb40aa85dd143c5454ccae : SUCCESS in 1h 45m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/bf26a15930dc4725aefd8a5c29f15ed8 : SUCCESS in 2h 23m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/84be3ca9c3854919a964cf3fe7ae7a51 : SUCCESS in 1h 24m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3c6a9359bebe437f998a250427bcf6b7 : SUCCESS in 2h 03m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1e2456edc8974ad79f3f252e095b34d9 : SUCCESS in 1h 44m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f59cfe95f75d4f36b8ff0519a8f2890e : SUCCESS in 1h 07m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/feea32d6679c4505b97789bd2bd923c6 : SUCCESS in 1h 44m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c5c933b53e3c4332b5941e5bdbc93102 : FAILURE in 1h 22m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/17524a9f8d4a4df992d6d53ddc0938e9 : FAILURE in 1h 05m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/afdcd8bf8a1746e49e0621bb4b20f225 : SUCCESS in 1h 35m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b7507e46617ce4cb08fc46a2cde42376dcdc2793","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-08 23:18:33.000000000","message":"Patch Set 3: Backport-Candidate-2 Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"99bf175f0c969e9ab2fff88c4dda782094e3b488","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-10-08 23:49:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dcbeb3ad02693679b7dbea20cb50fcd99bf6646b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 23:49:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"351b1e2371bddc2aa430c1d8c1e0d1f453ae8d7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 01:32:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/32978ee7a40c41b0b49a415d602b2ba0 : SUCCESS in 2m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/929fcbab84084922ab4369733aa98716 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/479b96eed86a4c38b3f65aa64e6fdfbe : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd424709b385434baa9984f041da22ca : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b270c8f274de419a9e605f12c1515b3a : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e880ac73972b4c12ac66eb90c34c9b01 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b02c66099ef42c388a595faeef3eb25 : SUCCESS in 9m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0f254b69ac0145078ff269b3daea1367 : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/76b76178e19242cbbaf9dcd50c79701e : SUCCESS in 5m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8602ea9682384d429843d341c8adf03f : SUCCESS in 47m 04s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a9dc3f786b7a4c73903d45917b343abd : SUCCESS in 44m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c06edaa8f37a4d9caa8ec4cb3626d0a7 : SUCCESS in 1h 32m 42s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3e207c03f0324222b28b97acce0cac6e : SUCCESS in 1h 38m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/264a82932f67448e86a67d6ce7a08f1c : SUCCESS in 1h 15m 35s","accounts_in_message":[],"_revision_number":3},{"id":"9ee4ebdbe5a73dcb77f7c94ce7c139364478447b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 01:32:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c1d82bf2b7b2c7773685d3de92f5af8d194d1de4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 01:34:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/334b0752021f4f90be5d0a3ec4ddd876 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":3},{"id":"ade5154f0c6f68c742c54bd856f9b365548d2486","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-10-09 16:13:30.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit e38f78bd20fd4ac51237424c00d960a8340aefa5","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1aeb416a7e72dad298acb7f11ad54eabfd9ea49e","revisions":{"ee209f0ab973f355a843b1e180751ae70a705c1a":{"kind":"REWORK","_number":1,"created":"2019-09-07 19:32:56.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/33/680833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/680833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/680833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/680833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/680833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/680833/1"}}},"commit":{"parents":[{"commit":"fa70e9759d52137e4ba847d20e9a941491f6455d","subject":"Merge \"Fix template that generates vrrp check script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa70e9759d52137e4ba847d20e9a941491f6455d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-07 19:30:19.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-07 19:32:55.000000000","tz":-420},"subject":"Bump diskimage-builder minimum to 2.24.0","message":"Bump diskimage-builder minimum to 2.24.0\n\nThere was a bug[1] in diskimage-builder when using pypi mirrors\nif the host doesn\u0027t have \"python\" available DIB will fail.\nSo, I am going to bump the requirements minimum in Octavia for\ndiskimage-builder to be the minimum version with this fix.\n\n[1] https://bugs.launchpad.net/diskimage-builder/+bug/1577105\n\nChange-Id: If0ff2a855ad5b9d9ef3742ad0596c97a6dbf81ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ee209f0ab973f355a843b1e180751ae70a705c1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ee209f0ab973f355a843b1e180751ae70a705c1a"}]},"branch":"refs/heads/master"},"329608ebf3a64e2bc0036242835cffcccb51b194":{"kind":"REWORK","_number":2,"created":"2019-09-07 19:35:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/33/680833/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/680833/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/680833/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/680833/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/680833/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/680833/2"}}},"commit":{"parents":[{"commit":"fa70e9759d52137e4ba847d20e9a941491f6455d","subject":"Merge \"Fix template that generates vrrp check script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa70e9759d52137e4ba847d20e9a941491f6455d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-07 19:30:19.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-07 19:35:44.000000000","tz":-420},"subject":"Bump diskimage-builder minimum to 2.24.0","message":"Bump diskimage-builder minimum to 2.24.0\n\nThere was a bug[1] in diskimage-builder when using pypi mirrors\nif the host doesn\u0027t have \"python\" available DIB will fail.\nSo, I am going to bump the requirements minimum in Octavia for\ndiskimage-builder to be the minimum version with this fix.\n\n[1] https://bugs.launchpad.net/diskimage-builder/+bug/1577105\n\nChange-Id: If0ff2a855ad5b9d9ef3742ad0596c97a6dbf81ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/329608ebf3a64e2bc0036242835cffcccb51b194"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/329608ebf3a64e2bc0036242835cffcccb51b194"}]},"branch":"refs/heads/master"},"1aeb416a7e72dad298acb7f11ad54eabfd9ea49e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-10 22:20:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/33/680833/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/680833/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/680833/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/680833/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/680833/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/680833/3"}}},"commit":{"parents":[{"commit":"09efc2a423c6ab033f9a648549aff87590be5990","subject":"Add get method support to the driver-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09efc2a423c6ab033f9a648549aff87590be5990"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-07 19:30:19.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-10 22:20:42.000000000","tz":0},"subject":"Bump diskimage-builder minimum to 2.24.0","message":"Bump diskimage-builder minimum to 2.24.0\n\nThere was a bug[1] in diskimage-builder when using pypi mirrors\nif the host doesn\u0027t have \"python\" available DIB will fail.\nSo, I am going to bump the requirements minimum in Octavia for\ndiskimage-builder to be the minimum version with this fix.\n\n[1] https://bugs.launchpad.net/diskimage-builder/+bug/1577105\n\nChange-Id: If0ff2a855ad5b9d9ef3742ad0596c97a6dbf81ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1aeb416a7e72dad298acb7f11ad54eabfd9ea49e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1aeb416a7e72dad298acb7f11ad54eabfd9ea49e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
