)]}'
{"id":"openstack%2Foctavia~681144","triplet_id":"openstack%2Foctavia~master~I5f9ce36c878973f4ed96527af6f1024a362421d8","project":"openstack/octavia","branch":"master","topic":"volume-based","hashtags":[],"change_id":"I5f9ce36c878973f4ed96527af6f1024a362421d8","subject":"Fix a few nits from the main volume-based patch","status":"MERGED","created":"2019-09-10 00:04:36.000000000","updated":"2019-09-12 15:44:20.000000000","submitted":"2019-09-12 15:42:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681144-1568302977953-62adfaea","meta_rev_id":"ce334e000eacef090f300e74ae2d733a443a6a83","_number":681144,"virtual_id_number":681144,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2019-09-12 15:42:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-09-12 10:32:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-09-12 10:32:52.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":1,"date":"2019-09-12 15:32:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2019-09-12 10:32:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-09-12 10:32:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-2,"date":"2019-09-12 14:01:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":1,"date":"2019-09-12 15:44:20.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-12 08:15:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2019-09-12 08:20:03.000000000","updated_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2019-09-12 08:32:05.000000000","updated_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2019-09-12 08:33:32.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-09-12 08:51:32.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-09-12 15:32:17.000000000","updated_by":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"reviewer":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"state":"REVIEWER"},{"updated":"2019-09-12 15:42:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f90a30350f486c2292d87da575d6dda31d7f843","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-10 00:04:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6337a4922eda1079c5779e5c11f5f171262f9c50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 03:35:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97a6d4e9e9c844fbb1800357b5c93b92 : FAILURE in 3m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86a78b213bf94d7aa5a05cdad7879b1f : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e173a229adc4a08baf52332801d33e7 : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f41981b067ac4ef4bf8a027679b906dc : FAILURE in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54a2e9d23b3944168018a9acd1aea876 : FAILURE in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48dd0a0f8b1d4c239283dea7cd8ac095 : FAILURE in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40a8b24c9ba146dbb5a029878167175c : SUCCESS in 6m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9f969a8f55a940cf8f30f84024147963 : FAILURE in 5m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f4ddedd1c8164a74b1996ae6abb2e54e : SUCCESS in 5m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b83d4d44f271413d8d71f1d62b9382ed : SUCCESS in 5m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b1fab5ca75004f1fa77e66ccfb7f5988 : SUCCESS in 5m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1f62bb8e44ff424880dfcc3d143f567f : SUCCESS in 40m 20s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4c0df7fcd4a64715add16ab8007bdd16 : SUCCESS in 40m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db5240b35a4d4271bf7ba489db260cc1 : RETRY_LIMIT in 57m 25s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9c410d0148a14feeaa37ce1dac97259a : SUCCESS in 1h 48m 59s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/39fe4e0587fc40fda5ddc2a172892859 : SUCCESS in 2h 25m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9ad963ffca844c8c9c8e11b77c005d85 : FAILURE in 2h 12m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ce9af6a717f5445d992966e826515bc4 : SUCCESS in 1h 57m 40s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a34c02e179f54b9381e1f9e64b4d892b : SUCCESS in 1h 19m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f65d250821dd44ac898f7748423f09ef : SUCCESS in 51m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/430ef38f8c2746e9b5f78c584638d26f : SUCCESS in 1h 36m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/820721141c7f479099156f0f55125fee : SUCCESS in 1h 14m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/af6fa4f028f74442adf37e9f11648353 : SUCCESS in 1h 12m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/55122b2eaab44044a70063ad93d5a562 : FAILURE in 1h 30m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc73afd7102ff71a17a92d21abe4005d8fec5390","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-10 22:21:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"60ec6a3362f1c26dbc041f805ba6f1b8d921c641","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 01:10:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92ba1d9425a44fa1afdc1be4c69f4d57 : SUCCESS in 9m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/393a439eb7c348ae87451fb05d44c347 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33011919239642f7ba2e840a0ff5e76a : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9adf64aed4f04d969fad9210e45e9e36 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8973f675190649fa8db69499651d8597 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9de8f49c52f0441b9f685d459efade6b : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9fca4c8506141d5aefb45b7f2c70f47 : SUCCESS in 9m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3ea27fd782024cf4aca44342f5d41d14 : SUCCESS in 5m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3dc1ad463bdf4c3689e8afce2487e7b5 : SUCCESS in 7m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0142e385fbf04957bb99721889076093 : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/85ab7b2b65f4420eb5695e3c819bc87a : SUCCESS in 6m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/353dfccde3a54bcf80c893c256e84483 : SUCCESS in 46m 16s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3c3f788a04a54f40bde0bbd499a12424 : SUCCESS in 46m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a5819c5c0234deaaaa8e35c7f6f199c : SUCCESS in 1h 33m 36s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4a66c9e16df64a368fd327b22799e4b0 : SUCCESS in 1h 42m 51s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2322900db87a4088baca6f217b697aca : SUCCESS in 2h 04m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cc37b31eeb894565985752d1a64268a7 : SUCCESS in 2h 00m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/46f16d9fbc824e19b9ca57df6dca367f : TIMED_OUT in 2h 33m 04s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e7d3a671b263404ba4f889fa2a6040d5 : SUCCESS in 1h 24m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b3e58364cece41d295c52fbdda9d79d4 : SUCCESS in 53m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/24cdb78648324a4f8bb30540b90c39db : SUCCESS in 1h 36m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d83e15ec2a64a4ba55c726b78c111fa : SUCCESS in 59m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b596272ffe144fdc9a93af49c59a0e2d : FAILURE in 1h 40m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/bc2ce70b63ca4fd2af72c6e7bbbd23d4 : FAILURE in 1h 34m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d13005d6197537c8b2b042499a4e027812c87c8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-11 21:22:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4eb04c8b455b9b41971254104eaaa34039af43f8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-11 21:57:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3fb771de69e143aa23e641fc9d5da3a541df3d73","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-11 22:37:06.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"574e25b6ca2296c248443f7cdf96d88d09a853ae","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-11 22:59:59.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"94a9cb82b10c243f9749091ec6787017853a22d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 23:49:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdd42b09c87642e28bbdbe05f40731c2 : SUCCESS in 7m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3b04f943baa43fea875d1d72592ff7a : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91b78a64f2404e9b84a60c176d5e9cc4 : SUCCESS in 7m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d27cef18f024906a3067898be393327 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/138253928ff34551b01003ff9f6362e0 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1bea0df07a7849e0ad295af08e70958d : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37dc8e6b760546ca98640d0e113291ed : SUCCESS in 8m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b1146c2c0d234c06953adacd1432c57b : SUCCESS in 6m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a224061dfb2044888d90375424c1a227 : SUCCESS in 7m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/62429573d91c44669d7755cf09d57691 : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1a52a839db104927a1ff0a4be525cfa8 : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a5ad8fc8f9464b5cb0e48f00eff33ab3 : SUCCESS in 46m 41s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/95069bb9d8a94a61b2a1f447096476f1 : SUCCESS in 53m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13a263c3f9a34f30955938fd7903dcd1 : SUCCESS in 1h 41m 43s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2e3a75ed9cd94622b981bf43f765d70d : FAILURE in 1h 47m 53s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f9bbb08812904ffb8f974d6f8bcef827 : SUCCESS in 2h 16m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e3f2d1acca30493a885cdc6a735c2db3 : SUCCESS in 1h 21m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/36b9ebe2ffdc42a59e459e690d059c2e : SUCCESS in 1h 30m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b5c7b0ade907411097b7d03751fb771f : SUCCESS in 1h 21m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/006e561e79a242ccae7fffef6be4d5c8 : SUCCESS in 55m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/47c98cc0d4a149e38d0e1c9c5c2902c4 : SUCCESS in 1h 37m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cac1bbc311124f89a5e4aeb60aaf2f4a : SUCCESS in 1h 15m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/254f4459f19c4b0dbef3f191a3f3683a : SUCCESS in 1h 12m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/64ff4449472c4c0aa075aad042151a01 : SUCCESS in 1h 19m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"807ace29c9fb7c0b622ed2ba0f41b147816fd935","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-12 00:36:40.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ec340048a3912674dd4d4a5e6469fcf145c7aa40","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-09-12 02:01:56.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b17897048468ea674b3bebc67fcfc322efab75c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 02:42:16.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb6352e9bf464899a57a2af07b11244c : SUCCESS in 8m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae91324b5eea45f48c1f7c0108b707c7 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/762c07c8de214e7ea936192e17e93522 : FAILURE in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9782d30ea93e46f8b53b82df302003dd : FAILURE in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9d315f514d940209f4c193372711817 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8f0227bf4254d6c9f50c86e430273ad : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/472c922502054043b963894e5286cedb : SUCCESS in 6m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fb74d35274c44bc5acc8aeedc27d4695 : SUCCESS in 6m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/385c2d44ec6d4d05b1698af04c1869de : SUCCESS in 5m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a003e3b5b26d496d9069b7ced2479209 : FAILURE in 3m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aed180f95db6414f86052867a0fb7dec : SUCCESS in 5m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c488c06bd7324052b5ab18aeb61a126a : SUCCESS in 45m 39s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d2db47e4439e4505a32ba99b90755705 : SUCCESS in 49m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/70a3c51575ad4a41a9d64856d93f8403 : SUCCESS in 1h 18m 27s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6ff775c175f64771926174018a4d4792 : FAILURE in 28m 59s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/467e701213cc4d20910830660a726c17 : FAILURE in 17m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ebe43a0c38b0400db95d0aba19a6bf70 : SUCCESS in 1h 19m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ed75a01a83354dd8b42dcee3066eb060 : FAILURE in 48m 16s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d6f7223169e640bf95ed883611d4e85e : FAILURE in 1h 05m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/08e92bb4ecda4a9ca0da434944e8b40b : FAILURE in 55m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fd527cc769514dd798b13c9aedcd4e78 : SUCCESS in 1h 41m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e41bedbf00547a8acfc4533b15b9fd9 : SUCCESS in 1h 17m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f19bf8f78ee54ea985e440108f8a55be : FAILURE in 33m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/850b5f12b98849df873b99eeff9aa75c : FAILURE in 18m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c58edec4988ed41259c68d1bdae0db3870309e02","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-12 03:01:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d4bbd22eeda44eb65bc41932b4ff017d61391c82","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-12 03:37:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8258726ba9ac8b86d343568502fd02f0fab8de58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 05:29:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4cdee01650f14df587764268f406301b : SUCCESS in 6m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31c10c6c803941eda14949196f06aff5 : SUCCESS in 9m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e841b847f03844ba8df4bf80d117413f : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8578f7d08d847ec85c72253a630d516 : FAILURE in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76edf10eaa4e4fb1bb3541a513708882 : FAILURE in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba8d1e0fa008489485585229cfdeb9df : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca25c95cb227407ea4f88b846566278f : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7883fcce2b4146dcaed06da79dea9394 : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81afc6010b11441082793660e6a9144e : SUCCESS in 6m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3a4cb009882f4f66b9e18ab2d3437853 : SUCCESS in 4m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/70503abbead9490bb561785dec8250aa : SUCCESS in 6m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/010ec807202b416ca45d42b4dd417ded : FAILURE in 4m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/370edf6ce6514981805a1b0507fed57a : SUCCESS in 7m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/04fef8af99c341b294f8cae7cec705cd : FAILURE in 22m 33s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8216436bdb6a4d01a54e7e39d6ce0ea3 : FAILURE in 14m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/059bbe1527214e6fbad914c0b03e8b1f : FAILURE in 21m 15s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e8b647a1656f443fa10d27a4ce9244d5 : FAILURE in 14m 52s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8490f134575140c5b614467af948db2a : FAILURE in 19m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c6e09a643c09437c8053c70c3d69edbe : FAILURE in 1h 33m 51s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/bf38deca72224e40ab3b262c6967f3cd : FAILURE in 52m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5bc39e16aec64e45a3683462bdc264aa : FAILURE in 1h 02m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8f80d435002f419d8e18e152709ed5b2 : FAILURE in 25m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1b9350f4b49c4d388686e84f8fdc3312 : FAILURE in 21m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eaa7a00855ae443e99da1c3340aaa7a2 : FAILURE in 19m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b59e3535face46c2a1ec92155426a795 : FAILURE in 13m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2505d2e55cea42eba90f85632bf87700 : FAILURE in 16m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9b729cf61a90b6253c756132975bf04fd418bfbc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-12 06:02:05.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"2d2767147cfc66bccd14b2ad44bb0c140e991dbf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 08:01:42.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3065bc4c0f105d2040f2aab27a996058c45a9017","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-12 08:05:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d2c6b35b1d598aa58e87270c4950cc589642726b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-12 08:15:45.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"9270e35ce4b134a1b518e9b8e40728a124e4c753","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-12 08:16:22.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"bbb097a245fbd71f77feeb5324b9e16b23101faf","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2019-09-12 08:18:46.000000000","message":"Patch Set 6: Backport-Candidate+1\n\nThank you !","accounts_in_message":[],"_revision_number":6},{"id":"94f6d127af672f7a85b117a015b6030ea922fdd6","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2019-09-12 08:19:39.000000000","message":"Removed Backport-Candidate+1 by Ilya Popov \u003cilya_p@hotmail.com\u003e\n","accounts_in_message":[],"_revision_number":6},{"id":"1b65df9094d4457dcefaa2ae5a61fbf03c89dd9c","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2019-09-12 08:20:03.000000000","message":"Patch Set 6: Code-Review+1\n\nThank you!","accounts_in_message":[],"_revision_number":6},{"id":"c70a0268e9b8df60d72b8730b0c930adc78de54f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 08:31:47.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"817c53b78df286f2b5165ee82fc9fa6a5030d765","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2019-09-12 08:32:05.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"b832b18097085e0595ded97f0f628b28d6e7dacc","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-12 08:32:20.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"5eb45c0a75124eec342c9ac944e40d5c7a28debb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 08:33:32.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"7192ce94498636f00180501fd98251991810f349","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-09-12 08:51:32.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"dcca2ebee48c1fa5d55bd3771e1dd1145f45bc87","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2019-09-12 09:03:33.000000000","message":"Patch Set 7: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"69b28f9e183b06d8e3c8b8070e499e19ccc04d71","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 09:11:54.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"17e4aa553537ad1eb4347f5cbe93d42c3fee57f8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 09:12:08.000000000","message":"Patch Set 8:\n\nfast failing...","accounts_in_message":[],"_revision_number":8},{"id":"0bf99148430cd0905ddea1cee2deeba072174999","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2019-09-12 09:54:58.000000000","message":"Patch Set 8: -Backport-Candidate Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"ce5d817cc954de44ac9ee12ae5731bb88691b3d2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 10:12:09.000000000","message":"Patch Set 8:\n\nFYI: Disable fortnebula cloud - https://review.opendev.org/#/c/681697/","accounts_in_message":[],"_revision_number":8},{"id":"29a4ade55dc828773433b86a2a115cfbd676c223","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 10:32:52.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"fd9668f2c13825c726bc7e5504ffcaa6415852f7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 10:33:24.000000000","message":"Patch Set 9:\n\nhttps://review.opendev.org/#/c/681697/ merged. Fast failing with new patch set.","accounts_in_message":[],"_revision_number":9},{"id":"b6b2d83b24181d5fe96812a28c4ab857bfafdcdd","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2019-09-12 10:42:24.000000000","message":"Patch Set 9: Backport-Candidate+1","accounts_in_message":[],"_revision_number":9},{"id":"e7c9e15e753219b42e320cfe745a4c198040a4d8","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2019-09-12 12:39:22.000000000","message":"Patch Set 9: -Backport-Candidate","accounts_in_message":[],"_revision_number":9},{"id":"1640695c9072bc1611391c19beb418d8a406cd81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 13:43:43.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/73fb941101c845f88c5984e1a1e4d80b : SUCCESS in 3m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32ecb74bd6d843d6abd459d470f3c8f9 : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21d59519e4d4403c9ca8a43dc637829a : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4977545db554efdbeaa83ca7d94cc36 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d2837fbad0b46f69924343c1d401ca8 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbf7fcf6be5840ce9fdf60a53d8deb86 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e73d2a5ad8e46c78ce9a0fe4d143bff : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53f41009d4fe4ec78f595317328dd529 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a57970c782a4f96a62276ee10c64434 : SUCCESS in 5m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d9f268f1f14f439e9f6670e0f712a967 : SUCCESS in 4m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4b573176e80b4d0e9c1612a812917086 : SUCCESS in 6m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2f8263e01d4046f1879457665417c7a2 : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f94023614b2e4befb7721ee8ff5b0eec : SUCCESS in 6m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/607d688fcac54670aca85a0dd29e1327 : SUCCESS in 47m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4f560adc59214194af0f8b0b15924347 : SUCCESS in 45m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dbc0ecfbc95d40769c536392f05400dc : SUCCESS in 1h 51m 46s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3fdfb4f324564c0dae27ba274ab8f484 : SUCCESS in 1h 36m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b4586ac04e5e4ecd9bd7afab114fd57e : FAILURE in 54m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fdf78657952e4b82a5d571602f4a80c2 : SUCCESS in 1h 18m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/722cdf14d4d54bffbf89841041510760 : FAILURE in 1h 25m 51s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2e29402221924a60b067df8a46924405 : SUCCESS in 1h 19m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ac2bb881171948d6a50da62217ad5eb5 : SUCCESS in 55m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fb312244bede43a7ab3900346b09fc89 : SUCCESS in 1h 37m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f1bc3ea9f9b4ebcbe0258c901f34e55 : FAILURE in 1h 59m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a5f5bc452d08490e962af2520a0ed4e2 : SUCCESS in 1h 13m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/67f8d863d6214b599de2a7f23a852314 : SUCCESS in 1h 37m 19s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"988ccbc5a7cd80818634e3504953baa352534d3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 13:43:57.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"4a80e49c6522fd16c7373ef848f74af689ebf950","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-09-12 14:01:20.000000000","message":"Patch Set 9: Backport-Candidate-2\n\nThe cinder volume feature patch was merged for Train. This patch does not need to be backported.","accounts_in_message":[],"_revision_number":9},{"id":"5e8f2b76b26c2fae6e6b0c60e2e3f84f54305930","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2019-09-12 15:32:17.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"9cbeaba53d3a24a57727bca1c5ab730a964c0b04","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2019-09-12 15:32:26.000000000","message":"Patch Set 9: -Backport-Candidate","accounts_in_message":[],"_revision_number":9},{"id":"8401a39c553d8f57da3afded79eca6d0fb509184","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 15:42:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"46ac48ef00b3bdc6102cc724ec377654b084b49e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 15:42:57.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ab4159c1269b4773b92ba82feb9019bf : SUCCESS in 3m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/967572f14bf64d6aa83997026f782083 : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cb534afe1ae46b99a9f76aae37260db : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3ccb51b3be04ee8b38cdbd8744ee63f : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ac7ce1e9797499ca6e73419ac22f7cb : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c98690f8f624bb5b53937a38f638360 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38a0631fbac643569073a683a5a29188 : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8720a98309f43caa6b3705f267335ec : SUCCESS in 5m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/50731b3f56a64e13b357c63349ef2d7c : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c4d01d9ca95046c4b073910c04210303 : SUCCESS in 5m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/959a5adf4c224c369b615b9969f0ec5c : SUCCESS in 43m 29s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8db03db95dce4c0599f6f7d02b1fb2ef : SUCCESS in 53m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/876a2e900a2b4336bb842cda31e5a1c0 : SUCCESS in 1h 52m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/606e954065a142dd8ff20af72ce2b80d : SUCCESS in 1h 47m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d14cb61d1cfe4cedbf0f891bdbef2851 : SUCCESS in 1h 26m 10s","accounts_in_message":[],"_revision_number":9},{"id":"f8f8c0b6cc062ba2cc82493924a31dfd54c11076","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 15:44:20.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/74f24e117b3c4c609fe7e2627cc7a48a : SUCCESS in 55s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6aa8e663f0845bd89c20fe4b3a4a399 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8e512e7d36f467e99c16e4e3ebccf08 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"05f5257075dda076b0960a319d49c970b0015022","revisions":{"8172d181d6e3559f287b4a349fde8963b92bda72":{"kind":"REWORK","_number":1,"created":"2019-09-10 00:04:36.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/44/681144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/681144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/681144/1"}}},"commit":{"parents":[{"commit":"67ec7e18a9a891e555a3a5ae827b3ad703133388","subject":"Support create amphora instance from volume based.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/67ec7e18a9a891e555a3a5ae827b3ad703133388"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-10 00:02:45.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-10 00:04:30.000000000","tz":-420},"subject":"Fix a few nits from the main volume-based patch","message":"Fix a few nits from the main volume-based patch\n\nI had a few minor nits on the volume-based patch. This patch\ncorrects those.\n\nChange-Id: I5f9ce36c878973f4ed96527af6f1024a362421d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8172d181d6e3559f287b4a349fde8963b92bda72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8172d181d6e3559f287b4a349fde8963b92bda72"}]},"branch":"refs/heads/master"},"1bfd8caf5f404d8dc8de8a16bd763071b69766f6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-10 22:21:56.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/44/681144/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/681144/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/681144/2"}}},"commit":{"parents":[{"commit":"93b509cfe5ccced92ab482fa789ed6cdb700f154","subject":"Support create amphora instance from volume based.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93b509cfe5ccced92ab482fa789ed6cdb700f154"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-10 00:02:45.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-10 22:21:56.000000000","tz":0},"subject":"Fix a few nits from the main volume-based patch","message":"Fix a few nits from the main volume-based patch\n\nI had a few minor nits on the volume-based patch. This patch\ncorrects those.\n\nChange-Id: I5f9ce36c878973f4ed96527af6f1024a362421d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1bfd8caf5f404d8dc8de8a16bd763071b69766f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1bfd8caf5f404d8dc8de8a16bd763071b69766f6"}]},"branch":"refs/heads/master"},"1d664425e9c17f23abd217fd006fd73a9ae92f90":{"kind":"REWORK","_number":3,"created":"2019-09-11 21:22:11.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/44/681144/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/681144/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/681144/3"}}},"commit":{"parents":[{"commit":"510525a92b6c6335f9ca28b0361e138edb52cd19","subject":"Merge \"Add `additive_only` parameter to Batch Member call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/510525a92b6c6335f9ca28b0361e138edb52cd19"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-10 00:02:45.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-09-11 21:21:05.000000000","tz":-420},"subject":"Fix a few nits from the main volume-based patch","message":"Fix a few nits from the main volume-based patch\n\nI had a few minor nits on the volume-based patch. This patch\ncorrects those.\n\nChange-Id: I5f9ce36c878973f4ed96527af6f1024a362421d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d664425e9c17f23abd217fd006fd73a9ae92f90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d664425e9c17f23abd217fd006fd73a9ae92f90"}]},"branch":"refs/heads/master"},"27c043fdc446619492a60c39feacee064da4828b":{"kind":"REWORK","_number":4,"created":"2019-09-12 03:01:25.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/44/681144/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/681144/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/681144/4"}}},"commit":{"parents":[{"commit":"510525a92b6c6335f9ca28b0361e138edb52cd19","subject":"Merge \"Add `additive_only` parameter to Batch Member call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/510525a92b6c6335f9ca28b0361e138edb52cd19"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-10 00:02:45.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-09-12 03:00:32.000000000","tz":-420},"subject":"Fix a few nits from the main volume-based patch","message":"Fix a few nits from the main volume-based patch\n\nI had a few minor nits on the volume-based patch. This patch\ncorrects those.\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I5f9ce36c878973f4ed96527af6f1024a362421d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27c043fdc446619492a60c39feacee064da4828b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27c043fdc446619492a60c39feacee064da4828b"}]},"branch":"refs/heads/master"},"afe648caef8fad86f4cb09e2292fab9a0ea669af":{"kind":"REWORK","_number":5,"created":"2019-09-12 03:37:02.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/44/681144/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/681144/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/681144/5"}}},"commit":{"parents":[{"commit":"510525a92b6c6335f9ca28b0361e138edb52cd19","subject":"Merge \"Add `additive_only` parameter to Batch Member call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/510525a92b6c6335f9ca28b0361e138edb52cd19"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-10 00:02:45.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-09-12 03:36:53.000000000","tz":-420},"subject":"Fix a few nits from the main volume-based patch","message":"Fix a few nits from the main volume-based patch\n\nI had a few minor nits on the volume-based patch. This patch\ncorrects those.\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I5f9ce36c878973f4ed96527af6f1024a362421d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/afe648caef8fad86f4cb09e2292fab9a0ea669af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/afe648caef8fad86f4cb09e2292fab9a0ea669af"}]},"branch":"refs/heads/master"},"e189acdfc48387c1cd84dfe6241ec22aade935f7":{"kind":"REWORK","_number":6,"created":"2019-09-12 08:05:06.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/44/681144/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/681144/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/681144/6"}}},"commit":{"parents":[{"commit":"510525a92b6c6335f9ca28b0361e138edb52cd19","subject":"Merge \"Add `additive_only` parameter to Batch Member call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/510525a92b6c6335f9ca28b0361e138edb52cd19"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-10 00:02:45.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-09-12 08:04:46.000000000","tz":-420},"subject":"Fix a few nits from the main volume-based patch","message":"Fix a few nits from the main volume-based patch\n\nI had a few minor nits on the volume-based patch. This patch\ncorrects those.\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I5f9ce36c878973f4ed96527af6f1024a362421d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e189acdfc48387c1cd84dfe6241ec22aade935f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e189acdfc48387c1cd84dfe6241ec22aade935f7"}]},"branch":"refs/heads/master"},"df7dc83fbbd7eb7e9aef13cd732e63fc05bad729":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-09-12 08:31:47.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/44/681144/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/681144/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/681144/7"}}},"commit":{"parents":[{"commit":"510525a92b6c6335f9ca28b0361e138edb52cd19","subject":"Merge \"Add `additive_only` parameter to Batch Member call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/510525a92b6c6335f9ca28b0361e138edb52cd19"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-10 00:02:45.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-12 08:31:43.000000000","tz":0},"subject":"Fix a few nits from the main volume-based patch","message":"Fix a few nits from the main volume-based patch\n\nI had a few minor nits on the volume-based patch. This patch\ncorrects those.\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n\nChange-Id: I5f9ce36c878973f4ed96527af6f1024a362421d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df7dc83fbbd7eb7e9aef13cd732e63fc05bad729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df7dc83fbbd7eb7e9aef13cd732e63fc05bad729"}]},"branch":"refs/heads/master"},"7012b3f34b7806b05ce29427ad4f91cc52e3e08d":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2019-09-12 09:11:54.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/44/681144/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/681144/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/681144/8"}}},"commit":{"parents":[{"commit":"510525a92b6c6335f9ca28b0361e138edb52cd19","subject":"Merge \"Add `additive_only` parameter to Batch Member call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/510525a92b6c6335f9ca28b0361e138edb52cd19"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-10 00:02:45.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-12 09:11:46.000000000","tz":0},"subject":"Fix a few nits from the main volume-based patch","message":"Fix a few nits from the main volume-based patch\n\nI had a few minor nits on the volume-based patch. This patch\ncorrects those.\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I5f9ce36c878973f4ed96527af6f1024a362421d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7012b3f34b7806b05ce29427ad4f91cc52e3e08d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7012b3f34b7806b05ce29427ad4f91cc52e3e08d"}]},"branch":"refs/heads/master"},"05f5257075dda076b0960a319d49c970b0015022":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2019-09-12 10:32:52.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/44/681144/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/681144/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/681144/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/681144/9"}}},"commit":{"parents":[{"commit":"510525a92b6c6335f9ca28b0361e138edb52cd19","subject":"Merge \"Add `additive_only` parameter to Batch Member call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/510525a92b6c6335f9ca28b0361e138edb52cd19"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-10 00:02:45.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-09-12 10:32:45.000000000","tz":0},"subject":"Fix a few nits from the main volume-based patch","message":"Fix a few nits from the main volume-based patch\n\nI had a few minor nits on the volume-based patch. This patch\ncorrects those.\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\n\nChange-Id: I5f9ce36c878973f4ed96527af6f1024a362421d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05f5257075dda076b0960a319d49c970b0015022"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05f5257075dda076b0960a319d49c970b0015022"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
